)]}'
{"id":"openstack%2Fopenstack-health~667289","triplet_id":"openstack%2Fopenstack-health~master~I4c973f62b728f3526fe8f45fe627e040702e3f04","project":"openstack/openstack-health","branch":"master","topic":"Bug#2","hashtags":[],"change_id":"I4c973f62b728f3526fe8f45fe627e040702e3f04","subject":"Drop nodejs4 job","status":"ABANDONED","created":"2019-06-25 08:41:04.000000000","updated":"2019-07-16 07:30:04.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"3b14c048e56f7dfc1147969338e219f21ace266c","_number":667289,"virtual_id_number":667289,"owner":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":-1,"date":"2019-06-25 08:57:29.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2019-06-28 01:16:43.000000000","_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-25 08:57:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"23add7a1edd4b588d371c3db567be4ed39b139b7","author":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"date":"2019-06-25 08:41:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a897cea54c847bccf44dde9b19facc212268c0fc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-25 08:57:29.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/89/667289/1/check/openstack-tox-pep8/1edcb54/ : SUCCESS in 3m 35s\n- openstack-tox-py27 http://logs.openstack.org/89/667289/1/check/openstack-tox-py27/2ce384a/ : SUCCESS in 2m 56s\n- nodejs10-npm-run-lint http://logs.openstack.org/89/667289/1/check/nodejs10-npm-run-lint/c851c37/ : SUCCESS in 7m 24s\n- nodejs10-npm-run-test http://logs.openstack.org/89/667289/1/check/nodejs10-npm-run-test/5c126b8/ : FAILURE in 7m 53s\n- openstack-tox-py36 http://logs.openstack.org/89/667289/1/check/openstack-tox-py36/6b1b73f/ : SUCCESS in 4m 01s\n- openstack-tox-docs http://logs.openstack.org/89/667289/1/check/openstack-tox-docs/6d18a84/html/ : SUCCESS in 3m 41s","accounts_in_message":[],"_revision_number":1},{"id":"1619d78b83bfb9490e36dbc3ddad542b9d2746da","author":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"date":"2019-06-27 03:21:00.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"11ae31cde156be2f1ec105e210294b4bdf789570","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 03:44:12.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/89/667289/1/check/openstack-tox-pep8/1d346f8/ : SUCCESS in 3m 33s\n- openstack-tox-py27 http://logs.openstack.org/89/667289/1/check/openstack-tox-py27/bf09e44/ : SUCCESS in 3m 19s\n- nodejs10-npm-run-lint http://logs.openstack.org/89/667289/1/check/nodejs10-npm-run-lint/392dfc0/ : SUCCESS in 7m 40s\n- nodejs10-npm-run-test http://logs.openstack.org/89/667289/1/check/nodejs10-npm-run-test/6318ab6/ : FAILURE in 8m 40s\n- openstack-tox-py36 http://logs.openstack.org/89/667289/1/check/openstack-tox-py36/60a1565/ : SUCCESS in 3m 16s\n- openstack-tox-docs http://logs.openstack.org/89/667289/1/check/openstack-tox-docs/46c62af/html/ : SUCCESS in 2m 52s","accounts_in_message":[],"_revision_number":1},{"id":"4c8ee3e31cf53768b8555e8a40f2adb23ea13ac4","author":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"date":"2019-06-28 01:16:43.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"41d53426505bfe8f0a175dacefe7605d25bcbeca","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-28 01:33:02.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/89/667289/1/check/openstack-tox-pep8/bdc7981/ : SUCCESS in 3m 30s\n- openstack-tox-py27 http://logs.openstack.org/89/667289/1/check/openstack-tox-py27/9acdb15/ : SUCCESS in 3m 29s\n- nodejs10-npm-run-lint http://logs.openstack.org/89/667289/1/check/nodejs10-npm-run-lint/13250fa/ : SUCCESS in 6m 57s\n- nodejs10-npm-run-test http://logs.openstack.org/89/667289/1/check/nodejs10-npm-run-test/cef051f/ : FAILURE in 8m 02s\n- openstack-tox-py36 http://logs.openstack.org/89/667289/1/check/openstack-tox-py36/66d53bf/ : SUCCESS in 3m 12s\n- openstack-tox-docs http://logs.openstack.org/89/667289/1/check/openstack-tox-docs/1bbcfc7/html/ : SUCCESS in 3m 15s","accounts_in_message":[],"_revision_number":1},{"id":"8365c31dc7f3e16ab29bf50a5a2b2ca61e393497","author":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"date":"2019-07-16 07:30:04.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"11cc51685f50ed24a10019f9df57e57dbcc1f905","revisions":{"11cc51685f50ed24a10019f9df57e57dbcc1f905":{"kind":"REWORK","_number":1,"created":"2019-06-25 08:41:04.000000000","uploader":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"ref":"refs/changes/89/667289/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-health","ref":"refs/changes/89/667289/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-health refs/changes/89/667289/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-health refs/changes/89/667289/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-health refs/changes/89/667289/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-health refs/changes/89/667289/1"}}},"commit":{"parents":[{"commit":"77a4dc7a0e6a23e1dc39705dc0a97df858e4bb19","subject":"Merge \"Update angularjs libraries version\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-health/commit/77a4dc7a0e6a23e1dc39705dc0a97df858e4bb19"}]}],"author":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-06-25 08:21:31.000000000","tz":480},"committer":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-06-25 08:21:35.000000000","tz":480},"subject":"Drop nodejs4 job","message":"Drop nodejs4 job\n\nAs Migration to bionic is completed or nodejs10 job running.\nSo we not required nodejs4 job anymore.\n\nChange-Id: I4c973f62b728f3526fe8f45fe627e040702e3f04\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-health/commit/11cc51685f50ed24a10019f9df57e57dbcc1f905"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-health/commit/11cc51685f50ed24a10019f9df57e57dbcc1f905"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
