)]}'
{"id":"openstack%2Fopenstack-health~711164","triplet_id":"openstack%2Fopenstack-health~master~I3201ef055995ce39fb5964dd5217d6b739e5c304","project":"openstack/openstack-health","branch":"master","hashtags":[],"change_id":"I3201ef055995ce39fb5964dd5217d6b739e5c304","subject":"[WIP] Update npm packages","status":"NEW","created":"2020-03-04 05:46:10.000000000","updated":"2020-03-04 06:27:28.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"f7fee9adb5f618f5932045d648df08fafd32df9c","_number":711164,"virtual_id_number":711164,"owner":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-03-04 06:27:28.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-04 06:27:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c841c4152c11f499c6bfdf00709d3723176a3c18","author":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"date":"2020-03-04 05:46:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8269f1c9f23522236f586daa58d9a21f0f417fa7","author":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"date":"2020-03-04 06:01:25.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f7fee9adb5f618f5932045d648df08fafd32df9c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-04 06:27:28.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- nodejs-npm-run-lint https://zuul.opendev.org/t/openstack/build/a7d8cb6afc844a6d9b5e1cb45615492c : FAILURE in 11m 25s\n- nodejs-npm-run-test https://zuul.opendev.org/t/openstack/build/309fb265e15c40c381061c0a512221d6 : FAILURE in 13m 02s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/c3b1f1d941aa4b8a8e8839560777ba05 : SUCCESS in 5m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8392706a1cb74ba89c7466727af3a31f : SUCCESS in 5m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2f6bc3ba657f463fa62b4c33bdf75b1a : SUCCESS in 6m 41s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/14b5c8f6fdc2400fb7a7410ce6c27dc0 : SUCCESS in 7m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/65f1c6943aa640908041c79272945487 : SUCCESS in 7m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ba8eda1b0d5438ab291d7d710d090fc : SUCCESS in 5m 52s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"e0cc7081f178c49938cd47f015f4517ef6b8f8f9","revisions":{"933faed2f99b0f9e91f25ca560adfb6dcdea3a9e":{"kind":"REWORK","_number":1,"created":"2020-03-04 05:46:10.000000000","uploader":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"ref":"refs/changes/64/711164/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-health","ref":"refs/changes/64/711164/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-health refs/changes/64/711164/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-health refs/changes/64/711164/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-health refs/changes/64/711164/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-health refs/changes/64/711164/1"}}},"commit":{"parents":[{"commit":"f9942b6ca5a05592092a638898ce71a657559864","subject":"[ussuri][goal] Drop python 2.7 support and testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-health/commit/f9942b6ca5a05592092a638898ce71a657559864"}]}],"author":{"name":"Masayuki Igawa","email":"masayuki@igawa.io","date":"2020-03-04 05:44:44.000000000","tz":540},"committer":{"name":"Masayuki Igawa","email":"masayuki@igawa.io","date":"2020-03-04 05:45:35.000000000","tz":540},"subject":"[WIP] Update npm packages","message":"[WIP] Update npm packages\n\nChange-Id: I3201ef055995ce39fb5964dd5217d6b739e5c304\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-health/commit/933faed2f99b0f9e91f25ca560adfb6dcdea3a9e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-health/commit/933faed2f99b0f9e91f25ca560adfb6dcdea3a9e"}]},"branch":"refs/heads/master"},"e0cc7081f178c49938cd47f015f4517ef6b8f8f9":{"kind":"REWORK","_number":2,"created":"2020-03-04 06:01:25.000000000","uploader":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"ref":"refs/changes/64/711164/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-health","ref":"refs/changes/64/711164/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-health refs/changes/64/711164/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-health refs/changes/64/711164/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-health refs/changes/64/711164/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-health refs/changes/64/711164/2"}}},"commit":{"parents":[{"commit":"f9942b6ca5a05592092a638898ce71a657559864","subject":"[ussuri][goal] Drop python 2.7 support and testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-health/commit/f9942b6ca5a05592092a638898ce71a657559864"}]}],"author":{"name":"Masayuki Igawa","email":"masayuki@igawa.io","date":"2020-03-04 05:44:44.000000000","tz":540},"committer":{"name":"Masayuki Igawa","email":"masayuki@igawa.io","date":"2020-03-04 06:00:55.000000000","tz":540},"subject":"[WIP] Update npm packages","message":"[WIP] Update npm packages\n\nChange-Id: I3201ef055995ce39fb5964dd5217d6b739e5c304\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-health/commit/e0cc7081f178c49938cd47f015f4517ef6b8f8f9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-health/commit/e0cc7081f178c49938cd47f015f4517ef6b8f8f9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
