)]}'
{"id":"openstack%2Fopenstack-health~715605","triplet_id":"openstack%2Fopenstack-health~master~I4418c77b622f0a0f2f729b957cceb577d20f30d9","project":"openstack/openstack-health","branch":"master","topic":"update-hacking","hashtags":[],"change_id":"I4418c77b622f0a0f2f729b957cceb577d20f30d9","subject":"Update hacking for Python3","status":"MERGED","created":"2020-03-28 11:16:52.000000000","updated":"2020-03-28 11:48:51.000000000","submitted":"2020-03-28 11:48:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"715605-1585396131071-046e596a","meta_rev_id":"aa3cd2f58b7cf7894684bf49a42faae735adbfa9","_number":715605,"virtual_id_number":715605,"owner":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-28 11:48:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"all":[{"value":2,"date":"2020-03-28 11:36:16.000000000","_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"all":[{"value":1,"date":"2020-03-28 11:36:16.000000000","_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-28 11:36:16.000000000","updated_by":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"reviewer":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"state":"REVIEWER"},{"updated":"2020-03-28 11:48:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7401128e0b61e68ec4c2dd306bb00e84e9aa2199","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-03-28 11:16:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"262478d3eef09db888ad6aef6a0c6250ddf90eb6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-28 11:27:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- nodejs10-npm-run-lint https://zuul.opendev.org/t/openstack/build/f7604af45d1c46f7bb2ec86f6641989d : SUCCESS in 7m 58s\n- nodejs10-npm-run-test https://zuul.opendev.org/t/openstack/build/204cb4ac4e4a48be84d0364ca6ae637a : SUCCESS in 9m 33s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/e32589b6ef354a489faf00777d3b6f6b : SUCCESS in 4m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/02be42113c784c1aa29322828ec7d7fa : SUCCESS in 3m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3bbb0900486248a9b1d5db39a7550eef : SUCCESS in 3m 15s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cfb5ae93929d4c168f4d9ad636491e6d : SUCCESS in 4m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c3db34bb886d4965858dbfa36715df46 : SUCCESS in 3m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6775e19c6eb24a43b02417a7ecff81cb : SUCCESS in 4m 00s","accounts_in_message":[],"_revision_number":1},{"id":"308c8c1af7d2fbc1b016f19bfc75eda996e256ab","author":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"date":"2020-03-28 11:36:16.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nThanks, LGTM","accounts_in_message":[],"_revision_number":1},{"id":"7af305fba385847f63dee6d8eaa81d49b5503a8f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-28 11:36:27.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a67e8e65d497f46137f8a99c6bd80c528d455a75","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-28 11:48:50.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- nodejs10-npm-run-lint https://zuul.opendev.org/t/openstack/build/649100d8424440d29dad1594ec49c61f : SUCCESS in 6m 55s\n- nodejs10-npm-run-test https://zuul.opendev.org/t/openstack/build/0895d6474afd4686a35e01e69e73cc8c : SUCCESS in 10m 58s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/68b49e6aae4143f4a09a3667e542f912 : SUCCESS in 3m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/35060dd0c2864fa1b336b0b489ff1e3a : SUCCESS in 3m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b0d4eaea05334aa0a3d854566ce301c9 : SUCCESS in 3m 41s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4cf4ed5910bc4b1abc26261fef5d260c : SUCCESS in 4m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c474ea9e81a643ee86ac2ad0aa003d88 : SUCCESS in 3m 54s","accounts_in_message":[],"_revision_number":1},{"id":"3c71eb03a2e88adcace5112ca547f186823e60cc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-28 11:48:51.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c889b33ecc68df56f321d9cc59861116340e6ba9","revisions":{"c889b33ecc68df56f321d9cc59861116340e6ba9":{"kind":"REWORK","_number":1,"created":"2020-03-28 11:16:52.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/05/715605/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-health","ref":"refs/changes/05/715605/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-health refs/changes/05/715605/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-health refs/changes/05/715605/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-health refs/changes/05/715605/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-health refs/changes/05/715605/1"}}},"commit":{"parents":[{"commit":"b15f68e52dd2661b825551dfef1ede954828f7eb","subject":"[ussuri][goal] Drop python 2.7 support and testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-health/commit/b15f68e52dd2661b825551dfef1ede954828f7eb"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-03-28 11:11:10.000000000","tz":60},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-03-28 11:16:40.000000000","tz":60},"subject":"Update hacking for Python3","message":"Update hacking for Python3\n\nThe repo is Python 3 now, so update hacking to version 2.0 which\nsupports Python 3.\n\nFix problems found.\n\nChange-Id: I4418c77b622f0a0f2f729b957cceb577d20f30d9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-health/commit/c889b33ecc68df56f321d9cc59861116340e6ba9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-health/commit/c889b33ecc68df56f321d9cc59861116340e6ba9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
