)]}'
{"id":"openstack%2Fopenstack-health~734454","triplet_id":"openstack%2Fopenstack-health~master~I534486bb82c723b63b990061b02048fab4ab84d7","project":"openstack/openstack-health","branch":"master","topic":"drop_mock","hashtags":[],"change_id":"I534486bb82c723b63b990061b02048fab4ab84d7","subject":"Use unittest.mock instead of mock","status":"ABANDONED","created":"2020-06-09 08:51:43.000000000","updated":"2020-06-19 13:47:17.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"a85aedee0072787bba7335094256b7f63b6af5c2","_number":734454,"virtual_id_number":734454,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2020-06-11 08:41:12.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-11 08:41:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"24afd5293c266bc7f085e434a50907cfc59dcb97","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-06-09 08:51:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4b7180e83586803e68f8332d750a2f2eab736480","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-09 09:12:55.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- horizon-nodejs10-run-lint https://zuul.opendev.org/t/openstack/build/09120a95a02a42d88ab5596933d52ddd : SUCCESS in 6m 45s\n- horizon-nodejs10-run-test https://zuul.opendev.org/t/openstack/build/d38f262aaf2048cbb54131f9ea3597c4 : SUCCESS in 12m 49s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/00005dde2b8f49df966f9ee2b971e484 : SUCCESS in 6m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/76c59bcf86fd476c9a55e15963da8a5a : FAILURE in 4m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d939f90f9a674ff7bca119a408b8a078 : SUCCESS in 4m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d391fda1ee8c4386b4e85e26215bfc25 : SUCCESS in 5m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2dc01a41554841c1b10b0f6376f58868 : SUCCESS in 4m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9475c4f873e64356b8c5717065b8c42b : SUCCESS in 5m 36s","accounts_in_message":[],"_revision_number":1},{"id":"85f02cc46106f64f5342383a38f40cc2aa558bb8","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-06-11 08:28:06.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c958de768c3b56f6f619671248814cf41d32217a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-11 08:41:12.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- horizon-nodejs10-run-lint https://zuul.opendev.org/t/openstack/build/3d7434b97cd5434c8f193d0c39aa3961 : SUCCESS in 7m 44s\n- horizon-nodejs10-run-test https://zuul.opendev.org/t/openstack/build/50c2596035db48e7a73a6287900377ec : SUCCESS in 11m 33s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/dd30673870d541fe916b47f3ab65775f : SUCCESS in 4m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2042d2c18c0948608f35389d47717879 : SUCCESS in 4m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4c90ec3698e548a99c615f1c2dbd61f0 : SUCCESS in 3m 49s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/be18d23adc194e00913dfcbd04cdcc61 : SUCCESS in 7m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a04a36d396ce44a88393e837f8035538 : SUCCESS in 10m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf8a755ca78f402e90ff5af3183708e8 : SUCCESS in 4m 09s","accounts_in_message":[],"_revision_number":2},{"id":"8c97f66613f4a9399836cf7beaf6d70f1e42d5c0","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-06-19 13:47:17.000000000","message":"Abandoned\n\nduplicated with https://review.opendev.org/#/c/717412/","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"6d791c0aa48917b02be00cfc906832bde27d7986","revisions":{"9a872effe68d3c392e1a2e83fed25f7e77868113":{"kind":"REWORK","_number":1,"created":"2020-06-09 08:51:43.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/54/734454/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-health","ref":"refs/changes/54/734454/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-health refs/changes/54/734454/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-health refs/changes/54/734454/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-health refs/changes/54/734454/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-health refs/changes/54/734454/1"}}},"commit":{"parents":[{"commit":"43a20b4f1b824a978a853308c560378f37c9e2a7","subject":"Switch to newer openstackdocstheme version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-health/commit/43a20b4f1b824a978a853308c560378f37c9e2a7"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-06-09 08:51:37.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-06-09 08:51:37.000000000","tz":120},"subject":"drop mock from lower-constraints and requirements","message":"drop mock from lower-constraints and requirements\n\nThe mock third party library was needed for mock support in py2\nruntimes. Since we now only support py36 and later, we don\u0027t need it\nin lower-constraints and requirements.\n\nThese changes will help us to drop `mock` from openstack/requirements\n\nChange-Id: I534486bb82c723b63b990061b02048fab4ab84d7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-health/commit/9a872effe68d3c392e1a2e83fed25f7e77868113"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-health/commit/9a872effe68d3c392e1a2e83fed25f7e77868113"}]},"branch":"refs/heads/master"},"6d791c0aa48917b02be00cfc906832bde27d7986":{"kind":"REWORK","_number":2,"created":"2020-06-11 08:28:06.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/54/734454/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-health","ref":"refs/changes/54/734454/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-health refs/changes/54/734454/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-health refs/changes/54/734454/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-health refs/changes/54/734454/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-health refs/changes/54/734454/2"}}},"commit":{"parents":[{"commit":"43a20b4f1b824a978a853308c560378f37c9e2a7","subject":"Switch to newer openstackdocstheme version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-health/commit/43a20b4f1b824a978a853308c560378f37c9e2a7"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-06-09 08:51:37.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-06-11 08:27:38.000000000","tz":120},"subject":"Use unittest.mock instead of mock","message":"Use unittest.mock instead of mock\n\nThe mock third party library was needed for mock support in py2\nruntimes. Since we now only support py36 and later, we can use the\nstandard lib unittest.mock module instead.\n\nChange-Id: I534486bb82c723b63b990061b02048fab4ab84d7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-health/commit/6d791c0aa48917b02be00cfc906832bde27d7986"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-health/commit/6d791c0aa48917b02be00cfc906832bde27d7986"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
