)]}'
{"id":"openstack%2Fopenstack-health~771976","triplet_id":"openstack%2Fopenstack-health~master~Icd33d84d94d58ba36b296d46f4397227fa36ef9c","project":"openstack/openstack-health","branch":"master","hashtags":[],"change_id":"Icd33d84d94d58ba36b296d46f4397227fa36ef9c","subject":"Cap pandas to \u003c 1.2.0 to fix gates","status":"MERGED","created":"2021-01-22 13:26:19.000000000","updated":"2021-01-22 14:57:55.000000000","submitted":"2021-01-22 14:57:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"771976","meta_rev_id":"43d85650a5290f159dcae1797adc899de91abd7d","_number":771976,"virtual_id_number":771976,"owner":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-22 14:57:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"all":[{"value":2,"date":"2021-01-22 14:47:17.000000000","_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"value":2,"date":"2021-01-22 13:52:09.000000000","_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"all":[{"value":1,"date":"2021-01-22 14:47:17.000000000","_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-22 13:37:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-22 13:39:44.000000000","updated_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"reviewer":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"state":"REVIEWER"},{"updated":"2021-01-22 13:39:44.000000000","updated_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"}],"messages":[{"id":"7b0dd065e6a2245160132477d8313e14d1a45d43","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2021-01-22 13:26:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d8271c1c72e98c94484c19b74d455a65c15383f3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2021-01-22 13:26:58.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"31e455dec9241d0064cd5bbc6882ac4116abba64","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-22 13:37:36.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/23ed481cd1eb4618b66df2277c58726c : SUCCESS in 4m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e228ac609bb740998c0b76395415880d : SUCCESS in 5m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3979eb6fd2d14a648e3c179500b0e9d7 : SUCCESS in 4m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/19bc1ddf9bbc4dda965dc15a6491ae88 : SUCCESS in 3m 53s\n- openstack-health-nodejs10-run-lint https://zuul.opendev.org/t/openstack/build/024b39268ec1430a9dc9eece32ea14f0 : SUCCESS in 6m 03s\n- openstack-health-nodejs10-run-test-browser https://zuul.opendev.org/t/openstack/build/3a07823385624e128b255d637a560a4e : SUCCESS in 9m 24s","accounts_in_message":[],"_revision_number":2},{"id":"9af548976775c903862d12d5e8422659d886cb98","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-01-22 13:52:09.000000000","message":"Patch Set 2: Code-Review+2\n\nlgtm","accounts_in_message":[],"_revision_number":2},{"id":"6a10c09038fe5a57d8fbe6f6f0293f524623e58d","author":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"date":"2021-01-22 14:47:17.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nMaybe, we should fix for the newer version but so far, I\u0027m OK with this change.","accounts_in_message":[],"_revision_number":2},{"id":"e198394dd60c0a0656b461aae3104ed329d14df4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-22 14:47:33.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"ad4d35e655bdeee3bb13f2f9fe110a76f60eb4df","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-22 14:57:54.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0635897bb6944e90a61f643d40b16341 : SUCCESS in 3m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1b4732e659e544da85b98917e42219f1 : SUCCESS in 3m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2eef309386fe4f238b5f42f691239c15 : SUCCESS in 3m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f6319c318d349cf91d50ee3cb2dd136 : SUCCESS in 4m 17s\n- openstack-health-nodejs10-run-lint https://zuul.opendev.org/t/openstack/build/9e0ece1918b841ecbbeb288c89fb7c8c : SUCCESS in 5m 26s\n- openstack-health-nodejs10-run-test-browser https://zuul.opendev.org/t/openstack/build/8849f311031349d9a09a937c1ecbd7b0 : SUCCESS in 8m 54s","accounts_in_message":[],"_revision_number":2},{"id":"43d85650a5290f159dcae1797adc899de91abd7d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-22 14:57:55.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"23f9274a1a0c4f3c1c5b6535c5c9b5b029f75a8c","revisions":{"9885fd2e9cb09c70a6d80d88139f51599a047945":{"kind":"REWORK","_number":1,"created":"2021-01-22 13:26:19.000000000","uploader":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"ref":"refs/changes/76/771976/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-health","ref":"refs/changes/76/771976/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-health refs/changes/76/771976/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-health refs/changes/76/771976/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-health refs/changes/76/771976/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-health refs/changes/76/771976/1"}}},"commit":{"parents":[{"commit":"1abd0617a36d397ea2a16b6cf2d0fbe65b2eb05f","subject":"Fix followup comment from review 768227","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-health/commit/1abd0617a36d397ea2a16b6cf2d0fbe65b2eb05f"}]}],"author":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2021-01-22 13:23:07.000000000","tz":0},"committer":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2021-01-22 13:23:07.000000000","tz":0},"subject":"Cap pandas to \u003c 1.2.0 to fix gates","message":"Cap pandas to \u003c 1.2.0 to fix gates\n\npandas\u003e\u003d1.2.0 contains some changes in rounding numbers which makes\nthe following tests fail\n\n - test_get_numeric_data_diff_build_name\n - test_get_numeric_data_diff_build_name\n\nwith the following assert error\n\n\u00272018-06-23T00:00:00\u0027: 4766.42225,\n\u00272018-06-23T00:00:00\u0027: 4766.4222500000005,\n                                 ++++++++\n\u00272018-06-24T00:00:00\u0027: 4899.55725,\n\u00272018-06-25T00:00:00\u0027: 5042.148499999999}}\n                               ^^^^^^^^^\n\nChange-Id: Icd33d84d94d58ba36b296d46f4397227fa36ef9c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-health/commit/9885fd2e9cb09c70a6d80d88139f51599a047945"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-health/commit/9885fd2e9cb09c70a6d80d88139f51599a047945"}]},"branch":"refs/heads/master"},"23f9274a1a0c4f3c1c5b6535c5c9b5b029f75a8c":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-01-22 13:26:58.000000000","uploader":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"ref":"refs/changes/76/771976/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-health","ref":"refs/changes/76/771976/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-health refs/changes/76/771976/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-health refs/changes/76/771976/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-health refs/changes/76/771976/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-health refs/changes/76/771976/2"}}},"commit":{"parents":[{"commit":"1abd0617a36d397ea2a16b6cf2d0fbe65b2eb05f","subject":"Fix followup comment from review 768227","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-health/commit/1abd0617a36d397ea2a16b6cf2d0fbe65b2eb05f"}]}],"author":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2021-01-22 13:23:07.000000000","tz":0},"committer":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2021-01-22 13:23:39.000000000","tz":0},"subject":"Cap pandas to \u003c 1.2.0 to fix gates","message":"Cap pandas to \u003c 1.2.0 to fix gates\n\npandas\u003e\u003d1.2.0 contains some changes in rounding numbers which makes\nthe following tests fail when executed by py38\n\n - test_get_numeric_data_diff_build_name\n - test_get_numeric_data_diff_build_name\n\nwith the following assert error\n\n\u00272018-06-23T00:00:00\u0027: 4766.42225,\n\u00272018-06-23T00:00:00\u0027: 4766.4222500000005,\n                                 ++++++++\n\u00272018-06-24T00:00:00\u0027: 4899.55725,\n\u00272018-06-25T00:00:00\u0027: 5042.148499999999}}\n                               ^^^^^^^^^\n\nChange-Id: Icd33d84d94d58ba36b296d46f4397227fa36ef9c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-health/commit/23f9274a1a0c4f3c1c5b6535c5c9b5b029f75a8c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-health/commit/23f9274a1a0c4f3c1c5b6535c5c9b5b029f75a8c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
