)]}'
{"id":"openstack%2Fopenstack-health~793851","triplet_id":"openstack%2Fopenstack-health~master~I748f813ae1b06ba31b30560d6d30c6528f47738f","project":"openstack/openstack-health","branch":"master","topic":"oftc","hashtags":[],"change_id":"I748f813ae1b06ba31b30560d6d30c6528f47738f","subject":"Update IRC network to OFTC","status":"MERGED","created":"2021-05-31 20:06:38.000000000","updated":"2021-06-01 16:41:55.000000000","submitted":"2021-06-01 16:41:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"793851-oftc","meta_rev_id":"415fc5365741881ea602fd15fc36f8eea0638b49","_number":793851,"virtual_id_number":793851,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-01 16:41:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"recommended":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"all":[{"value":1,"date":"2021-06-01 15:29:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":2,"date":"2021-06-01 07:32:06.000000000","_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"value":2,"date":"2021-06-01 16:29:37.000000000","_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"all":[{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"value":1,"date":"2021-06-01 16:29:37.000000000","_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-31 20:15:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-31 21:25:43.000000000","updated_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"reviewer":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"state":"CC"},{"updated":"2021-06-01 07:30:32.000000000","updated_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"reviewer":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"state":"REVIEWER"},{"updated":"2021-06-01 07:32:06.000000000","updated_by":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"reviewer":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"state":"REVIEWER"},{"updated":"2021-06-01 15:29:40.000000000","updated_by":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"reviewer":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"state":"REVIEWER"}],"messages":[{"id":"b85a297b6b9a962ce0c47ed94871b7b6460af527","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-05-31 20:06:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e50576c226238638e5a3e731479070df92cedea9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-31 20:15:52.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d047073d1a0940978dd0d325e5b67852 : SUCCESS in 4m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e5eb037fa2d1489e9e9ed3954028b6d1 : FAILURE in 6m 27s\n- openstack-health-nodejs10-run-lint https://zuul.opendev.org/t/openstack/build/33c16e85421b4a9095bec2778afa14ff : SUCCESS in 5m 02s\n- openstack-health-nodejs10-run-test-browser https://zuul.opendev.org/t/openstack/build/2b04252d09c945dc95ecbc3b0a6b0312 : SUCCESS in 7m 44s","accounts_in_message":[],"_revision_number":1},{"id":"bd0712e081f387bc9d59136d080a661a03c87ada","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2021-05-31 21:25:43.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"ec19352cc890613cb6833c665367402b77972faa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-31 21:48:39.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c9edc1051149449aa86cfc906feece3f : SUCCESS in 3m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a80db81f6f1941e59f16e9f2f199bf3d : FAILURE in 6m 06s\n- openstack-health-nodejs10-run-lint https://zuul.opendev.org/t/openstack/build/d11d18ad0df64565913292d238d4dae9 : SUCCESS in 5m 33s\n- openstack-health-nodejs10-run-test-browser https://zuul.opendev.org/t/openstack/build/4c5d8d79f81c462eabfc391950969c3b : SUCCESS in 8m 24s","accounts_in_message":[],"_revision_number":1},{"id":"adc1329c9f21726c988a83b04dc081c45f57643d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2021-06-01 07:30:32.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"35997c804ac524f1e4358044b678c16c0af26f55","author":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"date":"2021-06-01 07:32:06.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"0584474cd29af8fa714f3b0c681b3c99318fee10","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-01 07:44:27.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/43c34b33d98d4079bc7baa39d25fdb27 : SUCCESS in 4m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b64cf3976344e5f9db6ffa9a9f05351 : FAILURE in 9m 02s\n- openstack-health-nodejs10-run-lint https://zuul.opendev.org/t/openstack/build/bdaa94be2a824b4c9a5b2c66a785672a : SUCCESS in 6m 39s\n- openstack-health-nodejs10-run-test-browser https://zuul.opendev.org/t/openstack/build/44dabac81a9b4ac581c20e51f0361fb7 : SUCCESS in 11m 24s","accounts_in_message":[],"_revision_number":2},{"id":"0fb61f31d264879d48c156e466c1816b05b0ebb6","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2021-06-01 15:29:40.000000000","message":"Patch Set 2: Code-Review+1\n\nLGTM, doc job is fixed by https://review.opendev.org/c/openstack/openstack-health/+/793984","accounts_in_message":[],"_revision_number":2},{"id":"1523f40e48dd019046cff223b70f140b8704f983","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2021-06-01 16:02:20.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"22de732f7c9852170ce0ab4c040a33b762251134","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-06-01 16:03:23.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"dfb7617f8afdac8dcd4f1c945191bcb75ff54143","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-01 16:14:10.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf18f06b835c486aae3695433995c8a2 : SUCCESS in 5m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ecf163c54274d4da3118722fa89d232 : SUCCESS in 7m 41s\n- openstack-health-nodejs10-run-lint https://zuul.opendev.org/t/openstack/build/0ddac8f6d0a840579489699a6c818d6e : SUCCESS in 5m 56s\n- openstack-health-nodejs10-run-test-browser https://zuul.opendev.org/t/openstack/build/94b52063c42a43d991c84c70523ff672 : SUCCESS in 10m 13s","accounts_in_message":[],"_revision_number":2},{"id":"be5d823393d1f4e30108a455341e250e37d357fc","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2021-06-01 16:29:37.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"22f780e09a58a7d107b2143a4d253dd5087ba5f8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-01 16:29:51.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"16135511f874eb841ad9467d531d9cce48b110cb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-01 16:41:49.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6cf7bbd12970439894cf13685a960f04 : SUCCESS in 5m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/180f149577344d87b35903d7ff73f99b : SUCCESS in 7m 40s\n- openstack-health-nodejs10-run-lint https://zuul.opendev.org/t/openstack/build/99bd236ab1fd48839ff25ff907537f9f : SUCCESS in 5m 50s\n- openstack-health-nodejs10-run-test-browser https://zuul.opendev.org/t/openstack/build/46e6297accb54e54acbd89d9efbe75e2 : SUCCESS in 9m 57s","accounts_in_message":[],"_revision_number":2},{"id":"415fc5365741881ea602fd15fc36f8eea0638b49","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-01 16:41:55.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"c21da9a59343fecfb7d9e1152a34d02479e2cff6","revisions":{"ce4a871a067eee519d51982170750fc9e679e2d4":{"kind":"REWORK","_number":1,"created":"2021-05-31 20:06:38.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/51/793851/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-health","ref":"refs/changes/51/793851/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-health refs/changes/51/793851/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-health refs/changes/51/793851/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-health refs/changes/51/793851/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-health refs/changes/51/793851/1"}}},"commit":{"parents":[{"commit":"b91c0054e41934ebb5e1bfd4c94d6f09ed269c7e","subject":"Move openstack-health-nodejs10-run-test-browser to Firefox","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-health/commit/b91c0054e41934ebb5e1bfd4c94d6f09ed269c7e"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2021-05-31 20:05:36.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2021-05-31 20:05:36.000000000","tz":-300},"subject":"Update IRC network to OFTC","message":"Update IRC network to OFTC\n\nChange-Id: I748f813ae1b06ba31b30560d6d30c6528f47738f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-health/commit/ce4a871a067eee519d51982170750fc9e679e2d4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-health/commit/ce4a871a067eee519d51982170750fc9e679e2d4"}]},"branch":"refs/heads/master"},"c21da9a59343fecfb7d9e1152a34d02479e2cff6":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-06-01 07:30:32.000000000","uploader":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"ref":"refs/changes/51/793851/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-health","ref":"refs/changes/51/793851/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-health refs/changes/51/793851/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-health refs/changes/51/793851/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-health refs/changes/51/793851/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-health refs/changes/51/793851/2"}}},"commit":{"parents":[{"commit":"f0a1a438953a1729310b5e2f59d38b121848ddfc","subject":"Merge \"Switch testing to Xena testing runtime\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-health/commit/f0a1a438953a1729310b5e2f59d38b121848ddfc"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2021-05-31 20:05:36.000000000","tz":-300},"committer":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2021-06-01 07:26:55.000000000","tz":0},"subject":"Update IRC network to OFTC","message":"Update IRC network to OFTC\n\nChange-Id: I748f813ae1b06ba31b30560d6d30c6528f47738f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-health/commit/c21da9a59343fecfb7d9e1152a34d02479e2cff6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-health/commit/c21da9a59343fecfb7d9e1152a34d02479e2cff6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
