)]}'
{"id":"openstack%2Fopenstack-health~825336","triplet_id":"openstack%2Fopenstack-health~master~I368beadca9184654d1727579e7cd9274b53192d6","project":"openstack/openstack-health","branch":"master","topic":"fix/gulp","attention_set":{},"removed_from_attention_set":{"30674":{"account":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"last_update":"2022-10-07 09:31:50.000000000","reason":"Change was abandoned"},"22873":{"account":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"last_update":"2022-10-07 09:31:50.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I368beadca9184654d1727579e7cd9274b53192d6","subject":"WIP: Fix gulp","status":"ABANDONED","created":"2022-01-19 15:44:07.000000000","updated":"2022-10-07 09:31:50.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"158f799766f4c15ae82a51e3a40743ef4ff4bd02","_number":825336,"virtual_id_number":825336,"owner":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-01-19 15:57:17.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-19 15:44:33.000000000","updated_by":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"reviewer":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"state":"REVIEWER"},{"updated":"2022-01-19 15:57:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6b8517851557861449673ef5581b03812f074c27","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"date":"2022-01-19 15:44:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"947659b97c11c105857323a150415f787120a38d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-19 15:57:17.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0616cf877f248a5a0d859e7d8eb8a0a : SUCCESS in 6m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3caddcead9bc471c8a7b95c5b31b7057 : SUCCESS in 3m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3220d0d171754896aa1918bc8de63fc4 : SUCCESS in 4m 07s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1e47e3c2bcb04e6a84e8484c3d18543d : SUCCESS in 4m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f438e6e08a814a39a95cdffde7a8fc03 : SUCCESS in 4m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d924070f3aea4dcd9ab41734cbcec76d : SUCCESS in 4m 38s\n- openstack-health-nodejs10-run-lint https://zuul.opendev.org/t/openstack/build/d98675167cfe464492d188de7bbbc037 : FAILURE in 5m 07s\n- openstack-health-nodejs10-run-test-browser https://zuul.opendev.org/t/openstack/build/44b850758c46491bb3081953b267414e : FAILURE in 6m 25s","accounts_in_message":[],"_revision_number":1},{"id":"c1761395855e25d60c649ba5640c42207e9d2f09","author":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"date":"2022-03-10 13:19:19.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0987512b1e802ea3fba2ed69ca0aef5634cd58fd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-10 13:33:02.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b67773b88e654077ad1a6bdd6a7d640e : SUCCESS in 7m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eb662c54b5184a30b3e12f998bdb1281 : SUCCESS in 6m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5930eb55cd374e5fa8811e7ce5e8e020 : SUCCESS in 6m 51s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/db15ebf4fef34280b1dc30c5ea54ee6a : FAILURE in 6m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/39dc2db9bae844ddab9f9bb717846728 : FAILURE in 6m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9352420613214426a72d815ecec5905c : FAILURE in 6m 06s\n- openstack-health-nodejs10-run-lint https://zuul.opendev.org/t/openstack/build/1c062934b3c24943a6e47c35cdd80765 : FAILURE in 6m 26s\n- openstack-health-nodejs10-run-test-browser https://zuul.opendev.org/t/openstack/build/904c53b360044b6da1f9ed6ca3a60a41 : FAILURE in 8m 38s","accounts_in_message":[],"_revision_number":1},{"id":"158f799766f4c15ae82a51e3a40743ef4ff4bd02","tag":"autogenerated:gerrit:abandon","author":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"date":"2022-10-07 09:31:50.000000000","message":"Abandoned\n\nNo longer necessary","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e6a5a432472d8828f3aa2c9ff4a87ce775b52523","revisions":{"e6a5a432472d8828f3aa2c9ff4a87ce775b52523":{"kind":"REWORK","_number":1,"created":"2022-01-19 15:44:07.000000000","uploader":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"ref":"refs/changes/36/825336/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-health","ref":"refs/changes/36/825336/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-health refs/changes/36/825336/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-health refs/changes/36/825336/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-health refs/changes/36/825336/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-health refs/changes/36/825336/1"}}},"commit":{"parents":[{"commit":"a25e4d491ccf43aa5728f287ca00eba5df2ead5d","subject":"Test python3.9 as voting job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-health/commit/a25e4d491ccf43aa5728f287ca00eba5df2ead5d"}]}],"author":{"name":"lpiwowar","email":"lpiwowar@redhat.com","date":"2022-01-19 15:41:12.000000000","tz":60},"committer":{"name":"lpiwowar","email":"lpiwowar@redhat.com","date":"2022-01-19 15:43:00.000000000","tz":60},"subject":"WIP: Fix gulp","message":"WIP: Fix gulp\n\nChange-Id: I368beadca9184654d1727579e7cd9274b53192d6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-health/commit/e6a5a432472d8828f3aa2c9ff4a87ce775b52523"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-health/commit/e6a5a432472d8828f3aa2c9ff4a87ce775b52523"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
