)]}'
{"id":"openstack%2Fopenstack-helm-images~755173","triplet_id":"openstack%2Fopenstack-helm-images~master~I745945b403a5d959d51ac2520fc8ba840242cb17","project":"openstack/openstack-helm-images","branch":"master","hashtags":[],"change_id":"I745945b403a5d959d51ac2520fc8ba840242cb17","subject":"Add periodic build/test/promote for images","status":"MERGED","created":"2020-09-30 03:36:05.000000000","updated":"2020-09-30 16:41:00.000000000","submitted":"2020-09-30 16:30:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"755173-1601483419614-715396bc","meta_rev_id":"26d1165adbfc5df7b02ecab128988759add9c4be","_number":755173,"virtual_id_number":755173,"owner":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":20466,"name":"Tin Lam","email":"tin@lam.wtf","username":"tinlam"},{"value":0,"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},{"value":0,"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},{"value":0,"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-30 16:30:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},"all":[{"value":0,"_account_id":20466,"name":"Tin Lam","email":"tin@lam.wtf","username":"tinlam"},{"value":2,"date":"2020-09-30 16:07:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},{"value":0,"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},{"value":2,"date":"2020-09-30 16:07:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-09-30 15:58:54.000000000","permitted_voting_range":{"min":0,"max":2},"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"all":[{"value":0,"date":"2020-09-30 15:59:47.000000000","_account_id":20466,"name":"Tin Lam","email":"tin@lam.wtf","username":"tinlam"},{"value":0,"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},{"value":0,"date":"2020-09-30 15:59:49.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},{"value":1,"date":"2020-09-30 16:07:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},{"_account_id":20466,"name":"Tin Lam","email":"tin@lam.wtf","username":"tinlam"},{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-30 15:59:47.000000000","updated_by":{"_account_id":20466,"name":"Tin Lam","email":"tin@lam.wtf","username":"tinlam"},"reviewer":{"_account_id":20466,"name":"Tin Lam","email":"tin@lam.wtf","username":"tinlam"},"state":"REVIEWER"},{"updated":"2020-09-30 15:59:49.000000000","updated_by":{"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},"reviewer":{"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},"state":"REVIEWER"},{"updated":"2020-09-30 16:07:33.000000000","updated_by":{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},"reviewer":{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},"state":"REVIEWER"},{"updated":"2020-09-30 16:07:56.000000000","updated_by":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"reviewer":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"state":"REVIEWER"},{"updated":"2020-09-30 16:30:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"35ce5bb6c89ef8aca3fd6caefaf5e9622dfeb2d8","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2020-09-30 03:36:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c7000f3a2d855467b1517cacccfe90dcb79508d3","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2020-09-30 03:37:26.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"f5cad2cbac68797ce81aecc9ae7a49a59b6240bc","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2020-09-30 03:42:07.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"a4ff7ddf73bc7ef8a16fcfad5d78a078bee8e6de","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-30 05:19:14.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-helm-images-build-alerta https://zuul.opendev.org/t/openstack/build/2f781679810146ab89e2e22b11fab3da : SUCCESS in 6m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8344a7457043494083d68e49a2938cea : SUCCESS in 3m 18s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/15c21e21d62d45039fda91d063a491c6 : SUCCESS in 1h 35m 32s\n- openstack-helm-images-aio-monitoring https://zuul.opendev.org/t/openstack/build/975d0cb1347342669075f770bfa66e4a : SUCCESS in 35m 40s\n- openstack-helm-images-aio-logging https://zuul.opendev.org/t/openstack/build/efa19ebe1fd04ac09c7575407bfbbaec : SUCCESS in 34m 22s\n- openstack-helm-images-compute-kit-stein-ubuntu_bionic https://zuul.opendev.org/t/openstack/build/3a01ebbca9e4470eb07b17f937ee402e : SUCCESS in 52m 33s\n- openstack-helm-images-compute-kit-train-ubuntu_bionic https://zuul.opendev.org/t/openstack/build/e2bf33b41d2e4a0c9f33e4c5c7f0bbc3 : SUCCESS in 45m 21s\n- openstack-helm-images-cinder-stein-ubuntu_bionic https://zuul.opendev.org/t/openstack/build/f551dd04ef8044b7a71818b8e6667603 : SUCCESS in 47m 34s\n- openstack-helm-images-cinder-train-ubuntu_bionic https://zuul.opendev.org/t/openstack/build/67bf105d3ea34d15bcad13df50bb21b4 : SUCCESS in 43m 58s\n- openstack-helm-images-horizon-stein-ubuntu_bionic https://zuul.opendev.org/t/openstack/build/2f387f0e0b544bc787f456bd62f304f2 : SUCCESS in 26m 22s\n- openstack-helm-images-build-elasticsearch-s3 https://zuul.opendev.org/t/openstack/build/5cae21b000394e22b84e1463a79cf050 : SUCCESS in 9m 09s\n- openstack-helm-images-build-fluentd https://zuul.opendev.org/t/openstack/build/97906d12b91c42aba0a2863080fb61b4 : SUCCESS in 4m 22s\n- openstack-helm-images-build-libvirt https://zuul.opendev.org/t/openstack/build/9b35922be68d433e8daf0c21b9848034 : SUCCESS in 8m 03s\n- openstack-helm-images-build-mariadb https://zuul.opendev.org/t/openstack/build/68c0aa2bb78b426581272606e56ebf09 : SUCCESS in 5m 33s\n- openstack-helm-images-build-minikube-aio https://zuul.opendev.org/t/openstack/build/278eb3b5fa3a4754917cccb90b5124db : SUCCESS in 11m 15s\n- openstack-helm-images-build-nagios https://zuul.opendev.org/t/openstack/build/41d1752ec5884ebba01d0100bac5d65b : SUCCESS in 7m 27s\n- openstack-helm-images-build-node-problem-detector https://zuul.opendev.org/t/openstack/build/ef2b0647e5784c7887e0305eff953213 : SUCCESS in 3m 48s\n- openstack-helm-images-build-openstack-loci-ocata-ubuntu_xenial https://zuul.opendev.org/t/openstack/build/6649f8d6f4244743a3153df40912b52f : SUCCESS in 57m 00s\n- openstack-helm-images-build-openstack-loci-pike-ubuntu_xenial https://zuul.opendev.org/t/openstack/build/daf569c639f04654a808f7c07a612289 : SUCCESS in 30m 11s\n- openstack-helm-images-build-openstack-loci-queens-ubuntu_xenial https://zuul.opendev.org/t/openstack/build/7a276b2bf1514ded8568189553a3d9f4 : SUCCESS in 31m 07s\n- openstack-helm-images-build-openstack-loci-rocky-ubuntu_xenial https://zuul.opendev.org/t/openstack/build/0ed5fc27735144a4bc5be45f73a8d4de : SUCCESS in 35m 13s\n- openstack-helm-images-build-openstack-loci-rocky-ubuntu_bionic https://zuul.opendev.org/t/openstack/build/5c537948519f429da28300f6d6aa9570 : SUCCESS in 32m 10s\n- openstack-helm-images-build-openstack-loci-stein-ubuntu_bionic https://zuul.opendev.org/t/openstack/build/38954b34b19d42779c8e8130a004c861 : SUCCESS in 30m 38s\n- openstack-helm-images-build-openstack-loci-train-ubuntu_bionic https://zuul.opendev.org/t/openstack/build/890bb3a21f6a4465abff012b7154fafb : SUCCESS in 34m 16s\n- openstack-helm-images-build-openstack-loci-ussuri-ubuntu_bionic https://zuul.opendev.org/t/openstack/build/ea3a6b6747914c27ad4e03745a25b1e0 : SUCCESS in 38m 31s\n- openstack-helm-images-build-openstack-loci-master-ubuntu_bionic https://zuul.opendev.org/t/openstack/build/64e8e8c15d2e4d92aca4fe8a9b25d7d9 : SUCCESS in 32m 05s\n- openstack-helm-images-build-openvswitch https://zuul.opendev.org/t/openstack/build/e5d483922a3a400c8338431217cd77a3 : SUCCESS in 13m 36s\n- openstack-helm-images-build-osh-selenium https://zuul.opendev.org/t/openstack/build/7befb45956684534b8b9bb39640b1594 : SUCCESS in 4m 53s\n- openstack-helm-images-build-ospurge https://zuul.opendev.org/t/openstack/build/f0969656ef2c4843844ca4e0d1e67cb5 : SUCCESS in 5m 33s\n- openstack-helm-images-build-patroni https://zuul.opendev.org/t/openstack/build/bee0f159817d4879a47dfb6aef5ba3f4 : SUCCESS in 5m 18s\n- openstack-helm-images-build-prometheus-openstack-exporter https://zuul.opendev.org/t/openstack/build/a7b803cd77bf48088fc1b2c7f47e8657 : SUCCESS in 4m 18s","accounts_in_message":[],"_revision_number":3},{"id":"999665c48aaed0f19370635c067c421a6e0685f3","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2020-09-30 05:19:44.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"8d7bd105f3c43d21dd8c4a0d82760dc3e3a12746","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-30 05:20:04.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"3eaa643290245beba64541aee999e727490c5481","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2020-09-30 05:20:05.000000000","message":"Patch Set 3: Code-Review-2 Workflow-1","accounts_in_message":[],"_revision_number":3},{"id":"e3682432d40f4a6afa188e01d60fdf5205776278","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-30 05:24:08.000000000","message":"Patch Set 3: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":3},{"id":"c34ae767335718ebae5171563aff557fb1b29004","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2020-09-30 05:25:13.000000000","message":"Patch Set 3: -Code-Review -Workflow\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"56d998d439ab6699a61fce56d65438ffcc4ecb8d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-30 06:50:36.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-helm-images-build-alerta https://zuul.opendev.org/t/openstack/build/8cacdc563dc348d4aa86887be13d5efb : SUCCESS in 6m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a64427cfd1741118c48775f9da231d5 : SUCCESS in 3m 25s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/1b806ea91fad49b0b0543f1ebc62af79 : SUCCESS in 1h 24m 09s\n- openstack-helm-images-aio-monitoring https://zuul.opendev.org/t/openstack/build/263439ca195c4fad8a8b6d89a43d8b9b : SUCCESS in 31m 56s\n- openstack-helm-images-aio-logging https://zuul.opendev.org/t/openstack/build/168b64d7b9844f58aff1a198e8d205d8 : SUCCESS in 30m 19s\n- openstack-helm-images-compute-kit-stein-ubuntu_bionic https://zuul.opendev.org/t/openstack/build/498bc5211afb49b7ba75a6829b416634 : SUCCESS in 43m 09s\n- openstack-helm-images-compute-kit-train-ubuntu_bionic https://zuul.opendev.org/t/openstack/build/d86b7c33b9ea4830a671fc480cf0cea2 : SUCCESS in 41m 19s\n- openstack-helm-images-cinder-stein-ubuntu_bionic https://zuul.opendev.org/t/openstack/build/6962249e3b264c6fac8d89448d85f48c : SUCCESS in 39m 29s\n- openstack-helm-images-cinder-train-ubuntu_bionic https://zuul.opendev.org/t/openstack/build/596888c19da74cf89eabeb00fa85ad3b : SUCCESS in 38m 17s\n- openstack-helm-images-horizon-stein-ubuntu_bionic https://zuul.opendev.org/t/openstack/build/52cd98178bb54be9a06c16f8adf3b5c7 : SUCCESS in 39m 00s\n- openstack-helm-images-build-elasticsearch-s3 https://zuul.opendev.org/t/openstack/build/e01f25505c1d48868a167e700e80cf61 : SUCCESS in 10m 55s\n- openstack-helm-images-build-fluentd https://zuul.opendev.org/t/openstack/build/7511efdc418e447089a67fa3a20922e3 : SUCCESS in 6m 10s\n- openstack-helm-images-build-libvirt https://zuul.opendev.org/t/openstack/build/d502977abe2a4784aa0e9a0dd71a2cab : SUCCESS in 9m 09s\n- openstack-helm-images-build-mariadb https://zuul.opendev.org/t/openstack/build/41cf2a8f47414af5a9149dddfc4e2591 : SUCCESS in 5m 22s\n- openstack-helm-images-build-minikube-aio https://zuul.opendev.org/t/openstack/build/6f8d0c3cf5364e0f87a28d62eea4dba4 : SUCCESS in 10m 15s\n- openstack-helm-images-build-nagios https://zuul.opendev.org/t/openstack/build/06500a54d9854d629e8d337fb741ec0b : SUCCESS in 6m 31s\n- openstack-helm-images-build-node-problem-detector https://zuul.opendev.org/t/openstack/build/04f8184db2c641abad776a9ad39550b1 : SUCCESS in 6m 08s\n- openstack-helm-images-build-openstack-loci-ocata-ubuntu_xenial https://zuul.opendev.org/t/openstack/build/f72da13b05694efd9a7924ff8cd64deb : SUCCESS in 44m 35s\n- openstack-helm-images-build-openstack-loci-pike-ubuntu_xenial https://zuul.opendev.org/t/openstack/build/1ce6dcf7ba994a10a579fd3c066126ec : SUCCESS in 38m 20s\n- openstack-helm-images-build-openstack-loci-queens-ubuntu_xenial https://zuul.opendev.org/t/openstack/build/3e513942813e4f1483fe792e2a99b8c5 : SUCCESS in 1h 13m 34s\n- openstack-helm-images-build-openstack-loci-rocky-ubuntu_xenial https://zuul.opendev.org/t/openstack/build/caea0d1f0b144a01935337731bc4e378 : SUCCESS in 57m 24s\n- openstack-helm-images-build-openstack-loci-rocky-ubuntu_bionic https://zuul.opendev.org/t/openstack/build/4e7d360e4c464b28ae246fd7185d0fbb : SUCCESS in 42m 42s\n- openstack-helm-images-build-openstack-loci-stein-ubuntu_bionic https://zuul.opendev.org/t/openstack/build/ac467124812d42c19e0d0efeec5ff2cb : SUCCESS in 26m 20s\n- openstack-helm-images-build-openstack-loci-train-ubuntu_bionic https://zuul.opendev.org/t/openstack/build/bd19b7eb46bf4864acdc77814005dc6e : SUCCESS in 33m 31s\n- openstack-helm-images-build-openstack-loci-ussuri-ubuntu_bionic https://zuul.opendev.org/t/openstack/build/4d24cb4332b441e2ab1d91c34d2a616f : SUCCESS in 35m 32s\n- openstack-helm-images-build-openstack-loci-master-ubuntu_bionic https://zuul.opendev.org/t/openstack/build/6b53cb46b4084e65b1ae61e0eb60859d : SUCCESS in 1h 02m 39s\n- openstack-helm-images-build-openvswitch https://zuul.opendev.org/t/openstack/build/5f125efc70414d418cc885647466d0e8 : SUCCESS in 13m 08s\n- openstack-helm-images-build-osh-selenium https://zuul.opendev.org/t/openstack/build/f69decd9d13449f39b56c88d942dbd40 : SUCCESS in 5m 08s\n- openstack-helm-images-build-ospurge https://zuul.opendev.org/t/openstack/build/4471571711fa4b7385b7080ac9313e62 : SUCCESS in 6m 36s\n- openstack-helm-images-build-patroni https://zuul.opendev.org/t/openstack/build/ea4dc0be1cbd43f9af4d7d93ef82ebe1 : SUCCESS in 5m 43s\n- openstack-helm-images-build-prometheus-openstack-exporter https://zuul.opendev.org/t/openstack/build/caa3487bd8664f3ab952700883fa7958 : SUCCESS in 4m 51s","accounts_in_message":[],"_revision_number":3},{"id":"4bf4c5c29749b1c806178a08c4d27b0f0016541a","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2020-09-30 14:27:34.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"06612ee20e923a2ce7c5f6650158b1420cb99f38","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-30 14:27:49.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"5092ffaf3f3de4eded08f2e9ec979d70d0d298e2","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2020-09-30 14:27:52.000000000","message":"Patch Set 3: Code-Review-2 Workflow-1","accounts_in_message":[],"_revision_number":3},{"id":"752c8db1759c05d1c350e3ed7d50f6ac3b0a65ef","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2020-09-30 15:58:54.000000000","message":"Patch Set 3: -Code-Review -Workflow","accounts_in_message":[],"_revision_number":3},{"id":"44a844b0062945e0d8a22cc9bf09fe19cbad017f","author":{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},"date":"2020-09-30 16:07:33.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"3e41a1a55a2b8c7b842c1df4b5f25cba177c35b3","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2020-09-30 16:07:56.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"b7e340121d6599a83c962866b8dd71bf3cb1d2c7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-30 16:30:19.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"ccbb848cd09471e14998be3b18a9dea77a292eb0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-30 16:30:19.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-helm-images-upload-alerta https://zuul.opendev.org/t/openstack/build/4d5c0e084885477f9979f814727f5586 : SUCCESS in 52m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/baf733cc5b5844449087a7874eb6026b : SUCCESS in 5m 49s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/e04be26dc38a44ebbf6e3dcf4fb5197a : SUCCESS in 1h 56m 07s\n- openstack-helm-images-aio-monitoring https://zuul.opendev.org/t/openstack/build/c941489510ec4ea9a5fbe741f0740c6f : SUCCESS in 31m 08s\n- openstack-helm-images-aio-logging https://zuul.opendev.org/t/openstack/build/ea854ee441c04c21a2519166ce2a1b50 : SUCCESS in 33m 00s\n- openstack-helm-images-compute-kit-stein-ubuntu_bionic https://zuul.opendev.org/t/openstack/build/41a595c4b56d4850a6ee75cdc0b52cf4 : SUCCESS in 51m 18s\n- openstack-helm-images-compute-kit-train-ubuntu_bionic https://zuul.opendev.org/t/openstack/build/fa3d6bbd19fa4361b6fd5659afcd8220 : SUCCESS in 49m 25s\n- openstack-helm-images-cinder-stein-ubuntu_bionic https://zuul.opendev.org/t/openstack/build/7c63457dc2154cf48960548879fac84d : SUCCESS in 43m 12s\n- openstack-helm-images-cinder-train-ubuntu_bionic https://zuul.opendev.org/t/openstack/build/868cf523a7c849d6831e17ea839461f4 : SUCCESS in 44m 05s\n- openstack-helm-images-horizon-stein-ubuntu_bionic https://zuul.opendev.org/t/openstack/build/108062c4e00c4d46b9002ae339b3db6c : SUCCESS in 24m 34s\n- openstack-helm-images-upload-elasticsearch-s3 https://zuul.opendev.org/t/openstack/build/ee72c38f20c6438faac79e12efbcb5e1 : SUCCESS in 53m 40s\n- openstack-helm-images-upload-fluentd https://zuul.opendev.org/t/openstack/build/eb9fdf855c7c493dbc8f90afe458db4b : SUCCESS in 53m 34s\n- openstack-helm-images-upload-libvirt https://zuul.opendev.org/t/openstack/build/77ee03451a9c472a811759c55c129e53 : SUCCESS in 1h 41m 59s\n- openstack-helm-images-upload-mariadb https://zuul.opendev.org/t/openstack/build/aea2874878a449f88ecab651ed0a8d28 : SUCCESS in 1h 39m 39s\n- openstack-helm-images-upload-minikube-aio https://zuul.opendev.org/t/openstack/build/673cdc61c6d14cada51f256fec9e5cc8 : SUCCESS in 1h 41m 51s\n- openstack-helm-images-upload-nagios https://zuul.opendev.org/t/openstack/build/00e45d90f6dd4d5191beb8734d412435 : SUCCESS in 51m 38s\n- openstack-helm-images-upload-node-problem-detector https://zuul.opendev.org/t/openstack/build/61d8d97bd5de43fca735440c443f8c05 : SUCCESS in 49m 36s\n- openstack-helm-images-upload-openstack-loci-ocata-ubuntu_xenial https://zuul.opendev.org/t/openstack/build/ee637d9a282b4b69a9d4d96eb9f80580 : SUCCESS in 55m 28s\n- openstack-helm-images-upload-openstack-loci-pike-ubuntu_xenial https://zuul.opendev.org/t/openstack/build/14c1de6380e841169776f2bc0dac7dc1 : SUCCESS in 42m 01s\n- openstack-helm-images-upload-openstack-loci-queens-ubuntu_xenial https://zuul.opendev.org/t/openstack/build/2d602ccfa82e4eceb1a2d641885a90dd : SUCCESS in 44m 06s\n- openstack-helm-images-upload-openstack-loci-rocky-ubuntu_xenial https://zuul.opendev.org/t/openstack/build/a360551f432a4c2ea0b40a3da852834a : SUCCESS in 41m 10s\n- openstack-helm-images-upload-openstack-loci-rocky-ubuntu_bionic https://zuul.opendev.org/t/openstack/build/fdc29cffff3c4b83b25728ad52f32169 : SUCCESS in 40m 20s\n- openstack-helm-images-upload-openstack-loci-stein-ubuntu_bionic https://zuul.opendev.org/t/openstack/build/cab978d6c43f4ce9861be316442c33a2 : SUCCESS in 1h 39m 33s\n- openstack-helm-images-upload-openstack-loci-train-ubuntu_bionic https://zuul.opendev.org/t/openstack/build/46b827e0bf0c4a6abb63f288523c3486 : SUCCESS in 1h 42m 06s\n- openstack-helm-images-upload-openstack-loci-ussuri-ubuntu_bionic https://zuul.opendev.org/t/openstack/build/5ffa6fa154c645e29e3b7eeb5b238986 : SUCCESS in 57m 51s\n- openstack-helm-images-upload-openstack-loci-master-ubuntu_bionic https://zuul.opendev.org/t/openstack/build/753125b79b7c43dcb2869e28bb6e6335 : SUCCESS in 41m 10s\n- openstack-helm-images-upload-openvswitch https://zuul.opendev.org/t/openstack/build/b5688ef4b2224ae0a96ecf4acd85a35c : SUCCESS in 1h 36m 57s\n- openstack-helm-images-upload-osh-selenium https://zuul.opendev.org/t/openstack/build/959020ed03e240e8ad6fd74f6e0657e8 : SUCCESS in 1h 09m 34s\n- openstack-helm-images-upload-ospurge https://zuul.opendev.org/t/openstack/build/0f472146459a443c9843c578cf68d284 : SUCCESS in 1h 31m 16s\n- openstack-helm-images-upload-patroni https://zuul.opendev.org/t/openstack/build/b543ca4e9a9a47d7bfd5ad27cf334d60 : SUCCESS in 44m 53s\n- openstack-helm-images-upload-prometheus-openstack-exporter https://zuul.opendev.org/t/openstack/build/bbf81015b93b485b89d2adacf34ee594 : SUCCESS in 44m 17s","accounts_in_message":[],"_revision_number":3},{"id":"26d1165adbfc5df7b02ecab128988759add9c4be","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-30 16:41:00.000000000","message":"Patch Set 3:\n\nBuild failed (promote pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-helm-images-promote-alerta https://zuul.opendev.org/t/openstack/build/2c8a5159d2dd490db8ff7fb8af182ee4 : SUCCESS in 59s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e211df986d540d1a68a8ab4a2ecaee8 : SUCCESS in 1m 03s\n- openstack-helm-images-promote-elasticsearch-s3 https://zuul.opendev.org/t/openstack/build/7e84eb5a78214cfda3ebcbc595fb297f : SUCCESS in 1m 42s\n- openstack-helm-images-promote-fluentd https://zuul.opendev.org/t/openstack/build/d68719e712244197801c2b81823abe1b : SUCCESS in 1m 21s\n- openstack-helm-images-promote-libvirt https://zuul.opendev.org/t/openstack/build/2ed23509a39541dfb08ae248fd01f774 : SUCCESS in 1m 38s\n- openstack-helm-images-promote-mariadb https://zuul.opendev.org/t/openstack/build/378962c61af4490cb3766f0f18fc39b4 : SUCCESS in 1m 19s\n- openstack-helm-images-promote-minikube-aio https://zuul.opendev.org/t/openstack/build/2dfba628f4ff4a8a92165ee9d10674de : SUCCESS in 1m 11s\n- openstack-helm-images-promote-nagios https://zuul.opendev.org/t/openstack/build/8b64b738bb9b4ca990fc245a32a41522 : SUCCESS in 1m 12s\n- openstack-helm-images-promote-node-problem-detector https://zuul.opendev.org/t/openstack/build/86715f60408a48ea9ca5351dfad91a0b : SUCCESS in 1m 27s\n- openstack-helm-images-promote-openstack-loci-ocata-ubuntu_xenial https://zuul.opendev.org/t/openstack/build/e1267dcd76e24d288e51bace696a79b6 : SUCCESS in 5m 44s\n- openstack-helm-images-promote-openstack-loci-pike-ubuntu_xenial https://zuul.opendev.org/t/openstack/build/df4b799cc40945479de56c127e065c84 : SUCCESS in 5m 09s\n- openstack-helm-images-promote-openstack-loci-queens-ubuntu_xenial https://zuul.opendev.org/t/openstack/build/d5f13ff505d84d08b5bcf9e5e5713d52 : SUCCESS in 5m 10s\n- openstack-helm-images-promote-openstack-loci-rocky-ubuntu_xenial https://zuul.opendev.org/t/openstack/build/2c1b535befcd4800a5e4bf9c54c46d74 : SUCCESS in 5m 09s\n- openstack-helm-images-promote-openstack-loci-rocky-ubuntu_bionic https://zuul.opendev.org/t/openstack/build/8ea02b34341a4bbfbfee9719123faa50 : SUCCESS in 5m 17s\n- openstack-helm-images-promote-openstack-loci-stein-ubuntu_bionic https://zuul.opendev.org/t/openstack/build/661895539d5d43be8b6ddf32f033f728 : SUCCESS in 5m 09s\n- openstack-helm-images-promote-openstack-loci-train-ubuntu_bionic https://zuul.opendev.org/t/openstack/build/aa5369f5a2594d8bbd956459aec68ecc : SUCCESS in 5m 17s\n- openstack-helm-images-promote-openstack-loci-ussuri-ubuntu_bionic https://zuul.opendev.org/t/openstack/build/f7b6f4051ad541b692743948acfb5f5b : SUCCESS in 5m 19s\n- openstack-helm-images-promote-openstack-loci-master-ubuntu_bionic https://zuul.opendev.org/t/openstack/build/eb24fc793d8542f797949412e99bd40e : SUCCESS in 5m 26s\n- openstack-helm-images-promote-openvswitch https://zuul.opendev.org/t/openstack/build/2929e732f1ac486cba7b9adae602f39b : FAILURE in 1m 27s\n- openstack-helm-images-promote-osh-selenium https://zuul.opendev.org/t/openstack/build/1da14a23f58e4eabadc67c0aa2ad401f : SUCCESS in 1m 07s\n- openstack-helm-images-promote-ospurge https://zuul.opendev.org/t/openstack/build/ca7c2436f156461a96ad8dad22fe2901 : SUCCESS in 1m 07s\n- openstack-helm-images-promote-patroni https://zuul.opendev.org/t/openstack/build/293b5fc1b9994eaab82620cd2bb48951 : SUCCESS in 1m 16s\n- openstack-helm-images-promote-prometheus-openstack-exporter https://zuul.opendev.org/t/openstack/build/cc759031a8624653b340357fc41da502 : SUCCESS in 1m 15s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"2e02a617511246e04746919d077a232835f13d90","revisions":{"a362488391eb9176a125e061b1043d3fb7856311":{"kind":"REWORK","_number":1,"created":"2020-09-30 03:36:05.000000000","uploader":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"ref":"refs/changes/73/755173/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-helm-images","ref":"refs/changes/73/755173/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-helm-images refs/changes/73/755173/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-helm-images refs/changes/73/755173/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-helm-images refs/changes/73/755173/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-helm-images refs/changes/73/755173/1"}}},"commit":{"parents":[{"commit":"39f967781e699a7f70922ab589758356abb1409d","subject":"Brings back post-review buildset registry based image testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-images/commit/39f967781e699a7f70922ab589758356abb1409d"}]}],"author":{"name":"Andrii Ostapenko","email":"andrii.ostapenko@att.com","date":"2020-09-30 03:28:35.000000000","tz":-300},"committer":{"name":"Andrii Ostapenko","email":"andrii.ostapenko@att.com","date":"2020-09-30 03:28:35.000000000","tz":-300},"subject":"Add periodic build/test/promote for loci images","message":"Add periodic build/test/promote for loci images\n\nThis fuctionality allows to keep \u0027latest\u0027 images up to date having nightly\nbuilding/testing(Stein and Train only)/promotion for loci images.\n\nChange-Id: I745945b403a5d959d51ac2520fc8ba840242cb17\nSigned-off-by: Andrii Ostapenko \u003candrii.ostapenko@att.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-images/commit/a362488391eb9176a125e061b1043d3fb7856311"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-images/commit/a362488391eb9176a125e061b1043d3fb7856311"}]},"branch":"refs/heads/master"},"1f3995e1dbdfd9dbf779e9945894a025ed9f5d2a":{"kind":"REWORK","_number":2,"created":"2020-09-30 03:37:26.000000000","uploader":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"ref":"refs/changes/73/755173/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-helm-images","ref":"refs/changes/73/755173/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-helm-images refs/changes/73/755173/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-helm-images refs/changes/73/755173/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-helm-images refs/changes/73/755173/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-helm-images refs/changes/73/755173/2"}}},"commit":{"parents":[{"commit":"39f967781e699a7f70922ab589758356abb1409d","subject":"Brings back post-review buildset registry based image testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-images/commit/39f967781e699a7f70922ab589758356abb1409d"}]}],"author":{"name":"Andrii Ostapenko","email":"andrii.ostapenko@att.com","date":"2020-09-30 03:28:35.000000000","tz":-300},"committer":{"name":"Andrii Ostapenko","email":"anost1986@gmail.com","date":"2020-09-30 03:37:24.000000000","tz":0},"subject":"Add periodic build/test/promote for loci images","message":"Add periodic build/test/promote for loci images\n\nThis fuctionality allows to keep \u0027latest\u0027 images up to date having nightly\nbuilding/testing(Stein and Train only)/promotion for loci images.\n\nChange-Id: I745945b403a5d959d51ac2520fc8ba840242cb17\nSigned-off-by: Andrii Ostapenko \u003candrii.ostapenko@att.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-images/commit/1f3995e1dbdfd9dbf779e9945894a025ed9f5d2a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-images/commit/1f3995e1dbdfd9dbf779e9945894a025ed9f5d2a"}]},"branch":"refs/heads/master"},"2e02a617511246e04746919d077a232835f13d90":{"kind":"REWORK","_number":3,"created":"2020-09-30 03:42:07.000000000","uploader":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"ref":"refs/changes/73/755173/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-helm-images","ref":"refs/changes/73/755173/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-helm-images refs/changes/73/755173/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-helm-images refs/changes/73/755173/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-helm-images refs/changes/73/755173/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-helm-images refs/changes/73/755173/3"}}},"commit":{"parents":[{"commit":"39f967781e699a7f70922ab589758356abb1409d","subject":"Brings back post-review buildset registry based image testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-images/commit/39f967781e699a7f70922ab589758356abb1409d"}]}],"author":{"name":"Andrii Ostapenko","email":"andrii.ostapenko@att.com","date":"2020-09-30 03:28:35.000000000","tz":-300},"committer":{"name":"Andrii Ostapenko","email":"andrii.ostapenko@att.com","date":"2020-09-30 03:41:37.000000000","tz":-300},"subject":"Add periodic build/test/promote for images","message":"Add periodic build/test/promote for images\n\nThis fuctionality allows to keep \u0027latest\u0027 images up to date having nightly\nbuilding/testing/promotion for images.\n\nChange-Id: I745945b403a5d959d51ac2520fc8ba840242cb17\nSigned-off-by: Andrii Ostapenko \u003candrii.ostapenko@att.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-images/commit/2e02a617511246e04746919d077a232835f13d90"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-images/commit/2e02a617511246e04746919d077a232835f13d90"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
