)]}'
{"id":"openstack%2Fopenstack-helm-infra~657692","triplet_id":"openstack%2Fopenstack-helm-infra~master~I2cf271593591ce07435893336cff98a8b1c72166","project":"openstack/openstack-helm-infra","branch":"master","topic":"mon-upgrade","hashtags":[],"change_id":"I2cf271593591ce07435893336cff98a8b1c72166","subject":"Ceph: Dont attempt to reuse monmap if it exists","status":"MERGED","created":"2019-05-07 21:44:30.000000000","updated":"2019-05-15 16:35:16.000000000","submitted":"2019-05-15 16:35:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"657692-1557938116916-ec554898","meta_rev_id":"dab54f05459842e8642b9f859c2c732e0d576c44","_number":657692,"virtual_id_number":657692,"owner":{"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":20466,"name":"Tin Lam","email":"tin@lam.wtf","username":"tinlam"},{"value":0,"_account_id":28372,"name":"chinasubbareddy mallavarapu","email":"cr3938@att.com","username":"chinasubbareddy"},{"value":0,"_account_id":24165,"name":"Renis Makadia","email":"renis.makadia@att.com","username":"renmak"},{"value":0,"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},{"value":0,"_account_id":7769,"name":"Pentheus","display_name":"Alan Meadows","email":"alan.meadows@gmail.com","username":"alanmeadows"},{"value":2,"date":"2019-05-15 16:35:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29132,"name":"Krishna Venkata","email":"kvenkata986@gmail.com","username":"kvenkata986"},{"value":0,"_account_id":22477,"name":"Matt McEuen","email":"matt.mceuen@att.com","username":"mattmceuen"},{"value":0,"_account_id":8749,"name":"Evgeniy L","email":"eli@mirantis.com","username":"evgeniyl"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":20466,"name":"Tin Lam","email":"tin@lam.wtf","username":"tinlam"},"all":[{"value":2,"date":"2019-05-12 16:21:43.000000000","_account_id":20466,"name":"Tin Lam","email":"tin@lam.wtf","username":"tinlam"},{"value":1,"date":"2019-05-12 16:21:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28372,"name":"chinasubbareddy mallavarapu","email":"cr3938@att.com","username":"chinasubbareddy"},{"value":1,"date":"2019-05-14 18:15:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":24165,"name":"Renis Makadia","email":"renis.makadia@att.com","username":"renmak"},{"value":0,"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},{"value":0,"_account_id":7769,"name":"Pentheus","display_name":"Alan Meadows","email":"alan.meadows@gmail.com","username":"alanmeadows"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-05-12 16:21:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29132,"name":"Krishna Venkata","email":"kvenkata986@gmail.com","username":"kvenkata986"},{"value":2,"date":"2019-05-15 14:38:19.000000000","_account_id":22477,"name":"Matt McEuen","email":"matt.mceuen@att.com","username":"mattmceuen"},{"value":1,"date":"2019-05-12 16:21:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8749,"name":"Evgeniy L","email":"eli@mirantis.com","username":"evgeniyl"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22477,"name":"Matt McEuen","email":"matt.mceuen@att.com","username":"mattmceuen"},"all":[{"value":0,"_account_id":20466,"name":"Tin Lam","email":"tin@lam.wtf","username":"tinlam"},{"value":0,"_account_id":28372,"name":"chinasubbareddy mallavarapu","email":"cr3938@att.com","username":"chinasubbareddy"},{"value":0,"_account_id":24165,"name":"Renis Makadia","email":"renis.makadia@att.com","username":"renmak"},{"value":0,"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},{"value":0,"_account_id":7769,"name":"Pentheus","display_name":"Alan Meadows","email":"alan.meadows@gmail.com","username":"alanmeadows"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29132,"name":"Krishna Venkata","email":"kvenkata986@gmail.com","username":"kvenkata986"},{"value":1,"date":"2019-05-15 14:38:19.000000000","_account_id":22477,"name":"Matt McEuen","email":"matt.mceuen@att.com","username":"mattmceuen"},{"value":0,"_account_id":8749,"name":"Evgeniy L","email":"eli@mirantis.com","username":"evgeniyl"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7769,"name":"Pentheus","display_name":"Alan Meadows","email":"alan.meadows@gmail.com","username":"alanmeadows"},{"_account_id":8749,"name":"Evgeniy L","email":"eli@mirantis.com","username":"evgeniyl"},{"_account_id":20466,"name":"Tin Lam","email":"tin@lam.wtf","username":"tinlam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22477,"name":"Matt McEuen","email":"matt.mceuen@att.com","username":"mattmceuen"},{"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},{"_account_id":24165,"name":"Renis Makadia","email":"renis.makadia@att.com","username":"renmak"},{"_account_id":28372,"name":"chinasubbareddy mallavarapu","email":"cr3938@att.com","username":"chinasubbareddy"},{"_account_id":29132,"name":"Krishna Venkata","email":"kvenkata986@gmail.com","username":"kvenkata986"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-07 22:08:37.000000000","updated_by":{"_account_id":7769,"name":"Pentheus","display_name":"Alan Meadows","email":"alan.meadows@gmail.com","username":"alanmeadows"},"reviewer":{"_account_id":7769,"name":"Pentheus","display_name":"Alan Meadows","email":"alan.meadows@gmail.com","username":"alanmeadows"},"state":"REVIEWER"},{"updated":"2019-05-09 16:44:01.000000000","updated_by":{"_account_id":29132,"name":"Krishna Venkata","email":"kvenkata986@gmail.com","username":"kvenkata986"},"reviewer":{"_account_id":29132,"name":"Krishna Venkata","email":"kvenkata986@gmail.com","username":"kvenkata986"},"state":"REVIEWER"},{"updated":"2019-05-09 22:01:08.000000000","updated_by":{"_account_id":28372,"name":"chinasubbareddy mallavarapu","email":"cr3938@att.com","username":"chinasubbareddy"},"reviewer":{"_account_id":28372,"name":"chinasubbareddy mallavarapu","email":"cr3938@att.com","username":"chinasubbareddy"},"state":"REVIEWER"},{"updated":"2019-05-10 16:06:49.000000000","updated_by":{"_account_id":8749,"name":"Evgeniy L","email":"eli@mirantis.com","username":"evgeniyl"},"reviewer":{"_account_id":8749,"name":"Evgeniy L","email":"eli@mirantis.com","username":"evgeniyl"},"state":"REVIEWER"},{"updated":"2019-05-12 16:21:43.000000000","updated_by":{"_account_id":20466,"name":"Tin Lam","email":"tin@lam.wtf","username":"tinlam"},"reviewer":{"_account_id":20466,"name":"Tin Lam","email":"tin@lam.wtf","username":"tinlam"},"state":"REVIEWER"},{"updated":"2019-05-14 18:15:35.000000000","updated_by":{"_account_id":24165,"name":"Renis Makadia","email":"renis.makadia@att.com","username":"renmak"},"reviewer":{"_account_id":24165,"name":"Renis Makadia","email":"renis.makadia@att.com","username":"renmak"},"state":"REVIEWER"},{"updated":"2019-05-15 14:38:19.000000000","updated_by":{"_account_id":22477,"name":"Matt McEuen","email":"matt.mceuen@att.com","username":"mattmceuen"},"reviewer":{"_account_id":22477,"name":"Matt McEuen","email":"matt.mceuen@att.com","username":"mattmceuen"},"state":"REVIEWER"},{"updated":"2019-05-15 16:35:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0e49950d34efc7acf43671b553351d946388a9f3","author":{"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},"date":"2019-05-07 21:44:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"93ba72e852a415b9cc503d1f10c5634d51afa7d5","author":{"_account_id":22477,"name":"Matt McEuen","email":"matt.mceuen@att.com","username":"mattmceuen"},"date":"2019-05-07 21:53:55.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"413771b4b2576858c6023b48206866fe60ac7725","author":{"_account_id":7769,"name":"Pentheus","display_name":"Alan Meadows","email":"alan.meadows@gmail.com","username":"alanmeadows"},"date":"2019-05-07 22:08:37.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8d834d58847946da3efd81d72cee094e520da972","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-07 22:37:41.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/92/657692/1/check/openstack-tox-docs/a50401d/html/ : SUCCESS in 5m 43s\n- openstack-helm-lint http://logs.openstack.org/92/657692/1/check/openstack-helm-lint/00a9c84/ : SUCCESS in 2m 20s\n- openstack-helm-infra-aio-logging http://logs.openstack.org/92/657692/1/check/openstack-helm-infra-aio-logging/ae584cd/ : SUCCESS in 37m 29s\n- openstack-helm-infra-aio-monitoring http://logs.openstack.org/92/657692/1/check/openstack-helm-infra-aio-monitoring/0f365fd/ : FAILURE in 48m 04s\n- openstack-helm-infra-aio-network-policy http://logs.openstack.org/92/657692/1/check/openstack-helm-infra-aio-network-policy/1e605e9/ : FAILURE in 30m 52s (non-voting)\n- openstack-helm-infra-openstack-support http://logs.openstack.org/92/657692/1/check/openstack-helm-infra-openstack-support/96daed1/ : SUCCESS in 37m 30s\n- openstack-helm-infra-kubernetes-keystone-auth http://logs.openstack.org/92/657692/1/check/openstack-helm-infra-kubernetes-keystone-auth/3fcb6ab/ : SUCCESS in 37m 38s\n- openstack-helm-infra-airship-divingbell http://logs.openstack.org/92/657692/1/check/openstack-helm-infra-airship-divingbell/4d1c3d4/ : FAILURE in 9m 08s (non-voting)\n- openstack-helm-infra-aio-podsecuritypolicy http://logs.openstack.org/92/657692/1/check/openstack-helm-infra-aio-podsecuritypolicy/27bd482/ : FAILURE in 13m 50s (non-voting)\n- openstack-helm-infra-apparmor http://logs.openstack.org/92/657692/1/check/openstack-helm-infra-apparmor/6ece912/ : SUCCESS in 27m 08s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"43c88a45685fc0c1b21f9ed8c7c1cc0ffe0faff4","author":{"_account_id":29132,"name":"Krishna Venkata","email":"kvenkata986@gmail.com","username":"kvenkata986"},"date":"2019-05-08 15:23:15.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"98b92446fae963dfba73542644911ec93481c99e","author":{"_account_id":28372,"name":"chinasubbareddy mallavarapu","email":"cr3938@att.com","username":"chinasubbareddy"},"date":"2019-05-08 15:40:29.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"1c181a5d339e0a34597571f9db647167b735c07c","author":{"_account_id":24165,"name":"Renis Makadia","email":"renis.makadia@att.com","username":"renmak"},"date":"2019-05-08 16:32:25.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"6ed6785614586435a796d191363a2f6af206e819","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-08 16:44:02.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/92/657692/2/check/openstack-tox-docs/e49895b/html/ : SUCCESS in 3m 28s\n- openstack-helm-lint http://logs.openstack.org/92/657692/2/check/openstack-helm-lint/ec74eca/ : SUCCESS in 2m 09s\n- openstack-helm-infra-aio-logging http://logs.openstack.org/92/657692/2/check/openstack-helm-infra-aio-logging/7ad6299/ : SUCCESS in 40m 29s\n- openstack-helm-infra-aio-monitoring http://logs.openstack.org/92/657692/2/check/openstack-helm-infra-aio-monitoring/b0fb984/ : FAILURE in 32m 26s\n- openstack-helm-infra-aio-network-policy http://logs.openstack.org/92/657692/2/check/openstack-helm-infra-aio-network-policy/fd59810/ : FAILURE in 29m 09s (non-voting)\n- openstack-helm-infra-openstack-support http://logs.openstack.org/92/657692/2/check/openstack-helm-infra-openstack-support/1df66b4/ : SUCCESS in 37m 12s\n- openstack-helm-infra-kubernetes-keystone-auth http://logs.openstack.org/92/657692/2/check/openstack-helm-infra-kubernetes-keystone-auth/efb9b9a/ : SUCCESS in 37m 44s\n- openstack-helm-infra-airship-divingbell http://logs.openstack.org/92/657692/2/check/openstack-helm-infra-airship-divingbell/68d1414/ : SUCCESS in 10m 51s (non-voting)\n- openstack-helm-infra-aio-podsecuritypolicy http://logs.openstack.org/92/657692/2/check/openstack-helm-infra-aio-podsecuritypolicy/492be22/ : FAILURE in 13m 17s (non-voting)\n- openstack-helm-infra-apparmor http://logs.openstack.org/92/657692/2/check/openstack-helm-infra-apparmor/854ac9f/ : SUCCESS in 29m 50s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"6632d7cd8b159d164a3b2b9b2ac02008d4f23713","author":{"_account_id":29132,"name":"Krishna Venkata","email":"kvenkata986@gmail.com","username":"kvenkata986"},"date":"2019-05-08 21:51:58.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"588a563f627dd06fb14df12178a085c2681d7eec","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-08 22:55:25.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/92/657692/3/check/openstack-tox-docs/76aef97/html/ : SUCCESS in 6m 30s\n- openstack-helm-lint http://logs.openstack.org/92/657692/3/check/openstack-helm-lint/66cbe4d/ : SUCCESS in 1m 38s\n- openstack-helm-infra-aio-logging http://logs.openstack.org/92/657692/3/check/openstack-helm-infra-aio-logging/f828f84/ : SUCCESS in 44m 27s\n- openstack-helm-infra-aio-monitoring http://logs.openstack.org/92/657692/3/check/openstack-helm-infra-aio-monitoring/aca801b/ : SUCCESS in 1h 02m 35s\n- openstack-helm-infra-aio-network-policy http://logs.openstack.org/92/657692/3/check/openstack-helm-infra-aio-network-policy/9149f11/ : FAILURE in 28m 59s (non-voting)\n- openstack-helm-infra-openstack-support http://logs.openstack.org/92/657692/3/check/openstack-helm-infra-openstack-support/e88633d/ : FAILURE in 53m 45s\n- openstack-helm-infra-kubernetes-keystone-auth http://logs.openstack.org/92/657692/3/check/openstack-helm-infra-kubernetes-keystone-auth/bfed488/ : SUCCESS in 36m 42s\n- openstack-helm-infra-airship-divingbell http://logs.openstack.org/92/657692/3/check/openstack-helm-infra-airship-divingbell/37737d1/ : SUCCESS in 10m 23s (non-voting)\n- openstack-helm-infra-aio-podsecuritypolicy http://logs.openstack.org/92/657692/3/check/openstack-helm-infra-aio-podsecuritypolicy/aadb102/ : SUCCESS in 13m 12s (non-voting)\n- openstack-helm-infra-apparmor http://logs.openstack.org/92/657692/3/check/openstack-helm-infra-apparmor/4ffb66e/ : SUCCESS in 27m 56s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"b2cde8c41df54b1a0a9e2a72126a67ea86ba75b6","author":{"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},"date":"2019-05-09 15:24:34.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"56ebbafac3b3b47d8c82b4baf0b88f39030be24f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-09 16:09:16.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/92/657692/3/check/openstack-tox-docs/bff506f/html/ : SUCCESS in 3m 50s\n- openstack-helm-lint http://logs.openstack.org/92/657692/3/check/openstack-helm-lint/a5b246a/ : SUCCESS in 1m 36s\n- openstack-helm-infra-aio-logging http://logs.openstack.org/92/657692/3/check/openstack-helm-infra-aio-logging/d8df9d6/ : SUCCESS in 37m 50s\n- openstack-helm-infra-aio-monitoring http://logs.openstack.org/92/657692/3/check/openstack-helm-infra-aio-monitoring/25e266d/ : SUCCESS in 35m 52s\n- openstack-helm-infra-aio-network-policy http://logs.openstack.org/92/657692/3/check/openstack-helm-infra-aio-network-policy/b6f5bdc/ : FAILURE in 29m 03s (non-voting)\n- openstack-helm-infra-openstack-support http://logs.openstack.org/92/657692/3/check/openstack-helm-infra-openstack-support/c04239a/ : SUCCESS in 38m 04s\n- openstack-helm-infra-kubernetes-keystone-auth http://logs.openstack.org/92/657692/3/check/openstack-helm-infra-kubernetes-keystone-auth/00b73f1/ : SUCCESS in 37m 34s\n- openstack-helm-infra-airship-divingbell http://logs.openstack.org/92/657692/3/check/openstack-helm-infra-airship-divingbell/cc74012/ : SUCCESS in 10m 51s (non-voting)\n- openstack-helm-infra-aio-podsecuritypolicy http://logs.openstack.org/92/657692/3/check/openstack-helm-infra-aio-podsecuritypolicy/e500219/ : FAILURE in 12m 28s (non-voting)\n- openstack-helm-infra-apparmor http://logs.openstack.org/92/657692/3/check/openstack-helm-infra-apparmor/d98acd3/ : SUCCESS in 27m 17s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"3eb4a57a2eaf973026dc20611de707183a5689f9","author":{"_account_id":29132,"name":"Krishna Venkata","email":"kvenkata986@gmail.com","username":"kvenkata986"},"date":"2019-05-09 16:44:01.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"a514d5d35995ce838a68ea010b5c1b060e914cf9","author":{"_account_id":28372,"name":"chinasubbareddy mallavarapu","email":"cr3938@att.com","username":"chinasubbareddy"},"date":"2019-05-09 22:01:08.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"8f5a6ec70ce5f4ba88cbe649942302c389b9c44b","author":{"_account_id":8749,"name":"Evgeniy L","email":"eli@mirantis.com","username":"evgeniyl"},"date":"2019-05-10 16:06:49.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"3f376a2f610858b88a2544651b5cf74ef8fe43f1","author":{"_account_id":20466,"name":"Tin Lam","email":"tin@lam.wtf","username":"tinlam"},"date":"2019-05-12 16:21:39.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"3be32b91f0286582ce05163f1bde229eb610061f","author":{"_account_id":20466,"name":"Tin Lam","email":"tin@lam.wtf","username":"tinlam"},"date":"2019-05-12 16:21:43.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"803b88e1d027f148b2c737bab8dcde67a0544f2b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-12 17:11:51.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/92/657692/4/check/openstack-tox-docs/a2fdc72/html/ : SUCCESS in 3m 49s\n- openstack-helm-lint http://logs.openstack.org/92/657692/4/check/openstack-helm-lint/7b7524a/ : SUCCESS in 1m 29s\n- openstack-helm-infra-aio-logging http://logs.openstack.org/92/657692/4/check/openstack-helm-infra-aio-logging/04240e1/ : SUCCESS in 32m 05s\n- openstack-helm-infra-aio-monitoring http://logs.openstack.org/92/657692/4/check/openstack-helm-infra-aio-monitoring/458bff7/ : SUCCESS in 36m 09s\n- openstack-helm-infra-aio-network-policy http://logs.openstack.org/92/657692/4/check/openstack-helm-infra-aio-network-policy/0ec42e1/ : SUCCESS in 31m 11s (non-voting)\n- openstack-helm-infra-openstack-support http://logs.openstack.org/92/657692/4/check/openstack-helm-infra-openstack-support/a54772e/ : SUCCESS in 38m 19s\n- openstack-helm-infra-kubernetes-keystone-auth http://logs.openstack.org/92/657692/4/check/openstack-helm-infra-kubernetes-keystone-auth/7f4d33d/ : SUCCESS in 49m 16s\n- openstack-helm-infra-airship-divingbell http://logs.openstack.org/92/657692/4/check/openstack-helm-infra-airship-divingbell/b2294a2/ : SUCCESS in 9m 55s (non-voting)\n- openstack-helm-infra-aio-podsecuritypolicy http://logs.openstack.org/92/657692/4/check/openstack-helm-infra-aio-podsecuritypolicy/47030c3/ : SUCCESS in 12m 18s (non-voting)\n- openstack-helm-infra-apparmor http://logs.openstack.org/92/657692/4/check/openstack-helm-infra-apparmor/0590375/ : SUCCESS in 26m 59s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"c3f6571453c880e1be604070c1d44f1b2e891c83","author":{"_account_id":24165,"name":"Renis Makadia","email":"renis.makadia@att.com","username":"renmak"},"date":"2019-05-14 18:15:35.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"010284245e1682bd744e422a48d9129bf50359d4","author":{"_account_id":22477,"name":"Matt McEuen","email":"matt.mceuen@att.com","username":"mattmceuen"},"date":"2019-05-15 14:38:19.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"a1e85b3a9066268075ce062c9c2d6fb49508c8c2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-15 14:38:33.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"7ab0081bcdf9d6fca2ce8a5dfe8a480952a859b1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-15 16:35:16.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"d7dd6ce5ac70a7948d54573687370539738319ed","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-15 16:35:16.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/92/657692/4/gate/openstack-tox-docs/8a85838/html/ : SUCCESS in 3m 28s\n- openstack-helm-lint http://logs.openstack.org/92/657692/4/gate/openstack-helm-lint/46babf5/ : SUCCESS in 1m 30s\n- openstack-helm-infra-aio-logging http://logs.openstack.org/92/657692/4/gate/openstack-helm-infra-aio-logging/e09df8c/ : SUCCESS in 36m 55s\n- openstack-helm-infra-aio-monitoring http://logs.openstack.org/92/657692/4/gate/openstack-helm-infra-aio-monitoring/5c73a8f/ : SUCCESS in 31m 16s\n- openstack-helm-infra-openstack-support http://logs.openstack.org/92/657692/4/gate/openstack-helm-infra-openstack-support/4ca82eb/ : SUCCESS in 36m 58s\n- openstack-helm-infra-kubernetes-keystone-auth http://logs.openstack.org/92/657692/4/gate/openstack-helm-infra-kubernetes-keystone-auth/8e35cb3/ : SUCCESS in 38m 00s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"656f51d8d9885095af5c6bc45a451a727c83edff","revisions":{"1ee2eaef4fe63f39f342b08410f1dac3c39adb45":{"kind":"REWORK","_number":1,"created":"2019-05-07 21:44:30.000000000","uploader":{"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},"ref":"refs/changes/92/657692/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-helm-infra","ref":"refs/changes/92/657692/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/92/657692/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/92/657692/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/92/657692/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-helm-infra refs/changes/92/657692/1"}}},"commit":{"parents":[{"commit":"9771b16e8733c5824e2cca908a93da291434b04d","subject":"Merge \"Define test specific timeouts for Armada LMA components\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/9771b16e8733c5824e2cca908a93da291434b04d"}]}],"author":{"name":"Pete Birley","email":"pete@port.direct","date":"2019-05-07 21:43:31.000000000","tz":-300},"committer":{"name":"Pete Birley","email":"pete@port.direct","date":"2019-05-07 21:43:31.000000000","tz":-300},"subject":"Ceph: Dont attempt to reuse monmap if it exists","message":"Ceph: Dont attempt to reuse monmap if it exists\n\nThis PS forces the monmap to be clobbered each time the container starts\nwhich is required to recover from ome senarios when using an emptydir\nto back /etc/ceph\n\nChange-Id: I2cf271593591ce07435893336cff98a8b1c72166\nSigned-off-by: Pete Birley \u003cpete@port.direct\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/1ee2eaef4fe63f39f342b08410f1dac3c39adb45"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/1ee2eaef4fe63f39f342b08410f1dac3c39adb45"}]},"branch":"refs/heads/master"},"2133830845e5c792e16d22da51f8cfcd972361d9":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-05-08 15:23:15.000000000","uploader":{"_account_id":29132,"name":"Krishna Venkata","email":"kvenkata986@gmail.com","username":"kvenkata986"},"ref":"refs/changes/92/657692/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-helm-infra","ref":"refs/changes/92/657692/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/92/657692/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/92/657692/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/92/657692/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-helm-infra refs/changes/92/657692/2"}}},"commit":{"parents":[{"commit":"22dcebb93d154261f785fa57b09b23b78c9f2e14","subject":"This adds the ability to specify custom resource dependencies","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/22dcebb93d154261f785fa57b09b23b78c9f2e14"}]}],"author":{"name":"Pete Birley","email":"pete@port.direct","date":"2019-05-07 21:43:31.000000000","tz":-300},"committer":{"name":"Krishna Venkata","email":"kvenkata986@gmail.com","date":"2019-05-08 15:23:15.000000000","tz":0},"subject":"Ceph: Dont attempt to reuse monmap if it exists","message":"Ceph: Dont attempt to reuse monmap if it exists\n\nThis PS forces the monmap to be clobbered each time the container starts\nwhich is required to recover from ome senarios when using an emptydir\nto back /etc/ceph\n\nChange-Id: I2cf271593591ce07435893336cff98a8b1c72166\nSigned-off-by: Pete Birley \u003cpete@port.direct\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/2133830845e5c792e16d22da51f8cfcd972361d9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/2133830845e5c792e16d22da51f8cfcd972361d9"}]},"branch":"refs/heads/master"},"bfe57c2c1c89cf7006f600880bf9b7ba5f3f1524":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2019-05-08 21:51:58.000000000","uploader":{"_account_id":29132,"name":"Krishna Venkata","email":"kvenkata986@gmail.com","username":"kvenkata986"},"ref":"refs/changes/92/657692/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-helm-infra","ref":"refs/changes/92/657692/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/92/657692/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/92/657692/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/92/657692/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-helm-infra refs/changes/92/657692/3"}}},"commit":{"parents":[{"commit":"22dcebb93d154261f785fa57b09b23b78c9f2e14","subject":"This adds the ability to specify custom resource dependencies","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/22dcebb93d154261f785fa57b09b23b78c9f2e14"}]}],"author":{"name":"Pete Birley","email":"pete@port.direct","date":"2019-05-07 21:43:31.000000000","tz":-300},"committer":{"name":"Krishna Venkata","email":"kvenkata986@gmail.com","date":"2019-05-08 21:51:53.000000000","tz":0},"subject":"Ceph: Dont attempt to reuse monmap if it exists","message":"Ceph: Dont attempt to reuse monmap if it exists\n\nThis PS forces the monmap to be clobbered each time the container starts\nwhich is required to recover from ome senarios when using an emptydir\nto back /etc/ceph.\n\nChange-Id: I2cf271593591ce07435893336cff98a8b1c72166\nSigned-off-by: Pete Birley \u003cpete@port.direct\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/bfe57c2c1c89cf7006f600880bf9b7ba5f3f1524"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/bfe57c2c1c89cf7006f600880bf9b7ba5f3f1524"}]},"branch":"refs/heads/master"},"656f51d8d9885095af5c6bc45a451a727c83edff":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2019-05-12 16:21:39.000000000","uploader":{"_account_id":20466,"name":"Tin Lam","email":"tin@lam.wtf","username":"tinlam"},"ref":"refs/changes/92/657692/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-helm-infra","ref":"refs/changes/92/657692/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/92/657692/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/92/657692/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/92/657692/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-helm-infra refs/changes/92/657692/4"}}},"commit":{"parents":[{"commit":"93ae58c6d1b843c49be742970a719c3e4f9ef5ea","subject":"Merge \"ceph-osd: Fix security context\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/93ae58c6d1b843c49be742970a719c3e4f9ef5ea"}]}],"author":{"name":"Pete Birley","email":"pete@port.direct","date":"2019-05-07 21:43:31.000000000","tz":-300},"committer":{"name":"Tin Lam","email":"tin@irrational.io","date":"2019-05-12 16:21:39.000000000","tz":0},"subject":"Ceph: Dont attempt to reuse monmap if it exists","message":"Ceph: Dont attempt to reuse monmap if it exists\n\nThis PS forces the monmap to be clobbered each time the container starts\nwhich is required to recover from ome senarios when using an emptydir\nto back /etc/ceph.\n\nChange-Id: I2cf271593591ce07435893336cff98a8b1c72166\nSigned-off-by: Pete Birley \u003cpete@port.direct\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/656f51d8d9885095af5c6bc45a451a727c83edff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/656f51d8d9885095af5c6bc45a451a727c83edff"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
