)]}'
{"id":"openstack%2Fopenstack-helm-infra~658492","triplet_id":"openstack%2Fopenstack-helm-infra~master~I54e71d2a52bd639b3e93fc899c1bf2cd075b5396","project":"openstack/openstack-helm-infra","branch":"master","topic":"maxUnavailable","hashtags":[],"change_id":"I54e71d2a52bd639b3e93fc899c1bf2cd075b5396","subject":"Switch from default values being populated for upgrade strategy for ceph components","status":"MERGED","created":"2019-05-10 21:48:48.000000000","updated":"2019-06-18 21:18:15.000000000","submitted":"2019-06-18 21:18:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":21,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"658492-1560892695056-b3e15709","meta_rev_id":"9c3f078d67003df997cbd8cb40df7eb4e324e896","_number":658492,"virtual_id_number":658492,"owner":{"_account_id":29132,"name":"Krishna Venkata","email":"kvenkata986@gmail.com","username":"kvenkata986"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},{"value":0,"_account_id":28372,"name":"chinasubbareddy mallavarapu","email":"cr3938@att.com","username":"chinasubbareddy"},{"value":0,"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},{"value":0,"_account_id":17591,"name":"Steve Wilkerson","email":"wilkers.steve@gmail.com","username":"srwilkers"},{"value":0,"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},{"value":2,"date":"2019-06-18 21:18:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-06-18 17:01:58.000000000","_account_id":29132,"name":"Krishna Venkata","email":"kvenkata986@gmail.com","username":"kvenkata986"},{"value":0,"_account_id":24780,"name":"Sangeet Gupta","email":"sg774j@att.com","username":"sgupta"},{"value":0,"_account_id":8749,"name":"Evgeniy L","email":"eli@mirantis.com","username":"evgeniyl"},{"value":0,"_account_id":22477,"name":"Matt McEuen","email":"matt.mceuen@att.com","username":"mattmceuen"},{"value":0,"_account_id":17119,"name":"Dmitrii Kabanov","email":"dmitrii.kabanov@att.com","username":"dk370c"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17591,"name":"Steve Wilkerson","email":"wilkers.steve@gmail.com","username":"srwilkers"},"recommended":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"all":[{"value":0,"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},{"value":0,"date":"2019-06-18 14:52:57.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":28372,"name":"chinasubbareddy mallavarapu","email":"cr3938@att.com","username":"chinasubbareddy"},{"value":1,"date":"2019-06-18 01:42:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},{"value":2,"date":"2019-06-18 01:42:38.000000000","_account_id":17591,"name":"Steve Wilkerson","email":"wilkers.steve@gmail.com","username":"srwilkers"},{"value":2,"date":"2019-06-18 04:49:58.000000000","_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29132,"name":"Krishna Venkata","email":"kvenkata986@gmail.com","username":"kvenkata986"},{"value":1,"date":"2019-06-18 01:42:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":24780,"name":"Sangeet Gupta","email":"sg774j@att.com","username":"sgupta"},{"value":0,"_account_id":8749,"name":"Evgeniy L","email":"eli@mirantis.com","username":"evgeniyl"},{"value":0,"_account_id":22477,"name":"Matt McEuen","email":"matt.mceuen@att.com","username":"mattmceuen"},{"value":1,"date":"2019-06-18 01:42:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17119,"name":"Dmitrii Kabanov","email":"dmitrii.kabanov@att.com","username":"dk370c"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},"all":[{"value":0,"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},{"value":0,"_account_id":28372,"name":"chinasubbareddy mallavarapu","email":"cr3938@att.com","username":"chinasubbareddy"},{"value":0,"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},{"value":0,"_account_id":17591,"name":"Steve Wilkerson","email":"wilkers.steve@gmail.com","username":"srwilkers"},{"value":1,"date":"2019-06-18 04:49:58.000000000","_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29132,"name":"Krishna Venkata","email":"kvenkata986@gmail.com","username":"kvenkata986"},{"value":0,"_account_id":24780,"name":"Sangeet Gupta","email":"sg774j@att.com","username":"sgupta"},{"value":0,"_account_id":8749,"name":"Evgeniy L","email":"eli@mirantis.com","username":"evgeniyl"},{"value":0,"_account_id":22477,"name":"Matt McEuen","email":"matt.mceuen@att.com","username":"mattmceuen"},{"value":0,"_account_id":17119,"name":"Dmitrii Kabanov","email":"dmitrii.kabanov@att.com","username":"dk370c"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8749,"name":"Evgeniy L","email":"eli@mirantis.com","username":"evgeniyl"},{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},{"_account_id":17119,"name":"Dmitrii Kabanov","email":"dmitrii.kabanov@att.com","username":"dk370c"},{"_account_id":17591,"name":"Steve Wilkerson","email":"wilkers.steve@gmail.com","username":"srwilkers"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22477,"name":"Matt McEuen","email":"matt.mceuen@att.com","username":"mattmceuen"},{"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},{"_account_id":24780,"name":"Sangeet Gupta","email":"sg774j@att.com","username":"sgupta"},{"_account_id":28372,"name":"chinasubbareddy mallavarapu","email":"cr3938@att.com","username":"chinasubbareddy"},{"_account_id":29132,"name":"Krishna Venkata","email":"kvenkata986@gmail.com","username":"kvenkata986"},{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-17 22:03:36.000000000","updated_by":{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},"reviewer":{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},"state":"REVIEWER"},{"updated":"2019-05-29 14:02:58.000000000","updated_by":{"_account_id":8749,"name":"Evgeniy L","email":"eli@mirantis.com","username":"evgeniyl"},"reviewer":{"_account_id":8749,"name":"Evgeniy L","email":"eli@mirantis.com","username":"evgeniyl"},"state":"REVIEWER"},{"updated":"2019-05-29 16:58:48.000000000","updated_by":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"reviewer":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"state":"REVIEWER"},{"updated":"2019-05-29 17:53:39.000000000","updated_by":{"_account_id":17119,"name":"Dmitrii Kabanov","email":"dmitrii.kabanov@att.com","username":"dk370c"},"reviewer":{"_account_id":17119,"name":"Dmitrii Kabanov","email":"dmitrii.kabanov@att.com","username":"dk370c"},"state":"REVIEWER"},{"updated":"2019-06-12 00:09:12.000000000","updated_by":{"_account_id":22477,"name":"Matt McEuen","email":"matt.mceuen@att.com","username":"mattmceuen"},"reviewer":{"_account_id":22477,"name":"Matt McEuen","email":"matt.mceuen@att.com","username":"mattmceuen"},"state":"REVIEWER"},{"updated":"2019-06-12 15:10:51.000000000","updated_by":{"_account_id":17591,"name":"Steve Wilkerson","email":"wilkers.steve@gmail.com","username":"srwilkers"},"reviewer":{"_account_id":17591,"name":"Steve Wilkerson","email":"wilkers.steve@gmail.com","username":"srwilkers"},"state":"REVIEWER"},{"updated":"2019-06-13 13:44:39.000000000","updated_by":{"_account_id":24780,"name":"Sangeet Gupta","email":"sg774j@att.com","username":"sgupta"},"reviewer":{"_account_id":24780,"name":"Sangeet Gupta","email":"sg774j@att.com","username":"sgupta"},"state":"REVIEWER"},{"updated":"2019-06-18 04:49:58.000000000","updated_by":{"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},"reviewer":{"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},"state":"REVIEWER"},{"updated":"2019-06-18 14:52:57.000000000","updated_by":{"_account_id":28372,"name":"chinasubbareddy mallavarapu","email":"cr3938@att.com","username":"chinasubbareddy"},"reviewer":{"_account_id":28372,"name":"chinasubbareddy mallavarapu","email":"cr3938@att.com","username":"chinasubbareddy"},"state":"REVIEWER"},{"updated":"2019-06-18 21:18:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"282c1c12bb0b36264c2779d561576829151e3267","author":{"_account_id":29132,"name":"Krishna Venkata","email":"kvenkata986@gmail.com","username":"kvenkata986"},"date":"2019-05-10 21:48:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"303564851adcdbaf84bf1d6527663742fb7b2256","author":{"_account_id":29132,"name":"Krishna Venkata","email":"kvenkata986@gmail.com","username":"kvenkata986"},"date":"2019-05-10 22:15:30.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"43e985c54db1822d375370c40df60ea9f2c43dbf","author":{"_account_id":29132,"name":"Krishna Venkata","email":"kvenkata986@gmail.com","username":"kvenkata986"},"date":"2019-05-10 22:16:57.000000000","message":"Patch Set 2: Code-Review-1","accounts_in_message":[],"_revision_number":2},{"id":"4b55013b3e9b740a3bbc6af1cb591b5c6028bfdd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-10 23:05:44.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/92/658492/2/check/openstack-tox-docs/7771d56/html/ : SUCCESS in 5m 14s\n- openstack-helm-lint http://logs.openstack.org/92/658492/2/check/openstack-helm-lint/7fa5980/ : SUCCESS in 2m 22s\n- openstack-helm-infra-aio-logging http://logs.openstack.org/92/658492/2/check/openstack-helm-infra-aio-logging/9b1adf0/ : FAILURE in 41m 40s\n- openstack-helm-infra-aio-monitoring http://logs.openstack.org/92/658492/2/check/openstack-helm-infra-aio-monitoring/5e6e556/ : SUCCESS in 46m 48s\n- openstack-helm-infra-aio-network-policy http://logs.openstack.org/92/658492/2/check/openstack-helm-infra-aio-network-policy/214d8b7/ : FAILURE in 30m 53s (non-voting)\n- openstack-helm-infra-openstack-support http://logs.openstack.org/92/658492/2/check/openstack-helm-infra-openstack-support/f501931/ : SUCCESS in 46m 26s\n- openstack-helm-infra-kubernetes-keystone-auth http://logs.openstack.org/92/658492/2/check/openstack-helm-infra-kubernetes-keystone-auth/b4305d7/ : SUCCESS in 35m 01s\n- openstack-helm-infra-airship-divingbell http://logs.openstack.org/92/658492/2/check/openstack-helm-infra-airship-divingbell/0f77716/ : SUCCESS in 12m 39s (non-voting)\n- openstack-helm-infra-aio-podsecuritypolicy http://logs.openstack.org/92/658492/2/check/openstack-helm-infra-aio-podsecuritypolicy/1a4afd9/ : SUCCESS in 12m 18s (non-voting)\n- openstack-helm-infra-apparmor http://logs.openstack.org/92/658492/2/check/openstack-helm-infra-apparmor/287c54a/ : FAILURE in 29m 15s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"69f6ed1581cf191382195ab37c4d7e78b5e3d5f6","author":{"_account_id":29132,"name":"Krishna Venkata","email":"kvenkata986@gmail.com","username":"kvenkata986"},"date":"2019-05-13 15:01:20.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"0c5087a21eed74c6a58b850552bfe41588916a2d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-13 16:01:53.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/92/658492/2/check/openstack-tox-docs/e9219bd/html/ : SUCCESS in 4m 02s\n- openstack-helm-lint http://logs.openstack.org/92/658492/2/check/openstack-helm-lint/b6d9f22/ : SUCCESS in 1m 34s\n- openstack-helm-infra-aio-logging http://logs.openstack.org/92/658492/2/check/openstack-helm-infra-aio-logging/6fbe37e/ : SUCCESS in 34m 53s\n- openstack-helm-infra-aio-monitoring http://logs.openstack.org/92/658492/2/check/openstack-helm-infra-aio-monitoring/46d3650/ : SUCCESS in 38m 19s\n- openstack-helm-infra-aio-network-policy http://logs.openstack.org/92/658492/2/check/openstack-helm-infra-aio-network-policy/72e9709/ : SUCCESS in 40m 27s (non-voting)\n- openstack-helm-infra-openstack-support http://logs.openstack.org/92/658492/2/check/openstack-helm-infra-openstack-support/566e674/ : SUCCESS in 45m 15s\n- openstack-helm-infra-kubernetes-keystone-auth http://logs.openstack.org/92/658492/2/check/openstack-helm-infra-kubernetes-keystone-auth/1e263bc/ : SUCCESS in 47m 32s\n- openstack-helm-infra-airship-divingbell http://logs.openstack.org/92/658492/2/check/openstack-helm-infra-airship-divingbell/5956bc0/ : SUCCESS in 13m 43s (non-voting)\n- openstack-helm-infra-aio-podsecuritypolicy http://logs.openstack.org/92/658492/2/check/openstack-helm-infra-aio-podsecuritypolicy/4fc0247/ : FAILURE in 17m 39s (non-voting)\n- openstack-helm-infra-apparmor http://logs.openstack.org/92/658492/2/check/openstack-helm-infra-apparmor/a488c87/ : FAILURE in 11m 25s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"2b7a7eb14864c0ddf44d2bdacaff99ba9998860f","author":{"_account_id":29132,"name":"Krishna Venkata","email":"kvenkata986@gmail.com","username":"kvenkata986"},"date":"2019-05-13 16:38:25.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"6cd8a253bfd7bbcb4e2e3e5c9383525e682fcfdf","author":{"_account_id":29132,"name":"Krishna Venkata","email":"kvenkata986@gmail.com","username":"kvenkata986"},"date":"2019-05-13 16:42:07.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"d850b82307088b796593922fc9e8256a7be5d17a","author":{"_account_id":29132,"name":"Krishna Venkata","email":"kvenkata986@gmail.com","username":"kvenkata986"},"date":"2019-05-13 17:37:43.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"cd48d1564c73dbd32b174a4510ee625534909b2c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-13 18:21:48.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/92/658492/5/check/openstack-tox-docs/030568a/html/ : SUCCESS in 3m 30s\n- openstack-helm-lint http://logs.openstack.org/92/658492/5/check/openstack-helm-lint/f0c5552/ : SUCCESS in 1m 18s\n- openstack-helm-infra-aio-logging http://logs.openstack.org/92/658492/5/check/openstack-helm-infra-aio-logging/6fed501/ : FAILURE in 40m 32s\n- openstack-helm-infra-aio-monitoring http://logs.openstack.org/92/658492/5/check/openstack-helm-infra-aio-monitoring/9056e80/ : SUCCESS in 28m 49s\n- openstack-helm-infra-aio-network-policy http://logs.openstack.org/92/658492/5/check/openstack-helm-infra-aio-network-policy/b34211a/ : SUCCESS in 32m 04s (non-voting)\n- openstack-helm-infra-openstack-support http://logs.openstack.org/92/658492/5/check/openstack-helm-infra-openstack-support/f6eaf4e/ : SUCCESS in 38m 27s\n- openstack-helm-infra-kubernetes-keystone-auth http://logs.openstack.org/92/658492/5/check/openstack-helm-infra-kubernetes-keystone-auth/3eb5920/ : SUCCESS in 36m 52s\n- openstack-helm-infra-airship-divingbell http://logs.openstack.org/92/658492/5/check/openstack-helm-infra-airship-divingbell/bca4d66/ : SUCCESS in 10m 16s (non-voting)\n- openstack-helm-infra-aio-podsecuritypolicy http://logs.openstack.org/92/658492/5/check/openstack-helm-infra-aio-podsecuritypolicy/081091d/ : SUCCESS in 12m 23s (non-voting)\n- openstack-helm-infra-apparmor http://logs.openstack.org/92/658492/5/check/openstack-helm-infra-apparmor/2f7fb51/ : SUCCESS in 28m 52s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"67d3358d653aa59c799aae03186bd648df67976b","author":{"_account_id":29132,"name":"Krishna Venkata","email":"kvenkata986@gmail.com","username":"kvenkata986"},"date":"2019-05-13 18:46:46.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"d7d268be9a9954f7bf563291cf6922a34b8811a3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-13 19:39:59.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/92/658492/6/check/openstack-tox-docs/e22d068/html/ : SUCCESS in 4m 54s\n- openstack-helm-lint http://logs.openstack.org/92/658492/6/check/openstack-helm-lint/83735b5/ : SUCCESS in 1m 49s\n- openstack-helm-infra-aio-logging http://logs.openstack.org/92/658492/6/check/openstack-helm-infra-aio-logging/f69a4ae/ : SUCCESS in 42m 00s\n- openstack-helm-infra-aio-monitoring http://logs.openstack.org/92/658492/6/check/openstack-helm-infra-aio-monitoring/1fcb3d7/ : FAILURE in 48m 05s\n- openstack-helm-infra-aio-network-policy http://logs.openstack.org/92/658492/6/check/openstack-helm-infra-aio-network-policy/396f8b5/ : SUCCESS in 38m 11s (non-voting)\n- openstack-helm-infra-openstack-support http://logs.openstack.org/92/658492/6/check/openstack-helm-infra-openstack-support/e8aac44/ : SUCCESS in 44m 05s\n- openstack-helm-infra-kubernetes-keystone-auth http://logs.openstack.org/92/658492/6/check/openstack-helm-infra-kubernetes-keystone-auth/951d111/ : SUCCESS in 46m 44s\n- openstack-helm-infra-airship-divingbell http://logs.openstack.org/92/658492/6/check/openstack-helm-infra-airship-divingbell/b61b41a/ : SUCCESS in 12m 45s (non-voting)\n- openstack-helm-infra-aio-podsecuritypolicy http://logs.openstack.org/92/658492/6/check/openstack-helm-infra-aio-podsecuritypolicy/b5ad157/ : SUCCESS in 12m 50s (non-voting)\n- openstack-helm-infra-apparmor http://logs.openstack.org/92/658492/6/check/openstack-helm-infra-apparmor/8b9c4ee/ : SUCCESS in 35m 51s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"b5809d9749f4375b8cf88276c51658e682592ebd","author":{"_account_id":29132,"name":"Krishna Venkata","email":"kvenkata986@gmail.com","username":"kvenkata986"},"date":"2019-05-14 15:23:16.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"7dcd9799bec977a0dbefa53c47e044ab48c330e2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-14 16:09:56.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/92/658492/6/check/openstack-tox-docs/c591f8d/html/ : SUCCESS in 3m 45s\n- openstack-helm-lint http://logs.openstack.org/92/658492/6/check/openstack-helm-lint/fbea774/ : SUCCESS in 1m 45s\n- openstack-helm-infra-aio-logging http://logs.openstack.org/92/658492/6/check/openstack-helm-infra-aio-logging/ae969d8/ : SUCCESS in 39m 02s\n- openstack-helm-infra-aio-monitoring http://logs.openstack.org/92/658492/6/check/openstack-helm-infra-aio-monitoring/8153644/ : FAILURE in 20m 58s\n- openstack-helm-infra-aio-network-policy http://logs.openstack.org/92/658492/6/check/openstack-helm-infra-aio-network-policy/283c074/ : SUCCESS in 36m 06s (non-voting)\n- openstack-helm-infra-openstack-support http://logs.openstack.org/92/658492/6/check/openstack-helm-infra-openstack-support/28fdffc/ : SUCCESS in 43m 14s\n- openstack-helm-infra-kubernetes-keystone-auth http://logs.openstack.org/92/658492/6/check/openstack-helm-infra-kubernetes-keystone-auth/10b02c2/ : SUCCESS in 41m 10s\n- openstack-helm-infra-airship-divingbell http://logs.openstack.org/92/658492/6/check/openstack-helm-infra-airship-divingbell/7e01e61/ : SUCCESS in 11m 00s (non-voting)\n- openstack-helm-infra-aio-podsecuritypolicy http://logs.openstack.org/92/658492/6/check/openstack-helm-infra-aio-podsecuritypolicy/6399bdf/ : FAILURE in 12m 13s (non-voting)\n- openstack-helm-infra-apparmor http://logs.openstack.org/92/658492/6/check/openstack-helm-infra-apparmor/620eeef/ : SUCCESS in 27m 06s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"933c5bec8d19936cdd4d48f75deef229e28f47a9","author":{"_account_id":29132,"name":"Krishna Venkata","email":"kvenkata986@gmail.com","username":"kvenkata986"},"date":"2019-05-14 16:33:01.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"f5ee541584c3200e3b8176595f75e272a5f8970b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-14 17:29:12.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/92/658492/6/check/openstack-tox-docs/d376965/html/ : SUCCESS in 3m 53s\n- openstack-helm-lint http://logs.openstack.org/92/658492/6/check/openstack-helm-lint/0272b93/ : SUCCESS in 1m 37s\n- openstack-helm-infra-aio-logging http://logs.openstack.org/92/658492/6/check/openstack-helm-infra-aio-logging/b802b0f/ : SUCCESS in 42m 21s\n- openstack-helm-infra-aio-monitoring http://logs.openstack.org/92/658492/6/check/openstack-helm-infra-aio-monitoring/6eb23a7/ : SUCCESS in 32m 09s\n- openstack-helm-infra-aio-network-policy http://logs.openstack.org/92/658492/6/check/openstack-helm-infra-aio-network-policy/4228773/ : SUCCESS in 34m 49s (non-voting)\n- openstack-helm-infra-openstack-support http://logs.openstack.org/92/658492/6/check/openstack-helm-infra-openstack-support/31a6bc6/ : SUCCESS in 36m 44s\n- openstack-helm-infra-kubernetes-keystone-auth http://logs.openstack.org/92/658492/6/check/openstack-helm-infra-kubernetes-keystone-auth/518bb98/ : SUCCESS in 53m 54s\n- openstack-helm-infra-airship-divingbell http://logs.openstack.org/92/658492/6/check/openstack-helm-infra-airship-divingbell/4844b65/ : SUCCESS in 12m 02s (non-voting)\n- openstack-helm-infra-aio-podsecuritypolicy http://logs.openstack.org/92/658492/6/check/openstack-helm-infra-aio-podsecuritypolicy/a5c60dd/ : SUCCESS in 12m 45s (non-voting)\n- openstack-helm-infra-apparmor http://logs.openstack.org/92/658492/6/check/openstack-helm-infra-apparmor/46736c7/ : SUCCESS in 38m 04s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"66baefe44b053d7f75baa3dbd3c0aa47f3fb72c0","author":{"_account_id":29132,"name":"Krishna Venkata","email":"kvenkata986@gmail.com","username":"kvenkata986"},"date":"2019-05-14 20:53:31.000000000","message":"Topic set to maxUnavailable","accounts_in_message":[],"_revision_number":6},{"id":"84dfdaef2e532b0879c48234e0c34346753a52ee","author":{"_account_id":29132,"name":"Krishna Venkata","email":"kvenkata986@gmail.com","username":"kvenkata986"},"date":"2019-05-16 21:07:46.000000000","message":"Patch Set 7: Patch Set 6 was rebased","accounts_in_message":[],"_revision_number":7},{"id":"45472e59356a508cf4fc4bd19ccc13057f5880e0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-16 22:26:18.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/92/658492/7/check/openstack-tox-docs/3546faf/html/ : SUCCESS in 6m 04s\n- openstack-helm-lint http://logs.openstack.org/92/658492/7/check/openstack-helm-lint/bca3bf8/ : SUCCESS in 1m 34s\n- openstack-helm-infra-aio-logging http://logs.openstack.org/92/658492/7/check/openstack-helm-infra-aio-logging/cfa281a/ : FAILURE in 50m 18s\n- openstack-helm-infra-aio-monitoring http://logs.openstack.org/92/658492/7/check/openstack-helm-infra-aio-monitoring/2de8cd0/ : FAILURE in 54m 25s\n- openstack-helm-infra-aio-network-policy http://logs.openstack.org/92/658492/7/check/openstack-helm-infra-aio-network-policy/049937e/ : POST_FAILURE in 1h 11m 32s (non-voting)\n- openstack-helm-infra-openstack-support http://logs.openstack.org/92/658492/7/check/openstack-helm-infra-openstack-support/da94374/ : SUCCESS in 46m 12s\n- openstack-helm-infra-kubernetes-keystone-auth http://logs.openstack.org/92/658492/7/check/openstack-helm-infra-kubernetes-keystone-auth/14ad457/ : FAILURE in 54m 42s\n- openstack-helm-infra-airship-divingbell http://logs.openstack.org/92/658492/7/check/openstack-helm-infra-airship-divingbell/549a1da/ : SUCCESS in 10m 13s (non-voting)\n- openstack-helm-infra-aio-podsecuritypolicy http://logs.openstack.org/92/658492/7/check/openstack-helm-infra-aio-podsecuritypolicy/ff6bf54/ : SUCCESS in 10m 59s (non-voting)\n- openstack-helm-infra-apparmor http://logs.openstack.org/92/658492/7/check/openstack-helm-infra-apparmor/7172d05/ : SUCCESS in 27m 46s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"c29a78b81807108e8ee49728c992f49672c0d8c9","author":{"_account_id":29132,"name":"Krishna Venkata","email":"kvenkata986@gmail.com","username":"kvenkata986"},"date":"2019-05-17 17:34:47.000000000","message":"Patch Set 8: Published edit on patch set 7.","accounts_in_message":[],"_revision_number":8},{"id":"318bbf76999accb265a4ae56cde87607ccb57c58","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-17 18:54:18.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/92/658492/8/check/openstack-tox-docs/16af81b/html/ : SUCCESS in 3m 39s\n- openstack-helm-lint http://logs.openstack.org/92/658492/8/check/openstack-helm-lint/f607e3c/ : SUCCESS in 2m 08s\n- openstack-helm-infra-aio-logging http://logs.openstack.org/92/658492/8/check/openstack-helm-infra-aio-logging/d8602ba/ : FAILURE in 21m 35s\n- openstack-helm-infra-aio-monitoring http://logs.openstack.org/92/658492/8/check/openstack-helm-infra-aio-monitoring/9ce896a/ : SUCCESS in 32m 57s\n- openstack-helm-infra-aio-network-policy http://logs.openstack.org/92/658492/8/check/openstack-helm-infra-aio-network-policy/31ace65/ : FAILURE in 1h 15m 31s (non-voting)\n- openstack-helm-infra-openstack-support http://logs.openstack.org/92/658492/8/check/openstack-helm-infra-openstack-support/6915357/ : FAILURE in 20m 21s\n- openstack-helm-infra-kubernetes-keystone-auth http://logs.openstack.org/92/658492/8/check/openstack-helm-infra-kubernetes-keystone-auth/52290a0/ : SUCCESS in 38m 16s\n- openstack-helm-infra-airship-divingbell http://logs.openstack.org/92/658492/8/check/openstack-helm-infra-airship-divingbell/2365d6b/ : SUCCESS in 10m 22s (non-voting)\n- openstack-helm-infra-aio-podsecuritypolicy http://logs.openstack.org/92/658492/8/check/openstack-helm-infra-aio-podsecuritypolicy/e026e6d/ : FAILURE in 14m 23s (non-voting)\n- openstack-helm-infra-apparmor http://logs.openstack.org/92/658492/8/check/openstack-helm-infra-apparmor/65008b2/ : FAILURE in 14m 14s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"1d6b2b5937da4757788915a62d9da7e1a7e47b5d","author":{"_account_id":29132,"name":"Krishna Venkata","email":"kvenkata986@gmail.com","username":"kvenkata986"},"date":"2019-05-17 20:17:45.000000000","message":"Patch Set 9: Published edit on patch set 8.","accounts_in_message":[],"_revision_number":9},{"id":"fd0c1262eae1730500818fd9b05be80512b5e99e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-17 21:19:18.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/92/658492/9/check/openstack-tox-docs/beedddf/html/ : SUCCESS in 3m 29s\n- openstack-helm-lint http://logs.openstack.org/92/658492/9/check/openstack-helm-lint/8d1d1f3/ : SUCCESS in 1m 36s\n- openstack-helm-infra-aio-logging http://logs.openstack.org/92/658492/9/check/openstack-helm-infra-aio-logging/2b4091f/ : SUCCESS in 46m 20s\n- openstack-helm-infra-aio-monitoring http://logs.openstack.org/92/658492/9/check/openstack-helm-infra-aio-monitoring/fe6c953/ : SUCCESS in 31m 08s\n- openstack-helm-infra-aio-network-policy http://logs.openstack.org/92/658492/9/check/openstack-helm-infra-aio-network-policy/339a8d0/ : SUCCESS in 42m 30s (non-voting)\n- openstack-helm-infra-openstack-support http://logs.openstack.org/92/658492/9/check/openstack-helm-infra-openstack-support/421a439/ : FAILURE in 41m 46s\n- openstack-helm-infra-kubernetes-keystone-auth http://logs.openstack.org/92/658492/9/check/openstack-helm-infra-kubernetes-keystone-auth/63ce2b8/ : FAILURE in 45m 10s\n- openstack-helm-infra-airship-divingbell http://logs.openstack.org/92/658492/9/check/openstack-helm-infra-airship-divingbell/1ed2ef6/ : SUCCESS in 10m 52s (non-voting)\n- openstack-helm-infra-aio-podsecuritypolicy http://logs.openstack.org/92/658492/9/check/openstack-helm-infra-aio-podsecuritypolicy/00ca98c/ : SUCCESS in 13m 03s (non-voting)\n- openstack-helm-infra-apparmor http://logs.openstack.org/92/658492/9/check/openstack-helm-infra-apparmor/8f9f000/ : SUCCESS in 27m 00s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"f10917546b80dcd1515a0e71cf33951145c9fef6","author":{"_account_id":29132,"name":"Krishna Venkata","email":"kvenkata986@gmail.com","username":"kvenkata986"},"date":"2019-05-17 22:01:10.000000000","message":"Patch Set 10: Commit message was updated.","accounts_in_message":[],"_revision_number":10},{"id":"1b12a9341f9a177cc1e604662c6084ed17bb9f7e","author":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"date":"2019-05-17 22:05:52.000000000","message":"Patch Set 10: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"6044d84cb0d8bd6aa7d6ba54749873a463f78c8d","author":{"_account_id":29132,"name":"Krishna Venkata","email":"kvenkata986@gmail.com","username":"kvenkata986"},"date":"2019-05-17 22:33:57.000000000","message":"Patch Set 11: Commit message was updated.","accounts_in_message":[],"_revision_number":11},{"id":"20528a06079d976e48717a60348ab01ffd7d2a70","author":{"_account_id":29132,"name":"Krishna Venkata","email":"kvenkata986@gmail.com","username":"kvenkata986"},"date":"2019-05-17 22:34:14.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"be83bd3fbc26d6580962d9b274cdada8b3f02298","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-17 23:28:50.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/92/658492/11/check/openstack-tox-docs/fb8b6da/html/ : SUCCESS in 3m 24s\n- openstack-helm-lint http://logs.openstack.org/92/658492/11/check/openstack-helm-lint/a53f66c/ : SUCCESS in 1m 19s\n- openstack-helm-infra-aio-logging http://logs.openstack.org/92/658492/11/check/openstack-helm-infra-aio-logging/dd451ee/ : SUCCESS in 53m 53s\n- openstack-helm-infra-aio-monitoring http://logs.openstack.org/92/658492/11/check/openstack-helm-infra-aio-monitoring/b6e7635/ : FAILURE in 47m 09s\n- openstack-helm-infra-aio-network-policy http://logs.openstack.org/92/658492/11/check/openstack-helm-infra-aio-network-policy/1a20261/ : SUCCESS in 38m 08s (non-voting)\n- openstack-helm-infra-openstack-support http://logs.openstack.org/92/658492/11/check/openstack-helm-infra-openstack-support/fd268fe/ : SUCCESS in 36m 37s\n- openstack-helm-infra-kubernetes-keystone-auth http://logs.openstack.org/92/658492/11/check/openstack-helm-infra-kubernetes-keystone-auth/0dfa037/ : SUCCESS in 43m 57s\n- openstack-helm-infra-airship-divingbell http://logs.openstack.org/92/658492/11/check/openstack-helm-infra-airship-divingbell/b47b1b0/ : SUCCESS in 10m 00s (non-voting)\n- openstack-helm-infra-aio-podsecuritypolicy http://logs.openstack.org/92/658492/11/check/openstack-helm-infra-aio-podsecuritypolicy/d29dad1/ : SUCCESS in 14m 55s (non-voting)\n- openstack-helm-infra-apparmor http://logs.openstack.org/92/658492/11/check/openstack-helm-infra-apparmor/ea999a9/ : SUCCESS in 35m 20s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"61152130c1e03b512a010a844f892fba498218bf","author":{"_account_id":29132,"name":"Krishna Venkata","email":"kvenkata986@gmail.com","username":"kvenkata986"},"date":"2019-05-17 23:49:03.000000000","message":"Patch Set 11:\n\ncheck","accounts_in_message":[],"_revision_number":11},{"id":"f0ac48cd4b72c5cf60b9727cedd6054731172a06","author":{"_account_id":29132,"name":"Krishna Venkata","email":"kvenkata986@gmail.com","username":"kvenkata986"},"date":"2019-05-17 23:49:25.000000000","message":"Patch Set 11:\n\nrecheck","accounts_in_message":[],"_revision_number":11},{"id":"bc5760e90857986925ab38fb3a110f6686e125d5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-18 00:50:25.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/92/658492/11/check/openstack-tox-docs/b33ea14/html/ : SUCCESS in 3m 34s\n- openstack-helm-lint http://logs.openstack.org/92/658492/11/check/openstack-helm-lint/72c35c8/ : SUCCESS in 1m 20s\n- openstack-helm-infra-aio-logging http://logs.openstack.org/92/658492/11/check/openstack-helm-infra-aio-logging/92028b5/ : SUCCESS in 39m 16s\n- openstack-helm-infra-aio-monitoring http://logs.openstack.org/92/658492/11/check/openstack-helm-infra-aio-monitoring/102d0bf/ : SUCCESS in 29m 09s\n- openstack-helm-infra-aio-network-policy http://logs.openstack.org/92/658492/11/check/openstack-helm-infra-aio-network-policy/9b33e10/ : POST_FAILURE in 56m 19s (non-voting)\n- openstack-helm-infra-openstack-support http://logs.openstack.org/92/658492/11/check/openstack-helm-infra-openstack-support/fc6f64f/ : SUCCESS in 36m 58s\n- openstack-helm-infra-kubernetes-keystone-auth http://logs.openstack.org/92/658492/11/check/openstack-helm-infra-kubernetes-keystone-auth/3f40599/ : SUCCESS in 37m 50s\n- openstack-helm-infra-airship-divingbell http://logs.openstack.org/92/658492/11/check/openstack-helm-infra-airship-divingbell/3b05b82/ : SUCCESS in 10m 27s (non-voting)\n- openstack-helm-infra-aio-podsecuritypolicy http://logs.openstack.org/92/658492/11/check/openstack-helm-infra-aio-podsecuritypolicy/890b5f1/ : SUCCESS in 13m 21s (non-voting)\n- openstack-helm-infra-apparmor http://logs.openstack.org/92/658492/11/check/openstack-helm-infra-apparmor/bfe7124/ : SUCCESS in 27m 52s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"6c6936a45c555e0303d1c606f7ebcaf9bd69beaa","author":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"date":"2019-05-20 14:00:23.000000000","message":"Patch Set 11: Code-Review+1","accounts_in_message":[],"_revision_number":11},{"id":"d1bc742239322368703bfdf2982ae40d3d274024","author":{"_account_id":28372,"name":"chinasubbareddy mallavarapu","email":"cr3938@att.com","username":"chinasubbareddy"},"date":"2019-05-20 14:46:41.000000000","message":"Patch Set 11: Code-Review+1","accounts_in_message":[],"_revision_number":11},{"id":"603946a5747768c737492ed400b2d65c9549df44","author":{"_account_id":8749,"name":"Evgeniy L","email":"eli@mirantis.com","username":"evgeniyl"},"date":"2019-05-20 15:19:18.000000000","message":"Patch Set 11: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":11},{"id":"9f2d24b998f1aea7b9afe0238c5ff251e387f43d","author":{"_account_id":29132,"name":"Krishna Venkata","email":"kvenkata986@gmail.com","username":"kvenkata986"},"date":"2019-05-20 15:41:52.000000000","message":"Patch Set 11:\n\n(3 comments)","accounts_in_message":[],"_revision_number":11},{"id":"0e4ae405f5f1fa8d4eb118dbcd77abfefa5b0032","author":{"_account_id":8749,"name":"Evgeniy L","email":"eli@mirantis.com","username":"evgeniyl"},"date":"2019-05-20 15:43:27.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"86d5aae25ad2124fdeda2ce6d6efba46efe6fe05","author":{"_account_id":29132,"name":"Krishna Venkata","email":"kvenkata986@gmail.com","username":"kvenkata986"},"date":"2019-05-21 00:05:42.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"ef9a37565cf65c3a062f067d02139232924ce82f","author":{"_account_id":29132,"name":"Krishna Venkata","email":"kvenkata986@gmail.com","username":"kvenkata986"},"date":"2019-05-21 00:09:38.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"078093763387b7bf5e33188dcba9daa27e495b1b","author":{"_account_id":29132,"name":"Krishna Venkata","email":"kvenkata986@gmail.com","username":"kvenkata986"},"date":"2019-05-21 00:28:14.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"df2917e75c9da12c76926017abf9dab31e6348f4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-21 01:16:35.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/92/658492/14/check/openstack-tox-docs/4641913/html/ : SUCCESS in 3m 43s\n- openstack-helm-lint http://logs.openstack.org/92/658492/14/check/openstack-helm-lint/2d14dbe/ : SUCCESS in 1m 27s\n- openstack-helm-infra-aio-logging http://logs.openstack.org/92/658492/14/check/openstack-helm-infra-aio-logging/9f970af/ : SUCCESS in 40m 32s\n- openstack-helm-infra-aio-monitoring http://logs.openstack.org/92/658492/14/check/openstack-helm-infra-aio-monitoring/47275bb/ : SUCCESS in 29m 06s\n- openstack-helm-infra-aio-network-policy http://logs.openstack.org/92/658492/14/check/openstack-helm-infra-aio-network-policy/7d90e85/ : SUCCESS in 46m 47s (non-voting)\n- openstack-helm-infra-openstack-support http://logs.openstack.org/92/658492/14/check/openstack-helm-infra-openstack-support/659d4a7/ : SUCCESS in 39m 04s\n- openstack-helm-infra-kubernetes-keystone-auth http://logs.openstack.org/92/658492/14/check/openstack-helm-infra-kubernetes-keystone-auth/1e24b54/ : SUCCESS in 42m 49s\n- openstack-helm-infra-airship-divingbell http://logs.openstack.org/92/658492/14/check/openstack-helm-infra-airship-divingbell/93a7920/ : SUCCESS in 11m 16s (non-voting)\n- openstack-helm-infra-aio-podsecuritypolicy http://logs.openstack.org/92/658492/14/check/openstack-helm-infra-aio-podsecuritypolicy/f03bcc6/ : SUCCESS in 13m 43s (non-voting)\n- openstack-helm-infra-apparmor http://logs.openstack.org/92/658492/14/check/openstack-helm-infra-apparmor/914d4de/ : SUCCESS in 27m 39s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"63db1966900440c7e05e26c7b09eb067d555dc2d","author":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"date":"2019-05-21 13:14:20.000000000","message":"Patch Set 14: Code-Review+1","accounts_in_message":[],"_revision_number":14},{"id":"c187df0950148de1d4f9ab580320eee7efcf43ae","author":{"_account_id":8749,"name":"Evgeniy L","email":"eli@mirantis.com","username":"evgeniyl"},"date":"2019-05-21 15:10:19.000000000","message":"Patch Set 14: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":14},{"id":"37ce2ca06a075ec007fd3434779ea16472e3dd5b","author":{"_account_id":8749,"name":"Evgeniy L","email":"eli@mirantis.com","username":"evgeniyl"},"date":"2019-05-21 15:19:08.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"1ee3c80c97f6b3030646684851d2c4e0ba5a3229","author":{"_account_id":29132,"name":"Krishna Venkata","email":"kvenkata986@gmail.com","username":"kvenkata986"},"date":"2019-05-21 16:01:17.000000000","message":"Patch Set 14:\n\n(2 comments)","accounts_in_message":[],"_revision_number":14},{"id":"c442adcbb028eb8bf4c25fad14d3e3bd5c8672da","author":{"_account_id":29132,"name":"Krishna Venkata","email":"kvenkata986@gmail.com","username":"kvenkata986"},"date":"2019-05-21 19:45:48.000000000","message":"Patch Set 15: Published edit on patch set 14.","accounts_in_message":[],"_revision_number":15},{"id":"b1d3646bedbface265f983a4aba393474918331c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-21 20:32:51.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/92/658492/15/check/openstack-tox-docs/b02a876/html/ : SUCCESS in 4m 05s\n- openstack-helm-lint http://logs.openstack.org/92/658492/15/check/openstack-helm-lint/92ab5be/ : SUCCESS in 1m 46s\n- openstack-helm-infra-aio-logging http://logs.openstack.org/92/658492/15/check/openstack-helm-infra-aio-logging/6e8e239/ : SUCCESS in 37m 42s\n- openstack-helm-infra-aio-monitoring http://logs.openstack.org/92/658492/15/check/openstack-helm-infra-aio-monitoring/faeaeaf/ : SUCCESS in 29m 57s\n- openstack-helm-infra-aio-network-policy http://logs.openstack.org/92/658492/15/check/openstack-helm-infra-aio-network-policy/0fde247/ : SUCCESS in 35m 38s (non-voting)\n- openstack-helm-infra-openstack-support http://logs.openstack.org/92/658492/15/check/openstack-helm-infra-openstack-support/8982a8b/ : SUCCESS in 39m 31s\n- openstack-helm-infra-kubernetes-keystone-auth http://logs.openstack.org/92/658492/15/check/openstack-helm-infra-kubernetes-keystone-auth/aaf12e1/ : SUCCESS in 41m 12s\n- openstack-helm-infra-airship-divingbell http://logs.openstack.org/92/658492/15/check/openstack-helm-infra-airship-divingbell/bb3710f/ : SUCCESS in 11m 07s (non-voting)\n- openstack-helm-infra-aio-podsecuritypolicy http://logs.openstack.org/92/658492/15/check/openstack-helm-infra-aio-podsecuritypolicy/494eae6/ : SUCCESS in 14m 49s (non-voting)\n- openstack-helm-infra-apparmor http://logs.openstack.org/92/658492/15/check/openstack-helm-infra-apparmor/7cbcd66/ : SUCCESS in 27m 33s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"d9e3f893bcb9f720e4be50dae335e6eb21b16486","author":{"_account_id":29132,"name":"Krishna Venkata","email":"kvenkata986@gmail.com","username":"kvenkata986"},"date":"2019-05-22 15:21:15.000000000","message":"Patch Set 16: Commit message was updated.","accounts_in_message":[],"_revision_number":16},{"id":"26e783962cc873cf456b40e081df02793a4aea24","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-22 16:59:14.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/92/658492/16/check/openstack-tox-docs/5141a5c/html/ : SUCCESS in 4m 57s\n- openstack-helm-lint http://logs.openstack.org/92/658492/16/check/openstack-helm-lint/d1faf06/ : SUCCESS in 2m 33s\n- openstack-helm-infra-aio-logging http://logs.openstack.org/92/658492/16/check/openstack-helm-infra-aio-logging/1c32f00/ : SUCCESS in 40m 27s\n- openstack-helm-infra-aio-monitoring http://logs.openstack.org/92/658492/16/check/openstack-helm-infra-aio-monitoring/f32d72b/ : FAILURE in 32m 30s\n- openstack-helm-infra-aio-network-policy http://logs.openstack.org/92/658492/16/check/openstack-helm-infra-aio-network-policy/8eb45d8/ : SUCCESS in 36m 50s (non-voting)\n- openstack-helm-infra-openstack-support http://logs.openstack.org/92/658492/16/check/openstack-helm-infra-openstack-support/af36f95/ : SUCCESS in 38m 18s\n- openstack-helm-infra-kubernetes-keystone-auth http://logs.openstack.org/92/658492/16/check/openstack-helm-infra-kubernetes-keystone-auth/b89ad05/ : SUCCESS in 43m 17s\n- openstack-helm-infra-airship-divingbell http://logs.openstack.org/92/658492/16/check/openstack-helm-infra-airship-divingbell/0303d29/ : SUCCESS in 11m 00s (non-voting)\n- openstack-helm-infra-aio-podsecuritypolicy http://logs.openstack.org/92/658492/16/check/openstack-helm-infra-aio-podsecuritypolicy/0d7f6f9/ : SUCCESS in 12m 57s (non-voting)\n- openstack-helm-infra-apparmor http://logs.openstack.org/92/658492/16/check/openstack-helm-infra-apparmor/dfd113d/ : SUCCESS in 29m 00s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"c9062974a331714fb1851e129157e07bdcbc1a45","author":{"_account_id":29132,"name":"Krishna Venkata","email":"kvenkata986@gmail.com","username":"kvenkata986"},"date":"2019-05-22 17:07:45.000000000","message":"Patch Set 16:\n\nrecheck","accounts_in_message":[],"_revision_number":16},{"id":"e8ca9d9eb886ecc06e1ba81dabf509a541b63cd4","author":{"_account_id":29132,"name":"Krishna Venkata","email":"kvenkata986@gmail.com","username":"kvenkata986"},"date":"2019-05-22 19:05:58.000000000","message":"Patch Set 16:\n\nrecheck","accounts_in_message":[],"_revision_number":16},{"id":"e1561bc16739070bed5a5f4e424b662ac3f560fd","author":{"_account_id":29132,"name":"Krishna Venkata","email":"kvenkata986@gmail.com","username":"kvenkata986"},"date":"2019-05-22 19:41:30.000000000","message":"Patch Set 16:\n\nrecheck","accounts_in_message":[],"_revision_number":16},{"id":"ee534c86ec9262863cdc4325dd0522c4076993ba","author":{"_account_id":29132,"name":"Krishna Venkata","email":"kvenkata986@gmail.com","username":"kvenkata986"},"date":"2019-05-22 19:43:07.000000000","message":"Patch Set 17: Patch Set 16 was rebased","accounts_in_message":[],"_revision_number":17},{"id":"8cce441101a6e3fa51159416de3c55abf53eeb41","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-23 00:32:36.000000000","message":"Patch Set 17: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/92/658492/17/check/openstack-tox-docs/e8e80f9/html/ : SUCCESS in 3m 56s\n- openstack-helm-lint http://logs.openstack.org/92/658492/17/check/openstack-helm-lint/b719d6a/ : SUCCESS in 1m 26s\n- openstack-helm-infra-aio-logging http://logs.openstack.org/92/658492/17/check/openstack-helm-infra-aio-logging/98e74f2/ : FAILURE in 16m 44s\n- openstack-helm-infra-aio-monitoring http://logs.openstack.org/92/658492/17/check/openstack-helm-infra-aio-monitoring/2982dbf/ : SUCCESS in 30m 17s\n- openstack-helm-infra-aio-network-policy http://logs.openstack.org/92/658492/17/check/openstack-helm-infra-aio-network-policy/d5b735b/ : SUCCESS in 40m 44s (non-voting)\n- openstack-helm-infra-openstack-support http://logs.openstack.org/92/658492/17/check/openstack-helm-infra-openstack-support/1830e1b/ : SUCCESS in 37m 06s\n- openstack-helm-infra-kubernetes-keystone-auth http://logs.openstack.org/92/658492/17/check/openstack-helm-infra-kubernetes-keystone-auth/8217a30/ : SUCCESS in 41m 53s\n- openstack-helm-infra-airship-divingbell http://logs.openstack.org/92/658492/17/check/openstack-helm-infra-airship-divingbell/bad1634/ : SUCCESS in 12m 45s (non-voting)\n- openstack-helm-infra-aio-podsecuritypolicy http://logs.openstack.org/92/658492/17/check/openstack-helm-infra-aio-podsecuritypolicy/49e42f3/ : FAILURE in 12m 29s (non-voting)\n- openstack-helm-infra-apparmor http://logs.openstack.org/92/658492/17/check/openstack-helm-infra-apparmor/c9b7c0e/ : SUCCESS in 28m 52s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"d88c1cf310fa41f6acde84861c2160ebe6731723","author":{"_account_id":29132,"name":"Krishna Venkata","email":"kvenkata986@gmail.com","username":"kvenkata986"},"date":"2019-05-23 02:40:29.000000000","message":"Patch Set 17:\n\nrecheck","accounts_in_message":[],"_revision_number":17},{"id":"9bdbe21b962b314567d57ac09900e4f7d3f2475d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-23 03:26:51.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/92/658492/17/check/openstack-tox-docs/f6775df/html/ : SUCCESS in 3m 54s\n- openstack-helm-lint http://logs.openstack.org/92/658492/17/check/openstack-helm-lint/7a78919/ : SUCCESS in 1m 49s\n- openstack-helm-infra-aio-logging http://logs.openstack.org/92/658492/17/check/openstack-helm-infra-aio-logging/eb6ba75/ : SUCCESS in 37m 25s\n- openstack-helm-infra-aio-monitoring http://logs.openstack.org/92/658492/17/check/openstack-helm-infra-aio-monitoring/fac4858/ : SUCCESS in 28m 51s\n- openstack-helm-infra-aio-network-policy http://logs.openstack.org/92/658492/17/check/openstack-helm-infra-aio-network-policy/287d8e1/ : SUCCESS in 36m 04s (non-voting)\n- openstack-helm-infra-openstack-support http://logs.openstack.org/92/658492/17/check/openstack-helm-infra-openstack-support/b8943fb/ : SUCCESS in 38m 47s\n- openstack-helm-infra-kubernetes-keystone-auth http://logs.openstack.org/92/658492/17/check/openstack-helm-infra-kubernetes-keystone-auth/8bc2de2/ : SUCCESS in 38m 25s\n- openstack-helm-infra-airship-divingbell http://logs.openstack.org/92/658492/17/check/openstack-helm-infra-airship-divingbell/b8cddaa/ : SUCCESS in 10m 38s (non-voting)\n- openstack-helm-infra-aio-podsecuritypolicy http://logs.openstack.org/92/658492/17/check/openstack-helm-infra-aio-podsecuritypolicy/ef23317/ : SUCCESS in 13m 18s (non-voting)\n- openstack-helm-infra-apparmor http://logs.openstack.org/92/658492/17/check/openstack-helm-infra-apparmor/57f6d09/ : FAILURE in 10m 01s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"f1bc12d7c0232de4af7b4dd9d5dd8b871987f8eb","author":{"_account_id":8749,"name":"Evgeniy L","email":"eli@mirantis.com","username":"evgeniyl"},"date":"2019-05-29 14:02:58.000000000","message":"Patch Set 17: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"63e8564c569ba3634f2090cf78dd249fba7d060b","author":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"date":"2019-05-29 15:14:56.000000000","message":"Patch Set 17: Code-Review+1\n\n(3 comments)","accounts_in_message":[],"_revision_number":17},{"id":"3ca6787fe83ffcd494aef75698a28894ab9185d8","author":{"_account_id":29132,"name":"Krishna Venkata","email":"kvenkata986@gmail.com","username":"kvenkata986"},"date":"2019-05-29 15:21:15.000000000","message":"Patch Set 18: Commit message was updated.","accounts_in_message":[],"_revision_number":18},{"id":"2066c39709c2f95dfd473725d7e22cd29a9edac7","author":{"_account_id":29132,"name":"Krishna Venkata","email":"kvenkata986@gmail.com","username":"kvenkata986"},"date":"2019-05-29 15:22:02.000000000","message":"Patch Set 17:\n\n(3 comments)","accounts_in_message":[],"_revision_number":17},{"id":"b5776473c246c3ef63b2f15ea21f31f4f018d203","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-29 16:14:26.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/92/658492/18/check/openstack-tox-docs/af617e2/html/ : SUCCESS in 4m 32s\n- openstack-helm-lint http://logs.openstack.org/92/658492/18/check/openstack-helm-lint/7b668fd/ : SUCCESS in 1m 35s\n- openstack-helm-infra-aio-logging http://logs.openstack.org/92/658492/18/check/openstack-helm-infra-aio-logging/add5d49/ : SUCCESS in 40m 43s\n- openstack-helm-infra-aio-monitoring http://logs.openstack.org/92/658492/18/check/openstack-helm-infra-aio-monitoring/98afdff/ : SUCCESS in 30m 06s\n- openstack-helm-infra-aio-network-policy http://logs.openstack.org/92/658492/18/check/openstack-helm-infra-aio-network-policy/33e49a2/ : FAILURE in 27m 34s (non-voting)\n- openstack-helm-infra-openstack-support http://logs.openstack.org/92/658492/18/check/openstack-helm-infra-openstack-support/9aeab82/ : SUCCESS in 41m 50s\n- openstack-helm-infra-kubernetes-keystone-auth http://logs.openstack.org/92/658492/18/check/openstack-helm-infra-kubernetes-keystone-auth/22a1af6/ : SUCCESS in 44m 43s\n- openstack-helm-infra-airship-divingbell http://logs.openstack.org/92/658492/18/check/openstack-helm-infra-airship-divingbell/b316560/ : SUCCESS in 11m 11s (non-voting)\n- openstack-helm-infra-aio-podsecuritypolicy http://logs.openstack.org/92/658492/18/check/openstack-helm-infra-aio-podsecuritypolicy/5a756d9/ : SUCCESS in 16m 16s (non-voting)\n- openstack-helm-infra-apparmor http://logs.openstack.org/92/658492/18/check/openstack-helm-infra-apparmor/d9afa99/ : FAILURE in 24m 18s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"579211d1c8a9a0542ef40ce0b6e44284b4dc57c7","author":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"date":"2019-05-29 16:58:48.000000000","message":"Patch Set 18: Code-Review+1","accounts_in_message":[],"_revision_number":18},{"id":"817b92c5e7e53c463a34030e4369880ff23eae19","author":{"_account_id":28372,"name":"chinasubbareddy mallavarapu","email":"cr3938@att.com","username":"chinasubbareddy"},"date":"2019-05-29 17:43:47.000000000","message":"Patch Set 18: Code-Review+1","accounts_in_message":[],"_revision_number":18},{"id":"04967cdf3ac7c4c0aac2097eefb634b4d1edd6f2","author":{"_account_id":17119,"name":"Dmitrii Kabanov","email":"dmitrii.kabanov@att.com","username":"dk370c"},"date":"2019-05-29 17:53:39.000000000","message":"Patch Set 18: Code-Review+1","accounts_in_message":[],"_revision_number":18},{"id":"d79ed0170b9072ade442ac3d9656e2ba6f12ee13","author":{"_account_id":29132,"name":"Krishna Venkata","email":"kvenkata986@gmail.com","username":"kvenkata986"},"date":"2019-06-11 21:50:18.000000000","message":"Patch Set 18:\n\nrecheck","accounts_in_message":[],"_revision_number":18},{"id":"f0f3c3acd9ff18e0d826d8a9cf675b741286b269","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 22:54:28.000000000","message":"Patch Set 18:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/92/658492/18/check/openstack-tox-docs/782ec9c/html/ : SUCCESS in 4m 13s\n- openstack-helm-lint http://logs.openstack.org/92/658492/18/check/openstack-helm-lint/3470f19/ : SUCCESS in 1m 26s\n- openstack-helm-infra-aio-logging http://logs.openstack.org/92/658492/18/check/openstack-helm-infra-aio-logging/9417331/ : SUCCESS in 45m 58s\n- openstack-helm-infra-aio-monitoring http://logs.openstack.org/92/658492/18/check/openstack-helm-infra-aio-monitoring/ce49387/ : SUCCESS in 30m 36s\n- openstack-helm-infra-aio-network-policy http://logs.openstack.org/92/658492/18/check/openstack-helm-infra-aio-network-policy/4f1f72d/ : FAILURE in 27m 29s (non-voting)\n- openstack-helm-infra-openstack-support http://logs.openstack.org/92/658492/18/check/openstack-helm-infra-openstack-support/64579c8/ : SUCCESS in 39m 54s\n- openstack-helm-infra-kubernetes-keystone-auth http://logs.openstack.org/92/658492/18/check/openstack-helm-infra-kubernetes-keystone-auth/8c4d2d8/ : SUCCESS in 45m 25s\n- openstack-helm-infra-airship-divingbell http://logs.openstack.org/92/658492/18/check/openstack-helm-infra-airship-divingbell/7fd9f60/ : SUCCESS in 11m 14s (non-voting)\n- openstack-helm-infra-aio-podsecuritypolicy http://logs.openstack.org/92/658492/18/check/openstack-helm-infra-aio-podsecuritypolicy/1484e21/ : SUCCESS in 13m 02s (non-voting)\n- openstack-helm-infra-apparmor http://logs.openstack.org/92/658492/18/check/openstack-helm-infra-apparmor/0afece1/ : FAILURE in 20m 16s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"6885378d9a59c6326f98030ca6dff29f8d033c9b","author":{"_account_id":17591,"name":"Steve Wilkerson","email":"wilkers.steve@gmail.com","username":"srwilkers"},"date":"2019-06-12 15:10:51.000000000","message":"Patch Set 18: Code-Review+2\n\nLooks good to me.  Thanks Krishna","accounts_in_message":[],"_revision_number":18},{"id":"56c413a3b3f820cfd724f63399465f3517bdf02e","author":{"_account_id":24780,"name":"Sangeet Gupta","email":"sg774j@att.com","username":"sgupta"},"date":"2019-06-13 13:44:39.000000000","message":"Patch Set 18: Code-Review+1","accounts_in_message":[],"_revision_number":18},{"id":"b89787319843aa454c230739bee6936e06ff0c4d","author":{"_account_id":28372,"name":"chinasubbareddy mallavarapu","email":"cr3938@att.com","username":"chinasubbareddy"},"date":"2019-06-18 01:42:38.000000000","message":"Patch Set 19: Patch Set 18 was rebased","accounts_in_message":[],"_revision_number":19},{"id":"793b48702a99214e13595b6e61bf8c25f7851cd7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 02:37:24.000000000","message":"Patch Set 19: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/92/658492/19/check/openstack-tox-docs/68d386e/html/ : SUCCESS in 4m 06s\n- openstack-helm-lint http://logs.openstack.org/92/658492/19/check/openstack-helm-lint/2a63bcd/ : SUCCESS in 1m 21s\n- openstack-helm-infra-aio-logging http://logs.openstack.org/92/658492/19/check/openstack-helm-infra-aio-logging/e9b6588/ : SUCCESS in 44m 23s\n- openstack-helm-infra-aio-monitoring http://logs.openstack.org/92/658492/19/check/openstack-helm-infra-aio-monitoring/5cd5fad/ : SUCCESS in 30m 37s\n- openstack-helm-infra-aio-network-policy http://logs.openstack.org/92/658492/19/check/openstack-helm-infra-aio-network-policy/814a02f/ : FAILURE in 31m 59s (non-voting)\n- openstack-helm-infra-openstack-support http://logs.openstack.org/92/658492/19/check/openstack-helm-infra-openstack-support/56d65c0/ : SUCCESS in 43m 28s\n- openstack-helm-infra-kubernetes-keystone-auth http://logs.openstack.org/92/658492/19/check/openstack-helm-infra-kubernetes-keystone-auth/4ce660a/ : SUCCESS in 45m 16s\n- openstack-helm-infra-airship-divingbell http://logs.openstack.org/92/658492/19/check/openstack-helm-infra-airship-divingbell/ed7db02/ : SUCCESS in 11m 07s (non-voting)\n- openstack-helm-infra-aio-podsecuritypolicy http://logs.openstack.org/92/658492/19/check/openstack-helm-infra-aio-podsecuritypolicy/2a71d27/ : FAILURE in 12m 40s (non-voting)\n- openstack-helm-infra-apparmor http://logs.openstack.org/92/658492/19/check/openstack-helm-infra-apparmor/b34a90d/ : SUCCESS in 29m 11s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"688a36e70e65ec91d5edd3a1e0607a4eb8d63fe4","author":{"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},"date":"2019-06-18 04:49:58.000000000","message":"Patch Set 19: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":19},{"id":"691f1aa07b1ca2b3d396af28d77fb0e770277343","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 04:50:10.000000000","message":"Patch Set 19: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":19},{"id":"f4257e7f87f56a18ab11f5e21ea38942e4414d8b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 05:46:19.000000000","message":"Patch Set 19: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/92/658492/19/gate/openstack-tox-docs/11c8f3a/html/ : SUCCESS in 4m 22s\n- openstack-helm-lint http://logs.openstack.org/92/658492/19/gate/openstack-helm-lint/f7cb772/ : SUCCESS in 1m 49s\n- openstack-helm-infra-aio-logging http://logs.openstack.org/92/658492/19/gate/openstack-helm-infra-aio-logging/de0f1fe/ : SUCCESS in 42m 16s\n- openstack-helm-infra-aio-monitoring http://logs.openstack.org/92/658492/19/gate/openstack-helm-infra-aio-monitoring/a92da9c/ : FAILURE in 35m 06s\n- openstack-helm-infra-openstack-support http://logs.openstack.org/92/658492/19/gate/openstack-helm-infra-openstack-support/77229dd/ : SUCCESS in 45m 23s\n- openstack-helm-infra-kubernetes-keystone-auth http://logs.openstack.org/92/658492/19/gate/openstack-helm-infra-kubernetes-keystone-auth/c30360a/ : FAILURE in 53m 04s","accounts_in_message":[],"_revision_number":19},{"id":"d83df4ed8c93f4f7f11968d49b0814ffc12cc041","author":{"_account_id":28372,"name":"chinasubbareddy mallavarapu","email":"cr3938@att.com","username":"chinasubbareddy"},"date":"2019-06-18 14:52:57.000000000","message":"Patch Set 19: -Code-Review\n\nrecheck","accounts_in_message":[],"_revision_number":19},{"id":"fe1b339fb0fec20014b137a71e3fead28e31cec5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 15:53:20.000000000","message":"Patch Set 19: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/92/658492/19/check/openstack-tox-docs/b1c79ad/html/ : SUCCESS in 4m 13s\n- openstack-helm-lint http://logs.openstack.org/92/658492/19/check/openstack-helm-lint/28b043b/ : SUCCESS in 2m 39s\n- openstack-helm-infra-aio-logging http://logs.openstack.org/92/658492/19/check/openstack-helm-infra-aio-logging/4baa974/ : SUCCESS in 43m 14s\n- openstack-helm-infra-aio-monitoring http://logs.openstack.org/92/658492/19/check/openstack-helm-infra-aio-monitoring/c847599/ : SUCCESS in 31m 51s\n- openstack-helm-infra-aio-network-policy http://logs.openstack.org/92/658492/19/check/openstack-helm-infra-aio-network-policy/9ad8797/ : FAILURE in 30m 50s (non-voting)\n- openstack-helm-infra-openstack-support http://logs.openstack.org/92/658492/19/check/openstack-helm-infra-openstack-support/40adc0f/ : FAILURE in 53m 20s\n- openstack-helm-infra-kubernetes-keystone-auth http://logs.openstack.org/92/658492/19/check/openstack-helm-infra-kubernetes-keystone-auth/1aac2f3/ : SUCCESS in 48m 46s\n- openstack-helm-infra-airship-divingbell http://logs.openstack.org/92/658492/19/check/openstack-helm-infra-airship-divingbell/a58fb1a/ : SUCCESS in 11m 28s (non-voting)\n- openstack-helm-infra-aio-podsecuritypolicy http://logs.openstack.org/92/658492/19/check/openstack-helm-infra-aio-podsecuritypolicy/690bf77/ : SUCCESS in 14m 59s (non-voting)\n- openstack-helm-infra-apparmor http://logs.openstack.org/92/658492/19/check/openstack-helm-infra-apparmor/b6474c6/ : SUCCESS in 29m 44s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"e0ff8a35c5fd405c7d1e98869818a9e418c0db75","author":{"_account_id":29132,"name":"Krishna Venkata","email":"kvenkata986@gmail.com","username":"kvenkata986"},"date":"2019-06-18 15:59:06.000000000","message":"Patch Set 19:\n\nrecheck","accounts_in_message":[],"_revision_number":19},{"id":"12a0aa7cd17ee848c5e665cf783a11149d850306","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 16:58:00.000000000","message":"Patch Set 19:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/92/658492/19/check/openstack-tox-docs/1b41078/html/ : SUCCESS in 3m 55s\n- openstack-helm-lint http://logs.openstack.org/92/658492/19/check/openstack-helm-lint/d76fe1f/ : SUCCESS in 1m 43s\n- openstack-helm-infra-aio-logging http://logs.openstack.org/92/658492/19/check/openstack-helm-infra-aio-logging/e9be18a/ : SUCCESS in 41m 52s\n- openstack-helm-infra-aio-monitoring http://logs.openstack.org/92/658492/19/check/openstack-helm-infra-aio-monitoring/ab98165/ : SUCCESS in 29m 49s\n- openstack-helm-infra-aio-network-policy http://logs.openstack.org/92/658492/19/check/openstack-helm-infra-aio-network-policy/c5e3d3d/ : FAILURE in 27m 52s (non-voting)\n- openstack-helm-infra-openstack-support http://logs.openstack.org/92/658492/19/check/openstack-helm-infra-openstack-support/d46b739/ : SUCCESS in 44m 16s\n- openstack-helm-infra-kubernetes-keystone-auth http://logs.openstack.org/92/658492/19/check/openstack-helm-infra-kubernetes-keystone-auth/7bcf89e/ : FAILURE in 53m 05s\n- openstack-helm-infra-airship-divingbell http://logs.openstack.org/92/658492/19/check/openstack-helm-infra-airship-divingbell/96c034a/ : SUCCESS in 11m 26s (non-voting)\n- openstack-helm-infra-aio-podsecuritypolicy http://logs.openstack.org/92/658492/19/check/openstack-helm-infra-aio-podsecuritypolicy/6c9ead0/ : FAILURE in 13m 03s (non-voting)\n- openstack-helm-infra-apparmor http://logs.openstack.org/92/658492/19/check/openstack-helm-infra-apparmor/cb303ea/ : SUCCESS in 30m 43s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"450fd4f16191f2f116730e33067b2934bdc8d88b","author":{"_account_id":29132,"name":"Krishna Venkata","email":"kvenkata986@gmail.com","username":"kvenkata986"},"date":"2019-06-18 17:01:58.000000000","message":"Patch Set 19:\n\nrecheck","accounts_in_message":[],"_revision_number":19},{"id":"16f0c1a923f8abced3bcaab17dbf0a924cf604fc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 17:57:27.000000000","message":"Patch Set 19: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/92/658492/19/check/openstack-tox-docs/f94df24/html/ : SUCCESS in 4m 19s\n- openstack-helm-lint http://logs.openstack.org/92/658492/19/check/openstack-helm-lint/6b6a27e/ : SUCCESS in 1m 24s\n- openstack-helm-infra-aio-logging http://logs.openstack.org/92/658492/19/check/openstack-helm-infra-aio-logging/eea04fd/ : SUCCESS in 47m 35s\n- openstack-helm-infra-aio-monitoring http://logs.openstack.org/92/658492/19/check/openstack-helm-infra-aio-monitoring/276cd11/ : SUCCESS in 29m 40s\n- openstack-helm-infra-aio-network-policy http://logs.openstack.org/92/658492/19/check/openstack-helm-infra-aio-network-policy/eee7ee0/ : FAILURE in 29m 11s (non-voting)\n- openstack-helm-infra-openstack-support http://logs.openstack.org/92/658492/19/check/openstack-helm-infra-openstack-support/2a4e71a/ : SUCCESS in 45m 54s\n- openstack-helm-infra-kubernetes-keystone-auth http://logs.openstack.org/92/658492/19/check/openstack-helm-infra-kubernetes-keystone-auth/fb88f63/ : SUCCESS in 46m 11s\n- openstack-helm-infra-airship-divingbell http://logs.openstack.org/92/658492/19/check/openstack-helm-infra-airship-divingbell/625140e/ : SUCCESS in 11m 25s (non-voting)\n- openstack-helm-infra-aio-podsecuritypolicy http://logs.openstack.org/92/658492/19/check/openstack-helm-infra-aio-podsecuritypolicy/aba7e03/ : SUCCESS in 13m 48s (non-voting)\n- openstack-helm-infra-apparmor http://logs.openstack.org/92/658492/19/check/openstack-helm-infra-apparmor/fa91927/ : SUCCESS in 29m 55s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"e60dd1e01166dfc4c7835ef10ed7cfaea49d6a81","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 17:57:49.000000000","message":"Patch Set 19: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":19},{"id":"f83f0ed7bbd1daacaf05ab87225b5f26f66eb32c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 21:18:14.000000000","message":"Patch Set 19: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/92/658492/19/gate/openstack-tox-docs/73a56e7/html/ : SUCCESS in 3m 44s\n- openstack-helm-lint http://logs.openstack.org/92/658492/19/gate/openstack-helm-lint/e256cad/ : SUCCESS in 1m 42s\n- openstack-helm-infra-aio-logging http://logs.openstack.org/92/658492/19/gate/openstack-helm-infra-aio-logging/eeebf72/ : SUCCESS in 46m 15s\n- openstack-helm-infra-aio-monitoring http://logs.openstack.org/92/658492/19/gate/openstack-helm-infra-aio-monitoring/f6d44ee/ : SUCCESS in 31m 13s\n- openstack-helm-infra-openstack-support http://logs.openstack.org/92/658492/19/gate/openstack-helm-infra-openstack-support/4462209/ : SUCCESS in 43m 27s\n- openstack-helm-infra-kubernetes-keystone-auth http://logs.openstack.org/92/658492/19/gate/openstack-helm-infra-kubernetes-keystone-auth/907c949/ : SUCCESS in 48m 55s","accounts_in_message":[],"_revision_number":19},{"id":"9c3f078d67003df997cbd8cb40df7eb4e324e896","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 21:18:15.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":19}],"current_revision_number":19,"current_revision":"ed574f456fa01d67bd76a6c0ebd43046482821ef","revisions":{"6d5fae51e8689e0df526634b0d9cffe8d2908b74":{"kind":"REWORK","_number":1,"created":"2019-05-10 21:48:48.000000000","uploader":{"_account_id":29132,"name":"Krishna Venkata","email":"kvenkata986@gmail.com","username":"kvenkata986"},"ref":"refs/changes/92/658492/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-helm-infra","ref":"refs/changes/92/658492/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/92/658492/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/92/658492/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/92/658492/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-helm-infra refs/changes/92/658492/1"}}},"commit":{"parents":[{"commit":"93ae58c6d1b843c49be742970a719c3e4f9ef5ea","subject":"Merge \"ceph-osd: Fix security context\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/93ae58c6d1b843c49be742970a719c3e4f9ef5ea"}]}],"author":{"name":"Venkata, Krishna (kv988c)","email":"kv988c@att.com","date":"2019-05-10 21:46:43.000000000","tz":-300},"committer":{"name":"Venkata, Krishna (kv988c)","email":"kvenkata986@gmail.com","date":"2019-05-10 21:48:43.000000000","tz":-300},"subject":"[WIP] Implementing maxUnavailable value for Ceph components","message":"[WIP] Implementing maxUnavailable value for\nCeph components\n\nChange-Id: I54e71d2a52bd639b3e93fc899c1bf2cd075b5396\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/6d5fae51e8689e0df526634b0d9cffe8d2908b74"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/6d5fae51e8689e0df526634b0d9cffe8d2908b74"}]},"branch":"refs/heads/master"},"3276f9e7d433d144e7399d2a1784f0265b1aed82":{"kind":"REWORK","_number":2,"created":"2019-05-10 22:15:30.000000000","uploader":{"_account_id":29132,"name":"Krishna Venkata","email":"kvenkata986@gmail.com","username":"kvenkata986"},"ref":"refs/changes/92/658492/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-helm-infra","ref":"refs/changes/92/658492/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/92/658492/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/92/658492/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/92/658492/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-helm-infra refs/changes/92/658492/2"}}},"commit":{"parents":[{"commit":"93ae58c6d1b843c49be742970a719c3e4f9ef5ea","subject":"Merge \"ceph-osd: Fix security context\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/93ae58c6d1b843c49be742970a719c3e4f9ef5ea"}]}],"author":{"name":"Venkata, Krishna (kv988c)","email":"kv988c@att.com","date":"2019-05-10 21:46:43.000000000","tz":-300},"committer":{"name":"Venkata, Krishna (kv988c)","email":"kvenkata986@gmail.com","date":"2019-05-10 22:15:22.000000000","tz":-300},"subject":"[WIP] Implementing maxUnavailable value for Ceph components","message":"[WIP] Implementing maxUnavailable value for\nCeph components\n\nChange-Id: I54e71d2a52bd639b3e93fc899c1bf2cd075b5396\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/3276f9e7d433d144e7399d2a1784f0265b1aed82"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/3276f9e7d433d144e7399d2a1784f0265b1aed82"}]},"branch":"refs/heads/master"},"b5643065cf4b9ee896eb63781f3991e64d3e0764":{"kind":"REWORK","_number":3,"created":"2019-05-13 16:38:25.000000000","uploader":{"_account_id":29132,"name":"Krishna Venkata","email":"kvenkata986@gmail.com","username":"kvenkata986"},"ref":"refs/changes/92/658492/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-helm-infra","ref":"refs/changes/92/658492/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/92/658492/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/92/658492/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/92/658492/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-helm-infra refs/changes/92/658492/3"}}},"commit":{"parents":[{"commit":"93ae58c6d1b843c49be742970a719c3e4f9ef5ea","subject":"Merge \"ceph-osd: Fix security context\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/93ae58c6d1b843c49be742970a719c3e4f9ef5ea"}]}],"author":{"name":"Venkata, Krishna (kv988c)","email":"kv988c@att.com","date":"2019-05-10 21:46:43.000000000","tz":-300},"committer":{"name":"Venkata, Krishna (kv988c)","email":"kvenkata986@gmail.com","date":"2019-05-13 16:37:34.000000000","tz":-300},"subject":"[WIP] Implementing and parameterize Unavailable value for Ceph components","message":"[WIP] Implementing and parameterize Unavailable\nvalue for Ceph components\n\nChange-Id: I54e71d2a52bd639b3e93fc899c1bf2cd075b5396\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/b5643065cf4b9ee896eb63781f3991e64d3e0764"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/b5643065cf4b9ee896eb63781f3991e64d3e0764"}]},"branch":"refs/heads/master"},"7a1c9c3acd44a849058ce45f6c5348ebf208093a":{"kind":"REWORK","_number":4,"created":"2019-05-13 16:42:07.000000000","uploader":{"_account_id":29132,"name":"Krishna Venkata","email":"kvenkata986@gmail.com","username":"kvenkata986"},"ref":"refs/changes/92/658492/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-helm-infra","ref":"refs/changes/92/658492/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/92/658492/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/92/658492/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/92/658492/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-helm-infra refs/changes/92/658492/4"}}},"commit":{"parents":[{"commit":"93ae58c6d1b843c49be742970a719c3e4f9ef5ea","subject":"Merge \"ceph-osd: Fix security context\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/93ae58c6d1b843c49be742970a719c3e4f9ef5ea"}]}],"author":{"name":"Venkata, Krishna (kv988c)","email":"kv988c@att.com","date":"2019-05-10 21:46:43.000000000","tz":-300},"committer":{"name":"Venkata, Krishna (kv988c)","email":"kvenkata986@gmail.com","date":"2019-05-13 16:41:58.000000000","tz":-300},"subject":"[WIP] Implementing and parameterize Unavailable value for Ceph components","message":"[WIP] Implementing and parameterize Unavailable\nvalue for Ceph components\n\nChange-Id: I54e71d2a52bd639b3e93fc899c1bf2cd075b5396\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/7a1c9c3acd44a849058ce45f6c5348ebf208093a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/7a1c9c3acd44a849058ce45f6c5348ebf208093a"}]},"branch":"refs/heads/master"},"d35a35a04af1504266ee828e483d560dc41fbcc8":{"kind":"REWORK","_number":5,"created":"2019-05-13 17:37:43.000000000","uploader":{"_account_id":29132,"name":"Krishna Venkata","email":"kvenkata986@gmail.com","username":"kvenkata986"},"ref":"refs/changes/92/658492/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-helm-infra","ref":"refs/changes/92/658492/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/92/658492/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/92/658492/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/92/658492/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-helm-infra refs/changes/92/658492/5"}}},"commit":{"parents":[{"commit":"93ae58c6d1b843c49be742970a719c3e4f9ef5ea","subject":"Merge \"ceph-osd: Fix security context\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/93ae58c6d1b843c49be742970a719c3e4f9ef5ea"}]}],"author":{"name":"Venkata, Krishna (kv988c)","email":"kv988c@att.com","date":"2019-05-10 21:46:43.000000000","tz":-300},"committer":{"name":"Venkata, Krishna (kv988c)","email":"kvenkata986@gmail.com","date":"2019-05-13 17:37:34.000000000","tz":-300},"subject":"[WIP] Implementing and parameterize Unavailable value for Ceph components","message":"[WIP] Implementing and parameterize Unavailable\nvalue for Ceph components\n\nChange-Id: I54e71d2a52bd639b3e93fc899c1bf2cd075b5396\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/d35a35a04af1504266ee828e483d560dc41fbcc8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/d35a35a04af1504266ee828e483d560dc41fbcc8"}]},"branch":"refs/heads/master"},"c11964e3cc1e244b8326c4d7c23f5a1cf682e107":{"kind":"REWORK","_number":6,"created":"2019-05-13 18:46:46.000000000","uploader":{"_account_id":29132,"name":"Krishna Venkata","email":"kvenkata986@gmail.com","username":"kvenkata986"},"ref":"refs/changes/92/658492/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-helm-infra","ref":"refs/changes/92/658492/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/92/658492/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/92/658492/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/92/658492/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-helm-infra refs/changes/92/658492/6"}}},"commit":{"parents":[{"commit":"93ae58c6d1b843c49be742970a719c3e4f9ef5ea","subject":"Merge \"ceph-osd: Fix security context\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/93ae58c6d1b843c49be742970a719c3e4f9ef5ea"}]}],"author":{"name":"Venkata, Krishna (kv988c)","email":"kv988c@att.com","date":"2019-05-10 21:46:43.000000000","tz":-300},"committer":{"name":"Venkata, Krishna (kv988c)","email":"kvenkata986@gmail.com","date":"2019-05-13 18:46:34.000000000","tz":-300},"subject":"[WIP] Implementing and parameterize Unavailable value for Ceph components","message":"[WIP] Implementing and parameterize Unavailable\nvalue for Ceph components\n\nChange-Id: I54e71d2a52bd639b3e93fc899c1bf2cd075b5396\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/c11964e3cc1e244b8326c4d7c23f5a1cf682e107"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/c11964e3cc1e244b8326c4d7c23f5a1cf682e107"}]},"branch":"refs/heads/master"},"0f586a966c0680b017bdca77e81094d6e7368314":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2019-05-16 21:07:46.000000000","uploader":{"_account_id":29132,"name":"Krishna Venkata","email":"kvenkata986@gmail.com","username":"kvenkata986"},"ref":"refs/changes/92/658492/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-helm-infra","ref":"refs/changes/92/658492/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/92/658492/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/92/658492/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/92/658492/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-helm-infra refs/changes/92/658492/7"}}},"commit":{"parents":[{"commit":"b69584bd658ae5cb6744e499975f9c5a505774e5","subject":"Merge \"Expose Anti-Affinity Weight Setting\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/b69584bd658ae5cb6744e499975f9c5a505774e5"}]}],"author":{"name":"Venkata, Krishna (kv988c)","email":"kv988c@att.com","date":"2019-05-10 21:46:43.000000000","tz":-300},"committer":{"name":"Krishna Venkata","email":"kvenkata986@gmail.com","date":"2019-05-16 21:07:46.000000000","tz":0},"subject":"[WIP] Implementing and parameterize Unavailable value for Ceph components","message":"[WIP] Implementing and parameterize Unavailable\nvalue for Ceph components\n\nChange-Id: I54e71d2a52bd639b3e93fc899c1bf2cd075b5396\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/0f586a966c0680b017bdca77e81094d6e7368314"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/0f586a966c0680b017bdca77e81094d6e7368314"}]},"branch":"refs/heads/master"},"01bfffd8e54b0dec041283db3104d569d3add793":{"kind":"REWORK","_number":8,"created":"2019-05-17 17:34:47.000000000","uploader":{"_account_id":29132,"name":"Krishna Venkata","email":"kvenkata986@gmail.com","username":"kvenkata986"},"ref":"refs/changes/92/658492/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-helm-infra","ref":"refs/changes/92/658492/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/92/658492/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/92/658492/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/92/658492/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-helm-infra refs/changes/92/658492/8"}}},"commit":{"parents":[{"commit":"b69584bd658ae5cb6744e499975f9c5a505774e5","subject":"Merge \"Expose Anti-Affinity Weight Setting\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/b69584bd658ae5cb6744e499975f9c5a505774e5"}]}],"author":{"name":"Venkata, Krishna (kv988c)","email":"kv988c@att.com","date":"2019-05-10 21:46:43.000000000","tz":-300},"committer":{"name":"Krishna Venkata","email":"kvenkata986@gmail.com","date":"2019-05-17 17:34:30.000000000","tz":0},"subject":"[WIP] Implementing and parameterize Unavailable value for Ceph components","message":"[WIP] Implementing and parameterize Unavailable\nvalue for Ceph components\n\nChange-Id: I54e71d2a52bd639b3e93fc899c1bf2cd075b5396\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/01bfffd8e54b0dec041283db3104d569d3add793"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/01bfffd8e54b0dec041283db3104d569d3add793"}]},"branch":"refs/heads/master"},"bd57416ec13d66d52f06ad515ad5e1b7b80d2ae0":{"kind":"REWORK","_number":9,"created":"2019-05-17 20:17:45.000000000","uploader":{"_account_id":29132,"name":"Krishna Venkata","email":"kvenkata986@gmail.com","username":"kvenkata986"},"ref":"refs/changes/92/658492/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-helm-infra","ref":"refs/changes/92/658492/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/92/658492/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/92/658492/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/92/658492/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-helm-infra refs/changes/92/658492/9"}}},"commit":{"parents":[{"commit":"b69584bd658ae5cb6744e499975f9c5a505774e5","subject":"Merge \"Expose Anti-Affinity Weight Setting\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/b69584bd658ae5cb6744e499975f9c5a505774e5"}]}],"author":{"name":"Venkata, Krishna (kv988c)","email":"kv988c@att.com","date":"2019-05-10 21:46:43.000000000","tz":-300},"committer":{"name":"Krishna Venkata","email":"kvenkata986@gmail.com","date":"2019-05-17 20:17:37.000000000","tz":0},"subject":"[WIP] Implementing and parameterize Unavailable value for Ceph components","message":"[WIP] Implementing and parameterize Unavailable\nvalue for Ceph components\n\nChange-Id: I54e71d2a52bd639b3e93fc899c1bf2cd075b5396\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/bd57416ec13d66d52f06ad515ad5e1b7b80d2ae0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/bd57416ec13d66d52f06ad515ad5e1b7b80d2ae0"}]},"branch":"refs/heads/master"},"c25362fdb754093883642e562743bcd1e9ca25d2":{"kind":"NO_CODE_CHANGE","_number":10,"created":"2019-05-17 22:01:10.000000000","uploader":{"_account_id":29132,"name":"Krishna Venkata","email":"kvenkata986@gmail.com","username":"kvenkata986"},"ref":"refs/changes/92/658492/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-helm-infra","ref":"refs/changes/92/658492/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/92/658492/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/92/658492/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/92/658492/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-helm-infra refs/changes/92/658492/10"}}},"commit":{"parents":[{"commit":"b69584bd658ae5cb6744e499975f9c5a505774e5","subject":"Merge \"Expose Anti-Affinity Weight Setting\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/b69584bd658ae5cb6744e499975f9c5a505774e5"}]}],"author":{"name":"Venkata, Krishna (kv988c)","email":"kv988c@att.com","date":"2019-05-10 21:46:43.000000000","tz":-300},"committer":{"name":"Krishna Venkata","email":"kvenkata986@gmail.com","date":"2019-05-17 22:00:59.000000000","tz":0},"subject":"Parameterize updateStrategy values for  Ceph components. Making it customizable for future.","message":"Parameterize updateStrategy values for \nCeph components. Making it customizable for future.\n\nChange-Id: I54e71d2a52bd639b3e93fc899c1bf2cd075b5396\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/c25362fdb754093883642e562743bcd1e9ca25d2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/c25362fdb754093883642e562743bcd1e9ca25d2"}]},"branch":"refs/heads/master"},"32a14b7098d4844aa14fb305c0e6f3efa8c25f35":{"kind":"NO_CODE_CHANGE","_number":11,"created":"2019-05-17 22:33:57.000000000","uploader":{"_account_id":29132,"name":"Krishna Venkata","email":"kvenkata986@gmail.com","username":"kvenkata986"},"ref":"refs/changes/92/658492/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-helm-infra","ref":"refs/changes/92/658492/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/92/658492/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/92/658492/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/92/658492/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-helm-infra refs/changes/92/658492/11"}}},"commit":{"parents":[{"commit":"b69584bd658ae5cb6744e499975f9c5a505774e5","subject":"Merge \"Expose Anti-Affinity Weight Setting\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/b69584bd658ae5cb6744e499975f9c5a505774e5"}]}],"author":{"name":"Venkata, Krishna (kv988c)","email":"kv988c@att.com","date":"2019-05-10 21:46:43.000000000","tz":-300},"committer":{"name":"Krishna Venkata","email":"kvenkata986@gmail.com","date":"2019-05-17 22:33:55.000000000","tz":0},"subject":"Parameterize updateStrategy values for Ceph components. Making it customizable for future.","message":"Parameterize updateStrategy values for\nCeph components. Making it customizable for future.\n\nChange-Id: I54e71d2a52bd639b3e93fc899c1bf2cd075b5396\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/32a14b7098d4844aa14fb305c0e6f3efa8c25f35"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/32a14b7098d4844aa14fb305c0e6f3efa8c25f35"}]},"branch":"refs/heads/master"},"8e4698d4f61aeb4c478a66858adeaf11f715a12f":{"kind":"REWORK","_number":12,"created":"2019-05-21 00:05:42.000000000","uploader":{"_account_id":29132,"name":"Krishna Venkata","email":"kvenkata986@gmail.com","username":"kvenkata986"},"ref":"refs/changes/92/658492/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-helm-infra","ref":"refs/changes/92/658492/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/92/658492/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/92/658492/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/92/658492/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-helm-infra refs/changes/92/658492/12"}}},"commit":{"parents":[{"commit":"b69584bd658ae5cb6744e499975f9c5a505774e5","subject":"Merge \"Expose Anti-Affinity Weight Setting\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/b69584bd658ae5cb6744e499975f9c5a505774e5"}]}],"author":{"name":"Venkata, Krishna (kv988c)","email":"kv988c@att.com","date":"2019-05-10 21:46:43.000000000","tz":-300},"committer":{"name":"Venkata, Krishna (kv988c)","email":"kvenkata986@gmail.com","date":"2019-05-21 00:05:38.000000000","tz":-300},"subject":"updateStrategy values for Ceph components.","message":"updateStrategy values for Ceph components.\n\nChange-Id: I54e71d2a52bd639b3e93fc899c1bf2cd075b5396\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/8e4698d4f61aeb4c478a66858adeaf11f715a12f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/8e4698d4f61aeb4c478a66858adeaf11f715a12f"}]},"branch":"refs/heads/master"},"7b838679e833dd6e26da04635c827f71ac63ddb0":{"kind":"REWORK","_number":13,"created":"2019-05-21 00:09:38.000000000","uploader":{"_account_id":29132,"name":"Krishna Venkata","email":"kvenkata986@gmail.com","username":"kvenkata986"},"ref":"refs/changes/92/658492/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-helm-infra","ref":"refs/changes/92/658492/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/92/658492/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/92/658492/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/92/658492/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-helm-infra refs/changes/92/658492/13"}}},"commit":{"parents":[{"commit":"b69584bd658ae5cb6744e499975f9c5a505774e5","subject":"Merge \"Expose Anti-Affinity Weight Setting\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/b69584bd658ae5cb6744e499975f9c5a505774e5"}]}],"author":{"name":"Venkata, Krishna (kv988c)","email":"kv988c@att.com","date":"2019-05-10 21:46:43.000000000","tz":-300},"committer":{"name":"Venkata, Krishna (kv988c)","email":"kvenkata986@gmail.com","date":"2019-05-21 00:09:23.000000000","tz":-300},"subject":"updateStrategy values for Ceph components.","message":"updateStrategy values for Ceph components.\n\nChange-Id: I54e71d2a52bd639b3e93fc899c1bf2cd075b5396\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/7b838679e833dd6e26da04635c827f71ac63ddb0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/7b838679e833dd6e26da04635c827f71ac63ddb0"}]},"branch":"refs/heads/master"},"2f97b04f894ee20ae310deef3422df96644696ef":{"kind":"REWORK","_number":14,"created":"2019-05-21 00:28:14.000000000","uploader":{"_account_id":29132,"name":"Krishna Venkata","email":"kvenkata986@gmail.com","username":"kvenkata986"},"ref":"refs/changes/92/658492/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-helm-infra","ref":"refs/changes/92/658492/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/92/658492/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/92/658492/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/92/658492/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-helm-infra refs/changes/92/658492/14"}}},"commit":{"parents":[{"commit":"b69584bd658ae5cb6744e499975f9c5a505774e5","subject":"Merge \"Expose Anti-Affinity Weight Setting\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/b69584bd658ae5cb6744e499975f9c5a505774e5"}]}],"author":{"name":"Venkata, Krishna (kv988c)","email":"kv988c@att.com","date":"2019-05-10 21:46:43.000000000","tz":-300},"committer":{"name":"Venkata, Krishna (kv988c)","email":"kvenkata986@gmail.com","date":"2019-05-21 00:27:59.000000000","tz":-300},"subject":"updateStrategy values for Ceph components.","message":"updateStrategy values for Ceph components.\n\nChange-Id: I54e71d2a52bd639b3e93fc899c1bf2cd075b5396\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/2f97b04f894ee20ae310deef3422df96644696ef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/2f97b04f894ee20ae310deef3422df96644696ef"}]},"branch":"refs/heads/master"},"7b8acb49b52399534246efa47310a257719178ea":{"kind":"REWORK","_number":15,"created":"2019-05-21 19:45:48.000000000","uploader":{"_account_id":29132,"name":"Krishna Venkata","email":"kvenkata986@gmail.com","username":"kvenkata986"},"ref":"refs/changes/92/658492/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-helm-infra","ref":"refs/changes/92/658492/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/92/658492/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/92/658492/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/92/658492/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-helm-infra refs/changes/92/658492/15"}}},"commit":{"parents":[{"commit":"b69584bd658ae5cb6744e499975f9c5a505774e5","subject":"Merge \"Expose Anti-Affinity Weight Setting\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/b69584bd658ae5cb6744e499975f9c5a505774e5"}]}],"author":{"name":"Venkata, Krishna (kv988c)","email":"kv988c@att.com","date":"2019-05-10 21:46:43.000000000","tz":-300},"committer":{"name":"Krishna Venkata","email":"kvenkata986@gmail.com","date":"2019-05-21 19:45:45.000000000","tz":0},"subject":"updateStrategy values for Ceph components.","message":"updateStrategy values for Ceph components.\n\nChange-Id: I54e71d2a52bd639b3e93fc899c1bf2cd075b5396\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/7b8acb49b52399534246efa47310a257719178ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/7b8acb49b52399534246efa47310a257719178ea"}]},"branch":"refs/heads/master"},"438a62ddf6124414fced26f7afb57f815a80ea88":{"kind":"NO_CODE_CHANGE","_number":16,"created":"2019-05-22 15:21:15.000000000","uploader":{"_account_id":29132,"name":"Krishna Venkata","email":"kvenkata986@gmail.com","username":"kvenkata986"},"ref":"refs/changes/92/658492/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-helm-infra","ref":"refs/changes/92/658492/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/92/658492/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/92/658492/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/92/658492/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-helm-infra refs/changes/92/658492/16"}}},"commit":{"parents":[{"commit":"b69584bd658ae5cb6744e499975f9c5a505774e5","subject":"Merge \"Expose Anti-Affinity Weight Setting\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/b69584bd658ae5cb6744e499975f9c5a505774e5"}]}],"author":{"name":"Venkata, Krishna (kv988c)","email":"kv988c@att.com","date":"2019-05-10 21:46:43.000000000","tz":-300},"committer":{"name":"Krishna Venkata","email":"kvenkata986@gmail.com","date":"2019-05-22 15:21:12.000000000","tz":0},"subject":"Switch from default values being populated  for upgrade strategy for ceph components","message":"Switch from default values being populated \nfor upgrade strategy for ceph components\n\nThis PS uses HelmToolKit function to add \nupgrade strategy parameters to ceph Components \n\nChange-Id: I54e71d2a52bd639b3e93fc899c1bf2cd075b5396\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/438a62ddf6124414fced26f7afb57f815a80ea88"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/438a62ddf6124414fced26f7afb57f815a80ea88"}]},"branch":"refs/heads/master"},"dde88e29110eb43998c6f8d33f4e117fee6e3b24":{"kind":"TRIVIAL_REBASE","_number":17,"created":"2019-05-22 19:43:07.000000000","uploader":{"_account_id":29132,"name":"Krishna Venkata","email":"kvenkata986@gmail.com","username":"kvenkata986"},"ref":"refs/changes/92/658492/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-helm-infra","ref":"refs/changes/92/658492/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/92/658492/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/92/658492/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/92/658492/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-helm-infra refs/changes/92/658492/17"}}},"commit":{"parents":[{"commit":"c01098f39609389f8df773b2fa988d3f7de19fc5","subject":"Merge \"Fetch Total_OSD value from ceph-client/values.yaml\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/c01098f39609389f8df773b2fa988d3f7de19fc5"}]}],"author":{"name":"Venkata, Krishna (kv988c)","email":"kv988c@att.com","date":"2019-05-10 21:46:43.000000000","tz":-300},"committer":{"name":"Krishna Venkata","email":"kvenkata986@gmail.com","date":"2019-05-22 19:43:07.000000000","tz":0},"subject":"Switch from default values being populated  for upgrade strategy for ceph components","message":"Switch from default values being populated \nfor upgrade strategy for ceph components\n\nThis PS uses HelmToolKit function to add \nupgrade strategy parameters to ceph Components \n\nChange-Id: I54e71d2a52bd639b3e93fc899c1bf2cd075b5396\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/dde88e29110eb43998c6f8d33f4e117fee6e3b24"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/dde88e29110eb43998c6f8d33f4e117fee6e3b24"}]},"branch":"refs/heads/master"},"19fdb470c1bb1c8b4b9532091f02ff2c100647cb":{"kind":"NO_CODE_CHANGE","_number":18,"created":"2019-05-29 15:21:15.000000000","uploader":{"_account_id":29132,"name":"Krishna Venkata","email":"kvenkata986@gmail.com","username":"kvenkata986"},"ref":"refs/changes/92/658492/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-helm-infra","ref":"refs/changes/92/658492/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/92/658492/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/92/658492/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/92/658492/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-helm-infra refs/changes/92/658492/18"}}},"commit":{"parents":[{"commit":"c01098f39609389f8df773b2fa988d3f7de19fc5","subject":"Merge \"Fetch Total_OSD value from ceph-client/values.yaml\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/c01098f39609389f8df773b2fa988d3f7de19fc5"}]}],"author":{"name":"Venkata, Krishna (kv988c)","email":"kv988c@att.com","date":"2019-05-10 21:46:43.000000000","tz":-300},"committer":{"name":"Krishna Venkata","email":"kvenkata986@gmail.com","date":"2019-05-29 15:21:11.000000000","tz":0},"subject":"Switch from default values being populated for upgrade strategy for ceph components","message":"Switch from default values being populated\nfor upgrade strategy for ceph components\n\nThis PS uses HelmToolKit function to add\nupgrade strategy parameters to ceph Components\n\nChange-Id: I54e71d2a52bd639b3e93fc899c1bf2cd075b5396\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/19fdb470c1bb1c8b4b9532091f02ff2c100647cb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/19fdb470c1bb1c8b4b9532091f02ff2c100647cb"}]},"branch":"refs/heads/master"},"ed574f456fa01d67bd76a6c0ebd43046482821ef":{"kind":"TRIVIAL_REBASE","_number":19,"created":"2019-06-18 01:42:38.000000000","uploader":{"_account_id":28372,"name":"chinasubbareddy mallavarapu","email":"cr3938@att.com","username":"chinasubbareddy"},"ref":"refs/changes/92/658492/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-helm-infra","ref":"refs/changes/92/658492/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/92/658492/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/92/658492/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/92/658492/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-helm-infra refs/changes/92/658492/19"}}},"commit":{"parents":[{"commit":"146d3d5976e603ac9ef225a8c99ec09ed2b1b374","subject":"Ceph: fix schema error in OSD chart","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/146d3d5976e603ac9ef225a8c99ec09ed2b1b374"}]}],"author":{"name":"Venkata, Krishna (kv988c)","email":"kv988c@att.com","date":"2019-05-10 21:46:43.000000000","tz":-300},"committer":{"name":"chinasubbareddy mallavarapu","email":"cr3938@att.com","date":"2019-06-18 01:42:38.000000000","tz":0},"subject":"Switch from default values being populated for upgrade strategy for ceph components","message":"Switch from default values being populated\nfor upgrade strategy for ceph components\n\nThis PS uses HelmToolKit function to add\nupgrade strategy parameters to ceph Components\n\nChange-Id: I54e71d2a52bd639b3e93fc899c1bf2cd075b5396\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/ed574f456fa01d67bd76a6c0ebd43046482821ef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/ed574f456fa01d67bd76a6c0ebd43046482821ef"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
