)]}'
{"id":"openstack%2Fopenstack-helm-infra~659715","triplet_id":"openstack%2Fopenstack-helm-infra~master~Ic4b2b3e487968541f5f58644e98d4f19e55c203a","project":"openstack/openstack-helm-infra","branch":"master","hashtags":[],"change_id":"Ic4b2b3e487968541f5f58644e98d4f19e55c203a","subject":"[DONOT MERGE] test review","status":"ABANDONED","created":"2019-05-17 03:57:24.000000000","updated":"2019-05-17 06:24:41.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"a1d09a99c69384002f6c3bdffbc9be49f303befc","_number":659715,"virtual_id_number":659715,"owner":{"_account_id":30232,"name":"Prabhjot Singh Sethi","email":"prabhjot@atsgen.com","username":"psethi-ats"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":-1,"date":"2019-05-17 04:54:10.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-17 04:54:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"91f7278440d011efbf4840ff7bf56fa78f5c7849","author":{"_account_id":30232,"name":"Prabhjot Singh Sethi","email":"prabhjot@atsgen.com","username":"psethi-ats"},"date":"2019-05-17 03:57:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"627d19374051d0582b6dc8a9dc0040833b03548b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-17 04:54:10.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/15/659715/1/check/openstack-tox-docs/b9f4998/html/ : SUCCESS in 3m 24s\n- openstack-helm-lint http://logs.openstack.org/15/659715/1/check/openstack-helm-lint/89ae4c8/ : SUCCESS in 1m 18s\n- openstack-helm-infra-aio-logging http://logs.openstack.org/15/659715/1/check/openstack-helm-infra-aio-logging/29ee22c/ : FAILURE in 5m 33s\n- openstack-helm-infra-aio-monitoring http://logs.openstack.org/15/659715/1/check/openstack-helm-infra-aio-monitoring/0a7fee6/ : SUCCESS in 30m 21s\n- openstack-helm-infra-aio-network-policy http://logs.openstack.org/15/659715/1/check/openstack-helm-infra-aio-network-policy/04690a9/ : SUCCESS in 32m 42s (non-voting)\n- openstack-helm-infra-openstack-support http://logs.openstack.org/15/659715/1/check/openstack-helm-infra-openstack-support/385b1b8/ : FAILURE in 41m 33s\n- openstack-helm-infra-kubernetes-keystone-auth http://logs.openstack.org/15/659715/1/check/openstack-helm-infra-kubernetes-keystone-auth/55fe9cb/ : SUCCESS in 39m 25s\n- openstack-helm-infra-airship-divingbell http://logs.openstack.org/15/659715/1/check/openstack-helm-infra-airship-divingbell/9c4244f/ : SUCCESS in 10m 22s (non-voting)\n- openstack-helm-infra-aio-podsecuritypolicy http://logs.openstack.org/15/659715/1/check/openstack-helm-infra-aio-podsecuritypolicy/f0c7445/ : SUCCESS in 13m 12s (non-voting)\n- openstack-helm-infra-apparmor http://logs.openstack.org/15/659715/1/check/openstack-helm-infra-apparmor/f648d14/ : SUCCESS in 27m 03s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a1d09a99c69384002f6c3bdffbc9be49f303befc","author":{"_account_id":30232,"name":"Prabhjot Singh Sethi","email":"prabhjot@atsgen.com","username":"psethi-ats"},"date":"2019-05-17 06:24:41.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"33e25bc96a14045a465f5213d6a01f60bd870b61","revisions":{"33e25bc96a14045a465f5213d6a01f60bd870b61":{"kind":"REWORK","_number":1,"created":"2019-05-17 03:57:24.000000000","uploader":{"_account_id":30232,"name":"Prabhjot Singh Sethi","email":"prabhjot@atsgen.com","username":"psethi-ats"},"ref":"refs/changes/15/659715/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-helm-infra","ref":"refs/changes/15/659715/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/15/659715/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/15/659715/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/15/659715/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-helm-infra refs/changes/15/659715/1"}}},"commit":{"parents":[{"commit":"b69584bd658ae5cb6744e499975f9c5a505774e5","subject":"Merge \"Expose Anti-Affinity Weight Setting\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/b69584bd658ae5cb6744e499975f9c5a505774e5"}]}],"author":{"name":"Prabhjot Singh Sethi","email":"prabhjot@atsgen.com","date":"2019-05-17 03:54:56.000000000","tz":0},"committer":{"name":"Prabhjot Singh Sethi","email":"prabhjot@atsgen.com","date":"2019-05-17 03:57:23.000000000","tz":0},"subject":"[DONOT MERGE] test review","message":"[DONOT MERGE] test review\n\ntesting ovs image failure\n\nChange-Id: Ic4b2b3e487968541f5f58644e98d4f19e55c203a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/33e25bc96a14045a465f5213d6a01f60bd870b61"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/33e25bc96a14045a465f5213d6a01f60bd870b61"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
