)]}'
{"id":"openstack%2Fopenstack-helm-infra~699507","triplet_id":"openstack%2Fopenstack-helm-infra~master~I9c6570333d63101f5846f85861b004a30fa2fbe5","project":"openstack/openstack-helm-infra","branch":"master","topic":"fix-apparmor-job","hashtags":[],"change_id":"I9c6570333d63101f5846f85861b004a30fa2fbe5","subject":"WIP: Move apparmor job to use feature gates","status":"ABANDONED","created":"2019-12-17 21:54:36.000000000","updated":"2020-09-10 23:57:33.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"a06d17fb84060a6b39f42ff44642e9c49dee0d47","_number":699507,"virtual_id_number":699507,"owner":{"_account_id":17591,"name":"Steve Wilkerson","email":"wilkers.steve@gmail.com","username":"srwilkers"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2019-12-18 17:01:50.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-18 17:01:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a0bcdae9d90c2a02501ba0b1c5580d83c08088cf","author":{"_account_id":17591,"name":"Steve Wilkerson","email":"wilkers.steve@gmail.com","username":"srwilkers"},"date":"2019-12-17 21:54:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6c86393ab1bca86e22f54eb6fa1426696391bca8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-17 21:54:49.000000000","message":"Patch Set 1: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"517e6def9de75a860d59f1e4789432adbd0d3d5c","author":{"_account_id":17591,"name":"Steve Wilkerson","email":"wilkers.steve@gmail.com","username":"srwilkers"},"date":"2019-12-17 21:57:00.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"4d230ca81877b162b7639e36b82c263eb9b66e6f","author":{"_account_id":17591,"name":"Steve Wilkerson","email":"wilkers.steve@gmail.com","username":"srwilkers"},"date":"2019-12-17 22:37:03.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"de1ea5f761ce7757f07192f18cb773140a184cf7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-17 23:39:33.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b5d0800c42174e1c83c04095316a1e90 : SUCCESS in 4m 36s\n- openstack-helm-lint https://zuul.opendev.org/t/openstack/build/a4d9eb025fec4675aab2b32a3334e779 : SUCCESS in 2m 04s\n- openstack-helm-infra-aio-logging https://zuul.opendev.org/t/openstack/build/d2b7c22050f34564bbc4ab02e7f950d8 : FAILURE in 42m 57s\n- openstack-helm-infra-aio-monitoring https://zuul.opendev.org/t/openstack/build/63d759b0ea5d4f6eb6f7a9b970be89a2 : SUCCESS in 45m 40s\n- openstack-helm-infra-federated-monitoring https://zuul.opendev.org/t/openstack/build/20c6797c8e8f411a842aa09e8c5b106f : SUCCESS in 41m 31s (non-voting)\n- openstack-helm-infra-kafka https://zuul.opendev.org/t/openstack/build/c076d75000324f82a071a43c40349922 : SUCCESS in 48m 31s (non-voting)\n- openstack-helm-infra-aio-network-policy https://zuul.opendev.org/t/openstack/build/5b2ba9e5700d46e1aa62dcbf9333000a : FAILURE in 38m 13s (non-voting)\n- openstack-helm-infra-openstack-support https://zuul.opendev.org/t/openstack/build/8052c6a79ca6490a9809185aacf0f197 : SUCCESS in 59m 34s\n- openstack-helm-infra-kubernetes-keystone-auth https://zuul.opendev.org/t/openstack/build/4e1581698ae14e28b75405259f4925d3 : RETRY_LIMIT in 3m 57s (non-voting)\n- openstack-helm-infra-airship-divingbell https://zuul.opendev.org/t/openstack/build/1fc1bd6bdf1e437881d4ba0d07eb613b : FAILURE in 2m 38s (non-voting)\n- openstack-helm-infra-aio-podsecuritypolicy https://zuul.opendev.org/t/openstack/build/d8063148a5df4832a9cf9143893fac34 : FAILURE in 18m 12s (non-voting)\n- openstack-helm-infra-apparmor https://zuul.opendev.org/t/openstack/build/cd737fdea4ef4f5ea1bde1810aa8758f : SUCCESS in 45m 59s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"c3014a7b4e1ed297c87cc53e8f37fd7a56fa1fef","author":{"_account_id":17591,"name":"Steve Wilkerson","email":"wilkers.steve@gmail.com","username":"srwilkers"},"date":"2019-12-18 15:08:55.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"f3850b379cb91cc3178f37fb7b09c154b6c16bf8","author":{"_account_id":17591,"name":"Steve Wilkerson","email":"wilkers.steve@gmail.com","username":"srwilkers"},"date":"2019-12-18 16:01:13.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"4db6ca5f9838b58f7b9853678fe1800103ce8300","author":{"_account_id":17591,"name":"Steve Wilkerson","email":"wilkers.steve@gmail.com","username":"srwilkers"},"date":"2019-12-18 16:02:22.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"387483f2de1ca3eb8b1f306d87df505538cc6149","author":{"_account_id":17591,"name":"Steve Wilkerson","email":"wilkers.steve@gmail.com","username":"srwilkers"},"date":"2019-12-18 16:04:56.000000000","message":"Patch Set 7: Patch Set 6 was rebased","accounts_in_message":[],"_revision_number":7},{"id":"9e481612d27b9944045b3b41d89a17e5318fe4bd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-18 17:01:50.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/653587151a7948508b998956a9cb1860 : SUCCESS in 4m 15s\n- openstack-helm-lint https://zuul.opendev.org/t/openstack/build/57a76b124c374e8c96d25d9f82da64e5 : SUCCESS in 1m 43s\n- openstack-helm-infra-aio-logging https://zuul.opendev.org/t/openstack/build/3a196631e7f844ac8cd08e589af93713 : SUCCESS in 48m 16s\n- openstack-helm-infra-aio-monitoring https://zuul.opendev.org/t/openstack/build/3764b74b0e684d2eb5c325b39f429665 : SUCCESS in 36m 46s\n- openstack-helm-infra-federated-monitoring https://zuul.opendev.org/t/openstack/build/25fd9c1cd495406b98bbef73f6198ce2 : SUCCESS in 30m 37s (non-voting)\n- openstack-helm-infra-kafka https://zuul.opendev.org/t/openstack/build/dde47b929b2e42faa1b9ce48bdbe6308 : SUCCESS in 34m 22s (non-voting)\n- openstack-helm-infra-aio-network-policy https://zuul.opendev.org/t/openstack/build/cd27451493e64571b35603b14aac37d9 : FAILURE in 47m 21s (non-voting)\n- openstack-helm-infra-openstack-support https://zuul.opendev.org/t/openstack/build/34e37ba7498847a08e3be19eb5a61b7a : SUCCESS in 51m 34s\n- openstack-helm-infra-kubernetes-keystone-auth https://zuul.opendev.org/t/openstack/build/21cb156e052c40b3984d0f23195a6985 : RETRY_LIMIT in 4m 28s (non-voting)\n- openstack-helm-infra-airship-divingbell https://zuul.opendev.org/t/openstack/build/0721493ac10a4b86b2c8da680d3ccaf5 : SUCCESS in 15m 17s (non-voting)\n- openstack-helm-infra-aio-podsecuritypolicy https://zuul.opendev.org/t/openstack/build/a7f74b94be95435c836f8bd865f6b435 : FAILURE in 17m 01s (non-voting)\n- openstack-helm-infra-apparmor https://zuul.opendev.org/t/openstack/build/3b768e8727844e528b93403054f98b54 : SUCCESS in 37m 52s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"90b32ff48fa3ff66a43ff9d9efcad40fdd88c9d6","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2020-09-10 23:57:33.000000000","message":"Abandoned\n\nAbandoning since there hasn\u0027t been any recent activity, if anyone wants to continue this work, please feel free to restore this.","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"c3214c745abebeefc1de54b66a3f1318fd5014bd","revisions":{"b8c7f1f84febf9872af61c72fa114015e7e7a562":{"kind":"REWORK","_number":1,"created":"2019-12-17 21:54:36.000000000","uploader":{"_account_id":17591,"name":"Steve Wilkerson","email":"wilkers.steve@gmail.com","username":"srwilkers"},"ref":"refs/changes/07/699507/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-helm-infra","ref":"refs/changes/07/699507/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/07/699507/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/07/699507/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/07/699507/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-helm-infra refs/changes/07/699507/1"}}},"commit":{"parents":[{"commit":"3494fe8489a7f80cb7043d1bb7e574ea8f9c31db","subject":"Merge \"Add support to set vhost-iommu-support as global option\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/3494fe8489a7f80cb7043d1bb7e574ea8f9c31db"}]}],"author":{"name":"Steve Wilkerson","email":"sw5822@att.com","date":"2019-12-17 21:53:35.000000000","tz":-360},"committer":{"name":"Steve Wilkerson","email":"sw5822@att.com","date":"2019-12-17 21:54:32.000000000","tz":-360},"subject":"WIP: Move apparmor job to use feature gates","message":"WIP: Move apparmor job to use feature gates\n\nChange-Id: I9c6570333d63101f5846f85861b004a30fa2fbe5\nSigned-off-by: Steve Wilkerson \u003csw5822@att.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/b8c7f1f84febf9872af61c72fa114015e7e7a562"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/b8c7f1f84febf9872af61c72fa114015e7e7a562"}]},"branch":"refs/heads/master"},"e31d9b79dc2461805a678f5459221803b14e3f67":{"kind":"REWORK","_number":2,"created":"2019-12-17 21:57:00.000000000","uploader":{"_account_id":17591,"name":"Steve Wilkerson","email":"wilkers.steve@gmail.com","username":"srwilkers"},"ref":"refs/changes/07/699507/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-helm-infra","ref":"refs/changes/07/699507/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/07/699507/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/07/699507/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/07/699507/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-helm-infra refs/changes/07/699507/2"}}},"commit":{"parents":[{"commit":"016b56e5868d9cacaab5921f4458eba5f787a253","subject":"Ceph Nautilus compatibility","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/016b56e5868d9cacaab5921f4458eba5f787a253"}]}],"author":{"name":"Steve Wilkerson","email":"sw5822@att.com","date":"2019-12-17 21:53:35.000000000","tz":-360},"committer":{"name":"Steve Wilkerson","email":"sw5822@att.com","date":"2019-12-17 21:56:54.000000000","tz":-360},"subject":"WIP: Move apparmor job to use feature gates","message":"WIP: Move apparmor job to use feature gates\n\nChange-Id: I9c6570333d63101f5846f85861b004a30fa2fbe5\nSigned-off-by: Steve Wilkerson \u003csw5822@att.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/e31d9b79dc2461805a678f5459221803b14e3f67"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/e31d9b79dc2461805a678f5459221803b14e3f67"}]},"branch":"refs/heads/master"},"9618106ec27043baf67a8740c726142e872c6346":{"kind":"REWORK","_number":3,"created":"2019-12-17 22:37:03.000000000","uploader":{"_account_id":17591,"name":"Steve Wilkerson","email":"wilkers.steve@gmail.com","username":"srwilkers"},"ref":"refs/changes/07/699507/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-helm-infra","ref":"refs/changes/07/699507/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/07/699507/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/07/699507/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/07/699507/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-helm-infra refs/changes/07/699507/3"}}},"commit":{"parents":[{"commit":"016b56e5868d9cacaab5921f4458eba5f787a253","subject":"Ceph Nautilus compatibility","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/016b56e5868d9cacaab5921f4458eba5f787a253"}]}],"author":{"name":"Steve Wilkerson","email":"sw5822@att.com","date":"2019-12-17 21:53:35.000000000","tz":-360},"committer":{"name":"Steve Wilkerson","email":"sw5822@att.com","date":"2019-12-17 22:36:57.000000000","tz":-360},"subject":"WIP: Move apparmor job to use feature gates","message":"WIP: Move apparmor job to use feature gates\n\nChange-Id: I9c6570333d63101f5846f85861b004a30fa2fbe5\nSigned-off-by: Steve Wilkerson \u003csw5822@att.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/9618106ec27043baf67a8740c726142e872c6346"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/9618106ec27043baf67a8740c726142e872c6346"}]},"branch":"refs/heads/master"},"7fc8818f68c2fbc20ec50482cd774054034db78b":{"kind":"REWORK","_number":4,"created":"2019-12-18 15:08:55.000000000","uploader":{"_account_id":17591,"name":"Steve Wilkerson","email":"wilkers.steve@gmail.com","username":"srwilkers"},"ref":"refs/changes/07/699507/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-helm-infra","ref":"refs/changes/07/699507/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/07/699507/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/07/699507/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/07/699507/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-helm-infra refs/changes/07/699507/4"}}},"commit":{"parents":[{"commit":"016b56e5868d9cacaab5921f4458eba5f787a253","subject":"Ceph Nautilus compatibility","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/016b56e5868d9cacaab5921f4458eba5f787a253"}]}],"author":{"name":"Steve Wilkerson","email":"sw5822@att.com","date":"2019-12-17 21:53:35.000000000","tz":-360},"committer":{"name":"Steve Wilkerson","email":"sw5822@att.com","date":"2019-12-18 15:08:49.000000000","tz":-360},"subject":"WIP: Move apparmor job to use feature gates","message":"WIP: Move apparmor job to use feature gates\n\nChange-Id: I9c6570333d63101f5846f85861b004a30fa2fbe5\nSigned-off-by: Steve Wilkerson \u003csw5822@att.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/7fc8818f68c2fbc20ec50482cd774054034db78b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/7fc8818f68c2fbc20ec50482cd774054034db78b"}]},"branch":"refs/heads/master"},"fe968e48cfe4b6f5acd1d5ae5eb67dab8d753e64":{"kind":"REWORK","_number":5,"created":"2019-12-18 16:01:13.000000000","uploader":{"_account_id":17591,"name":"Steve Wilkerson","email":"wilkers.steve@gmail.com","username":"srwilkers"},"ref":"refs/changes/07/699507/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-helm-infra","ref":"refs/changes/07/699507/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/07/699507/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/07/699507/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/07/699507/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-helm-infra refs/changes/07/699507/5"}}},"commit":{"parents":[{"commit":"016b56e5868d9cacaab5921f4458eba5f787a253","subject":"Ceph Nautilus compatibility","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/016b56e5868d9cacaab5921f4458eba5f787a253"}]}],"author":{"name":"Steve Wilkerson","email":"sw5822@att.com","date":"2019-12-17 21:53:35.000000000","tz":-360},"committer":{"name":"Steve Wilkerson","email":"sw5822@att.com","date":"2019-12-18 16:01:08.000000000","tz":-360},"subject":"WIP: Move apparmor job to use feature gates","message":"WIP: Move apparmor job to use feature gates\n\nChange-Id: I9c6570333d63101f5846f85861b004a30fa2fbe5\nSigned-off-by: Steve Wilkerson \u003csw5822@att.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/fe968e48cfe4b6f5acd1d5ae5eb67dab8d753e64"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/fe968e48cfe4b6f5acd1d5ae5eb67dab8d753e64"}]},"branch":"refs/heads/master"},"756d190277a6475649e3d76286b215d6cf0dd48f":{"kind":"REWORK","_number":6,"created":"2019-12-18 16:02:22.000000000","uploader":{"_account_id":17591,"name":"Steve Wilkerson","email":"wilkers.steve@gmail.com","username":"srwilkers"},"ref":"refs/changes/07/699507/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-helm-infra","ref":"refs/changes/07/699507/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/07/699507/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/07/699507/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/07/699507/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-helm-infra refs/changes/07/699507/6"}}},"commit":{"parents":[{"commit":"016b56e5868d9cacaab5921f4458eba5f787a253","subject":"Ceph Nautilus compatibility","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/016b56e5868d9cacaab5921f4458eba5f787a253"}]}],"author":{"name":"Steve Wilkerson","email":"sw5822@att.com","date":"2019-12-17 21:53:35.000000000","tz":-360},"committer":{"name":"Steve Wilkerson","email":"sw5822@att.com","date":"2019-12-18 16:02:17.000000000","tz":-360},"subject":"WIP: Move apparmor job to use feature gates","message":"WIP: Move apparmor job to use feature gates\n\nChange-Id: I9c6570333d63101f5846f85861b004a30fa2fbe5\nSigned-off-by: Steve Wilkerson \u003csw5822@att.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/756d190277a6475649e3d76286b215d6cf0dd48f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/756d190277a6475649e3d76286b215d6cf0dd48f"}]},"branch":"refs/heads/master"},"c3214c745abebeefc1de54b66a3f1318fd5014bd":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2019-12-18 16:04:56.000000000","uploader":{"_account_id":17591,"name":"Steve Wilkerson","email":"wilkers.steve@gmail.com","username":"srwilkers"},"ref":"refs/changes/07/699507/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-helm-infra","ref":"refs/changes/07/699507/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/07/699507/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/07/699507/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/07/699507/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-helm-infra refs/changes/07/699507/7"}}},"commit":{"parents":[{"commit":"13b23841342a4ed6a65a010a7a51ace9ded6b066","subject":"Merge \"[ceph-client] update pool validation logic\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/13b23841342a4ed6a65a010a7a51ace9ded6b066"}]}],"author":{"name":"Steve Wilkerson","email":"sw5822@att.com","date":"2019-12-17 21:53:35.000000000","tz":-360},"committer":{"name":"Steve Wilkerson","email":"wilkers.steve@gmail.com","date":"2019-12-18 16:04:56.000000000","tz":0},"subject":"WIP: Move apparmor job to use feature gates","message":"WIP: Move apparmor job to use feature gates\n\nChange-Id: I9c6570333d63101f5846f85861b004a30fa2fbe5\nSigned-off-by: Steve Wilkerson \u003csw5822@att.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/c3214c745abebeefc1de54b66a3f1318fd5014bd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/c3214c745abebeefc1de54b66a3f1318fd5014bd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
