)]}'
{"id":"openstack%2Fopenstack-helm-infra~705858","triplet_id":"openstack%2Fopenstack-helm-infra~master~I7508a4ccb0f896311a27f9d188cead3cdab7c29b","project":"openstack/openstack-helm-infra","branch":"master","topic":"netpol","hashtags":[],"change_id":"I7508a4ccb0f896311a27f9d188cead3cdab7c29b","subject":"[WIP] Move netpol job to use feature gates","status":"ABANDONED","created":"2020-02-04 22:18:39.000000000","updated":"2020-02-18 21:48:31.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"3603538bd7f47f65b2da64da5738c0b832c87862","_number":705858,"virtual_id_number":705858,"owner":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-02-10 19:06:39.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-10 19:06:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e952a0879a9924f670ea0ef9c952815e5e501eb9","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2020-02-04 22:18:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"88596fd7cd26e89370f1d3512302717944a3214f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-04 22:49:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-helm-infra-aio-network-policy https://zuul.opendev.org/t/openstack/build/00ba7a172c184142a489a109e3f75053 : SUCCESS in 30m 42s","accounts_in_message":[],"_revision_number":1},{"id":"8b65cf2c857623065c24c94621dda83dd4312d8e","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2020-02-10 18:20:16.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"4eac237423f1d9dd4959c4b78d66d0a8a3276b3b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-10 19:06:39.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-helm-infra-aio-network-policy https://zuul.opendev.org/t/openstack/build/e51a59d7cafb465cac4d134a67afc648 : FAILURE in 44m 26s","accounts_in_message":[],"_revision_number":2},{"id":"7c7b0615009af132c719ad028e08fc9acf336a26","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2020-02-18 21:48:31.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"4d6fcd134a52d73557f9103b43269cacf7c6ca56","revisions":{"fa61e755a70de52bbb25c4394684b1ade25ede1c":{"kind":"REWORK","_number":1,"created":"2020-02-04 22:18:39.000000000","uploader":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"ref":"refs/changes/58/705858/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-helm-infra","ref":"refs/changes/58/705858/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/58/705858/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/58/705858/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/58/705858/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-helm-infra refs/changes/58/705858/1"}}},"commit":{"parents":[{"commit":"92dfac645a564f46fe6c74a3c706a3598a758b8a","subject":"[Ceph Nautilus] Fix _checkPGs.py.tpl for Nautilus compatibility","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/92dfac645a564f46fe6c74a3c706a3598a758b8a"}]}],"author":{"name":"Gage Hugo","email":"gagehugo@gmail.com","date":"2020-02-04 22:18:25.000000000","tz":-360},"committer":{"name":"Gage Hugo","email":"gagehugo@gmail.com","date":"2020-02-04 22:18:25.000000000","tz":-360},"subject":"[WIP] Move netpol job to use feature gates","message":"[WIP] Move netpol job to use feature gates\n\nChange-Id: I7508a4ccb0f896311a27f9d188cead3cdab7c29b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/fa61e755a70de52bbb25c4394684b1ade25ede1c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/fa61e755a70de52bbb25c4394684b1ade25ede1c"}]},"branch":"refs/heads/master"},"4d6fcd134a52d73557f9103b43269cacf7c6ca56":{"kind":"REWORK","_number":2,"created":"2020-02-10 18:20:16.000000000","uploader":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"ref":"refs/changes/58/705858/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-helm-infra","ref":"refs/changes/58/705858/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/58/705858/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/58/705858/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/58/705858/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-helm-infra refs/changes/58/705858/2"}}},"commit":{"parents":[{"commit":"aa48b168967402451891f6d18f90705d08c26d85","subject":"Add train release support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/aa48b168967402451891f6d18f90705d08c26d85"}]}],"author":{"name":"Gage Hugo","email":"gagehugo@gmail.com","date":"2020-02-04 22:18:25.000000000","tz":-360},"committer":{"name":"Gage Hugo","email":"gagehugo@gmail.com","date":"2020-02-10 18:20:12.000000000","tz":-360},"subject":"[WIP] Move netpol job to use feature gates","message":"[WIP] Move netpol job to use feature gates\n\nChange-Id: I7508a4ccb0f896311a27f9d188cead3cdab7c29b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/4d6fcd134a52d73557f9103b43269cacf7c6ca56"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/4d6fcd134a52d73557f9103b43269cacf7c6ca56"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
