)]}'
{"id":"openstack%2Fopenstack-helm-infra~725096","triplet_id":"openstack%2Fopenstack-helm-infra~master~Ib31adb98e822b1b57acd8fd2f3f338e6cfe24c23","project":"openstack/openstack-helm-infra","branch":"master","topic":"tox","hashtags":[],"change_id":"Ib31adb98e822b1b57acd8fd2f3f338e6cfe24c23","subject":"chore(tox): update tox.ini","status":"MERGED","created":"2020-05-03 11:46:53.000000000","updated":"2020-06-04 23:16:47.000000000","submitted":"2020-06-04 23:15:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"725096-1591312516359-5027401d","meta_rev_id":"afe726e8dfe287a3b01044dc798ed78969c9dda1","_number":725096,"virtual_id_number":725096,"owner":{"_account_id":20466,"name":"Tin Lam","email":"tin@lam.wtf","username":"tinlam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},{"value":0,"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"_account_id":30777,"name":"Steven Fitzpatrick","email":"steven@fitzpatrick.wtf","username":"sf280x"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-04 23:15:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},"all":[{"value":2,"date":"2020-05-31 17:05:46.000000000","_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},{"value":1,"date":"2020-05-31 17:05:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":2,"date":"2020-06-04 22:25:40.000000000","_account_id":30777,"name":"Steven Fitzpatrick","email":"steven@fitzpatrick.wtf","username":"sf280x"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30777,"name":"Steven Fitzpatrick","email":"steven@fitzpatrick.wtf","username":"sf280x"},"all":[{"value":0,"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},{"value":0,"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":1,"date":"2020-06-04 22:25:40.000000000","_account_id":30777,"name":"Steven Fitzpatrick","email":"steven@fitzpatrick.wtf","username":"sf280x"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30777,"name":"Steven Fitzpatrick","email":"steven@fitzpatrick.wtf","username":"sf280x"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-19 18:53:05.000000000","updated_by":{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},"reviewer":{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},"state":"REVIEWER"},{"updated":"2020-05-20 01:27:16.000000000","updated_by":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"reviewer":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"state":"REVIEWER"},{"updated":"2020-06-04 22:25:40.000000000","updated_by":{"_account_id":30777,"name":"Steven Fitzpatrick","email":"steven@fitzpatrick.wtf","username":"sf280x"},"reviewer":{"_account_id":30777,"name":"Steven Fitzpatrick","email":"steven@fitzpatrick.wtf","username":"sf280x"},"state":"REVIEWER"},{"updated":"2020-06-04 23:15:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"56c49b826d9311ebef98921749a2494b0fce5e65","author":{"_account_id":20466,"name":"Tin Lam","email":"tin@lam.wtf","username":"tinlam"},"date":"2020-05-03 11:46:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4cfc2930f6e02e7fb331a2e2e2ab122771233a9d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-03 12:51:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1545a46f7ed048a9bb1011511cce2dd0 : SUCCESS in 3m 53s\n- openstack-helm-lint https://zuul.opendev.org/t/openstack/build/237dde33eaa547f5ae25e70b6cc45b4b : SUCCESS in 1m 44s\n- openstack-helm-infra-aio-logging https://zuul.opendev.org/t/openstack/build/36c60b70d2684f68b368d09b874fcfa9 : SUCCESS in 40m 51s\n- openstack-helm-infra-aio-monitoring https://zuul.opendev.org/t/openstack/build/eb6853f8782e448f97b4048fec3e3007 : SUCCESS in 42m 22s\n- openstack-helm-infra-openstack-support https://zuul.opendev.org/t/openstack/build/383acf1fd4cf4a099cbf1173d2f0b9b9 : SUCCESS in 1h 04m 03s\n- openstack-helm-infra-airship-divingbell https://zuul.opendev.org/t/openstack/build/83d787b50b46490392ae7d46b3f6ef83 : SUCCESS in 13m 15s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"cdc6e4bc582ced53bef8c6791281460377b3e428","author":{"_account_id":20466,"name":"Tin Lam","email":"tin@lam.wtf","username":"tinlam"},"date":"2020-05-10 22:07:19.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"f538bf2356fdfd71849401ce4792d17bb50d5c99","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-10 22:55:33.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ed67071b2d840bfa7760863ee014f8b : SUCCESS in 4m 18s\n- openstack-helm-lint https://zuul.opendev.org/t/openstack/build/988c1b997e0d4be284fb1df69edb8cbb : SUCCESS in 2m 08s\n- openstack-helm-infra-aio-logging https://zuul.opendev.org/t/openstack/build/86ce1d20b71f43f7be360965fc47da6f : SUCCESS in 44m 24s\n- openstack-helm-infra-aio-monitoring https://zuul.opendev.org/t/openstack/build/ffcc4a7324604503a6e164c98faa4bbd : SUCCESS in 39m 40s\n- openstack-helm-infra-openstack-support https://zuul.opendev.org/t/openstack/build/f99333dc97504a58b425be00d152cbf6 : SUCCESS in 47m 24s\n- openstack-helm-infra-airship-divingbell https://zuul.opendev.org/t/openstack/build/58e09be8916842f399dd44b71145d8d1 : SUCCESS in 12m 59s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c77e2a48117b167b48e52c7c107fe3879eab1c39","author":{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},"date":"2020-05-19 18:53:05.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"7a3d89d7d53632b78afe25512b46061a1177014e","author":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"date":"2020-05-20 01:27:16.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"3a7937c1805170e3f5e7cc5b56f1eb52287bd032","author":{"_account_id":20466,"name":"Tin Lam","email":"tin@lam.wtf","username":"tinlam"},"date":"2020-05-23 03:50:47.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"a2d419a4f31cfba58ae2dab508b47001e119492b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-23 04:56:27.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/11001a30de0e44fea278f797fbfdf3f7 : SUCCESS in 4m 37s\n- openstack-helm-lint https://zuul.opendev.org/t/openstack/build/18345f6e7cf247a1870a9d5b1eb2a8c9 : SUCCESS in 2m 09s\n- openstack-helm-infra-aio-logging https://zuul.opendev.org/t/openstack/build/431d3e78d5d2485a850cf838831add27 : SUCCESS in 42m 40s\n- openstack-helm-infra-aio-monitoring https://zuul.opendev.org/t/openstack/build/a7f3fd694cb74f5ba551929bb987356b : SUCCESS in 57m 52s\n- openstack-helm-infra-openstack-support https://zuul.opendev.org/t/openstack/build/da733452a0c24c7382de94cc391d8812 : SUCCESS in 49m 44s\n- openstack-helm-infra-airship-divingbell https://zuul.opendev.org/t/openstack/build/9bdbb1f91db2469da83e599e20422e25 : SUCCESS in 13m 09s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"ac945e2ae7467595dcb4c870a6a8391961f63f92","author":{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},"date":"2020-05-24 02:07:14.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"db9c2628c42f57ec8261d5f45a69aea9ba6ce2de","author":{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},"date":"2020-05-24 02:07:17.000000000","message":"Removed Workflow+1 by Chris Wedgwood \u003ccw@f00f.org\u003e\n","accounts_in_message":[],"_revision_number":3},{"id":"279e0a46fb23c024237ff64e57957bfb5c6be7a3","author":{"_account_id":20466,"name":"Tin Lam","email":"tin@lam.wtf","username":"tinlam"},"date":"2020-05-31 17:05:46.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"71701b66cbd99da2598587654b9a588d786c69a3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-31 17:53:46.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/afba30d711494fd09b39fe1fe4b5a835 : SUCCESS in 3m 00s\n- openstack-helm-lint https://zuul.opendev.org/t/openstack/build/7f5a1b15f53b4e89a98ebe7e89ecedf7 : SUCCESS in 1m 45s\n- openstack-helm-infra-aio-logging https://zuul.opendev.org/t/openstack/build/4f14bb6a50074e52ab907765380e0b37 : SUCCESS in 42m 56s\n- openstack-helm-infra-aio-monitoring https://zuul.opendev.org/t/openstack/build/77356148b3aa435caf97c241dfbb7e43 : SUCCESS in 34m 53s\n- openstack-helm-infra-openstack-support https://zuul.opendev.org/t/openstack/build/f80e08442a874e7692729350421f5bb0 : SUCCESS in 47m 10s\n- openstack-helm-infra-airship-divingbell https://zuul.opendev.org/t/openstack/build/a340a070acb54b45badf380ae47ca6d6 : SUCCESS in 11m 37s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"5a2b80b5befb9596c45e60479c6d24f3bbd53ab7","author":{"_account_id":30777,"name":"Steven Fitzpatrick","email":"steven@fitzpatrick.wtf","username":"sf280x"},"date":"2020-06-04 22:25:40.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"a52c8f9cbe19af8dc526c4ef115db1d82482c410","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 22:25:52.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"707c922ec4a547032e831e0400a2760c2c370d97","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 23:15:16.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/83ec3941cdf04b59ab1ff5a17baef577 : SUCCESS in 3m 40s\n- openstack-helm-lint https://zuul.opendev.org/t/openstack/build/be5e43cf9b2a47df8c18a722c6813f3a : SUCCESS in 2m 23s\n- openstack-helm-infra-aio-logging https://zuul.opendev.org/t/openstack/build/67ed60c6da0640949db20032de53622a : SUCCESS in 34m 37s\n- openstack-helm-infra-aio-monitoring https://zuul.opendev.org/t/openstack/build/0360916b0caa4b339235987d6fcdcbe0 : SUCCESS in 37m 05s\n- openstack-helm-infra-openstack-support https://zuul.opendev.org/t/openstack/build/6d755dc67a8a4b9faa15e51150552bf2 : SUCCESS in 48m 12s","accounts_in_message":[],"_revision_number":4},{"id":"19366a7769d092f9fb2f4b1b619123b37b8e1d31","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 23:15:16.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"346076631ea14c28dad311113bffe171893ec91d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 23:16:47.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57b35421820a414bba6c9db8055ca964 : SUCCESS in 1m 16s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"e12310088809d3a2468339683b65d19612514077","revisions":{"719ad03f63910036ddf66a03be26ab984e6e5d96":{"kind":"REWORK","_number":1,"created":"2020-05-03 11:46:53.000000000","uploader":{"_account_id":20466,"name":"Tin Lam","email":"tin@lam.wtf","username":"tinlam"},"ref":"refs/changes/96/725096/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-helm-infra","ref":"refs/changes/96/725096/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/96/725096/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/96/725096/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/96/725096/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-helm-infra refs/changes/96/725096/1"}}},"commit":{"parents":[{"commit":"a12ea0244e75804793d3b1e2bde714643caf1edf","subject":"[ceph-osd] Use lvm commands instead of ceph-volume to get OSD properties","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/a12ea0244e75804793d3b1e2bde714643caf1edf"}]}],"author":{"name":"Tin Lam","email":"tin@irrational.io","date":"2020-05-03 11:46:39.000000000","tz":-300},"committer":{"name":"Tin Lam","email":"tin@irrational.io","date":"2020-05-03 11:46:39.000000000","tz":-300},"subject":"chore(tox): update tox.ini","message":"chore(tox): update tox.ini\n\nThis updates to use TOX_CONSTRAINTS_FILE instead of\nUPPER_CONSTRAINTS_FILE since the latter is obsolete.\n\nChange-Id: Ib31adb98e822b1b57acd8fd2f3f338e6cfe24c23\nSigned-off-by: Tin Lam \u003ctin@irrational.io\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/719ad03f63910036ddf66a03be26ab984e6e5d96"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/719ad03f63910036ddf66a03be26ab984e6e5d96"}]},"branch":"refs/heads/master"},"0915de099751109a8356ef765d739db02752abed":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-05-10 22:07:19.000000000","uploader":{"_account_id":20466,"name":"Tin Lam","email":"tin@lam.wtf","username":"tinlam"},"ref":"refs/changes/96/725096/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-helm-infra","ref":"refs/changes/96/725096/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/96/725096/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/96/725096/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/96/725096/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-helm-infra refs/changes/96/725096/2"}}},"commit":{"parents":[{"commit":"f8c5d0676904cdb05de80234cb8659bde856420a","subject":"Merge \"Settings for Remote Elasticsearch Clusters\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/f8c5d0676904cdb05de80234cb8659bde856420a"}]}],"author":{"name":"Tin Lam","email":"tin@irrational.io","date":"2020-05-03 11:46:39.000000000","tz":-300},"committer":{"name":"Tin Lam","email":"tin@irrational.io","date":"2020-05-10 22:07:19.000000000","tz":0},"subject":"chore(tox): update tox.ini","message":"chore(tox): update tox.ini\n\nThis updates to use TOX_CONSTRAINTS_FILE instead of\nUPPER_CONSTRAINTS_FILE since the latter is obsolete.\n\nChange-Id: Ib31adb98e822b1b57acd8fd2f3f338e6cfe24c23\nSigned-off-by: Tin Lam \u003ctin@irrational.io\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/0915de099751109a8356ef765d739db02752abed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/0915de099751109a8356ef765d739db02752abed"}]},"branch":"refs/heads/master"},"293f497480b9d8196348dcd0e2541a51468a8d45":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-05-23 03:50:47.000000000","uploader":{"_account_id":20466,"name":"Tin Lam","email":"tin@lam.wtf","username":"tinlam"},"ref":"refs/changes/96/725096/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-helm-infra","ref":"refs/changes/96/725096/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/96/725096/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/96/725096/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/96/725096/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-helm-infra refs/changes/96/725096/3"}}},"commit":{"parents":[{"commit":"3691681d84d51e2019c24f0dad9f42962b06f10f","subject":"Merge \"Enable yamllint checks\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/3691681d84d51e2019c24f0dad9f42962b06f10f"}]}],"author":{"name":"Tin Lam","email":"tin@irrational.io","date":"2020-05-03 11:46:39.000000000","tz":-300},"committer":{"name":"Tin Lam","email":"tin@irrational.io","date":"2020-05-23 03:50:47.000000000","tz":0},"subject":"chore(tox): update tox.ini","message":"chore(tox): update tox.ini\n\nThis updates to use TOX_CONSTRAINTS_FILE instead of\nUPPER_CONSTRAINTS_FILE since the latter is obsolete.\n\nChange-Id: Ib31adb98e822b1b57acd8fd2f3f338e6cfe24c23\nSigned-off-by: Tin Lam \u003ctin@irrational.io\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/293f497480b9d8196348dcd0e2541a51468a8d45"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/293f497480b9d8196348dcd0e2541a51468a8d45"}]},"branch":"refs/heads/master"},"e12310088809d3a2468339683b65d19612514077":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2020-05-31 17:05:46.000000000","uploader":{"_account_id":20466,"name":"Tin Lam","email":"tin@lam.wtf","username":"tinlam"},"ref":"refs/changes/96/725096/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-helm-infra","ref":"refs/changes/96/725096/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/96/725096/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/96/725096/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/96/725096/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-helm-infra refs/changes/96/725096/4"}}},"commit":{"parents":[{"commit":"934d7946bed2c314857f3ee3e9550b39e02bd1d7","subject":"Merge \"Enable yamllint checks\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/934d7946bed2c314857f3ee3e9550b39e02bd1d7"}]}],"author":{"name":"Tin Lam","email":"tin@irrational.io","date":"2020-05-03 11:46:39.000000000","tz":-300},"committer":{"name":"Tin Lam","email":"tin@irrational.io","date":"2020-05-31 17:05:46.000000000","tz":0},"subject":"chore(tox): update tox.ini","message":"chore(tox): update tox.ini\n\nThis updates to use TOX_CONSTRAINTS_FILE instead of\nUPPER_CONSTRAINTS_FILE since the latter is obsolete.\n\nChange-Id: Ib31adb98e822b1b57acd8fd2f3f338e6cfe24c23\nSigned-off-by: Tin Lam \u003ctin@irrational.io\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/e12310088809d3a2468339683b65d19612514077"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/e12310088809d3a2468339683b65d19612514077"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
