)]}'
{"id":"openstack%2Fopenstack-helm-infra~741037","triplet_id":"openstack%2Fopenstack-helm-infra~master~Id38c9fb0b18dd8ba164a69f179d940192efc3247","project":"openstack/openstack-helm-infra","branch":"master","hashtags":[],"change_id":"Id38c9fb0b18dd8ba164a69f179d940192efc3247","subject":"feat(tls): add tls to mariadb chart","status":"MERGED","created":"2020-07-14 19:35:53.000000000","updated":"2020-07-14 22:44:15.000000000","submitted":"2020-07-14 22:42:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"741037-1594766554774-98bbf7d3","meta_rev_id":"cfdf55f17fb8c735b15a3ed90f1d17bd65990c91","_number":741037,"virtual_id_number":741037,"owner":{"_account_id":24780,"name":"Sangeet Gupta","email":"sg774j@att.com","username":"sgupta"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":20466,"name":"Tin Lam","email":"tin@lam.wtf","username":"tinlam"},{"value":0,"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-14 22:42:34.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":30495,"name":"Nafiz Haider","email":"nafizhaider22@gmail.com","username":"nafh27"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":20466,"name":"Tin Lam","email":"tin@lam.wtf","username":"tinlam"},"all":[{"value":2,"date":"2020-07-14 19:47:21.000000000","_account_id":20466,"name":"Tin Lam","email":"tin@lam.wtf","username":"tinlam"},{"value":2,"date":"2020-07-14 21:28:42.000000000","_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-07-14 20:48:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30495,"name":"Nafiz Haider","email":"nafizhaider22@gmail.com","username":"nafh27"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},"all":[{"value":0,"_account_id":20466,"name":"Tin Lam","email":"tin@lam.wtf","username":"tinlam"},{"value":1,"date":"2020-07-14 21:28:42.000000000","_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":30495,"name":"Nafiz Haider","email":"nafizhaider22@gmail.com","username":"nafh27"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},{"_account_id":20466,"name":"Tin Lam","email":"tin@lam.wtf","username":"tinlam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30495,"name":"Nafiz Haider","email":"nafizhaider22@gmail.com","username":"nafh27"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-14 19:47:21.000000000","updated_by":{"_account_id":20466,"name":"Tin Lam","email":"tin@lam.wtf","username":"tinlam"},"reviewer":{"_account_id":20466,"name":"Tin Lam","email":"tin@lam.wtf","username":"tinlam"},"state":"REVIEWER"},{"updated":"2020-07-14 20:48:36.000000000","updated_by":{"_account_id":30495,"name":"Nafiz Haider","email":"nafizhaider22@gmail.com","username":"nafh27"},"reviewer":{"_account_id":30495,"name":"Nafiz Haider","email":"nafizhaider22@gmail.com","username":"nafh27"},"state":"REVIEWER"},{"updated":"2020-07-14 21:28:42.000000000","updated_by":{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},"reviewer":{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},"state":"REVIEWER"},{"updated":"2020-07-14 22:42:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fba5b016e57dced1a551c130d6cdc7634cd0d213","author":{"_account_id":24780,"name":"Sangeet Gupta","email":"sg774j@att.com","username":"sgupta"},"date":"2020-07-14 19:35:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a3bddd49260be659af63849d4b937b3ebd690f12","author":{"_account_id":20466,"name":"Tin Lam","email":"tin@lam.wtf","username":"tinlam"},"date":"2020-07-14 19:47:21.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b872f4ecdcf56b009017b48f9e4abdd8b0816345","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-14 20:34:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd7b2545f3674a5aa49cdd0e55120ac7 : SUCCESS in 4m 00s\n- openstack-helm-lint https://zuul.opendev.org/t/openstack/build/bee20dcca8e44551be9f658b7b0fd84b : SUCCESS in 3m 05s\n- openstack-helm-infra-bandit https://zuul.opendev.org/t/openstack/build/f16d86c05f2c45daa6b6ecb029dd04eb : SUCCESS in 7m 18s\n- openstack-helm-infra-aio-logging https://zuul.opendev.org/t/openstack/build/860a3e9a9b81449aa0c67ae1bc68b18b : SUCCESS in 36m 33s\n- openstack-helm-infra-aio-monitoring https://zuul.opendev.org/t/openstack/build/040e9d96055742c4af225f9a2b053a3a : SUCCESS in 40m 33s\n- openstack-helm-infra-openstack-support https://zuul.opendev.org/t/openstack/build/58615cd3f1414d99b0d1e45eea22c622 : SUCCESS in 58m 07s\n- openstack-helm-infra-airship-divingbell https://zuul.opendev.org/t/openstack/build/323ca486c858439f9c51592a5bed7e87 : SUCCESS in 13m 27s (non-voting)\n- openstack-helm-infra-metacontroller https://zuul.opendev.org/t/openstack/build/9477df7839b84399b61eb9c8f5685eb6 : SUCCESS in 18m 18s","accounts_in_message":[],"_revision_number":1},{"id":"da997487df030e9cf6b5780b0b86f5e1e877fe8a","author":{"_account_id":30495,"name":"Nafiz Haider","email":"nafizhaider22@gmail.com","username":"nafh27"},"date":"2020-07-14 20:48:36.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"fdb22d896db44928cb45c70e5e928e92eaf8f824","author":{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},"date":"2020-07-14 21:28:42.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"161c1bf3f40c5b190436648225dafe77d8dad324","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-14 21:28:54.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"68e96a2665d5f4c5e1bdad2701b0fe49d1b1212b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-14 22:42:34.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"e33c7fb0d382ae04f4d6398abb0b14473ea10a51","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-14 22:42:34.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af3b3d93e4954aa7ad28ca24d8cc93c0 : SUCCESS in 4m 22s\n- openstack-helm-lint https://zuul.opendev.org/t/openstack/build/0f8db7f997a94e4d9d4962f8749edc09 : SUCCESS in 2m 57s\n- openstack-helm-infra-aio-logging https://zuul.opendev.org/t/openstack/build/6468ed9ec9c143edb5515fb7891c9149 : SUCCESS in 50m 46s\n- openstack-helm-infra-aio-monitoring https://zuul.opendev.org/t/openstack/build/c34e8d8172d44b26bf2d04119252b72a : SUCCESS in 58m 15s\n- openstack-helm-infra-openstack-support https://zuul.opendev.org/t/openstack/build/2c9a20f2e0114685b5edd1845f32baf8 : SUCCESS in 1h 13m 00s","accounts_in_message":[],"_revision_number":1},{"id":"cfdf55f17fb8c735b15a3ed90f1d17bd65990c91","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-14 22:44:15.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f1297af7aa54120bd012c9f4be51e51 : SUCCESS in 1m 20s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6e13d74c876a5c3eaf03b664f3df538972c54452","revisions":{"6e13d74c876a5c3eaf03b664f3df538972c54452":{"kind":"REWORK","_number":1,"created":"2020-07-14 19:35:53.000000000","uploader":{"_account_id":24780,"name":"Sangeet Gupta","email":"sg774j@att.com","username":"sgupta"},"ref":"refs/changes/37/741037/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-helm-infra","ref":"refs/changes/37/741037/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/37/741037/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/37/741037/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/37/741037/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-helm-infra refs/changes/37/741037/1"}}},"commit":{"parents":[{"commit":"b1fc699808c903e57e0fdf32230bd9fab377faaa","subject":"Merge \"Elasticsearch: Improve logging in cluster wait\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/b1fc699808c903e57e0fdf32230bd9fab377faaa"}]}],"author":{"name":"sgupta","email":"sg774j@att.com","date":"2020-07-14 19:32:57.000000000","tz":0},"committer":{"name":"sgupta","email":"sg774j@att.com","date":"2020-07-14 19:32:57.000000000","tz":0},"subject":"feat(tls): add tls to mariadb chart","message":"feat(tls): add tls to mariadb chart\n\nThis patch set makes changes for maraidb certs to be used\nby all users when connecting to MariaDB.\n\nChange-Id: Id38c9fb0b18dd8ba164a69f179d940192efc3247\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/6e13d74c876a5c3eaf03b664f3df538972c54452"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/6e13d74c876a5c3eaf03b664f3df538972c54452"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
