)]}'
{"id":"openstack%2Fopenstack-helm-infra~759701","triplet_id":"openstack%2Fopenstack-helm-infra~master~I4f6f015bcd803ad636f55b191c90d1447bef51ab","project":"openstack/openstack-helm-infra","branch":"master","hashtags":[],"change_id":"I4f6f015bcd803ad636f55b191c90d1447bef51ab","subject":"Introduce prometheus-webhook-snmp","status":"ABANDONED","created":"2020-10-26 14:14:16.000000000","updated":"2021-02-04 01:53:02.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"379e38772220cefc075b267e3f5c9be5c57cfb87","_number":759701,"virtual_id_number":759701,"owner":{"_account_id":31713,"name":"William Zhang","email":"xz8905@att.com","username":"willxz"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":31713,"name":"William Zhang","email":"xz8905@att.com","username":"willxz"},{"_account_id":31026,"name":"Akshay","display_name":"Akshay Wasankar","email":"akshaywasankar51@gmail.com","username":"akshay_wasankar"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-11-24 01:00:51.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":31026,"name":"Akshay","display_name":"Akshay Wasankar","email":"akshaywasankar51@gmail.com","username":"akshay_wasankar"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":31713,"name":"William Zhang","email":"xz8905@att.com","username":"willxz"},{"value":1,"date":"2021-02-04 01:53:02.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":31026,"name":"Akshay","display_name":"Akshay Wasankar","email":"akshaywasankar51@gmail.com","username":"akshay_wasankar"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":31713,"name":"William Zhang","email":"xz8905@att.com","username":"willxz"},{"_account_id":31026,"name":"Akshay","display_name":"Akshay Wasankar","email":"akshaywasankar51@gmail.com","username":"akshay_wasankar"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31026,"name":"Akshay","display_name":"Akshay Wasankar","email":"akshaywasankar51@gmail.com","username":"akshay_wasankar"},{"_account_id":31713,"name":"William Zhang","email":"xz8905@att.com","username":"willxz"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-26 16:59:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-10-26 19:26:41.000000000","updated_by":{"_account_id":30582,"name":"RADHIKA PAI","email":"rp592h@att.com","username":"rp592h"},"reviewer":{"_account_id":30582,"name":"RADHIKA PAI","email":"rp592h@att.com","username":"rp592h"},"state":"REVIEWER"},{"updated":"2020-10-28 18:38:24.000000000","updated_by":{"_account_id":31026,"name":"Akshay","display_name":"Akshay Wasankar","email":"akshaywasankar51@gmail.com","username":"akshay_wasankar"},"reviewer":{"_account_id":31026,"name":"Akshay","display_name":"Akshay Wasankar","email":"akshaywasankar51@gmail.com","username":"akshay_wasankar"},"state":"REVIEWER"},{"updated":"2021-02-03 15:29:44.000000000","updated_by":{"_account_id":30582,"name":"RADHIKA PAI","email":"rp592h@att.com","username":"rp592h"},"reviewer":{"_account_id":30582,"name":"RADHIKA PAI","email":"rp592h@att.com","username":"rp592h"},"state":"REMOVED"}],"messages":[{"id":"1425dbedf6cff9b4da6949381bfc25521388ebe4","author":{"_account_id":31713,"name":"William Zhang","email":"xz8905@att.com","username":"willxz"},"date":"2020-10-26 14:14:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9ec3a639dba4c633c79c4bfcadccbe5887385197","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-26 15:56:13.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/935bf54f01534b22a97271c1a3fb72c8 : SUCCESS in 3m 01s\n- openstack-helm-lint https://zuul.opendev.org/t/openstack/build/4a5c30c7163047a7a220505889a76834 : SUCCESS in 3m 46s\n- openstack-helm-infra-aio-logging https://zuul.opendev.org/t/openstack/build/ca4f70b0936c4c47a4008363ba2572eb : SUCCESS in 33m 47s\n- openstack-helm-infra-aio-monitoring https://zuul.opendev.org/t/openstack/build/1f9ac53cc96645b8bf2fd2d3a134be48 : SUCCESS in 30m 35s\n- openstack-helm-infra-openstack-support https://zuul.opendev.org/t/openstack/build/ef43c90135644a62a86aba56800b0dae : FAILURE in 1h 18m 09s\n- openstack-helm-infra-openstack-support-ssl https://zuul.opendev.org/t/openstack/build/ae54d349acf24621a7549f6487f1592a : SUCCESS in 44m 47s\n- openstack-helm-infra-airship-divingbell https://zuul.opendev.org/t/openstack/build/604b542520af465aa9122e8f742cff2e : SUCCESS in 10m 43s (non-voting)\n- openstack-helm-infra-metacontroller https://zuul.opendev.org/t/openstack/build/cf0a698e8d2746099cdef458c25151c7 : SUCCESS in 13m 32s","accounts_in_message":[],"_revision_number":1},{"id":"a120ac0de8e692070128206e001bd3c31b07d014","author":{"_account_id":31713,"name":"William Zhang","email":"xz8905@att.com","username":"willxz"},"date":"2020-10-26 16:10:35.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"05f240ccef90c3f97354aef53ed9d379e8c65dd2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-26 16:59:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cea6357f59484e528b381eedcef9b3e0 : SUCCESS in 3m 26s\n- openstack-helm-lint https://zuul.opendev.org/t/openstack/build/a035051bc93244aea65e47dc727dd8d1 : SUCCESS in 3m 47s\n- openstack-helm-infra-aio-logging https://zuul.opendev.org/t/openstack/build/85a2679a10974846803e0fd4d5977bc0 : SUCCESS in 34m 30s\n- openstack-helm-infra-aio-monitoring https://zuul.opendev.org/t/openstack/build/aad13ef7d0d4471994b749af3d499907 : SUCCESS in 33m 22s\n- openstack-helm-infra-openstack-support https://zuul.opendev.org/t/openstack/build/3141098d3fa846ac971e791f61bff641 : SUCCESS in 46m 38s\n- openstack-helm-infra-openstack-support-ssl https://zuul.opendev.org/t/openstack/build/d506c780e8fd40dda9ec3c2c19121fb3 : SUCCESS in 46m 59s\n- openstack-helm-infra-airship-divingbell https://zuul.opendev.org/t/openstack/build/4ad3ca74e7fa491381e70f9eb06de95c : SUCCESS in 11m 18s (non-voting)\n- openstack-helm-infra-metacontroller https://zuul.opendev.org/t/openstack/build/6d30151511e347b1b76a7a5964f5ac4c : SUCCESS in 13m 24s","accounts_in_message":[],"_revision_number":1},{"id":"a192b591c9873b3addd18555237b688fa81bdb5a","author":{"_account_id":30582,"name":"RADHIKA PAI","email":"rp592h@att.com","username":"rp592h"},"date":"2020-10-26 19:26:41.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"72a07fc1152d057223e2c95ab10dcfef203562f3","author":{"_account_id":31026,"name":"Akshay","display_name":"Akshay Wasankar","email":"akshaywasankar51@gmail.com","username":"akshay_wasankar"},"date":"2020-10-28 18:38:24.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"978ec7886254b63d43037a2611f9ab83f775e728","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},"date":"2020-11-24 00:10:25.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"1e4bb5d94daa255a9a0ea3a1db1f8bab5ab0b068","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-24 01:00:51.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7188724e6468416a9c933cf21be692ca : SUCCESS in 2m 45s\n- openstack-helm-lint https://zuul.opendev.org/t/openstack/build/7fb53739ce46465a832feb4eaa769e7b : SUCCESS in 3m 24s\n- openstack-helm-infra-aio-logging https://zuul.opendev.org/t/openstack/build/1f8fae055dad4ac8ba6f71504ccb7921 : SUCCESS in 48m 21s\n- openstack-helm-infra-aio-monitoring https://zuul.opendev.org/t/openstack/build/f741559a2cbb44e7807de2b83306d27b : SUCCESS in 32m 03s\n- openstack-helm-infra-openstack-support https://zuul.opendev.org/t/openstack/build/d28d568effe94b02a623f97e3dcf5d0b : SUCCESS in 44m 56s\n- openstack-helm-infra-openstack-support-ssl https://zuul.opendev.org/t/openstack/build/346078f8428b449a867332fbe838d5bf : SUCCESS in 48m 45s\n- openstack-helm-infra-metacontroller https://zuul.opendev.org/t/openstack/build/b4efd3b9154d473d83d4a6107bb3a37d : SUCCESS in 16m 38s","accounts_in_message":[],"_revision_number":2},{"id":"15167cfb24743fc116c7f140e438ee58c569aa45","tag":"autogenerated:gerrit:deleteReviewer","author":{"_account_id":30582,"name":"RADHIKA PAI","email":"rp592h@att.com","username":"rp592h"},"date":"2021-02-03 15:29:44.000000000","message":"Removed reviewer RADHIKA PAI.","accounts_in_message":[],"_revision_number":2},{"id":"475f69718b293f179754a06ebdda02f861771949","tag":"autogenerated:gerrit:abandon","author":{"_account_id":31713,"name":"William Zhang","email":"xz8905@att.com","username":"willxz"},"date":"2021-02-04 01:53:02.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"2c0bbbf766c5267ec39dd29e1702a0efd12aeb8f","revisions":{"97c1612b5984650ff21323f3715d2da830b8750e":{"kind":"REWORK","_number":1,"created":"2020-10-26 14:14:16.000000000","uploader":{"_account_id":31713,"name":"William Zhang","email":"xz8905@att.com","username":"willxz"},"ref":"refs/changes/01/759701/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-helm-infra","ref":"refs/changes/01/759701/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/01/759701/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/01/759701/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/01/759701/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-helm-infra refs/changes/01/759701/1"}}},"commit":{"parents":[{"commit":"757a353b70e8f1333c664502f8abc76a801e3844","subject":"Merge \"postgresql: Revert \"Add default reject rule ...\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/757a353b70e8f1333c664502f8abc76a801e3844"}]}],"author":{"name":"Xiaoguang(William) Zhang","email":"xz8905@att.com","date":"2020-10-26 14:12:35.000000000","tz":-240},"committer":{"name":"Xiaoguang(William) Zhang","email":"xz8905@att.com","date":"2020-10-26 14:14:06.000000000","tz":-240},"subject":"Introduce prometheus-webhook-snmp","message":"Introduce prometheus-webhook-snmp\n\nAdd prometheus-webhook-snmp to alertmnager\ndisable snmp_notifier from alertmanager\n\nChange-Id: I4f6f015bcd803ad636f55b191c90d1447bef51ab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/97c1612b5984650ff21323f3715d2da830b8750e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/97c1612b5984650ff21323f3715d2da830b8750e"}]},"branch":"refs/heads/master"},"2c0bbbf766c5267ec39dd29e1702a0efd12aeb8f":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-11-24 00:10:25.000000000","uploader":{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},"ref":"refs/changes/01/759701/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-helm-infra","ref":"refs/changes/01/759701/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/01/759701/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/01/759701/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/01/759701/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-helm-infra refs/changes/01/759701/2"}}},"commit":{"parents":[{"commit":"15ad6e9a6cdd1dd0c80be4672fd4efcb5002fda2","subject":"Merge \"fix(secret): changes rmq-exporter secret src\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/15ad6e9a6cdd1dd0c80be4672fd4efcb5002fda2"}]}],"author":{"name":"Xiaoguang(William) Zhang","email":"xz8905@att.com","date":"2020-10-26 14:12:35.000000000","tz":-240},"committer":{"name":"Chris Wedgwood","email":"cw@f00f.org","date":"2020-11-24 00:10:25.000000000","tz":0},"subject":"Introduce prometheus-webhook-snmp","message":"Introduce prometheus-webhook-snmp\n\nAdd prometheus-webhook-snmp to alertmnager\ndisable snmp_notifier from alertmanager\n\nChange-Id: I4f6f015bcd803ad636f55b191c90d1447bef51ab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/2c0bbbf766c5267ec39dd29e1702a0efd12aeb8f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/2c0bbbf766c5267ec39dd29e1702a0efd12aeb8f"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
