)]}'
{"id":"openstack%2Fopenstack-helm-infra~762709","triplet_id":"openstack%2Fopenstack-helm-infra~master~I954972e5810b80a5abe8b882838b9a3accafe880","project":"openstack/openstack-helm-infra","branch":"master","hashtags":[],"change_id":"I954972e5810b80a5abe8b882838b9a3accafe880","subject":"[CEPH-POOL] Read the quota per pool from pool spec","status":"ABANDONED","created":"2020-11-13 19:14:24.000000000","updated":"2020-11-24 16:09:20.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"7baefab099b7e5904253903e20af5fb3fe3d531d","_number":762709,"virtual_id_number":762709,"owner":{"_account_id":28372,"name":"chinasubbareddy mallavarapu","email":"cr3938@att.com","username":"chinasubbareddy"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},{"_account_id":28372,"name":"chinasubbareddy mallavarapu","email":"cr3938@att.com","username":"chinasubbareddy"},{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-11-24 00:54:06.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},"all":[{"value":2,"date":"2020-11-24 04:55:55.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28372,"name":"chinasubbareddy mallavarapu","email":"cr3938@att.com","username":"chinasubbareddy"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":28372,"name":"chinasubbareddy mallavarapu","email":"cr3938@att.com","username":"chinasubbareddy"},{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28372,"name":"chinasubbareddy mallavarapu","email":"cr3938@att.com","username":"chinasubbareddy"},{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-16 23:56:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-17 00:02:54.000000000","updated_by":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"reviewer":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"state":"REVIEWER"},{"updated":"2020-11-24 04:55:55.000000000","updated_by":{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},"reviewer":{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},"state":"REVIEWER"}],"messages":[{"id":"ec8936c9063b52a663c55f6a75764fe90b809136","author":{"_account_id":28372,"name":"chinasubbareddy mallavarapu","email":"cr3938@att.com","username":"chinasubbareddy"},"date":"2020-11-13 19:14:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"88455cb3272da8226b98d9b21a566a5ea392e14d","author":{"_account_id":28372,"name":"chinasubbareddy mallavarapu","email":"cr3938@att.com","username":"chinasubbareddy"},"date":"2020-11-13 19:15:31.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"cf4a507f008514109ade0fc759e9ba051b18641b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-13 20:00:00.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/29143d37778641f0b43b682a56be5e3d : SUCCESS in 4m 42s\n- openstack-helm-lint https://zuul.opendev.org/t/openstack/build/7e194e43453c4adaad9be4be0cc75d64 : SUCCESS in 3m 47s\n- openstack-helm-infra-aio-logging https://zuul.opendev.org/t/openstack/build/b48ce396121d4737bf361599ba1bcd38 : FAILURE in 31m 58s\n- openstack-helm-infra-aio-monitoring https://zuul.opendev.org/t/openstack/build/694aa55567aa49d2bf271c55144dca3c : SUCCESS in 30m 00s\n- openstack-helm-infra-openstack-support https://zuul.opendev.org/t/openstack/build/8997184bbb4d49539648e47af0005852 : FAILURE in 30m 30s\n- openstack-helm-infra-openstack-support-ssl https://zuul.opendev.org/t/openstack/build/5ac9f0ee89f942479142a5b5084c6c60 : FAILURE in 30m 59s\n- openstack-helm-infra-metacontroller https://zuul.opendev.org/t/openstack/build/3bd5cd532d6944f8a8a77963a4361757 : SUCCESS in 15m 02s","accounts_in_message":[],"_revision_number":2},{"id":"9ba59d947fda2b847c2bcebeacc00cebc0369ccc","author":{"_account_id":28372,"name":"chinasubbareddy mallavarapu","email":"cr3938@att.com","username":"chinasubbareddy"},"date":"2020-11-13 20:59:30.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"3d889758edfce875c222774122df25b4af719379","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-13 21:49:39.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bdd3bd0c014045fc989894abb9c14493 : SUCCESS in 4m 24s\n- openstack-helm-lint https://zuul.opendev.org/t/openstack/build/1cb5e24ef35c41a581e4a796cf4a4d68 : SUCCESS in 4m 22s\n- openstack-helm-infra-aio-logging https://zuul.opendev.org/t/openstack/build/f14196d462624937b84c9497b39aa3fa : SUCCESS in 32m 10s\n- openstack-helm-infra-aio-monitoring https://zuul.opendev.org/t/openstack/build/774f877670de42af882a8445c9c12978 : SUCCESS in 33m 40s\n- openstack-helm-infra-openstack-support https://zuul.opendev.org/t/openstack/build/6accf9b2faef4f3492f5dd32989dd90b : SUCCESS in 46m 17s\n- openstack-helm-infra-openstack-support-ssl https://zuul.opendev.org/t/openstack/build/280825fdedd94b9c980eb2fb1493e6bd : SUCCESS in 46m 26s\n- openstack-helm-infra-metacontroller https://zuul.opendev.org/t/openstack/build/5356f12307d944adac4bd52c7ecd3dd6 : SUCCESS in 14m 12s","accounts_in_message":[],"_revision_number":3},{"id":"3e2e5cd66d4dd711cb88eb6c00a0ca7c03cddae5","author":{"_account_id":28372,"name":"chinasubbareddy mallavarapu","email":"cr3938@att.com","username":"chinasubbareddy"},"date":"2020-11-16 20:42:39.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"557db53d540bb22f4726d140b9dade709b755d02","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-16 20:57:30.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/33f07f5df5034369b718eda20e70e07c : SUCCESS in 5m 06s\n- openstack-helm-lint https://zuul.opendev.org/t/openstack/build/4ee8898f839d4b22b0ab009e2fd8588f : FAILURE in 2m 04s\n- openstack-helm-infra-aio-logging https://zuul.opendev.org/t/openstack/build/3c99bf721db345879fd9d423549396e8 : FAILURE in 9m 47s\n- openstack-helm-infra-aio-monitoring https://zuul.opendev.org/t/openstack/build/92eed816487348ff9804d1840af6021d : FAILURE in 9m 57s\n- openstack-helm-infra-openstack-support https://zuul.opendev.org/t/openstack/build/a6b7ff4f0b37456790bd09dbeb0222dd : FAILURE in 9m 34s\n- openstack-helm-infra-openstack-support-ssl https://zuul.opendev.org/t/openstack/build/607c369382ea43b0832f5fa0de8a7eba : FAILURE in 8m 22s\n- openstack-helm-infra-metacontroller https://zuul.opendev.org/t/openstack/build/22a4e50382a240d5bd3119804b1c5f9a : FAILURE in 8m 38s","accounts_in_message":[],"_revision_number":4},{"id":"c557e203b09f1b4eb25676e11d90b3070ddee9ff","author":{"_account_id":28372,"name":"chinasubbareddy mallavarapu","email":"cr3938@att.com","username":"chinasubbareddy"},"date":"2020-11-16 21:23:49.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"a5edc36853cdc43bc8c8951d8bf295452096c179","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-16 21:57:33.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aaeb02c4a92d4814b1939d008555d4c7 : SUCCESS in 3m 48s\n- openstack-helm-lint https://zuul.opendev.org/t/openstack/build/9cdcae27504e4661a07325880d4415f5 : SUCCESS in 3m 06s\n- openstack-helm-infra-aio-logging https://zuul.opendev.org/t/openstack/build/33526790c427472eac0991d74a6367f8 : FAILURE in 32m 39s\n- openstack-helm-infra-aio-monitoring https://zuul.opendev.org/t/openstack/build/6f65788d853c4d55967614d0ae65c430 : FAILURE in 26m 53s\n- openstack-helm-infra-openstack-support https://zuul.opendev.org/t/openstack/build/2f50c992855044df947a25e5391b7189 : FAILURE in 32m 09s\n- openstack-helm-infra-openstack-support-ssl https://zuul.opendev.org/t/openstack/build/b632478d6b504a1a86acd6309bf40dcf : FAILURE in 30m 09s\n- openstack-helm-infra-metacontroller https://zuul.opendev.org/t/openstack/build/fc22d0cee6b74c569f3c8b0d799b0512 : SUCCESS in 22m 23s","accounts_in_message":[],"_revision_number":5},{"id":"da0bb1dd5a2b125ecdd95fcf73122d0b64f2b015","author":{"_account_id":28372,"name":"chinasubbareddy mallavarapu","email":"cr3938@att.com","username":"chinasubbareddy"},"date":"2020-11-16 22:04:25.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"636fbbefceaa89eafcfdba243c8cb85cad318cf9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-16 22:53:11.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a3e674b9bd3a47dfa0de149a32c9a916 : SUCCESS in 4m 01s\n- openstack-helm-lint https://zuul.opendev.org/t/openstack/build/5838522f0e5c4fc0b61c766fbe086edd : SUCCESS in 4m 36s\n- openstack-helm-infra-aio-logging https://zuul.opendev.org/t/openstack/build/502745838ac34673a86f89c4fa4727b3 : SUCCESS in 34m 18s\n- openstack-helm-infra-aio-monitoring https://zuul.opendev.org/t/openstack/build/65e2ef5f7ea5487f803ae4725af51551 : SUCCESS in 30m 43s\n- openstack-helm-infra-openstack-support https://zuul.opendev.org/t/openstack/build/bdabe2e7e8ed42709adc9ed11c2ccee9 : SUCCESS in 46m 55s\n- openstack-helm-infra-openstack-support-ssl https://zuul.opendev.org/t/openstack/build/9d03063431bc477cbc30744b9e3a6c47 : FAILURE in 5m 02s\n- openstack-helm-infra-metacontroller https://zuul.opendev.org/t/openstack/build/df68cf1ad72846f8ac2b48212295b10a : SUCCESS in 24m 47s","accounts_in_message":[],"_revision_number":6},{"id":"d359958cea94c0fa341713019e64146df1215178","author":{"_account_id":28372,"name":"chinasubbareddy mallavarapu","email":"cr3938@att.com","username":"chinasubbareddy"},"date":"2020-11-16 23:03:33.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"9e2c96a4328ba9bceb9ad2e452c45236afafface","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-16 23:56:34.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57d434ae76b4453e96848f2f321cf640 : SUCCESS in 2m 52s\n- openstack-helm-lint https://zuul.opendev.org/t/openstack/build/68e9e4ff4af347a6b5c4b60cdd533d83 : SUCCESS in 4m 37s\n- openstack-helm-infra-aio-logging https://zuul.opendev.org/t/openstack/build/10605945aa344f9da4abfcda10c10359 : SUCCESS in 32m 06s\n- openstack-helm-infra-aio-monitoring https://zuul.opendev.org/t/openstack/build/d0806cf5945249379bf3a860b901462d : SUCCESS in 29m 59s\n- openstack-helm-infra-openstack-support https://zuul.opendev.org/t/openstack/build/f12e684e941e4a8d8a3db4e1cdb5eaa7 : SUCCESS in 46m 57s\n- openstack-helm-infra-openstack-support-ssl https://zuul.opendev.org/t/openstack/build/2d63b2c4c6454ac3a1e37adeb98d6a45 : SUCCESS in 45m 10s\n- openstack-helm-infra-metacontroller https://zuul.opendev.org/t/openstack/build/11c6da55d0c04df995c9a5d47247fd91 : SUCCESS in 13m 17s","accounts_in_message":[],"_revision_number":6},{"id":"3d628f2f5c9d89634635b6a7fdf7090f8336bd0b","author":{"_account_id":28372,"name":"chinasubbareddy mallavarapu","email":"cr3938@att.com","username":"chinasubbareddy"},"date":"2020-11-17 19:25:51.000000000","message":"Patch Set 6: Workflow-1","accounts_in_message":[],"_revision_number":6},{"id":"a84214d753f3ac80dae835f5b293fc617251374d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},"date":"2020-11-24 00:07:38.000000000","message":"Patch Set 7: Patch Set 6 was rebased","accounts_in_message":[],"_revision_number":7},{"id":"d35770509a57e739d0581560f6264204184d8797","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-24 00:54:06.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/85cfa0bd375746d0bac0e95a7a917b6e : SUCCESS in 3m 24s\n- openstack-helm-lint https://zuul.opendev.org/t/openstack/build/081ad3d80ec54ff5be6d4d522e2c9fa2 : SUCCESS in 4m 36s\n- openstack-helm-infra-aio-logging https://zuul.opendev.org/t/openstack/build/12653693b210405683bc20f667106e4c : SUCCESS in 32m 31s\n- openstack-helm-infra-aio-monitoring https://zuul.opendev.org/t/openstack/build/31875fc517b646449fd88f9aab72a011 : SUCCESS in 32m 36s\n- openstack-helm-infra-openstack-support https://zuul.opendev.org/t/openstack/build/7149c7b108be488f997d514f853a8da0 : SUCCESS in 43m 04s\n- openstack-helm-infra-openstack-support-ssl https://zuul.opendev.org/t/openstack/build/429881d0ac8c44f98cdf3f0244b0400f : SUCCESS in 45m 05s\n- openstack-helm-infra-metacontroller https://zuul.opendev.org/t/openstack/build/11c4f5b341874004a5ecca970897b4c4 : SUCCESS in 13m 06s","accounts_in_message":[],"_revision_number":7},{"id":"df41ae0290a527cffe1eda29b8229a3827e5eec9","author":{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},"date":"2020-11-24 04:55:55.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"7baefab099b7e5904253903e20af5fb3fe3d531d","tag":"autogenerated:gerrit:abandon","author":{"_account_id":28372,"name":"chinasubbareddy mallavarapu","email":"cr3938@att.com","username":"chinasubbareddy"},"date":"2020-11-24 16:09:20.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"bb643d14807dbe23334549d15d87599b048934ec","revisions":{"8e46d614b03757d29da0f20059338027ce2fcbb9":{"kind":"REWORK","_number":1,"created":"2020-11-13 19:14:24.000000000","uploader":{"_account_id":28372,"name":"chinasubbareddy mallavarapu","email":"cr3938@att.com","username":"chinasubbareddy"},"ref":"refs/changes/09/762709/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-helm-infra","ref":"refs/changes/09/762709/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/09/762709/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/09/762709/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/09/762709/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-helm-infra refs/changes/09/762709/1"}}},"commit":{"parents":[{"commit":"ca372bfea6affbe1b79dd79311cf19ce88c7244a","subject":"Fix typo in check inactive PGs logic","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/ca372bfea6affbe1b79dd79311cf19ce88c7244a"}]}],"author":{"name":"Chinasubbareddy Mallavarapu","email":"cr3938@att.com","date":"2020-11-13 19:12:38.000000000","tz":0},"committer":{"name":"Chinasubbareddy Mallavarapu","email":"cr3938@att.com","date":"2020-11-13 19:12:38.000000000","tz":0},"subject":"[CEPH-POOL] Read the quota per pool from pool spec","message":"[CEPH-POOL] Read the quota per pool from pool spec\n\nThis is to read pool quota from pool ssec so that each pool\ncan have own quota.\n\nChange-Id: I954972e5810b80a5abe8b882838b9a3accafe880\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/8e46d614b03757d29da0f20059338027ce2fcbb9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/8e46d614b03757d29da0f20059338027ce2fcbb9"}]},"branch":"refs/heads/master"},"3f657b298f7eaf325a78e4c644f279e6b7681ea8":{"kind":"REWORK","_number":2,"created":"2020-11-13 19:15:31.000000000","uploader":{"_account_id":28372,"name":"chinasubbareddy mallavarapu","email":"cr3938@att.com","username":"chinasubbareddy"},"ref":"refs/changes/09/762709/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-helm-infra","ref":"refs/changes/09/762709/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/09/762709/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/09/762709/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/09/762709/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-helm-infra refs/changes/09/762709/2"}}},"commit":{"parents":[{"commit":"ca372bfea6affbe1b79dd79311cf19ce88c7244a","subject":"Fix typo in check inactive PGs logic","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/ca372bfea6affbe1b79dd79311cf19ce88c7244a"}]}],"author":{"name":"Chinasubbareddy Mallavarapu","email":"cr3938@att.com","date":"2020-11-13 19:12:38.000000000","tz":0},"committer":{"name":"Chinasubbareddy Mallavarapu","email":"cr3938@att.com","date":"2020-11-13 19:15:26.000000000","tz":0},"subject":"[CEPH-POOL] Read the quota per pool from pool spec","message":"[CEPH-POOL] Read the quota per pool from pool spec\n\nThis is to read pool quota from pool ssec so that each pool\ncan have own quota.\n\nChange-Id: I954972e5810b80a5abe8b882838b9a3accafe880\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/3f657b298f7eaf325a78e4c644f279e6b7681ea8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/3f657b298f7eaf325a78e4c644f279e6b7681ea8"}]},"branch":"refs/heads/master"},"c122a06ef2537d155c2ec8a05f129e27e4698f91":{"kind":"REWORK","_number":3,"created":"2020-11-13 20:59:30.000000000","uploader":{"_account_id":28372,"name":"chinasubbareddy mallavarapu","email":"cr3938@att.com","username":"chinasubbareddy"},"ref":"refs/changes/09/762709/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-helm-infra","ref":"refs/changes/09/762709/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/09/762709/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/09/762709/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/09/762709/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-helm-infra refs/changes/09/762709/3"}}},"commit":{"parents":[{"commit":"ca372bfea6affbe1b79dd79311cf19ce88c7244a","subject":"Fix typo in check inactive PGs logic","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/ca372bfea6affbe1b79dd79311cf19ce88c7244a"}]}],"author":{"name":"Chinasubbareddy Mallavarapu","email":"cr3938@att.com","date":"2020-11-13 19:12:38.000000000","tz":0},"committer":{"name":"chinasubbareddy mallavarapu","email":"cr3938@att.com","date":"2020-11-13 20:59:28.000000000","tz":0},"subject":"[CEPH-POOL] Read the quota per pool from pool spec","message":"[CEPH-POOL] Read the quota per pool from pool spec\n\nThis is to read pool quota from pool ssec so that each pool\ncan have own quota.\n\nChange-Id: I954972e5810b80a5abe8b882838b9a3accafe880\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/c122a06ef2537d155c2ec8a05f129e27e4698f91"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/c122a06ef2537d155c2ec8a05f129e27e4698f91"}]},"branch":"refs/heads/master"},"d521467012187a453caf758717a040c5f49c97ac":{"kind":"REWORK","_number":4,"created":"2020-11-16 20:42:39.000000000","uploader":{"_account_id":28372,"name":"chinasubbareddy mallavarapu","email":"cr3938@att.com","username":"chinasubbareddy"},"ref":"refs/changes/09/762709/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-helm-infra","ref":"refs/changes/09/762709/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/09/762709/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/09/762709/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/09/762709/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-helm-infra refs/changes/09/762709/4"}}},"commit":{"parents":[{"commit":"ca372bfea6affbe1b79dd79311cf19ce88c7244a","subject":"Fix typo in check inactive PGs logic","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/ca372bfea6affbe1b79dd79311cf19ce88c7244a"}]}],"author":{"name":"Chinasubbareddy Mallavarapu","email":"cr3938@att.com","date":"2020-11-13 19:12:38.000000000","tz":0},"committer":{"name":"Chinasubbareddy Mallavarapu","email":"cr3938@att.com","date":"2020-11-16 20:42:32.000000000","tz":0},"subject":"[CEPH-POOL] Read the quota per pool from pool spec","message":"[CEPH-POOL] Read the quota per pool from pool spec\n\nThis is to read pool quota from pool ssec so that each pool\ncan have own quota.\n\nChange-Id: I954972e5810b80a5abe8b882838b9a3accafe880\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/d521467012187a453caf758717a040c5f49c97ac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/d521467012187a453caf758717a040c5f49c97ac"}]},"branch":"refs/heads/master"},"da70fb663c58e04aee8fd9d7c8a124d0209fd4f9":{"kind":"REWORK","_number":5,"created":"2020-11-16 21:23:49.000000000","uploader":{"_account_id":28372,"name":"chinasubbareddy mallavarapu","email":"cr3938@att.com","username":"chinasubbareddy"},"ref":"refs/changes/09/762709/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-helm-infra","ref":"refs/changes/09/762709/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/09/762709/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/09/762709/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/09/762709/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-helm-infra refs/changes/09/762709/5"}}},"commit":{"parents":[{"commit":"ca372bfea6affbe1b79dd79311cf19ce88c7244a","subject":"Fix typo in check inactive PGs logic","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/ca372bfea6affbe1b79dd79311cf19ce88c7244a"}]}],"author":{"name":"Chinasubbareddy Mallavarapu","email":"cr3938@att.com","date":"2020-11-13 19:12:38.000000000","tz":0},"committer":{"name":"Chinasubbareddy Mallavarapu","email":"cr3938@att.com","date":"2020-11-16 21:23:45.000000000","tz":0},"subject":"[CEPH-POOL] Read the quota per pool from pool spec","message":"[CEPH-POOL] Read the quota per pool from pool spec\n\nThis is to read pool quota from pool ssec so that each pool\ncan have own quota.\n\nChange-Id: I954972e5810b80a5abe8b882838b9a3accafe880\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/da70fb663c58e04aee8fd9d7c8a124d0209fd4f9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/da70fb663c58e04aee8fd9d7c8a124d0209fd4f9"}]},"branch":"refs/heads/master"},"3bd723b067558926eb1b553e3eb0aa5bca01ffb2":{"kind":"REWORK","_number":6,"created":"2020-11-16 22:04:25.000000000","uploader":{"_account_id":28372,"name":"chinasubbareddy mallavarapu","email":"cr3938@att.com","username":"chinasubbareddy"},"ref":"refs/changes/09/762709/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-helm-infra","ref":"refs/changes/09/762709/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/09/762709/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/09/762709/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/09/762709/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-helm-infra refs/changes/09/762709/6"}}},"commit":{"parents":[{"commit":"ca372bfea6affbe1b79dd79311cf19ce88c7244a","subject":"Fix typo in check inactive PGs logic","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/ca372bfea6affbe1b79dd79311cf19ce88c7244a"}]}],"author":{"name":"Chinasubbareddy Mallavarapu","email":"cr3938@att.com","date":"2020-11-13 19:12:38.000000000","tz":0},"committer":{"name":"Chinasubbareddy Mallavarapu","email":"cr3938@att.com","date":"2020-11-16 22:04:20.000000000","tz":0},"subject":"[CEPH-POOL] Read the quota per pool from pool spec","message":"[CEPH-POOL] Read the quota per pool from pool spec\n\nThis is to read pool quota from pool ssec so that each pool\ncan have own quota.\n\nChange-Id: I954972e5810b80a5abe8b882838b9a3accafe880\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/3bd723b067558926eb1b553e3eb0aa5bca01ffb2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/3bd723b067558926eb1b553e3eb0aa5bca01ffb2"}]},"branch":"refs/heads/master"},"bb643d14807dbe23334549d15d87599b048934ec":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2020-11-24 00:07:38.000000000","uploader":{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},"ref":"refs/changes/09/762709/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-helm-infra","ref":"refs/changes/09/762709/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/09/762709/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/09/762709/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/09/762709/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-helm-infra refs/changes/09/762709/7"}}},"commit":{"parents":[{"commit":"15ad6e9a6cdd1dd0c80be4672fd4efcb5002fda2","subject":"Merge \"fix(secret): changes rmq-exporter secret src\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/15ad6e9a6cdd1dd0c80be4672fd4efcb5002fda2"}]}],"author":{"name":"Chinasubbareddy Mallavarapu","email":"cr3938@att.com","date":"2020-11-13 19:12:38.000000000","tz":0},"committer":{"name":"Chris Wedgwood","email":"cw@f00f.org","date":"2020-11-24 00:07:38.000000000","tz":0},"subject":"[CEPH-POOL] Read the quota per pool from pool spec","message":"[CEPH-POOL] Read the quota per pool from pool spec\n\nThis is to read pool quota from pool ssec so that each pool\ncan have own quota.\n\nChange-Id: I954972e5810b80a5abe8b882838b9a3accafe880\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/bb643d14807dbe23334549d15d87599b048934ec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/bb643d14807dbe23334549d15d87599b048934ec"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
