)]}'
{"id":"openstack%2Fopenstack-helm-infra~767386","triplet_id":"openstack%2Fopenstack-helm-infra~master~If32ab354670166a3c844991de1744de63a508303","project":"openstack/openstack-helm-infra","branch":"master","topic":"change-767386-2","hashtags":[],"change_id":"If32ab354670166a3c844991de1744de63a508303","subject":"[ceph-osd] Correct naming convention for logical volumes in disk_zap()","status":"MERGED","created":"2020-12-16 17:39:26.000000000","updated":"2020-12-17 19:37:16.000000000","submitted":"2020-12-17 19:35:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"767386-change-767386-2","meta_rev_id":"b3a785163c4aecfbf27e3ed7250afc349879ea6c","_number":767386,"virtual_id_number":767386,"owner":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31728,"name":"Oleksii Shcherba","display_name":"Oleksii Shcherba","email":"v-oshcherba@microsoft.com","username":"mihalych"},{"value":0,"_account_id":32433,"name":"Vladimir Sigunov","email":"vladimir.sigunov@att.com","username":"vs422h"},{"value":0,"_account_id":28372,"name":"chinasubbareddy mallavarapu","email":"cr3938@att.com","username":"chinasubbareddy"},{"value":0,"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},{"value":0,"_account_id":29144,"name":"PRATEEK REDDY DODDA","email":"prateek.reddy.dodda@att.com","username":"pd2839"},{"value":0,"_account_id":30777,"name":"Steven Fitzpatrick","email":"steven@fitzpatrick.wtf","username":"sf280x"},{"value":0,"_account_id":18250,"name":"Roy Tang","email":"roy.s.tang@att.com","username":"rt7380"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-12-17 19:35:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22636,"name":"Cliff Parsons","email":"cliffhparsons@aol.com","username":"cliffparsons"},{"value":0,"_account_id":32190,"name":"Frank Ritchie","email":"fr801x@att.com","username":"fr801x"},{"value":0,"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},{"value":0,"_account_id":18511,"name":"Brian Wickersham","email":"bkw86@bellsouth.net","username":"bw6938"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30777,"name":"Steven Fitzpatrick","email":"steven@fitzpatrick.wtf","username":"sf280x"},"recommended":{"_account_id":28372,"name":"chinasubbareddy mallavarapu","email":"cr3938@att.com","username":"chinasubbareddy"},"all":[{"value":1,"date":"2020-12-17 17:50:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31728,"name":"Oleksii Shcherba","display_name":"Oleksii Shcherba","email":"v-oshcherba@microsoft.com","username":"mihalych"},{"value":0,"_account_id":32433,"name":"Vladimir Sigunov","email":"vladimir.sigunov@att.com","username":"vs422h"},{"value":1,"date":"2020-12-17 18:11:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28372,"name":"chinasubbareddy mallavarapu","email":"cr3938@att.com","username":"chinasubbareddy"},{"value":0,"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},{"value":0,"_account_id":29144,"name":"PRATEEK REDDY DODDA","email":"prateek.reddy.dodda@att.com","username":"pd2839"},{"value":2,"date":"2020-12-17 18:21:52.000000000","_account_id":30777,"name":"Steven Fitzpatrick","email":"steven@fitzpatrick.wtf","username":"sf280x"},{"value":0,"_account_id":18250,"name":"Roy Tang","email":"roy.s.tang@att.com","username":"rt7380"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22636,"name":"Cliff Parsons","email":"cliffhparsons@aol.com","username":"cliffparsons"},{"value":1,"date":"2020-12-17 17:51:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32190,"name":"Frank Ritchie","email":"fr801x@att.com","username":"fr801x"},{"value":2,"date":"2020-12-17 17:44:54.000000000","_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},{"value":0,"_account_id":18511,"name":"Brian Wickersham","email":"bkw86@bellsouth.net","username":"bw6938"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30777,"name":"Steven Fitzpatrick","email":"steven@fitzpatrick.wtf","username":"sf280x"},"all":[{"value":0,"_account_id":31728,"name":"Oleksii Shcherba","display_name":"Oleksii Shcherba","email":"v-oshcherba@microsoft.com","username":"mihalych"},{"value":0,"_account_id":32433,"name":"Vladimir Sigunov","email":"vladimir.sigunov@att.com","username":"vs422h"},{"value":0,"_account_id":28372,"name":"chinasubbareddy mallavarapu","email":"cr3938@att.com","username":"chinasubbareddy"},{"value":0,"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},{"value":0,"_account_id":29144,"name":"PRATEEK REDDY DODDA","email":"prateek.reddy.dodda@att.com","username":"pd2839"},{"value":1,"date":"2020-12-17 18:21:52.000000000","_account_id":30777,"name":"Steven Fitzpatrick","email":"steven@fitzpatrick.wtf","username":"sf280x"},{"value":0,"_account_id":18250,"name":"Roy Tang","email":"roy.s.tang@att.com","username":"rt7380"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22636,"name":"Cliff Parsons","email":"cliffhparsons@aol.com","username":"cliffparsons"},{"value":0,"_account_id":32190,"name":"Frank Ritchie","email":"fr801x@att.com","username":"fr801x"},{"value":0,"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},{"value":0,"_account_id":18511,"name":"Brian Wickersham","email":"bkw86@bellsouth.net","username":"bw6938"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},{"_account_id":32260,"name":"Frank Ritchie","email":"fr801x@att.com","username":"frankritchie"},{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"_account_id":20466,"name":"Tin Lam","email":"tin@lam.wtf","username":"tinlam"},{"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"}],"reviewers":{"REVIEWER":[{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},{"_account_id":18250,"name":"Roy Tang","email":"roy.s.tang@att.com","username":"rt7380"},{"_account_id":18511,"name":"Brian Wickersham","email":"bkw86@bellsouth.net","username":"bw6938"},{"_account_id":20466,"name":"Tin Lam","email":"tin@lam.wtf","username":"tinlam"},{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22636,"name":"Cliff Parsons","email":"cliffhparsons@aol.com","username":"cliffparsons"},{"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},{"_account_id":28372,"name":"chinasubbareddy mallavarapu","email":"cr3938@att.com","username":"chinasubbareddy"},{"_account_id":29144,"name":"PRATEEK REDDY DODDA","email":"prateek.reddy.dodda@att.com","username":"pd2839"},{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},{"_account_id":30777,"name":"Steven Fitzpatrick","email":"steven@fitzpatrick.wtf","username":"sf280x"},{"_account_id":31728,"name":"Oleksii Shcherba","display_name":"Oleksii Shcherba","email":"v-oshcherba@microsoft.com","username":"mihalych"},{"_account_id":32190,"name":"Frank Ritchie","email":"fr801x@att.com","username":"fr801x"},{"_account_id":32260,"name":"Frank Ritchie","email":"fr801x@att.com","username":"frankritchie"},{"_account_id":32433,"name":"Vladimir Sigunov","email":"vladimir.sigunov@att.com","username":"vs422h"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-16 18:46:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-16 21:16:38.000000000","updated_by":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"reviewer":{"_account_id":32433,"name":"Vladimir Sigunov","email":"vladimir.sigunov@att.com","username":"vs422h"},"state":"REVIEWER"},{"updated":"2020-12-16 21:16:38.000000000","updated_by":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"reviewer":{"_account_id":32260,"name":"Frank Ritchie","email":"fr801x@att.com","username":"frankritchie"},"state":"REVIEWER"},{"updated":"2020-12-16 21:16:38.000000000","updated_by":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"reviewer":{"_account_id":18511,"name":"Brian Wickersham","email":"bkw86@bellsouth.net","username":"bw6938"},"state":"REVIEWER"},{"updated":"2020-12-16 21:16:38.000000000","updated_by":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"reviewer":{"_account_id":28372,"name":"chinasubbareddy mallavarapu","email":"cr3938@att.com","username":"chinasubbareddy"},"state":"REVIEWER"},{"updated":"2020-12-16 21:22:25.000000000","updated_by":{"_account_id":32190,"name":"Frank Ritchie","email":"fr801x@att.com","username":"fr801x"},"reviewer":{"_account_id":32190,"name":"Frank Ritchie","email":"fr801x@att.com","username":"fr801x"},"state":"REVIEWER"},{"updated":"2020-12-16 22:19:27.000000000","updated_by":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"reviewer":{"_account_id":30777,"name":"Steven Fitzpatrick","email":"steven@fitzpatrick.wtf","username":"sf280x"},"state":"REVIEWER"},{"updated":"2020-12-16 22:19:27.000000000","updated_by":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"reviewer":{"_account_id":20466,"name":"Tin Lam","email":"tin@lam.wtf","username":"tinlam"},"state":"REVIEWER"},{"updated":"2020-12-16 22:19:27.000000000","updated_by":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"reviewer":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"state":"REVIEWER"},{"updated":"2020-12-16 22:19:27.000000000","updated_by":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"reviewer":{"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},"state":"REVIEWER"},{"updated":"2020-12-16 22:19:27.000000000","updated_by":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"reviewer":{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},"state":"REVIEWER"},{"updated":"2020-12-16 22:21:09.000000000","updated_by":{"_account_id":29144,"name":"PRATEEK REDDY DODDA","email":"prateek.reddy.dodda@att.com","username":"pd2839"},"reviewer":{"_account_id":29144,"name":"PRATEEK REDDY DODDA","email":"prateek.reddy.dodda@att.com","username":"pd2839"},"state":"REVIEWER"},{"updated":"2020-12-16 22:22:59.000000000","updated_by":{"_account_id":22636,"name":"Cliff Parsons","email":"cliffhparsons@aol.com","username":"cliffparsons"},"reviewer":{"_account_id":22636,"name":"Cliff Parsons","email":"cliffhparsons@aol.com","username":"cliffparsons"},"state":"REVIEWER"},{"updated":"2020-12-17 06:42:06.000000000","updated_by":{"_account_id":18250,"name":"Roy Tang","email":"roy.s.tang@att.com","username":"rt7380"},"reviewer":{"_account_id":18250,"name":"Roy Tang","email":"roy.s.tang@att.com","username":"rt7380"},"state":"REVIEWER"},{"updated":"2020-12-17 15:49:41.000000000","updated_by":{"_account_id":31728,"name":"Oleksii Shcherba","display_name":"Oleksii Shcherba","email":"v-oshcherba@microsoft.com","username":"mihalych"},"reviewer":{"_account_id":31728,"name":"Oleksii Shcherba","display_name":"Oleksii Shcherba","email":"v-oshcherba@microsoft.com","username":"mihalych"},"state":"REVIEWER"},{"updated":"2020-12-17 16:26:06.000000000","updated_by":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"reviewer":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"state":"REVIEWER"}],"messages":[{"id":"5abed47632e7f9ea9693b64ead9248300d08c0d0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"date":"2020-12-16 17:39:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2be3ae54f0ee7811e313dbc036eae6792bbc5cff","author":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"date":"2020-12-16 17:40:06.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"e7b0528e487029a2133cc1f65975f823a2ac2b85","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"date":"2020-12-16 17:45:28.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e2264613181b8fc584a1664a1d04a1b07b332e26","author":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"date":"2020-12-16 17:45:51.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"a097f57e144c5bbd51c2164d52b3563680997f84","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-16 18:46:08.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9495ce21f0ad4e26a6769098ca1f1587 : SUCCESS in 4m 20s\n- openstack-helm-lint https://zuul.opendev.org/t/openstack/build/65e0b997f2bd4c599dda6d4edc8a9a75 : SUCCESS in 4m 10s\n- openstack-helm-infra-aio-logging https://zuul.opendev.org/t/openstack/build/77599e89ae2441ed81eabb2ba555c65c : SUCCESS in 41m 34s\n- openstack-helm-infra-aio-monitoring https://zuul.opendev.org/t/openstack/build/0672932b4b904bbebe09b082ecab15be : SUCCESS in 32m 09s\n- openstack-helm-infra-openstack-support https://zuul.opendev.org/t/openstack/build/e4e875e3d2b34640a70ae8c2b95508fe : SUCCESS in 53m 55s\n- openstack-helm-infra-openstack-support-ssl https://zuul.opendev.org/t/openstack/build/d3859b5db44b4bd6b5f3daab5aef702f : SUCCESS in 45m 52s\n- openstack-helm-infra-metacontroller https://zuul.opendev.org/t/openstack/build/da2ff8ccae594bc4bfb672a654c5a525 : SUCCESS in 12m 13s","accounts_in_message":[],"_revision_number":2},{"id":"5640ddbd95473c77cf786e5d0bb38e14c41b369d","author":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"date":"2020-12-16 18:47:41.000000000","message":"Patch Set 2:\n\nWaiting for https://review.opendev.org/c/openstack/openstack-helm-infra/+/767009 to merge as it is a higher priority and conflicts with this change.","accounts_in_message":[],"_revision_number":2},{"id":"c93a04dc245dec78ff463e360579227795c10bbd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"date":"2020-12-16 21:15:08.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"f7ac79f6ab6628368773550bb470b4d73c426616","author":{"_account_id":32190,"name":"Frank Ritchie","email":"fr801x@att.com","username":"fr801x"},"date":"2020-12-16 21:22:25.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"3632094d69e3eba5d8c9bdf1a2d5c63764db993e","author":{"_account_id":32433,"name":"Vladimir Sigunov","email":"vladimir.sigunov@att.com","username":"vs422h"},"date":"2020-12-16 21:52:12.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"c321bdcf6333ca5967739f6bead0099645c31dcd","author":{"_account_id":28372,"name":"chinasubbareddy mallavarapu","email":"cr3938@att.com","username":"chinasubbareddy"},"date":"2020-12-16 22:06:13.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"87488697aba65689968862df3d4dcafcaa78d5ff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-16 22:19:05.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/82737a6e9c9a46ff941078e4927b5ffb : SUCCESS in 4m 25s\n- openstack-helm-lint https://zuul.opendev.org/t/openstack/build/b19528bdfbe54397b94da3af737d88cb : SUCCESS in 3m 24s\n- openstack-helm-infra-aio-logging https://zuul.opendev.org/t/openstack/build/7f8c6616c8d94a09afbc63564e4f4d83 : SUCCESS in 34m 56s\n- openstack-helm-infra-aio-monitoring https://zuul.opendev.org/t/openstack/build/e7a22f6690684e1a80b6f9cde65f0919 : SUCCESS in 34m 30s\n- openstack-helm-infra-openstack-support https://zuul.opendev.org/t/openstack/build/49580664212e44988eabd48f78e345e7 : SUCCESS in 45m 11s\n- openstack-helm-infra-openstack-support-ssl https://zuul.opendev.org/t/openstack/build/fe71871e4d304d35ab52e515def09a11 : SUCCESS in 56m 49s\n- openstack-helm-infra-metacontroller https://zuul.opendev.org/t/openstack/build/313d3452df9d4f688bbfeaca2e3bbfac : SUCCESS in 15m 30s","accounts_in_message":[],"_revision_number":3},{"id":"9f8859a0a0e9c413fe64145cb3de38f7bdb8aeb8","author":{"_account_id":29144,"name":"PRATEEK REDDY DODDA","email":"prateek.reddy.dodda@att.com","username":"pd2839"},"date":"2020-12-16 22:21:09.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"d94500fcb2d35a9dab08511e4567e741bf388211","author":{"_account_id":22636,"name":"Cliff Parsons","email":"cliffhparsons@aol.com","username":"cliffparsons"},"date":"2020-12-16 22:22:59.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"c7fd62794c36e2b6e2509ebf8981b524c6d100de","author":{"_account_id":30777,"name":"Steven Fitzpatrick","email":"steven@fitzpatrick.wtf","username":"sf280x"},"date":"2020-12-16 22:23:33.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"a40e0940c53367778136f5cb8868dc7c1d12e676","author":{"_account_id":18511,"name":"Brian Wickersham","email":"bkw86@bellsouth.net","username":"bw6938"},"date":"2020-12-17 01:06:29.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"c6dcba0aa4f19129bc1e134e2af5fe92bbec4a41","author":{"_account_id":18250,"name":"Roy Tang","email":"roy.s.tang@att.com","username":"rt7380"},"date":"2020-12-17 06:42:06.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"6c51b0e76f5e7dc755c54c135ea6f001d9c21e82","author":{"_account_id":31728,"name":"Oleksii Shcherba","display_name":"Oleksii Shcherba","email":"v-oshcherba@microsoft.com","username":"mihalych"},"date":"2020-12-17 15:49:41.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"dfcd91ddb38e2604ad3d207127adb4121df2c89a","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2020-12-17 16:28:59.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"fb9ecf2882db11a0753f52875727479bdf3b7901","author":{"_account_id":28372,"name":"chinasubbareddy mallavarapu","email":"cr3938@att.com","username":"chinasubbareddy"},"date":"2020-12-17 16:30:39.000000000","message":"Patch Set 3: -Code-Review","accounts_in_message":[],"_revision_number":3},{"id":"5b64481d6abf2653b1416c961d36e19480d88977","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"date":"2020-12-17 16:31:41.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"4c759648d4293dfd128f9c88a0d449e79fefe197","author":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"date":"2020-12-17 16:32:14.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"014bd9c5c4c96368d589c6f720cfa33f66306cef","author":{"_account_id":28372,"name":"chinasubbareddy mallavarapu","email":"cr3938@att.com","username":"chinasubbareddy"},"date":"2020-12-17 16:35:34.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"f86e7a7aca9fa267a2c2d4a28704a2cfb6b0a48c","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2020-12-17 17:16:26.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"3f571aaddabe507871f71ce6af2518e02e849843","author":{"_account_id":28372,"name":"chinasubbareddy mallavarapu","email":"cr3938@att.com","username":"chinasubbareddy"},"date":"2020-12-17 17:21:15.000000000","message":"Patch Set 4: Code-Review-1","accounts_in_message":[],"_revision_number":4},{"id":"ccaab63b8ab06b87a261b3426bcc7e6ad77f2b45","author":{"_account_id":28372,"name":"chinasubbareddy mallavarapu","email":"cr3938@att.com","username":"chinasubbareddy"},"date":"2020-12-17 17:21:20.000000000","message":"Patch Set 4: -Code-Review","accounts_in_message":[],"_revision_number":4},{"id":"39352caa84b616f8b2fddc4363296fc903ae5237","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-17 17:25:42.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/777b894843ab4f5e88a6cc189274ebe9 : SUCCESS in 2m 32s\n- openstack-helm-lint https://zuul.opendev.org/t/openstack/build/baa1a7d7b0d6490eb05cce88ba8ecf0b : SUCCESS in 3m 41s\n- openstack-helm-infra-aio-logging https://zuul.opendev.org/t/openstack/build/d6661617c6e942d191a57fb5f249d7cd : SUCCESS in 32m 21s\n- openstack-helm-infra-aio-monitoring https://zuul.opendev.org/t/openstack/build/3d290b1e702b4285aad3f37f3f5ce16b : SUCCESS in 29m 11s\n- openstack-helm-infra-openstack-support https://zuul.opendev.org/t/openstack/build/d7728c40212646fe86782653de52b681 : SUCCESS in 51m 12s\n- openstack-helm-infra-openstack-support-ssl https://zuul.opendev.org/t/openstack/build/a27dfb79c554453f95b698862dfe21a0 : SUCCESS in 43m 51s\n- openstack-helm-infra-metacontroller https://zuul.opendev.org/t/openstack/build/3474165ce3094355b52fe3e3f417076b : SUCCESS in 14m 05s","accounts_in_message":[],"_revision_number":4},{"id":"f77d3485239496b739bb46edeb8c793ed5d122a6","author":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"date":"2020-12-17 17:27:24.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"b8209cec22aaeb5d9226236d4596a943d9b7f677","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2020-12-17 17:44:54.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"dc98423150758e6664dd8d849e2853ed22f480cb","author":{"_account_id":31728,"name":"Oleksii Shcherba","display_name":"Oleksii Shcherba","email":"v-oshcherba@microsoft.com","username":"mihalych"},"date":"2020-12-17 17:50:44.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"657e414e90568bd790774811e276ae59db7c3f52","author":{"_account_id":32190,"name":"Frank Ritchie","email":"fr801x@att.com","username":"fr801x"},"date":"2020-12-17 17:51:32.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"5ed91dd9407e4f24f3c270dc190a2bb7c2c9a5db","author":{"_account_id":28372,"name":"chinasubbareddy mallavarapu","email":"cr3938@att.com","username":"chinasubbareddy"},"date":"2020-12-17 18:11:34.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"b46886debc4b626f371ce7d200d5e03ae93a4d7b","author":{"_account_id":30777,"name":"Steven Fitzpatrick","email":"steven@fitzpatrick.wtf","username":"sf280x"},"date":"2020-12-17 18:21:52.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"88c2c0f9dd69862fbffacff9de768159e302a163","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-17 18:22:31.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"ec36c45a348cd3c728c2b639a3b0215b871219fe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-17 19:35:51.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/32fecb5fd6a14d2abf6d5fb42f37df57 : SUCCESS in 3m 25s\n- openstack-helm-lint https://zuul.opendev.org/t/openstack/build/8ff4ecfb6f11422581cadf77b00c453b : SUCCESS in 3m 44s\n- openstack-helm-infra-aio-logging https://zuul.opendev.org/t/openstack/build/ad3490a982434439a88b7bad469b615c : SUCCESS in 38m 25s\n- openstack-helm-infra-aio-monitoring https://zuul.opendev.org/t/openstack/build/08a138eaf42b4610a9e0e7f240f509b4 : SUCCESS in 36m 54s\n- openstack-helm-infra-openstack-support https://zuul.opendev.org/t/openstack/build/f10c2af887284c8ca2f11cee26d4993a : SUCCESS in 48m 29s\n- openstack-helm-infra-openstack-support-ssl https://zuul.opendev.org/t/openstack/build/82cbf0c4ec3b42a5a386afdf82daa28b : SUCCESS in 45m 29s","accounts_in_message":[],"_revision_number":4},{"id":"5580fcb079ce204668633fb4a5fd3ec46f5a6e92","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-17 19:35:52.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"b3a785163c4aecfbf27e3ed7250afc349879ea6c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-17 19:37:16.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b8b523ef5be346118f36fbad3fd966ea : SUCCESS in 54s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"213596d71cc3cb3fca7a78e6fef12479f36b89a0","revisions":{"b7710ae525cd9c653353cbaedf1db4c37a9a5369":{"kind":"REWORK","_number":1,"created":"2020-12-16 17:39:26.000000000","uploader":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"ref":"refs/changes/86/767386/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-helm-infra","ref":"refs/changes/86/767386/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/86/767386/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/86/767386/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/86/767386/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-helm-infra refs/changes/86/767386/1"}}},"commit":{"parents":[{"commit":"9b1ac0ffcb5d0a7d10670074e181ef32a6372b15","subject":"Enable shareProcessNamespace in mon daemonset","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/9b1ac0ffcb5d0a7d10670074e181ef32a6372b15"}]}],"author":{"name":"Stephen Taylor","email":"st053q@att.com","date":"2020-12-16 17:35:53.000000000","tz":-420},"committer":{"name":"Stephen Taylor","email":"st053q@att.com","date":"2020-12-16 17:35:53.000000000","tz":-420},"subject":"[ceph-osd] Correct naming convention for logical volumes in disk_zap()","message":"[ceph-osd] Correct naming convention for logical volumes in disk_zap()\n\nOSD logical volume names used to be based on the logical disk path,\ni.e. /dev/sdb, but that has changed. The lvremove logic in disk_zap()\nis still using the old naming convention. This change fixes that.\n\nChange-Id: If32ab354670166a3c844991de1744de63a508303\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/b7710ae525cd9c653353cbaedf1db4c37a9a5369"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/b7710ae525cd9c653353cbaedf1db4c37a9a5369"}]},"branch":"refs/heads/master"},"313464cdbc2d08987b9d26e2ac9d72f77d47151c":{"kind":"REWORK","_number":2,"created":"2020-12-16 17:45:28.000000000","uploader":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"ref":"refs/changes/86/767386/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-helm-infra","ref":"refs/changes/86/767386/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/86/767386/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/86/767386/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/86/767386/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-helm-infra refs/changes/86/767386/2"}}},"commit":{"parents":[{"commit":"9b1ac0ffcb5d0a7d10670074e181ef32a6372b15","subject":"Enable shareProcessNamespace in mon daemonset","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/9b1ac0ffcb5d0a7d10670074e181ef32a6372b15"}]}],"author":{"name":"Stephen Taylor","email":"st053q@att.com","date":"2020-12-16 17:35:53.000000000","tz":-420},"committer":{"name":"Stephen Taylor","email":"st053q@att.com","date":"2020-12-16 17:45:13.000000000","tz":-420},"subject":"[ceph-osd] Correct naming convention for logical volumes in disk_zap()","message":"[ceph-osd] Correct naming convention for logical volumes in disk_zap()\n\nOSD logical volume names used to be based on the logical disk path,\ni.e. /dev/sdb, but that has changed. The lvremove logic in disk_zap()\nis still using the old naming convention. This change fixes that.\n\nChange-Id: If32ab354670166a3c844991de1744de63a508303\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/313464cdbc2d08987b9d26e2ac9d72f77d47151c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/313464cdbc2d08987b9d26e2ac9d72f77d47151c"}]},"branch":"refs/heads/master"},"7dae5f148e0b58439aafe4e2f3d010710830e4a9":{"kind":"REWORK","_number":3,"created":"2020-12-16 21:15:08.000000000","uploader":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"ref":"refs/changes/86/767386/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-helm-infra","ref":"refs/changes/86/767386/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/86/767386/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/86/767386/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/86/767386/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-helm-infra refs/changes/86/767386/3"}}},"commit":{"parents":[{"commit":"81f928544b2403962088c2f01362836c2f2348b6","subject":"Merge \"[ceph-osd] Alias synchronized commands and fix descriptor leak\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/81f928544b2403962088c2f01362836c2f2348b6"}]}],"author":{"name":"Stephen Taylor","email":"st053q@att.com","date":"2020-12-16 17:35:53.000000000","tz":-420},"committer":{"name":"Stephen Taylor","email":"st053q@att.com","date":"2020-12-16 21:14:35.000000000","tz":-420},"subject":"[ceph-osd] Correct naming convention for logical volumes in disk_zap()","message":"[ceph-osd] Correct naming convention for logical volumes in disk_zap()\n\nOSD logical volume names used to be based on the logical disk path,\ni.e. /dev/sdb, but that has changed. The lvremove logic in disk_zap()\nis still using the old naming convention. This change fixes that.\n\nChange-Id: If32ab354670166a3c844991de1744de63a508303\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/7dae5f148e0b58439aafe4e2f3d010710830e4a9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/7dae5f148e0b58439aafe4e2f3d010710830e4a9"}]},"branch":"refs/heads/master"},"213596d71cc3cb3fca7a78e6fef12479f36b89a0":{"kind":"REWORK","_number":4,"created":"2020-12-17 16:31:41.000000000","uploader":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"ref":"refs/changes/86/767386/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-helm-infra","ref":"refs/changes/86/767386/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/86/767386/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/86/767386/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/86/767386/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-helm-infra refs/changes/86/767386/4"}}},"commit":{"parents":[{"commit":"81f928544b2403962088c2f01362836c2f2348b6","subject":"Merge \"[ceph-osd] Alias synchronized commands and fix descriptor leak\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/81f928544b2403962088c2f01362836c2f2348b6"}]}],"author":{"name":"Stephen Taylor","email":"st053q@att.com","date":"2020-12-16 17:35:53.000000000","tz":-420},"committer":{"name":"Stephen Taylor","email":"st053q@att.com","date":"2020-12-17 16:29:51.000000000","tz":-420},"subject":"[ceph-osd] Correct naming convention for logical volumes in disk_zap()","message":"[ceph-osd] Correct naming convention for logical volumes in disk_zap()\n\nOSD logical volume names used to be based on the logical disk path,\ni.e. /dev/sdb, but that has changed. The lvremove logic in disk_zap()\nis still using the old naming convention. This change fixes that.\n\nChange-Id: If32ab354670166a3c844991de1744de63a508303\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/213596d71cc3cb3fca7a78e6fef12479f36b89a0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/213596d71cc3cb3fca7a78e6fef12479f36b89a0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30777,"name":"Steven Fitzpatrick","email":"steven@fitzpatrick.wtf","username":"sf280x"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30777,"name":"Steven Fitzpatrick","email":"steven@fitzpatrick.wtf","username":"sf280x"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
