)]}'
{"id":"openstack%2Fopenstack-helm-infra~777711","triplet_id":"openstack%2Fopenstack-helm-infra~master~Ie08fb367ec8b1803fcc6e8cd22dc8da43c90e5c4","project":"openstack/openstack-helm-infra","branch":"master","hashtags":[],"change_id":"Ie08fb367ec8b1803fcc6e8cd22dc8da43c90e5c4","subject":"[ceph-client] enhance logic to enable and disable the autoscaler","status":"MERGED","created":"2021-02-26 03:59:17.000000000","updated":"2021-03-10 18:42:02.000000000","submitted":"2021-03-10 18:39:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"777711","meta_rev_id":"de824bd796747ba57ffae8acb9778ca575aa3323","_number":777711,"virtual_id_number":777711,"owner":{"_account_id":18511,"name":"Brian Wickersham","email":"bkw86@bellsouth.net","username":"bw6938"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30449,"name":"jinyuanliu","email":"liujinyuan@inspur.com","username":"jinyuanliu"},{"value":0,"_account_id":28372,"name":"chinasubbareddy mallavarapu","email":"cr3938@att.com","username":"chinasubbareddy"},{"value":0,"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},{"value":0,"_account_id":18250,"name":"Roy Tang","email":"roy.s.tang@att.com","username":"rt7380"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-10 18:39:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22636,"name":"Cliff Parsons","email":"cliffhparsons@aol.com","username":"cliffparsons"},{"value":0,"_account_id":24780,"name":"Sangeet Gupta","email":"sg774j@att.com","username":"sgupta"},{"value":0,"_account_id":32190,"name":"Frank Ritchie","email":"fr801x@att.com","username":"fr801x"},{"value":0,"_account_id":18511,"name":"Brian Wickersham","email":"bkw86@bellsouth.net","username":"bw6938"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28372,"name":"chinasubbareddy mallavarapu","email":"cr3938@att.com","username":"chinasubbareddy"},"recommended":{"_account_id":30449,"name":"jinyuanliu","email":"liujinyuan@inspur.com","username":"jinyuanliu"},"all":[{"value":1,"date":"2021-03-10 00:38:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30449,"name":"jinyuanliu","email":"liujinyuan@inspur.com","username":"jinyuanliu"},{"value":2,"date":"2021-03-10 14:33:37.000000000","_account_id":28372,"name":"chinasubbareddy mallavarapu","email":"cr3938@att.com","username":"chinasubbareddy"},{"value":1,"date":"2021-03-10 14:32:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},{"value":2,"date":"2021-03-10 15:05:53.000000000","_account_id":18250,"name":"Roy Tang","email":"roy.s.tang@att.com","username":"rt7380"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-03-09 22:53:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22636,"name":"Cliff Parsons","email":"cliffhparsons@aol.com","username":"cliffparsons"},{"value":1,"date":"2021-03-10 14:40:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":24780,"name":"Sangeet Gupta","email":"sg774j@att.com","username":"sgupta"},{"value":1,"date":"2021-03-09 22:18:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32190,"name":"Frank Ritchie","email":"fr801x@att.com","username":"fr801x"},{"value":0,"_account_id":18511,"name":"Brian Wickersham","email":"bkw86@bellsouth.net","username":"bw6938"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":18250,"name":"Roy Tang","email":"roy.s.tang@att.com","username":"rt7380"},"all":[{"value":0,"_account_id":30449,"name":"jinyuanliu","email":"liujinyuan@inspur.com","username":"jinyuanliu"},{"value":0,"_account_id":28372,"name":"chinasubbareddy mallavarapu","email":"cr3938@att.com","username":"chinasubbareddy"},{"value":0,"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},{"value":1,"date":"2021-03-10 15:05:53.000000000","_account_id":18250,"name":"Roy Tang","email":"roy.s.tang@att.com","username":"rt7380"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22636,"name":"Cliff Parsons","email":"cliffhparsons@aol.com","username":"cliffparsons"},{"value":0,"_account_id":24780,"name":"Sangeet Gupta","email":"sg774j@att.com","username":"sgupta"},{"value":0,"_account_id":32190,"name":"Frank Ritchie","email":"fr801x@att.com","username":"fr801x"},{"value":0,"_account_id":18511,"name":"Brian Wickersham","email":"bkw86@bellsouth.net","username":"bw6938"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":33135,"name":"kiran surapathi","email":"ks342f@att.com","username":"kiransurapathi"}],"reviewers":{"REVIEWER":[{"_account_id":18250,"name":"Roy Tang","email":"roy.s.tang@att.com","username":"rt7380"},{"_account_id":18511,"name":"Brian Wickersham","email":"bkw86@bellsouth.net","username":"bw6938"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22636,"name":"Cliff Parsons","email":"cliffhparsons@aol.com","username":"cliffparsons"},{"_account_id":24780,"name":"Sangeet Gupta","email":"sg774j@att.com","username":"sgupta"},{"_account_id":28372,"name":"chinasubbareddy mallavarapu","email":"cr3938@att.com","username":"chinasubbareddy"},{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},{"_account_id":30449,"name":"jinyuanliu","email":"liujinyuan@inspur.com","username":"jinyuanliu"},{"_account_id":32190,"name":"Frank Ritchie","email":"fr801x@att.com","username":"fr801x"},{"_account_id":33135,"name":"kiran surapathi","email":"ks342f@att.com","username":"kiransurapathi"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-26 04:45:13.000000000","updated_by":{"_account_id":18511,"name":"Brian Wickersham","email":"bkw86@bellsouth.net","username":"bw6938"},"reviewer":{"_account_id":28372,"name":"chinasubbareddy mallavarapu","email":"cr3938@att.com","username":"chinasubbareddy"},"state":"REVIEWER"},{"updated":"2021-02-26 04:45:13.000000000","updated_by":{"_account_id":18511,"name":"Brian Wickersham","email":"bkw86@bellsouth.net","username":"bw6938"},"reviewer":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"state":"REVIEWER"},{"updated":"2021-02-26 05:37:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-26 19:17:01.000000000","updated_by":{"_account_id":18511,"name":"Brian Wickersham","email":"bkw86@bellsouth.net","username":"bw6938"},"reviewer":{"_account_id":22636,"name":"Cliff Parsons","email":"cliffhparsons@aol.com","username":"cliffparsons"},"state":"CC"},{"updated":"2021-02-26 19:17:46.000000000","updated_by":{"_account_id":18511,"name":"Brian Wickersham","email":"bkw86@bellsouth.net","username":"bw6938"},"reviewer":{"_account_id":32190,"name":"Frank Ritchie","email":"fr801x@att.com","username":"fr801x"},"state":"REVIEWER"},{"updated":"2021-02-26 19:17:46.000000000","updated_by":{"_account_id":18511,"name":"Brian Wickersham","email":"bkw86@bellsouth.net","username":"bw6938"},"reviewer":{"_account_id":22636,"name":"Cliff Parsons","email":"cliffhparsons@aol.com","username":"cliffparsons"},"state":"REVIEWER"},{"updated":"2021-02-26 21:01:13.000000000","updated_by":{"_account_id":33135,"name":"kiran surapathi","email":"ks342f@att.com","username":"kiransurapathi"},"reviewer":{"_account_id":33135,"name":"kiran surapathi","email":"ks342f@att.com","username":"kiransurapathi"},"state":"REVIEWER"},{"updated":"2021-03-02 13:24:47.000000000","updated_by":{"_account_id":24780,"name":"Sangeet Gupta","email":"sg774j@att.com","username":"sgupta"},"reviewer":{"_account_id":24780,"name":"Sangeet Gupta","email":"sg774j@att.com","username":"sgupta"},"state":"REVIEWER"},{"updated":"2021-03-03 00:35:30.000000000","updated_by":{"_account_id":30449,"name":"jinyuanliu","email":"liujinyuan@inspur.com","username":"jinyuanliu"},"reviewer":{"_account_id":30449,"name":"jinyuanliu","email":"liujinyuan@inspur.com","username":"jinyuanliu"},"state":"REVIEWER"},{"updated":"2021-03-10 15:05:53.000000000","updated_by":{"_account_id":18250,"name":"Roy Tang","email":"roy.s.tang@att.com","username":"rt7380"},"reviewer":{"_account_id":18250,"name":"Roy Tang","email":"roy.s.tang@att.com","username":"rt7380"},"state":"REVIEWER"}],"messages":[{"id":"fd8e0a8335c3c2d7726e4e6d71b22bc411ba0f90","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":18511,"name":"Brian Wickersham","email":"bkw86@bellsouth.net","username":"bw6938"},"date":"2021-02-26 03:59:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"214c136a7b77cab4de187e41ace774fe78c6bb21","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":18511,"name":"Brian Wickersham","email":"bkw86@bellsouth.net","username":"bw6938"},"date":"2021-02-26 04:03:40.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"ebc4b867d7461fa0ee4830eef37f8ddadaed17d2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":18511,"name":"Brian Wickersham","email":"bkw86@bellsouth.net","username":"bw6938"},"date":"2021-02-26 04:07:27.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"ba58a238e2d3896daf58917e59df6f6fc1e8714c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":18511,"name":"Brian Wickersham","email":"bkw86@bellsouth.net","username":"bw6938"},"date":"2021-02-26 04:11:11.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"870f8ba282f68889244690d4197bfaeaf416d7ab","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":18511,"name":"Brian Wickersham","email":"bkw86@bellsouth.net","username":"bw6938"},"date":"2021-02-26 04:23:25.000000000","message":"Patch Set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"68d8769bf7966d57b44b7052e8e666d52a1e6d87","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":18511,"name":"Brian Wickersham","email":"bkw86@bellsouth.net","username":"bw6938"},"date":"2021-02-26 04:32:58.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":5},{"id":"ab364d3ab8602189e9f235a24cc81c700a6ba979","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":18511,"name":"Brian Wickersham","email":"bkw86@bellsouth.net","username":"bw6938"},"date":"2021-02-26 04:43:35.000000000","message":"Uploaded patch set 6: Commit message was updated.","accounts_in_message":[],"_revision_number":6},{"id":"e9cf17d43283e54058272ec36c2b955d9dff5e4d","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":18511,"name":"Brian Wickersham","email":"bkw86@bellsouth.net","username":"bw6938"},"date":"2021-02-26 04:44:36.000000000","message":"Patch Set 7: Patch Set 6 was rebased","accounts_in_message":[],"_revision_number":7},{"id":"e577f97cf0a9aa21f9c92271a7ccb261c356114e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-26 05:37:25.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b17b4de55d5346979687aba8bbe94c94 : SUCCESS in 3m 33s\n- openstack-helm-lint https://zuul.opendev.org/t/openstack/build/94da44e042b24aeab3473da850e0b8d9 : SUCCESS in 3m 44s\n- openstack-helm-infra-aio-logging https://zuul.opendev.org/t/openstack/build/81744a13644340d480710099703947ac : SUCCESS in 30m 20s\n- openstack-helm-infra-aio-monitoring https://zuul.opendev.org/t/openstack/build/f9cdfffb8ff844fbb8b9bf1f7e2f2d05 : SUCCESS in 29m 34s\n- openstack-helm-infra-openstack-support https://zuul.opendev.org/t/openstack/build/3e6bbeba60c848baaf2ae1e55b40be2d : SUCCESS in 51m 35s\n- openstack-helm-infra-openstack-support-ssl https://zuul.opendev.org/t/openstack/build/986577d2f98c45f5ab27eb04960e09ea : SUCCESS in 46m 39s\n- openstack-helm-infra-metacontroller https://zuul.opendev.org/t/openstack/build/310bcfd5b60a42eea6c8f4722094200d : SUCCESS in 14m 19s","accounts_in_message":[],"_revision_number":7},{"id":"6a5c98803f30463396867ea410c4be30db56c6bb","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":18511,"name":"Brian Wickersham","email":"bkw86@bellsouth.net","username":"bw6938"},"date":"2021-02-26 14:45:08.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"b04554642023b3978666f6571c6973486ddf9ea0","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":18511,"name":"Brian Wickersham","email":"bkw86@bellsouth.net","username":"bw6938"},"date":"2021-02-26 14:45:42.000000000","message":"Patch Set 9: Patch Set 8 was rebased","accounts_in_message":[],"_revision_number":9},{"id":"19f48722ffd09807192ee01f54495b9ba77ff7fa","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":18511,"name":"Brian Wickersham","email":"bkw86@bellsouth.net","username":"bw6938"},"date":"2021-02-26 15:34:30.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":9},{"id":"0bbb0b3516d9d2fe40f600ffcae889a7d9b6fa60","author":{"_account_id":18511,"name":"Brian Wickersham","email":"bkw86@bellsouth.net","username":"bw6938"},"date":"2021-02-26 15:34:41.000000000","message":"Patch Set 9: Workflow-1","accounts_in_message":[],"_revision_number":9},{"id":"aa0735071b5ef08aaaae084082458544fb7e9b77","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-26 15:36:50.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bfe1576d92864905bcf5610128dc4404 : SUCCESS in 4m 13s\n- openstack-helm-lint https://zuul.opendev.org/t/openstack/build/548730f39fec4b938cf83570f684f4d8 : SUCCESS in 4m 32s\n- openstack-helm-infra-aio-logging https://zuul.opendev.org/t/openstack/build/96012d1296a94d47b48d36299a5e5904 : SUCCESS in 44m 14s\n- openstack-helm-infra-aio-monitoring https://zuul.opendev.org/t/openstack/build/96ea829d03714cfb81677465ff10cce5 : SUCCESS in 32m 52s\n- openstack-helm-infra-openstack-support https://zuul.opendev.org/t/openstack/build/3d444069eca248428e589fd2bd08844e : SUCCESS in 48m 13s\n- openstack-helm-infra-openstack-support-ssl https://zuul.opendev.org/t/openstack/build/af87afe331c14cf28981baa502ae72c5 : SUCCESS in 44m 38s\n- openstack-helm-infra-metacontroller https://zuul.opendev.org/t/openstack/build/1c210314079349779e861393949a3705 : SUCCESS in 17m 41s","accounts_in_message":[],"_revision_number":9},{"id":"961f466936068387538e490d6e6264a6441d90f8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":18511,"name":"Brian Wickersham","email":"bkw86@bellsouth.net","username":"bw6938"},"date":"2021-02-26 15:56:30.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"813e9ebdb1149518878359825d1da49741244333","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":18511,"name":"Brian Wickersham","email":"bkw86@bellsouth.net","username":"bw6938"},"date":"2021-02-26 15:58:12.000000000","message":"Patch Set 11: Patch Set 10 was rebased","accounts_in_message":[],"_revision_number":11},{"id":"0d4deb567fcd84176e103a55124df6d5ad534223","author":{"_account_id":18511,"name":"Brian Wickersham","email":"bkw86@bellsouth.net","username":"bw6938"},"date":"2021-02-26 15:58:57.000000000","message":"Patch Set 11: Workflow-1","accounts_in_message":[],"_revision_number":11},{"id":"98f8d38a152c4813e74fffe228549b231546de89","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":18511,"name":"Brian Wickersham","email":"bkw86@bellsouth.net","username":"bw6938"},"date":"2021-02-26 16:16:34.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"b9335b70720b5d8ebf30d22dfaf295401c16dd78","author":{"_account_id":18511,"name":"Brian Wickersham","email":"bkw86@bellsouth.net","username":"bw6938"},"date":"2021-02-26 16:18:37.000000000","message":"Patch Set 12: Workflow-1","accounts_in_message":[],"_revision_number":12},{"id":"3f5b1c55191d13d92ffecedf3d3cba5b24e548a2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":18511,"name":"Brian Wickersham","email":"bkw86@bellsouth.net","username":"bw6938"},"date":"2021-02-26 16:31:37.000000000","message":"Uploaded patch set 13: Commit message was updated.","accounts_in_message":[],"_revision_number":13},{"id":"a085b7ed9a42876f084d44ba72372bc1020e7d29","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-26 17:12:11.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/316641772d82410b83595809554cb64f : SUCCESS in 3m 14s\n- openstack-helm-lint https://zuul.opendev.org/t/openstack/build/0da4531f5d364ac597fef56eab55b70f : SUCCESS in 4m 13s\n- openstack-helm-infra-aio-logging https://zuul.opendev.org/t/openstack/build/8d19ef9246584214aad309441c70f40f : FAILURE in 31m 48s\n- openstack-helm-infra-aio-monitoring https://zuul.opendev.org/t/openstack/build/3081ee2fe2a0465c940d28998c49d3a8 : SUCCESS in 28m 17s\n- openstack-helm-infra-openstack-support https://zuul.opendev.org/t/openstack/build/263ef836296b48a2a1e46caafa2b555e : FAILURE in 31m 58s\n- openstack-helm-infra-openstack-support-ssl https://zuul.opendev.org/t/openstack/build/5786de6e2e18424f948ec2d84a831a05 : FAILURE in 32m 09s\n- openstack-helm-infra-metacontroller https://zuul.opendev.org/t/openstack/build/cb85a2f48a22434f8880c285daff2512 : SUCCESS in 14m 28s","accounts_in_message":[],"_revision_number":13},{"id":"49f9f574800a75509a1c438662980f017d038b1c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":18511,"name":"Brian Wickersham","email":"bkw86@bellsouth.net","username":"bw6938"},"date":"2021-02-26 19:15:51.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"1fc16e093250d64f046e114ca9901068c9157a9e","author":{"_account_id":18511,"name":"Brian Wickersham","email":"bkw86@bellsouth.net","username":"bw6938"},"date":"2021-02-26 19:16:49.000000000","message":"Patch Set 14: Workflow-1","accounts_in_message":[],"_revision_number":14},{"id":"b0c01c52f66488df3a2cc5c124908439e99a1419","tag":"autogenerated:gerrit:setAssignee","author":{"_account_id":18511,"name":"Brian Wickersham","email":"bkw86@bellsouth.net","username":"bw6938"},"date":"2021-02-26 19:17:01.000000000","message":"Assignee added: Cliff Parsons \u003ccliff.parsons@att.com\u003e","accounts_in_message":[],"_revision_number":14},{"id":"8c4bc230180258a4295067c38245bfc8ba0f4b98","tag":"autogenerated:gerrit:deleteAssignee","author":{"_account_id":18511,"name":"Brian Wickersham","email":"bkw86@bellsouth.net","username":"bw6938"},"date":"2021-02-26 19:17:05.000000000","message":"Assignee deleted: Cliff Parsons \u003ccliff.parsons@att.com\u003e","accounts_in_message":[],"_revision_number":14},{"id":"fe3dd5019e491ba819e7b6b4bcad795f66c9aa32","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-26 19:52:42.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/efb55ee2a8f94304961ed020c0a3f026 : SUCCESS in 3m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e16524779c2e427d881e29451ac9d13f : SUCCESS in 3m 15s\n- openstack-helm-lint https://zuul.opendev.org/t/openstack/build/f536368b99f041de973c229932f3d0d6 : SUCCESS in 3m 56s\n- openstack-helm-infra-aio-logging https://zuul.opendev.org/t/openstack/build/6b3eb6a5c2b141269b4a64b91fd784d0 : FAILURE in 32m 17s\n- openstack-helm-infra-aio-monitoring https://zuul.opendev.org/t/openstack/build/c44e0b3699734c79899daf589a49a476 : SUCCESS in 35m 32s\n- openstack-helm-infra-openstack-support https://zuul.opendev.org/t/openstack/build/bda91d0990a648f594799754da8fc72d : FAILURE in 31m 22s\n- openstack-helm-infra-openstack-support-ssl https://zuul.opendev.org/t/openstack/build/fe50c806fbf24de286cdbeb3569254c2 : FAILURE in 30m 10s\n- openstack-helm-infra-metacontroller https://zuul.opendev.org/t/openstack/build/cfcd8619f0874826bd5fe2fa18949e99 : SUCCESS in 14m 11s","accounts_in_message":[],"_revision_number":14},{"id":"daf09f5dc27a18f00ebadde0e1d6c65bafeda463","author":{"_account_id":18511,"name":"Brian Wickersham","email":"bkw86@bellsouth.net","username":"bw6938"},"date":"2021-02-26 20:34:54.000000000","message":"Patch Set 14:\n\nrecheck","accounts_in_message":[],"_revision_number":14},{"id":"4628382729acdd021bffe5c9618e7eff5d78aced","tag":"autogenerated:gerrit:deleteVote","author":{"_account_id":18511,"name":"Brian Wickersham","email":"bkw86@bellsouth.net","username":"bw6938"},"date":"2021-02-26 20:36:43.000000000","message":"Removed Workflow-1 by Brian Wickersham \u003cbrian.wickersham@att.com\u003e\n","accounts_in_message":[],"_revision_number":14},{"id":"c8bcfefe822e7b2632f3815a6733591b4c5339c1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-26 21:11:54.000000000","message":"Patch Set 14:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/720c26526103499b84e56062a6a39430 : SUCCESS in 3m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/96f5a79432d5406085a5e0c7372fb092 : SUCCESS in 3m 20s\n- openstack-helm-lint https://zuul.opendev.org/t/openstack/build/91c000252948427e9dd2edd952457b62 : SUCCESS in 4m 07s\n- openstack-helm-infra-aio-logging https://zuul.opendev.org/t/openstack/build/ab643f47c59047a091b5cfc0c37ef4dc : FAILURE in 32m 31s\n- openstack-helm-infra-aio-monitoring https://zuul.opendev.org/t/openstack/build/be0b831cfbb84215a652973aa7f0743a : SUCCESS in 27m 37s\n- openstack-helm-infra-openstack-support https://zuul.opendev.org/t/openstack/build/778e4aeb4f91455aa5334b3fa9af6009 : FAILURE in 11m 32s\n- openstack-helm-infra-openstack-support-ssl https://zuul.opendev.org/t/openstack/build/d0e1a5965d7742debaa5e39a8f13b46f : FAILURE in 34m 33s\n- openstack-helm-infra-metacontroller https://zuul.opendev.org/t/openstack/build/014c4c71aa5644fe9cbde9b428681258 : SUCCESS in 15m 05s","accounts_in_message":[],"_revision_number":14},{"id":"b45293bf109faa2338c6e6c6c93a70cb2ea865f9","author":{"_account_id":18511,"name":"Brian Wickersham","email":"bkw86@bellsouth.net","username":"bw6938"},"date":"2021-02-28 05:22:28.000000000","message":"Patch Set 14:\n\nrecheck","accounts_in_message":[],"_revision_number":14},{"id":"9385a059cb61562cfd560e1f0ff7e066da2fd8b8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-28 05:55:03.000000000","message":"Patch Set 14:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e8e0f9a5300e4daf9ddc83da09413a25 : SUCCESS in 3m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9c01e2c48e40451faf1f0c39a751583b : SUCCESS in 3m 09s\n- openstack-helm-lint https://zuul.opendev.org/t/openstack/build/bcd54ab668cc4499b02024f8e45e3b21 : SUCCESS in 3m 55s\n- openstack-helm-infra-aio-logging https://zuul.opendev.org/t/openstack/build/1f7c3938f31245109c036b8ab5deafae : FAILURE in 31m 40s\n- openstack-helm-infra-aio-monitoring https://zuul.opendev.org/t/openstack/build/3232e7437fec4fb4b593aeec7550cfc1 : SUCCESS in 24m 57s\n- openstack-helm-infra-openstack-support https://zuul.opendev.org/t/openstack/build/bcd23b4c4f6840fea7e9cd76f843d0d6 : FAILURE in 30m 53s\n- openstack-helm-infra-openstack-support-ssl https://zuul.opendev.org/t/openstack/build/0c49454d0b9e4f6ba143df6b878112a0 : FAILURE in 29m 24s\n- openstack-helm-infra-metacontroller https://zuul.opendev.org/t/openstack/build/a7b204380b984cf481fe0ed1d3c00758 : SUCCESS in 13m 06s","accounts_in_message":[],"_revision_number":14},{"id":"9d3175c68f3d9a0c715bd2310f16e42b389b8e0a","author":{"_account_id":18511,"name":"Brian Wickersham","email":"bkw86@bellsouth.net","username":"bw6938"},"date":"2021-02-28 14:49:50.000000000","message":"Patch Set 14:\n\nrecheck","accounts_in_message":[],"_revision_number":14},{"id":"766629e9dc61231b5e96723054be363d8416a867","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-28 15:28:53.000000000","message":"Patch Set 14:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/74e22a19ca4f4c7ebbf14e0c2866db6a : SUCCESS in 3m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/da549672b71545b0b04a607d39a0ff48 : SUCCESS in 4m 40s\n- openstack-helm-lint https://zuul.opendev.org/t/openstack/build/190a5cb44b3f4cd7838835419d96ab22 : SUCCESS in 3m 36s\n- openstack-helm-infra-aio-logging https://zuul.opendev.org/t/openstack/build/527fc6ed44f84752942d0a8f00312a1c : FAILURE in 28m 53s\n- openstack-helm-infra-aio-monitoring https://zuul.opendev.org/t/openstack/build/de9dd81b5ce240e482b86ea05744151a : SUCCESS in 36m 45s\n- openstack-helm-infra-openstack-support https://zuul.opendev.org/t/openstack/build/e5e1ec279f6640e1837945b81426a647 : FAILURE in 29m 50s\n- openstack-helm-infra-openstack-support-ssl https://zuul.opendev.org/t/openstack/build/b6437e3e25fc4d82ad81c60364ce91cd : FAILURE in 29m 03s\n- openstack-helm-infra-metacontroller https://zuul.opendev.org/t/openstack/build/ac86302222984219a5d18454235c6eaf : SUCCESS in 12m 54s","accounts_in_message":[],"_revision_number":14},{"id":"157a2a0cad0b28fe4a8a4fae6d17fd54279f793e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":18511,"name":"Brian Wickersham","email":"bkw86@bellsouth.net","username":"bw6938"},"date":"2021-03-01 01:34:40.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"2dbfeb412687f481c05642d0498e30422be6a167","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-01 02:08:50.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2260688a7b864339b41a5234836a091a : SUCCESS in 3m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d32bcfc0017046659032afce3dd5c7a2 : SUCCESS in 3m 14s\n- openstack-helm-lint https://zuul.opendev.org/t/openstack/build/1b42b729591e47e79ecdb9c6e0ee0cf6 : SUCCESS in 3m 53s\n- openstack-helm-infra-aio-logging https://zuul.opendev.org/t/openstack/build/12fc0ec7bf92443bae7768d1405f6b64 : FAILURE in 30m 24s\n- openstack-helm-infra-aio-monitoring https://zuul.opendev.org/t/openstack/build/8b182d8102f34b098991b8cd29c73dff : SUCCESS in 33m 13s\n- openstack-helm-infra-openstack-support https://zuul.opendev.org/t/openstack/build/8f5c3df56e8342508ecf886064fc4204 : FAILURE in 27m 52s\n- openstack-helm-infra-openstack-support-ssl https://zuul.opendev.org/t/openstack/build/1bb445df0cf74d5f9f59e0c32b4ba5dd : FAILURE in 31m 09s\n- openstack-helm-infra-metacontroller https://zuul.opendev.org/t/openstack/build/78a78598a90e43d4b80d8309965d5d59 : SUCCESS in 13m 09s","accounts_in_message":[],"_revision_number":15},{"id":"0f30d9f411755a514c42177f8070b59cddfe8b24","author":{"_account_id":18511,"name":"Brian Wickersham","email":"bkw86@bellsouth.net","username":"bw6938"},"date":"2021-03-01 03:12:19.000000000","message":"Patch Set 15:\n\nrecheck","accounts_in_message":[],"_revision_number":15},{"id":"47c267f8ee8ea189aaf78efd538a8abcb55dd2f9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-01 03:49:26.000000000","message":"Patch Set 15:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5cd8d20208714550ac1281ad7474c183 : SUCCESS in 3m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a60809cde5544ca3adc6ae4f1e389bcc : SUCCESS in 3m 18s\n- openstack-helm-lint https://zuul.opendev.org/t/openstack/build/f27be41acc934ea5b51c782f9c5517fb : SUCCESS in 3m 56s\n- openstack-helm-infra-aio-logging https://zuul.opendev.org/t/openstack/build/b949bf2163e34d88abb30176509797c7 : FAILURE in 31m 38s\n- openstack-helm-infra-aio-monitoring https://zuul.opendev.org/t/openstack/build/7e6bd47b7b1141f7949781b49235d03e : SUCCESS in 28m 29s\n- openstack-helm-infra-openstack-support https://zuul.opendev.org/t/openstack/build/ba07c5c689ac4b3cbba92797711a8955 : FAILURE in 31m 09s\n- openstack-helm-infra-openstack-support-ssl https://zuul.opendev.org/t/openstack/build/a55eac5635c24d2eac68311b88270401 : FAILURE in 30m 52s\n- openstack-helm-infra-metacontroller https://zuul.opendev.org/t/openstack/build/1e598263fe2242d6aa273a44d0ee5365 : SUCCESS in 13m 28s","accounts_in_message":[],"_revision_number":15},{"id":"bb6c1f57d6bf33c4c5a15aa6475fda55263a813c","author":{"_account_id":18511,"name":"Brian Wickersham","email":"bkw86@bellsouth.net","username":"bw6938"},"date":"2021-03-01 15:06:09.000000000","message":"Patch Set 15: Code-Review+1","accounts_in_message":[],"_revision_number":15},{"id":"a1775135d73d3f2bfdb9c3851b523ff44168cf6c","tag":"autogenerated:gerrit:deleteVote","author":{"_account_id":18511,"name":"Brian Wickersham","email":"bkw86@bellsouth.net","username":"bw6938"},"date":"2021-03-01 15:06:26.000000000","message":"Removed Code-Review+1 by Brian Wickersham \u003cbrian.wickersham@att.com\u003e\n","accounts_in_message":[],"_revision_number":15},{"id":"e011250399d072740d936bc8f8a9abb3a0443505","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":18511,"name":"Brian Wickersham","email":"bkw86@bellsouth.net","username":"bw6938"},"date":"2021-03-01 17:47:07.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"4d9601c83eb71f38f15f0b4642ad93ae13bb9a7d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-01 18:27:36.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0c9a920829ac437094b34a0cfa85557d : SUCCESS in 3m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2770fda509924dc194b0c8a07e959468 : SUCCESS in 4m 06s\n- openstack-helm-lint https://zuul.opendev.org/t/openstack/build/fd82ba2a2fc846c897e31ff30d00cb6c : SUCCESS in 4m 52s\n- openstack-helm-infra-aio-logging https://zuul.opendev.org/t/openstack/build/3e8f2da26f484984abf23a9a29ff7205 : FAILURE in 31m 58s\n- openstack-helm-infra-aio-monitoring https://zuul.opendev.org/t/openstack/build/baacbc7a9dd44b5fabc737ffdfdd6b2f : SUCCESS in 26m 47s\n- openstack-helm-infra-openstack-support https://zuul.opendev.org/t/openstack/build/76cf306dfe9f45fd872fefb87c5fb0cf : FAILURE in 31m 27s\n- openstack-helm-infra-openstack-support-ssl https://zuul.opendev.org/t/openstack/build/b01a2524f9294f33870ed7bc0ee7a987 : FAILURE in 32m 55s\n- openstack-helm-infra-metacontroller https://zuul.opendev.org/t/openstack/build/8f66c47f52e14ca990f9d5ba545a9801 : SUCCESS in 15m 04s","accounts_in_message":[],"_revision_number":16},{"id":"f364984573c7f788430e3e62efd9524898435c12","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":18511,"name":"Brian Wickersham","email":"bkw86@bellsouth.net","username":"bw6938"},"date":"2021-03-01 18:59:42.000000000","message":"Uploaded patch set 17.","accounts_in_message":[],"_revision_number":17},{"id":"11a6b035f34f725200b6df3a51f62b511b96087f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-01 19:50:17.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a60e46ff3d9449e8a30587bdaec00e43 : SUCCESS in 4m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ec91ac2325bd4039b90da38205982dd0 : SUCCESS in 3m 36s\n- openstack-helm-lint https://zuul.opendev.org/t/openstack/build/64ca54f2fdf949edb33300a5adf9a6c3 : SUCCESS in 4m 54s\n- openstack-helm-infra-aio-logging https://zuul.opendev.org/t/openstack/build/3c65655aa7584d1bbaddc3677fea774a : SUCCESS in 34m 25s\n- openstack-helm-infra-aio-monitoring https://zuul.opendev.org/t/openstack/build/dea8c4e0b9d6464f89b5cb66771fb959 : SUCCESS in 25m 23s\n- openstack-helm-infra-openstack-support https://zuul.opendev.org/t/openstack/build/3f167f0bf4024aa49e2db0a9e721819f : SUCCESS in 45m 00s\n- openstack-helm-infra-openstack-support-ssl https://zuul.opendev.org/t/openstack/build/cad345dfdb35413fb18bce3ba4540851 : SUCCESS in 43m 43s\n- openstack-helm-infra-metacontroller https://zuul.opendev.org/t/openstack/build/cc4e4aebb5de41fd8f12602d282ba8e5 : SUCCESS in 13m 26s","accounts_in_message":[],"_revision_number":17},{"id":"31be88893ed3addf522fa79f04deb33536cb34f4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":18511,"name":"Brian Wickersham","email":"bkw86@bellsouth.net","username":"bw6938"},"date":"2021-03-01 20:47:27.000000000","message":"Uploaded patch set 18.","accounts_in_message":[],"_revision_number":18},{"id":"baceee56082e31a879ecb2b6b2185cbb763c71c0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-01 21:26:00.000000000","message":"Patch Set 18: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e15c9be09179499599667cb363a4ba99 : SUCCESS in 3m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1781101774b7417cabf51457cf77af40 : SUCCESS in 4m 05s\n- openstack-helm-lint https://zuul.opendev.org/t/openstack/build/d81a721dc919456a980e3bb4d5504395 : SUCCESS in 4m 30s\n- openstack-helm-infra-aio-logging https://zuul.opendev.org/t/openstack/build/3a2bb51c9f6440558c6d15d822dcf8d3 : FAILURE in 34m 37s\n- openstack-helm-infra-aio-monitoring https://zuul.opendev.org/t/openstack/build/ad16cecebca54e849447eb0bfc4a545b : SUCCESS in 27m 28s\n- openstack-helm-infra-openstack-support https://zuul.opendev.org/t/openstack/build/d147899df3d84114ba54b90ef2fb2807 : FAILURE in 32m 42s\n- openstack-helm-infra-openstack-support-ssl https://zuul.opendev.org/t/openstack/build/e2d4c7a4668b4c58a01d7edef55d170c : FAILURE in 31m 03s\n- openstack-helm-infra-metacontroller https://zuul.opendev.org/t/openstack/build/c0560105f5a94d8c979a24dcdfa15406 : SUCCESS in 14m 47s","accounts_in_message":[],"_revision_number":18},{"id":"81a05b458c985ba3a52f6a25395d4a5757aaa13b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":18511,"name":"Brian Wickersham","email":"bkw86@bellsouth.net","username":"bw6938"},"date":"2021-03-02 03:00:18.000000000","message":"Uploaded patch set 19.","accounts_in_message":[],"_revision_number":19},{"id":"3589d65500f527289cf31f63b962a85ef2d2d964","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-02 03:33:30.000000000","message":"Patch Set 19: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b3f84b5ee7e3416c99486c51866a99eb : SUCCESS in 4m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6d30f3de7999487eb64ea9a2d18e474b : SUCCESS in 3m 39s\n- openstack-helm-lint https://zuul.opendev.org/t/openstack/build/00eecad782ff410285b9b44cf866034b : SUCCESS in 4m 55s\n- openstack-helm-infra-aio-logging https://zuul.opendev.org/t/openstack/build/08321256ec634e1c8e91f921dbc4a8a8 : FAILURE in 32m 36s\n- openstack-helm-infra-aio-monitoring https://zuul.opendev.org/t/openstack/build/b4629201b2b0459f8eda6109df82e2cf : SUCCESS in 30m 54s\n- openstack-helm-infra-openstack-support https://zuul.opendev.org/t/openstack/build/6eb0ff124d4f4b739507b994508a85ea : FAILURE in 30m 59s\n- openstack-helm-infra-openstack-support-ssl https://zuul.opendev.org/t/openstack/build/2ab0633d27b84a4e8141b9ca7f903788 : FAILURE in 31m 21s\n- openstack-helm-infra-metacontroller https://zuul.opendev.org/t/openstack/build/aded8c84c1db4765853947076c48655c : SUCCESS in 13m 56s","accounts_in_message":[],"_revision_number":19},{"id":"0f37d68da2fe6cd160643e66e1261399c710e73b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":18511,"name":"Brian Wickersham","email":"bkw86@bellsouth.net","username":"bw6938"},"date":"2021-03-02 03:44:09.000000000","message":"Uploaded patch set 20.","accounts_in_message":[],"_revision_number":20},{"id":"3cdde7d0fbe159795bb0d881fcf73f53dc00de71","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-02 04:29:39.000000000","message":"Patch Set 20: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/213f6f01f90247b2b0cef2e77c67532d : SUCCESS in 4m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b1e383ef4669409f8dd4865ee0a10776 : SUCCESS in 3m 47s\n- openstack-helm-lint https://zuul.opendev.org/t/openstack/build/d3a8f720a5f842598e01d747eb52b3d6 : SUCCESS in 4m 34s\n- openstack-helm-infra-aio-logging https://zuul.opendev.org/t/openstack/build/6d5c068d00de4474b18cdd33fe3c848f : SUCCESS in 31m 28s\n- openstack-helm-infra-aio-monitoring https://zuul.opendev.org/t/openstack/build/6b805bfb624441f3afcbadc414f93a23 : SUCCESS in 28m 51s\n- openstack-helm-infra-openstack-support https://zuul.opendev.org/t/openstack/build/d1a7bebc1e874c438d4ba65bb8594c74 : SUCCESS in 42m 20s\n- openstack-helm-infra-openstack-support-ssl https://zuul.opendev.org/t/openstack/build/f01aae3b78834caba7b2d80ae068883a : SUCCESS in 45m 00s\n- openstack-helm-infra-metacontroller https://zuul.opendev.org/t/openstack/build/e0e7548861bf49ae87112deaacee79ac : SUCCESS in 13m 27s","accounts_in_message":[],"_revision_number":20},{"id":"4bd36bec488b7bdba42f1497ecadb60b967d47c0","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":18511,"name":"Brian Wickersham","email":"bkw86@bellsouth.net","username":"bw6938"},"date":"2021-03-02 05:24:08.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":20},{"id":"8c6af6289ba35bed6c1f5b1c9d4a1c8d78f5ecd6","author":{"_account_id":24780,"name":"Sangeet Gupta","email":"sg774j@att.com","username":"sgupta"},"date":"2021-03-02 13:24:47.000000000","message":"Patch Set 20: Code-Review+1","accounts_in_message":[],"_revision_number":20},{"id":"86d6622eef7e3aad89d0760a76c474627d995a39","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":18511,"name":"Brian Wickersham","email":"bkw86@bellsouth.net","username":"bw6938"},"date":"2021-03-02 19:59:05.000000000","message":"Uploaded patch set 21.","accounts_in_message":[],"_revision_number":21},{"id":"6f8b014b0cbd2789d1b07a8a84ab361c0907a6d1","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":18511,"name":"Brian Wickersham","email":"bkw86@bellsouth.net","username":"bw6938"},"date":"2021-03-02 20:39:34.000000000","message":"Uploaded patch set 22.","accounts_in_message":[],"_revision_number":22},{"id":"dccf72c74d3fd898000b6f737f5bbd47855e0f29","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-02 21:21:06.000000000","message":"Patch Set 22: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad0217be160348b198c9b1e80c71ac2c : SUCCESS in 4m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5bca0af9afa244368720c8d7dc64f0fa : SUCCESS in 4m 26s\n- openstack-helm-lint https://zuul.opendev.org/t/openstack/build/748b67f74ed64cdabfb34921c0c020c7 : FAILURE in 3m 50s\n- openstack-helm-infra-aio-logging https://zuul.opendev.org/t/openstack/build/b02eb4c517c64d9685dbb2dc874a8597 : SUCCESS in 34m 18s\n- openstack-helm-infra-aio-monitoring https://zuul.opendev.org/t/openstack/build/3e701e55fcbe49748f54c633ea5282c6 : FAILURE in 17m 05s\n- openstack-helm-infra-openstack-support https://zuul.opendev.org/t/openstack/build/87160058ce3747cfbe0dfb2a6689a7e9 : FAILURE in 40m 16s\n- openstack-helm-infra-openstack-support-ssl https://zuul.opendev.org/t/openstack/build/c6c8b8502c1547e894186a288f4cb769 : FAILURE in 40m 10s\n- openstack-helm-infra-metacontroller https://zuul.opendev.org/t/openstack/build/7d6bffff53cb4de09e8c63fb29340704 : SUCCESS in 16m 07s","accounts_in_message":[],"_revision_number":22},{"id":"0cb88a34ed66f67f0bfcb1155d82172250d9ecab","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":18511,"name":"Brian Wickersham","email":"bkw86@bellsouth.net","username":"bw6938"},"date":"2021-03-02 22:33:33.000000000","message":"Uploaded patch set 23.","accounts_in_message":[],"_revision_number":23},{"id":"033ad1119625d5c34ea304ed15b4d064ba237035","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-02 23:23:09.000000000","message":"Patch Set 23: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/daebe758c7664accbbdac970942670d6 : SUCCESS in 4m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6911490ca84540bab079e098d4afa068 : SUCCESS in 4m 39s\n- openstack-helm-lint https://zuul.opendev.org/t/openstack/build/84022f16c9fd41b3b43d41b911d50175 : FAILURE in 3m 14s\n- openstack-helm-infra-aio-logging https://zuul.opendev.org/t/openstack/build/d6486bb42c46481a9cd2d1d8c395d12e : SUCCESS in 35m 06s\n- openstack-helm-infra-aio-monitoring https://zuul.opendev.org/t/openstack/build/6bcac543a27a4647b315377767ff1dc0 : SUCCESS in 48m 54s\n- openstack-helm-infra-openstack-support https://zuul.opendev.org/t/openstack/build/44dc914111504d5e87782abb74a42064 : SUCCESS in 47m 17s\n- openstack-helm-infra-openstack-support-ssl https://zuul.opendev.org/t/openstack/build/b5c5272320024850b8c3c775231777fa : SUCCESS in 42m 18s\n- openstack-helm-infra-metacontroller https://zuul.opendev.org/t/openstack/build/da27a9066559413fb764d2cb016b320b : SUCCESS in 14m 55s","accounts_in_message":[],"_revision_number":23},{"id":"655c35ed6a2c8ed395cd205dcffee61d87813e21","author":{"_account_id":30449,"name":"jinyuanliu","email":"liujinyuan@inspur.com","username":"jinyuanliu"},"date":"2021-03-03 00:35:30.000000000","message":"Patch Set 23: Code-Review+1","accounts_in_message":[],"_revision_number":23},{"id":"686ae7fe7db4fd218d5bb81768384a383adaadee","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":18511,"name":"Brian Wickersham","email":"bkw86@bellsouth.net","username":"bw6938"},"date":"2021-03-03 04:05:51.000000000","message":"Uploaded patch set 24.","accounts_in_message":[],"_revision_number":24},{"id":"682546bc1aa507563d3663a8469a632cd25f241c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-03 04:59:23.000000000","message":"Patch Set 24: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c58710e814c745e19189147d1f68a91b : SUCCESS in 4m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2ed02d54c6c34dbd8e315f63c841781c : SUCCESS in 3m 11s\n- openstack-helm-lint https://zuul.opendev.org/t/openstack/build/9bb964dcecee4394880997fb71b70503 : SUCCESS in 3m 44s\n- openstack-helm-infra-aio-logging https://zuul.opendev.org/t/openstack/build/b10e2303a22f46dc8a3a748789062a27 : SUCCESS in 31m 03s\n- openstack-helm-infra-aio-monitoring https://zuul.opendev.org/t/openstack/build/4f849a4a55b54f6ea9a94db1a9784e54 : SUCCESS in 38m 53s\n- openstack-helm-infra-openstack-support https://zuul.opendev.org/t/openstack/build/0aa4af2fbca94739b53cb749c2591c4e : SUCCESS in 45m 08s\n- openstack-helm-infra-openstack-support-ssl https://zuul.opendev.org/t/openstack/build/b5bfa7676c084fc3a49e6b5fe8fa17cd : SUCCESS in 44m 31s\n- openstack-helm-infra-metacontroller https://zuul.opendev.org/t/openstack/build/964aed74acd8480c824f6b74b19d3f28 : SUCCESS in 17m 57s","accounts_in_message":[],"_revision_number":24},{"id":"e7716b298ad7766452c8425ad1920dd69a1721d1","author":{"_account_id":18511,"name":"Brian Wickersham","email":"bkw86@bellsouth.net","username":"bw6938"},"date":"2021-03-09 22:11:35.000000000","message":"Patch Set 24:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":24},{"id":"b2a1e2e27184d421c5c15322153717a4ce686a17","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":18511,"name":"Brian Wickersham","email":"bkw86@bellsouth.net","username":"bw6938"},"date":"2021-03-09 22:11:47.000000000","message":"Patch Set 25: Patch Set 24 was rebased","accounts_in_message":[],"_revision_number":25},{"id":"ace551e353e58f86c208dd965b37c97d431d1961","author":{"_account_id":32190,"name":"Frank Ritchie","email":"fr801x@att.com","username":"fr801x"},"date":"2021-03-09 22:18:31.000000000","message":"Patch Set 25: Code-Review+1","accounts_in_message":[],"_revision_number":25},{"id":"2f201457c2216531a76f37110735eaa71a0da7d3","author":{"_account_id":22636,"name":"Cliff Parsons","email":"cliffhparsons@aol.com","username":"cliffparsons"},"date":"2021-03-09 22:53:57.000000000","message":"Patch Set 25: Code-Review+1","accounts_in_message":[],"_revision_number":25},{"id":"2928fb47ce863fdbc594860ae34ce792fbe19a76","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-09 23:06:47.000000000","message":"Patch Set 25: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9996656ce43a40f9b77837c97432d02f : SUCCESS in 4m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4c044d88283b40888ed276fcf8cad0ff : SUCCESS in 4m 16s\n- openstack-helm-lint https://zuul.opendev.org/t/openstack/build/1844c9cee32e40c9bbf25816d9430448 : SUCCESS in 4m 04s\n- openstack-helm-infra-aio-logging https://zuul.opendev.org/t/openstack/build/7a9b34783a024899a7a3936525775003 : SUCCESS in 35m 38s\n- openstack-helm-infra-aio-monitoring https://zuul.opendev.org/t/openstack/build/977b480d04bd4a43ac42252918965db9 : SUCCESS in 27m 26s\n- openstack-helm-infra-openstack-support https://zuul.opendev.org/t/openstack/build/583ec969aa13412b8fd47a68bd0f5289 : SUCCESS in 47m 52s\n- openstack-helm-infra-openstack-support-ssl https://zuul.opendev.org/t/openstack/build/5127a41dc95d4146b99646b34e5f8827 : SUCCESS in 42m 58s\n- openstack-helm-infra-metacontroller https://zuul.opendev.org/t/openstack/build/2f704f7e7c45475fb9768e9c7c1fd71f : SUCCESS in 13m 31s","accounts_in_message":[],"_revision_number":25},{"id":"9901d28e5c00b068dac5c2898b78e23bddcd9f4f","author":{"_account_id":30449,"name":"jinyuanliu","email":"liujinyuan@inspur.com","username":"jinyuanliu"},"date":"2021-03-10 00:38:54.000000000","message":"Patch Set 25: Code-Review+1","accounts_in_message":[],"_revision_number":25},{"id":"cb3120ba193e927c5d6878810c68215fd5ca8889","author":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"date":"2021-03-10 14:32:22.000000000","message":"Patch Set 25: Code-Review+1","accounts_in_message":[],"_revision_number":25},{"id":"95837716ef46157d2b11af0edaeed1db929867d6","author":{"_account_id":28372,"name":"chinasubbareddy mallavarapu","email":"cr3938@att.com","username":"chinasubbareddy"},"date":"2021-03-10 14:33:37.000000000","message":"Patch Set 25: Code-Review+2","accounts_in_message":[],"_revision_number":25},{"id":"b070f7a8e0ba3ea89835f22d5e9281c4f2a97e08","author":{"_account_id":24780,"name":"Sangeet Gupta","email":"sg774j@att.com","username":"sgupta"},"date":"2021-03-10 14:40:31.000000000","message":"Patch Set 25: Code-Review+1","accounts_in_message":[],"_revision_number":25},{"id":"b38230f94a982ca1212e1809b1a677350c2ad752","author":{"_account_id":18250,"name":"Roy Tang","email":"roy.s.tang@att.com","username":"rt7380"},"date":"2021-03-10 15:05:53.000000000","message":"Patch Set 25: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":25},{"id":"0546b6a4a29a3e1b618c8f2ef1b03f0155f79460","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-10 15:06:05.000000000","message":"Patch Set 25: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":25},{"id":"67c9458c4306a00faf7b634f481002a2a4b523d2","author":{"_account_id":18511,"name":"Brian Wickersham","email":"bkw86@bellsouth.net","username":"bw6938"},"date":"2021-03-10 16:01:40.000000000","message":"Patch Set 25:\n\nrecheck","accounts_in_message":[],"_revision_number":25},{"id":"ecea70ecb7c54991ee9f9d0c12fb883a40484594","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-10 16:04:19.000000000","message":"Patch Set 25: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7efd1581c4524b8a9f61c5f0b01ecca0 : SUCCESS in 4m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a8a35dc13e4a4879af8beaf0a64aa6bb : SUCCESS in 4m 09s\n- openstack-helm-lint https://zuul.opendev.org/t/openstack/build/bffe0bc6e975424d80f7b1961322bdef : SUCCESS in 4m 03s\n- openstack-helm-infra-aio-logging https://zuul.opendev.org/t/openstack/build/f8d9beb2806f44e9aeac44e45a2fe23a : SUCCESS in 37m 02s\n- openstack-helm-infra-aio-monitoring https://zuul.opendev.org/t/openstack/build/a3c69ccd26404db8806fe46da24299d7 : SUCCESS in 28m 16s\n- openstack-helm-infra-openstack-support https://zuul.opendev.org/t/openstack/build/b29a42d18b0940d0a8b80e9065670960 : FAILURE in 43m 11s\n- openstack-helm-infra-openstack-support-ssl https://zuul.opendev.org/t/openstack/build/a20d4aee446a40eaac1ae5c502247a48 : SUCCESS in 53m 12s","accounts_in_message":[],"_revision_number":25},{"id":"20f48042691695f964901669465f472fe19eb50e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-10 17:36:44.000000000","message":"Patch Set 25: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b3fef87034804c1d9315a175f22e4e29 : SUCCESS in 4m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6b364bfa559e451fa17274d7014747da : SUCCESS in 6m 12s\n- openstack-helm-lint https://zuul.opendev.org/t/openstack/build/71af710db2d24c21aee00bc7b952d2c7 : SUCCESS in 5m 43s\n- openstack-helm-infra-aio-logging https://zuul.opendev.org/t/openstack/build/7f138c7c94734aaba2a768ce5a7c394f : SUCCESS in 37m 29s\n- openstack-helm-infra-aio-monitoring https://zuul.opendev.org/t/openstack/build/97eb0fde2ee04768a2196cb737a5d198 : SUCCESS in 30m 41s\n- openstack-helm-infra-openstack-support https://zuul.opendev.org/t/openstack/build/01d09ce3289444f7ac63fc3f92a7b86c : SUCCESS in 1h 10m 51s\n- openstack-helm-infra-openstack-support-ssl https://zuul.opendev.org/t/openstack/build/04ddd85a2b354f0eaac931d84feb0a8c : SUCCESS in 48m 28s\n- openstack-helm-infra-metacontroller https://zuul.opendev.org/t/openstack/build/db319d06aa8f42d7804e4c4334f7f803 : SUCCESS in 14m 46s","accounts_in_message":[],"_revision_number":25},{"id":"4ea5402edc187fdb04a4b7090b65ddce4178f058","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-10 17:44:19.000000000","message":"Patch Set 25: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":25},{"id":"024889d60401fb3ffecb0ae6a125ca24ce698045","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-10 18:39:03.000000000","message":"Patch Set 25: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/50fafab17a8b4cb38bd64a9c5567417f : SUCCESS in 5m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/53ee751ec3c24b20b7f22c195e155415 : SUCCESS in 5m 00s\n- openstack-helm-lint https://zuul.opendev.org/t/openstack/build/98cfea6e338a4601a2246681c195fc56 : SUCCESS in 6m 39s\n- openstack-helm-infra-aio-logging https://zuul.opendev.org/t/openstack/build/c5d1cd6b2bce4673931e2104f84b335d : SUCCESS in 48m 01s\n- openstack-helm-infra-aio-monitoring https://zuul.opendev.org/t/openstack/build/45932c3aac50473b97120a7f37bd71be : SUCCESS in 31m 29s\n- openstack-helm-infra-openstack-support https://zuul.opendev.org/t/openstack/build/f30d09aca75945d0a2ef2537a8e6e792 : SUCCESS in 52m 26s\n- openstack-helm-infra-openstack-support-ssl https://zuul.opendev.org/t/openstack/build/ae8e386f13ad45828481bec46de6f31f : SUCCESS in 50m 06s","accounts_in_message":[],"_revision_number":25},{"id":"48105ae04655438d4f0e1ff90a902a5aac0a3ede","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-10 18:39:04.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":25},{"id":"de824bd796747ba57ffae8acb9778ca575aa3323","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-10 18:42:02.000000000","message":"Patch Set 25:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2cc3c0e454c9414ba8b49e54da7b520e : SUCCESS in 1m 52s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b7b1dd95793e477b9c74d71a08528a8f : SUCCESS in 1m 50s","accounts_in_message":[],"_revision_number":25}],"current_revision_number":25,"current_revision":"bb3ce70a1034b8815028468780438eddcca04119","revisions":{"ac3b6b56f85dc6929e35ba5d50e80d57a25e3cad":{"kind":"REWORK","_number":1,"created":"2021-02-26 03:59:17.000000000","uploader":{"_account_id":18511,"name":"Brian Wickersham","email":"bkw86@bellsouth.net","username":"bw6938"},"ref":"refs/changes/11/777711/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-helm-infra","ref":"refs/changes/11/777711/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/11/777711/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/11/777711/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/11/777711/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-helm-infra refs/changes/11/777711/1"}}},"commit":{"parents":[{"commit":"128c3f748aee3ae72cccca0ff17a71ac1ed5c2fb","subject":"Merge \"Revert \"[ceph-client] enhance logic to enable the autoscaler for Octopus\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/128c3f748aee3ae72cccca0ff17a71ac1ed5c2fb"}]}],"author":{"name":"bw6938","email":"brian.wickersham@att.com","date":"2021-02-26 03:43:06.000000000","tz":0},"committer":{"name":"bw6938","email":"brian.wickersham@att.com","date":"2021-02-26 03:56:00.000000000","tz":0},"subject":"[ceph-client] enhance logic to enable and disable the autoscaler","message":"[ceph-client] enhance logic to enable and disable the autoscaler\n\nThe autoscaler was introduced in the Nautilus release. This\nchange only sets the pg num values for a pool if the autoscaler\nis disabled or the Ceph release is earlier than Nautilus. Also,\nwhen pools are created with the autoscaler enabled a pg_num_min\nvalue specifies the minimum value of pg_num that the autoscaler\nwill target. That default was recently changed from 8 to 32\nwhich severely limits the number of pools in a small cluster per\nhttps://github.com/rook/rook/issues/5091. So, we are specifying\nthe pg_num_value as 1 instead of 32.\n\nChange-Id: Ie08fb367ec8b1803fcc6e8cd22dc8da43c90e5c4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/ac3b6b56f85dc6929e35ba5d50e80d57a25e3cad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/ac3b6b56f85dc6929e35ba5d50e80d57a25e3cad"}]},"branch":"refs/heads/master"},"b0c5c38264d93a16442077abb195ef16cf116e34":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-02-26 04:03:40.000000000","uploader":{"_account_id":18511,"name":"Brian Wickersham","email":"bkw86@bellsouth.net","username":"bw6938"},"ref":"refs/changes/11/777711/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-helm-infra","ref":"refs/changes/11/777711/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/11/777711/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/11/777711/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/11/777711/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-helm-infra refs/changes/11/777711/2"}}},"commit":{"parents":[{"commit":"128c3f748aee3ae72cccca0ff17a71ac1ed5c2fb","subject":"Merge \"Revert \"[ceph-client] enhance logic to enable the autoscaler for Octopus\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/128c3f748aee3ae72cccca0ff17a71ac1ed5c2fb"}]}],"author":{"name":"bw6938","email":"brian.wickersham@att.com","date":"2021-02-26 03:43:06.000000000","tz":0},"committer":{"name":"Brian Wickersham","email":"brian.wickersham@att.com","date":"2021-02-26 04:03:30.000000000","tz":0},"subject":"[ceph-client] enhance logic to enable and disable the autoscaler","message":"[ceph-client] enhance logic to enable and disable the autoscaler\n\nThe autoscaler was introduced in the Nautilus release. This\nchange only sets the pg_num values for a pool if the autoscaler\nis disabled or the Ceph release is earlier than Nautilus. Also,\nwhen pools are created with the autoscaler enabled a pg_num_min\nvalue specifies the minimum value of pg_num that the autoscaler\nwill target. That default was recently changed from 8 to 32\nwhich severely limits the number of pools in a small cluster per\nhttps://github.com/rook/rook/issues/5091. So, we are specifying\nthe pg_num_min value of 1 at pool creation to override the default\nof 32 that the autoscaler would target.\n\nChange-Id: Ie08fb367ec8b1803fcc6e8cd22dc8da43c90e5c4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/b0c5c38264d93a16442077abb195ef16cf116e34"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/b0c5c38264d93a16442077abb195ef16cf116e34"}]},"branch":"refs/heads/master","description":"Edit commit message"},"02a1c7de749502a8df774e2a8b93cc562f7b1251":{"kind":"REWORK","_number":3,"created":"2021-02-26 04:07:27.000000000","uploader":{"_account_id":18511,"name":"Brian Wickersham","email":"bkw86@bellsouth.net","username":"bw6938"},"ref":"refs/changes/11/777711/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-helm-infra","ref":"refs/changes/11/777711/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/11/777711/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/11/777711/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/11/777711/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-helm-infra refs/changes/11/777711/3"}}},"commit":{"parents":[{"commit":"128c3f748aee3ae72cccca0ff17a71ac1ed5c2fb","subject":"Merge \"Revert \"[ceph-client] enhance logic to enable the autoscaler for Octopus\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/128c3f748aee3ae72cccca0ff17a71ac1ed5c2fb"}]}],"author":{"name":"bw6938","email":"brian.wickersham@att.com","date":"2021-02-26 03:43:06.000000000","tz":0},"committer":{"name":"bw6938","email":"brian.wickersham@att.com","date":"2021-02-26 04:07:19.000000000","tz":0},"subject":"[ceph-client] enhance logic to enable and disable the autoscaler","message":"[ceph-client] enhance logic to enable and disable the autoscaler\n\nThe autoscaler was introduced in the Nautilus release. This\nchange only sets the pg num values for a pool if the autoscaler\nis disabled or the Ceph release is earlier than Nautilus. Also,\nwhen pools are created with the autoscaler enabled a pg_num_min\nvalue specifies the minimum value of pg_num that the autoscaler\nwill target. That default was recently changed from 8 to 32\nwhich severely limits the number of pools in a small cluster per\nhttps://github.com/rook/rook/issues/5091. So, we are specifying\nthe pg_num_value as 1 instead of 32.\n\nChange-Id: Ie08fb367ec8b1803fcc6e8cd22dc8da43c90e5c4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/02a1c7de749502a8df774e2a8b93cc562f7b1251"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/02a1c7de749502a8df774e2a8b93cc562f7b1251"}]},"branch":"refs/heads/master"},"d2652babfaeb497e68cc7c481b7b34fe61f9c35f":{"kind":"REWORK","_number":4,"created":"2021-02-26 04:11:11.000000000","uploader":{"_account_id":18511,"name":"Brian Wickersham","email":"bkw86@bellsouth.net","username":"bw6938"},"ref":"refs/changes/11/777711/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-helm-infra","ref":"refs/changes/11/777711/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/11/777711/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/11/777711/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/11/777711/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-helm-infra refs/changes/11/777711/4"}}},"commit":{"parents":[{"commit":"128c3f748aee3ae72cccca0ff17a71ac1ed5c2fb","subject":"Merge \"Revert \"[ceph-client] enhance logic to enable the autoscaler for Octopus\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/128c3f748aee3ae72cccca0ff17a71ac1ed5c2fb"}]}],"author":{"name":"bw6938","email":"brian.wickersham@att.com","date":"2021-02-26 03:43:06.000000000","tz":0},"committer":{"name":"bw6938","email":"brian.wickersham@att.com","date":"2021-02-26 04:11:04.000000000","tz":0},"subject":"[ceph-client] enhance logic to enable and disable the autoscaler","message":"[ceph-client] enhance logic to enable and disable the autoscaler\n\nThe autoscaler was introduced in the Nautilus release. This\nchange only sets the pg num values for a pool if the autoscaler\nis disabled or the Ceph release is earlier than Nautilus. Also,\nwhen pools are created with the autoscaler enabled a pg_num_min\nvalue specifies the minimum value of pg_num that the autoscaler\nwill target. That default was recently changed from 8 to 32\nwhich severely limits the number of pools in a small cluster per\nhttps://github.com/rook/rook/issues/5091. So, we are specifying\nthe pg_num_value as 1 instead of 32.\n\nChange-Id: Ie08fb367ec8b1803fcc6e8cd22dc8da43c90e5c4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/d2652babfaeb497e68cc7c481b7b34fe61f9c35f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/d2652babfaeb497e68cc7c481b7b34fe61f9c35f"}]},"branch":"refs/heads/master"},"c41739f39006ac22f5fb58528db7925f04505cca":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2021-02-26 04:23:25.000000000","uploader":{"_account_id":18511,"name":"Brian Wickersham","email":"bkw86@bellsouth.net","username":"bw6938"},"ref":"refs/changes/11/777711/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-helm-infra","ref":"refs/changes/11/777711/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/11/777711/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/11/777711/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/11/777711/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-helm-infra refs/changes/11/777711/5"}}},"commit":{"parents":[{"commit":"128c3f748aee3ae72cccca0ff17a71ac1ed5c2fb","subject":"Merge \"Revert \"[ceph-client] enhance logic to enable the autoscaler for Octopus\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/128c3f748aee3ae72cccca0ff17a71ac1ed5c2fb"}]}],"author":{"name":"bw6938","email":"brian.wickersham@att.com","date":"2021-02-26 03:43:06.000000000","tz":0},"committer":{"name":"Brian Wickersham","email":"brian.wickersham@att.com","date":"2021-02-26 04:23:14.000000000","tz":0},"subject":"[ceph-client] enhance logic to enable and disable the autoscaler","message":"[ceph-client] enhance logic to enable and disable the autoscaler\n\nThe autoscaler was introduced in the Nautilus release. This\nchange only sets the pg_num value for a pool if the autoscaler\nis disabled or the Ceph release is earlier than Nautilus. Also,\nwhen pools are created with the autoscaler enabled a pg_num_min\nvalue specifies the minimum value of pg_num that the autoscaler\nwill target. That default was recently changed from 8 to 32\nwhich severely limits the number of pools in a small cluster per\nhttps://github.com/rook/rook/issues/5091. So, we are specifying\nthe pg_num_min value of 1 at pool creation to override the default\nof 32 that the autoscaler would target.\n\nChange-Id: Ie08fb367ec8b1803fcc6e8cd22dc8da43c90e5c4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/c41739f39006ac22f5fb58528db7925f04505cca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/c41739f39006ac22f5fb58528db7925f04505cca"}]},"branch":"refs/heads/master","description":"Edit commit message"},"44e6627a4e0965dd2f9932fb841614d71142c037":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2021-02-26 04:43:35.000000000","uploader":{"_account_id":18511,"name":"Brian Wickersham","email":"bkw86@bellsouth.net","username":"bw6938"},"ref":"refs/changes/11/777711/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-helm-infra","ref":"refs/changes/11/777711/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/11/777711/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/11/777711/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/11/777711/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-helm-infra refs/changes/11/777711/6"}}},"commit":{"parents":[{"commit":"128c3f748aee3ae72cccca0ff17a71ac1ed5c2fb","subject":"Merge \"Revert \"[ceph-client] enhance logic to enable the autoscaler for Octopus\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/128c3f748aee3ae72cccca0ff17a71ac1ed5c2fb"}]}],"author":{"name":"bw6938","email":"brian.wickersham@att.com","date":"2021-02-26 03:43:06.000000000","tz":0},"committer":{"name":"bw6938","email":"brian.wickersham@att.com","date":"2021-02-26 04:33:41.000000000","tz":0},"subject":"[ceph-client] enhance logic to enable and disable the autoscaler","message":"[ceph-client] enhance logic to enable and disable the autoscaler\n\nThe autoscaler was introduced in the Nautilus release. This\nchange only sets the pg_num value for a pool if the autoscaler\nis disabled or the Ceph release is earlier than Nautilus.\n\nWhen pools are created with the autoscaler enabled, a pg_num_min\nvalue specifies the minimum value of pg_num that the autoscaler\nwill target. That default was recently changed from 8 to 32\nwhich severely limits the number of pools in a small cluster per\nhttps://github.com/rook/rook/issues/5091. This change overrides\nthe default pg_num_min value of 32 with a value of 1 using the\noptional --pg-num-min \u003cvalue\u003e argument at pool creation.\n\nChange-Id: Ie08fb367ec8b1803fcc6e8cd22dc8da43c90e5c4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/44e6627a4e0965dd2f9932fb841614d71142c037"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/44e6627a4e0965dd2f9932fb841614d71142c037"}]},"branch":"refs/heads/master"},"a05330a695edc99a55ae77f016e2c68f223d6f6a":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2021-02-26 04:44:36.000000000","uploader":{"_account_id":18511,"name":"Brian Wickersham","email":"bkw86@bellsouth.net","username":"bw6938"},"ref":"refs/changes/11/777711/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-helm-infra","ref":"refs/changes/11/777711/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/11/777711/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/11/777711/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/11/777711/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-helm-infra refs/changes/11/777711/7"}}},"commit":{"parents":[{"commit":"937c98476600d1b7ca2d8d9fc68b11a3f292efb8","subject":"Merge \"[ca-issuer]: Support different versions of cert-manager\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/937c98476600d1b7ca2d8d9fc68b11a3f292efb8"}]}],"author":{"name":"bw6938","email":"brian.wickersham@att.com","date":"2021-02-26 03:43:06.000000000","tz":0},"committer":{"name":"Brian Wickersham","email":"brian.wickersham@att.com","date":"2021-02-26 04:44:36.000000000","tz":0},"subject":"[ceph-client] enhance logic to enable and disable the autoscaler","message":"[ceph-client] enhance logic to enable and disable the autoscaler\n\nThe autoscaler was introduced in the Nautilus release. This\nchange only sets the pg_num value for a pool if the autoscaler\nis disabled or the Ceph release is earlier than Nautilus.\n\nWhen pools are created with the autoscaler enabled, a pg_num_min\nvalue specifies the minimum value of pg_num that the autoscaler\nwill target. That default was recently changed from 8 to 32\nwhich severely limits the number of pools in a small cluster per\nhttps://github.com/rook/rook/issues/5091. This change overrides\nthe default pg_num_min value of 32 with a value of 1 using the\noptional --pg-num-min \u003cvalue\u003e argument at pool creation.\n\nChange-Id: Ie08fb367ec8b1803fcc6e8cd22dc8da43c90e5c4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/a05330a695edc99a55ae77f016e2c68f223d6f6a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/a05330a695edc99a55ae77f016e2c68f223d6f6a"}]},"branch":"refs/heads/master","description":"Rebase"},"74b93319d75b73f61cf0f1cd5cdd329719df11a3":{"kind":"REWORK","_number":8,"created":"2021-02-26 14:45:08.000000000","uploader":{"_account_id":18511,"name":"Brian Wickersham","email":"bkw86@bellsouth.net","username":"bw6938"},"ref":"refs/changes/11/777711/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-helm-infra","ref":"refs/changes/11/777711/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/11/777711/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/11/777711/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/11/777711/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-helm-infra refs/changes/11/777711/8"}}},"commit":{"parents":[{"commit":"128c3f748aee3ae72cccca0ff17a71ac1ed5c2fb","subject":"Merge \"Revert \"[ceph-client] enhance logic to enable the autoscaler for Octopus\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/128c3f748aee3ae72cccca0ff17a71ac1ed5c2fb"}]}],"author":{"name":"bw6938","email":"brian.wickersham@att.com","date":"2021-02-26 03:43:06.000000000","tz":0},"committer":{"name":"bw6938","email":"brian.wickersham@att.com","date":"2021-02-26 14:45:01.000000000","tz":0},"subject":"[ceph-client] enhance logic to enable and disable the autoscaler","message":"[ceph-client] enhance logic to enable and disable the autoscaler\n\nThe autoscaler was introduced in the Nautilus release. This\nchange only sets the pg_num value for a pool if the autoscaler\nis disabled or the Ceph release is earlier than Nautilus.\n\nWhen pools are created with the autoscaler enabled, a pg_num_min\nvalue specifies the minimum value of pg_num that the autoscaler\nwill target. That default was recently changed from 8 to 32\nwhich severely limits the number of pools in a small cluster per\nhttps://github.com/rook/rook/issues/5091. This change overrides\nthe default pg_num_min value of 32 with a value of 1 using the\noptional --pg-num-min \u003cvalue\u003e argument at pool creation.\n\nChange-Id: Ie08fb367ec8b1803fcc6e8cd22dc8da43c90e5c4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/74b93319d75b73f61cf0f1cd5cdd329719df11a3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/74b93319d75b73f61cf0f1cd5cdd329719df11a3"}]},"branch":"refs/heads/master"},"4db180fd4c4f19c8e6b877f22620720394db8ada":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2021-02-26 14:45:42.000000000","uploader":{"_account_id":18511,"name":"Brian Wickersham","email":"bkw86@bellsouth.net","username":"bw6938"},"ref":"refs/changes/11/777711/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-helm-infra","ref":"refs/changes/11/777711/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/11/777711/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/11/777711/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/11/777711/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-helm-infra refs/changes/11/777711/9"}}},"commit":{"parents":[{"commit":"937c98476600d1b7ca2d8d9fc68b11a3f292efb8","subject":"Merge \"[ca-issuer]: Support different versions of cert-manager\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/937c98476600d1b7ca2d8d9fc68b11a3f292efb8"}]}],"author":{"name":"bw6938","email":"brian.wickersham@att.com","date":"2021-02-26 03:43:06.000000000","tz":0},"committer":{"name":"Brian Wickersham","email":"brian.wickersham@att.com","date":"2021-02-26 14:45:42.000000000","tz":0},"subject":"[ceph-client] enhance logic to enable and disable the autoscaler","message":"[ceph-client] enhance logic to enable and disable the autoscaler\n\nThe autoscaler was introduced in the Nautilus release. This\nchange only sets the pg_num value for a pool if the autoscaler\nis disabled or the Ceph release is earlier than Nautilus.\n\nWhen pools are created with the autoscaler enabled, a pg_num_min\nvalue specifies the minimum value of pg_num that the autoscaler\nwill target. That default was recently changed from 8 to 32\nwhich severely limits the number of pools in a small cluster per\nhttps://github.com/rook/rook/issues/5091. This change overrides\nthe default pg_num_min value of 32 with a value of 1 using the\noptional --pg-num-min \u003cvalue\u003e argument at pool creation.\n\nChange-Id: Ie08fb367ec8b1803fcc6e8cd22dc8da43c90e5c4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/4db180fd4c4f19c8e6b877f22620720394db8ada"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/4db180fd4c4f19c8e6b877f22620720394db8ada"}]},"branch":"refs/heads/master","description":"Rebase"},"d51afd38a71cea34cd33ee1717a8d102b5546337":{"kind":"REWORK","_number":10,"created":"2021-02-26 15:56:30.000000000","uploader":{"_account_id":18511,"name":"Brian Wickersham","email":"bkw86@bellsouth.net","username":"bw6938"},"ref":"refs/changes/11/777711/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-helm-infra","ref":"refs/changes/11/777711/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/11/777711/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/11/777711/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/11/777711/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-helm-infra refs/changes/11/777711/10"}}},"commit":{"parents":[{"commit":"128c3f748aee3ae72cccca0ff17a71ac1ed5c2fb","subject":"Merge \"Revert \"[ceph-client] enhance logic to enable the autoscaler for Octopus\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/128c3f748aee3ae72cccca0ff17a71ac1ed5c2fb"}]}],"author":{"name":"bw6938","email":"brian.wickersham@att.com","date":"2021-02-26 03:43:06.000000000","tz":0},"committer":{"name":"bw6938","email":"brian.wickersham@att.com","date":"2021-02-26 15:56:22.000000000","tz":0},"subject":"[ceph-client] enhance logic to enable and disable the autoscaler","message":"[ceph-client] enhance logic to enable and disable the autoscaler\n\nThe autoscaler was introduced in the Nautilus release. This\nchange only sets the pg_num value for a pool if the autoscaler\nis disabled or the Ceph release is earlier than Nautilus.\n\nWhen pools are created with the autoscaler enabled, a pg_num_min\nvalue specifies the minimum value of pg_num that the autoscaler\nwill target. That default was recently changed from 8 to 32\nwhich severely limits the number of pools in a small cluster per\nhttps://github.com/rook/rook/issues/5091. This change overrides\nthe default pg_num_min value of 32 with a value of 1 using the\noptional --pg-num-min \u003cvalue\u003e argument at pool creation.\n\nChange-Id: Ie08fb367ec8b1803fcc6e8cd22dc8da43c90e5c4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/d51afd38a71cea34cd33ee1717a8d102b5546337"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/d51afd38a71cea34cd33ee1717a8d102b5546337"}]},"branch":"refs/heads/master"},"a855a4f0d2c2d599c058ceae8c98c79be76d9546":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2021-02-26 15:58:12.000000000","uploader":{"_account_id":18511,"name":"Brian Wickersham","email":"bkw86@bellsouth.net","username":"bw6938"},"ref":"refs/changes/11/777711/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-helm-infra","ref":"refs/changes/11/777711/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/11/777711/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/11/777711/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/11/777711/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-helm-infra refs/changes/11/777711/11"}}},"commit":{"parents":[{"commit":"937c98476600d1b7ca2d8d9fc68b11a3f292efb8","subject":"Merge \"[ca-issuer]: Support different versions of cert-manager\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/937c98476600d1b7ca2d8d9fc68b11a3f292efb8"}]}],"author":{"name":"bw6938","email":"brian.wickersham@att.com","date":"2021-02-26 03:43:06.000000000","tz":0},"committer":{"name":"Brian Wickersham","email":"brian.wickersham@att.com","date":"2021-02-26 15:58:12.000000000","tz":0},"subject":"[ceph-client] enhance logic to enable and disable the autoscaler","message":"[ceph-client] enhance logic to enable and disable the autoscaler\n\nThe autoscaler was introduced in the Nautilus release. This\nchange only sets the pg_num value for a pool if the autoscaler\nis disabled or the Ceph release is earlier than Nautilus.\n\nWhen pools are created with the autoscaler enabled, a pg_num_min\nvalue specifies the minimum value of pg_num that the autoscaler\nwill target. That default was recently changed from 8 to 32\nwhich severely limits the number of pools in a small cluster per\nhttps://github.com/rook/rook/issues/5091. This change overrides\nthe default pg_num_min value of 32 with a value of 1 using the\noptional --pg-num-min \u003cvalue\u003e argument at pool creation.\n\nChange-Id: Ie08fb367ec8b1803fcc6e8cd22dc8da43c90e5c4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/a855a4f0d2c2d599c058ceae8c98c79be76d9546"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/a855a4f0d2c2d599c058ceae8c98c79be76d9546"}]},"branch":"refs/heads/master","description":"Rebase"},"0d4bcc9251ea2e48e58b1f379909c29ded731051":{"kind":"REWORK","_number":12,"created":"2021-02-26 16:16:34.000000000","uploader":{"_account_id":18511,"name":"Brian Wickersham","email":"bkw86@bellsouth.net","username":"bw6938"},"ref":"refs/changes/11/777711/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-helm-infra","ref":"refs/changes/11/777711/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/11/777711/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/11/777711/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/11/777711/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-helm-infra refs/changes/11/777711/12"}}},"commit":{"parents":[{"commit":"128c3f748aee3ae72cccca0ff17a71ac1ed5c2fb","subject":"Merge \"Revert \"[ceph-client] enhance logic to enable the autoscaler for Octopus\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/128c3f748aee3ae72cccca0ff17a71ac1ed5c2fb"}]}],"author":{"name":"bw6938","email":"brian.wickersham@att.com","date":"2021-02-26 03:43:06.000000000","tz":0},"committer":{"name":"bw6938","email":"brian.wickersham@att.com","date":"2021-02-26 16:16:27.000000000","tz":0},"subject":"[ceph-client] enhance logic to enable and disable the autoscaler","message":"[ceph-client] enhance logic to enable and disable the autoscaler\n\nThe autoscaler was introduced in the Nautilus release. This\nchange only sets the pg_num value for a pool if the autoscaler\nis disabled or the Ceph release is earlier than Nautilus.\n\nWhen pools are created with the autoscaler enabled, a pg_num_min\nvalue specifies the minimum value of pg_num that the autoscaler\nwill target. That default was recently changed from 8 to 32\nwhich severely limits the number of pools in a small cluster per\nhttps://github.com/rook/rook/issues/5091. This change overrides\nthe default pg_num_min value of 32 with a value of 1 using the\noptional --pg-num-min \u003cvalue\u003e argument at pool creation.\n\nChange-Id: Ie08fb367ec8b1803fcc6e8cd22dc8da43c90e5c4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/0d4bcc9251ea2e48e58b1f379909c29ded731051"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/0d4bcc9251ea2e48e58b1f379909c29ded731051"}]},"branch":"refs/heads/master"},"e2faaa411e9aec9715d20df81888ef432a785925":{"kind":"NO_CODE_CHANGE","_number":13,"created":"2021-02-26 16:31:37.000000000","uploader":{"_account_id":18511,"name":"Brian Wickersham","email":"bkw86@bellsouth.net","username":"bw6938"},"ref":"refs/changes/11/777711/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-helm-infra","ref":"refs/changes/11/777711/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/11/777711/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/11/777711/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/11/777711/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-helm-infra refs/changes/11/777711/13"}}},"commit":{"parents":[{"commit":"128c3f748aee3ae72cccca0ff17a71ac1ed5c2fb","subject":"Merge \"Revert \"[ceph-client] enhance logic to enable the autoscaler for Octopus\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/128c3f748aee3ae72cccca0ff17a71ac1ed5c2fb"}]}],"author":{"name":"bw6938","email":"brian.wickersham@att.com","date":"2021-02-26 03:43:06.000000000","tz":0},"committer":{"name":"bw6938","email":"brian.wickersham@att.com","date":"2021-02-26 16:25:56.000000000","tz":0},"subject":"[ceph-client] enhance logic to enable and disable the autoscaler","message":"[ceph-client] enhance logic to enable and disable the autoscaler\n\nThe autoscaler was introduced in the Nautilus release. This\nchange only sets the pg_num value for a pool if the autoscaler\nis disabled or the Ceph release is earlier than Nautilus.\n\nWhen pools are created with the autoscaler enabled, a pg_num_min\nvalue specifies the minimum value of pg_num that the autoscaler\nwill target. That default was recently changed from 8 to 32\nwhich severely limits the number of pools in a small cluster per\nhttps://github.com/rook/rook/issues/5091. This change overrides\nthe default pg_num_min value of 32 with a value of 8 (matching\nthe default pg_num value of 8) using the optional --pg-num-min\n\u003cvalue\u003e argument at pool creation and pg_num_min value for\nexisting pools.\n\nChange-Id: Ie08fb367ec8b1803fcc6e8cd22dc8da43c90e5c4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/e2faaa411e9aec9715d20df81888ef432a785925"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/e2faaa411e9aec9715d20df81888ef432a785925"}]},"branch":"refs/heads/master"},"479417f3a8cb3324e78a97722c99a60649ac5022":{"kind":"REWORK","_number":14,"created":"2021-02-26 19:15:51.000000000","uploader":{"_account_id":18511,"name":"Brian Wickersham","email":"bkw86@bellsouth.net","username":"bw6938"},"ref":"refs/changes/11/777711/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-helm-infra","ref":"refs/changes/11/777711/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/11/777711/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/11/777711/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/11/777711/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-helm-infra refs/changes/11/777711/14"}}},"commit":{"parents":[{"commit":"cf7d665e793349f331b3eb76aea5c6d07c2b033a","subject":"[ceph-client] Separate pool quotas from pg_num calculations","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/cf7d665e793349f331b3eb76aea5c6d07c2b033a"}]}],"author":{"name":"bw6938","email":"brian.wickersham@att.com","date":"2021-02-26 03:43:06.000000000","tz":0},"committer":{"name":"bw6938","email":"brian.wickersham@att.com","date":"2021-02-26 19:15:40.000000000","tz":0},"subject":"[ceph-client] enhance logic to enable and disable the autoscaler","message":"[ceph-client] enhance logic to enable and disable the autoscaler\n\nThe autoscaler was introduced in the Nautilus release. This\nchange only sets the pg_num value for a pool if the autoscaler\nis disabled or the Ceph release is earlier than Nautilus.\n\nWhen pools are created with the autoscaler enabled, a pg_num_min\nvalue specifies the minimum value of pg_num that the autoscaler\nwill target. That default was recently changed from 8 to 32\nwhich severely limits the number of pools in a small cluster per\nhttps://github.com/rook/rook/issues/5091. This change overrides\nthe default pg_num_min value of 32 with a value of 8 (matching\nthe default pg_num value of 8) using the optional --pg-num-min\n\u003cvalue\u003e argument at pool creation and pg_num_min value for\nexisting pools.\n\nChange-Id: Ie08fb367ec8b1803fcc6e8cd22dc8da43c90e5c4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/479417f3a8cb3324e78a97722c99a60649ac5022"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/479417f3a8cb3324e78a97722c99a60649ac5022"}]},"branch":"refs/heads/master"},"3643b531f2306ae26267d1e5e00953a5a32ccc8a":{"kind":"REWORK","_number":15,"created":"2021-03-01 01:34:40.000000000","uploader":{"_account_id":18511,"name":"Brian Wickersham","email":"bkw86@bellsouth.net","username":"bw6938"},"ref":"refs/changes/11/777711/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-helm-infra","ref":"refs/changes/11/777711/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/11/777711/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/11/777711/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/11/777711/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-helm-infra refs/changes/11/777711/15"}}},"commit":{"parents":[{"commit":"cf7d665e793349f331b3eb76aea5c6d07c2b033a","subject":"[ceph-client] Separate pool quotas from pg_num calculations","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/cf7d665e793349f331b3eb76aea5c6d07c2b033a"}]}],"author":{"name":"bw6938","email":"brian.wickersham@att.com","date":"2021-02-26 03:43:06.000000000","tz":0},"committer":{"name":"bw6938","email":"brian.wickersham@att.com","date":"2021-03-01 01:29:39.000000000","tz":0},"subject":"[ceph-client] enhance logic to enable and disable the autoscaler","message":"[ceph-client] enhance logic to enable and disable the autoscaler\n\nThe autoscaler was introduced in the Nautilus release. This\nchange only sets the pg_num value for a pool if the autoscaler\nis disabled or the Ceph release is earlier than Nautilus.\n\nWhen pools are created with the autoscaler enabled, a pg_num_min\nvalue specifies the minimum value of pg_num that the autoscaler\nwill target. That default was recently changed from 8 to 32\nwhich severely limits the number of pools in a small cluster per\nhttps://github.com/rook/rook/issues/5091. This change overrides\nthe default pg_num_min value of 32 with a value of 8 (matching\nthe default pg_num value of 8) using the optional --pg-num-min\n\u003cvalue\u003e argument at pool creation and pg_num_min value for\nexisting pools.\n\nChange-Id: Ie08fb367ec8b1803fcc6e8cd22dc8da43c90e5c4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/3643b531f2306ae26267d1e5e00953a5a32ccc8a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/3643b531f2306ae26267d1e5e00953a5a32ccc8a"}]},"branch":"refs/heads/master"},"75b25dc3372170c7aa22c303b63b8a2239ed42a0":{"kind":"REWORK","_number":16,"created":"2021-03-01 17:47:07.000000000","uploader":{"_account_id":18511,"name":"Brian Wickersham","email":"bkw86@bellsouth.net","username":"bw6938"},"ref":"refs/changes/11/777711/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-helm-infra","ref":"refs/changes/11/777711/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/11/777711/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/11/777711/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/11/777711/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-helm-infra refs/changes/11/777711/16"}}},"commit":{"parents":[{"commit":"cf7d665e793349f331b3eb76aea5c6d07c2b033a","subject":"[ceph-client] Separate pool quotas from pg_num calculations","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/cf7d665e793349f331b3eb76aea5c6d07c2b033a"}]}],"author":{"name":"bw6938","email":"brian.wickersham@att.com","date":"2021-02-26 03:43:06.000000000","tz":0},"committer":{"name":"bw6938","email":"brian.wickersham@att.com","date":"2021-03-01 17:47:00.000000000","tz":0},"subject":"[ceph-client] enhance logic to enable and disable the autoscaler","message":"[ceph-client] enhance logic to enable and disable the autoscaler\n\nThe autoscaler was introduced in the Nautilus release. This\nchange only sets the pg_num value for a pool if the autoscaler\nis disabled or the Ceph release is earlier than Nautilus.\n\nWhen pools are created with the autoscaler enabled, a pg_num_min\nvalue specifies the minimum value of pg_num that the autoscaler\nwill target. That default was recently changed from 8 to 32\nwhich severely limits the number of pools in a small cluster per\nhttps://github.com/rook/rook/issues/5091. This change overrides\nthe default pg_num_min value of 32 with a value of 8 (matching\nthe default pg_num value of 8) using the optional --pg-num-min\n\u003cvalue\u003e argument at pool creation and pg_num_min value for\nexisting pools.\n\nChange-Id: Ie08fb367ec8b1803fcc6e8cd22dc8da43c90e5c4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/75b25dc3372170c7aa22c303b63b8a2239ed42a0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/75b25dc3372170c7aa22c303b63b8a2239ed42a0"}]},"branch":"refs/heads/master"},"61f21eef7e1b9cd123357d3dfa4e3e663eb64598":{"kind":"REWORK","_number":17,"created":"2021-03-01 18:59:42.000000000","uploader":{"_account_id":18511,"name":"Brian Wickersham","email":"bkw86@bellsouth.net","username":"bw6938"},"ref":"refs/changes/11/777711/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-helm-infra","ref":"refs/changes/11/777711/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/11/777711/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/11/777711/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/11/777711/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-helm-infra refs/changes/11/777711/17"}}},"commit":{"parents":[{"commit":"cf7d665e793349f331b3eb76aea5c6d07c2b033a","subject":"[ceph-client] Separate pool quotas from pg_num calculations","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/cf7d665e793349f331b3eb76aea5c6d07c2b033a"}]}],"author":{"name":"bw6938","email":"brian.wickersham@att.com","date":"2021-02-26 03:43:06.000000000","tz":0},"committer":{"name":"bw6938","email":"brian.wickersham@att.com","date":"2021-03-01 18:59:19.000000000","tz":0},"subject":"[ceph-client] enhance logic to enable and disable the autoscaler","message":"[ceph-client] enhance logic to enable and disable the autoscaler\n\nThe autoscaler was introduced in the Nautilus release. This\nchange only sets the pg_num value for a pool if the autoscaler\nis disabled or the Ceph release is earlier than Nautilus.\n\nWhen pools are created with the autoscaler enabled, a pg_num_min\nvalue specifies the minimum value of pg_num that the autoscaler\nwill target. That default was recently changed from 8 to 32\nwhich severely limits the number of pools in a small cluster per\nhttps://github.com/rook/rook/issues/5091. This change overrides\nthe default pg_num_min value of 32 with a value of 8 (matching\nthe default pg_num value of 8) using the optional --pg-num-min\n\u003cvalue\u003e argument at pool creation and pg_num_min value for\nexisting pools.\n\nChange-Id: Ie08fb367ec8b1803fcc6e8cd22dc8da43c90e5c4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/61f21eef7e1b9cd123357d3dfa4e3e663eb64598"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/61f21eef7e1b9cd123357d3dfa4e3e663eb64598"}]},"branch":"refs/heads/master"},"bb6deac25f23947e185adee56787798cb119126c":{"kind":"REWORK","_number":18,"created":"2021-03-01 20:47:27.000000000","uploader":{"_account_id":18511,"name":"Brian Wickersham","email":"bkw86@bellsouth.net","username":"bw6938"},"ref":"refs/changes/11/777711/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-helm-infra","ref":"refs/changes/11/777711/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/11/777711/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/11/777711/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/11/777711/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-helm-infra refs/changes/11/777711/18"}}},"commit":{"parents":[{"commit":"cf7d665e793349f331b3eb76aea5c6d07c2b033a","subject":"[ceph-client] Separate pool quotas from pg_num calculations","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/cf7d665e793349f331b3eb76aea5c6d07c2b033a"}]}],"author":{"name":"bw6938","email":"brian.wickersham@att.com","date":"2021-02-26 03:43:06.000000000","tz":0},"committer":{"name":"bw6938","email":"brian.wickersham@att.com","date":"2021-03-01 20:47:21.000000000","tz":0},"subject":"[ceph-client] enhance logic to enable and disable the autoscaler","message":"[ceph-client] enhance logic to enable and disable the autoscaler\n\nThe autoscaler was introduced in the Nautilus release. This\nchange only sets the pg_num value for a pool if the autoscaler\nis disabled or the Ceph release is earlier than Nautilus.\n\nWhen pools are created with the autoscaler enabled, a pg_num_min\nvalue specifies the minimum value of pg_num that the autoscaler\nwill target. That default was recently changed from 8 to 32\nwhich severely limits the number of pools in a small cluster per\nhttps://github.com/rook/rook/issues/5091. This change overrides\nthe default pg_num_min value of 32 with a value of 8 (matching\nthe default pg_num value of 8) using the optional --pg-num-min\n\u003cvalue\u003e argument at pool creation and pg_num_min value for\nexisting pools.\n\nChange-Id: Ie08fb367ec8b1803fcc6e8cd22dc8da43c90e5c4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/bb6deac25f23947e185adee56787798cb119126c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/bb6deac25f23947e185adee56787798cb119126c"}]},"branch":"refs/heads/master"},"0b09d6cf004cd40bc93ecedc0be82afe076a4e11":{"kind":"REWORK","_number":19,"created":"2021-03-02 03:00:18.000000000","uploader":{"_account_id":18511,"name":"Brian Wickersham","email":"bkw86@bellsouth.net","username":"bw6938"},"ref":"refs/changes/11/777711/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-helm-infra","ref":"refs/changes/11/777711/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/11/777711/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/11/777711/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/11/777711/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-helm-infra refs/changes/11/777711/19"}}},"commit":{"parents":[{"commit":"cf7d665e793349f331b3eb76aea5c6d07c2b033a","subject":"[ceph-client] Separate pool quotas from pg_num calculations","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/cf7d665e793349f331b3eb76aea5c6d07c2b033a"}]}],"author":{"name":"bw6938","email":"brian.wickersham@att.com","date":"2021-02-26 03:43:06.000000000","tz":0},"committer":{"name":"bw6938","email":"brian.wickersham@att.com","date":"2021-03-02 03:00:08.000000000","tz":0},"subject":"[ceph-client] enhance logic to enable and disable the autoscaler","message":"[ceph-client] enhance logic to enable and disable the autoscaler\n\nThe autoscaler was introduced in the Nautilus release. This\nchange only sets the pg_num value for a pool if the autoscaler\nis disabled or the Ceph release is earlier than Nautilus.\n\nWhen pools are created with the autoscaler enabled, a pg_num_min\nvalue specifies the minimum value of pg_num that the autoscaler\nwill target. That default was recently changed from 8 to 32\nwhich severely limits the number of pools in a small cluster per\nhttps://github.com/rook/rook/issues/5091. This change overrides\nthe default pg_num_min value of 32 with a value of 8 (matching\nthe default pg_num value of 8) using the optional --pg-num-min\n\u003cvalue\u003e argument at pool creation and pg_num_min value for\nexisting pools.\n\nChange-Id: Ie08fb367ec8b1803fcc6e8cd22dc8da43c90e5c4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/0b09d6cf004cd40bc93ecedc0be82afe076a4e11"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/0b09d6cf004cd40bc93ecedc0be82afe076a4e11"}]},"branch":"refs/heads/master"},"ba9f5c076376bcfeab1b2f4e7ae79ef91011deeb":{"kind":"REWORK","_number":20,"created":"2021-03-02 03:44:09.000000000","uploader":{"_account_id":18511,"name":"Brian Wickersham","email":"bkw86@bellsouth.net","username":"bw6938"},"ref":"refs/changes/11/777711/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-helm-infra","ref":"refs/changes/11/777711/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/11/777711/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/11/777711/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/11/777711/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-helm-infra refs/changes/11/777711/20"}}},"commit":{"parents":[{"commit":"cf7d665e793349f331b3eb76aea5c6d07c2b033a","subject":"[ceph-client] Separate pool quotas from pg_num calculations","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/cf7d665e793349f331b3eb76aea5c6d07c2b033a"}]}],"author":{"name":"bw6938","email":"brian.wickersham@att.com","date":"2021-02-26 03:43:06.000000000","tz":0},"committer":{"name":"bw6938","email":"brian.wickersham@att.com","date":"2021-03-02 03:44:02.000000000","tz":0},"subject":"[ceph-client] enhance logic to enable and disable the autoscaler","message":"[ceph-client] enhance logic to enable and disable the autoscaler\n\nThe autoscaler was introduced in the Nautilus release. This\nchange only sets the pg_num value for a pool if the autoscaler\nis disabled or the Ceph release is earlier than Nautilus.\n\nWhen pools are created with the autoscaler enabled, a pg_num_min\nvalue specifies the minimum value of pg_num that the autoscaler\nwill target. That default was recently changed from 8 to 32\nwhich severely limits the number of pools in a small cluster per\nhttps://github.com/rook/rook/issues/5091. This change overrides\nthe default pg_num_min value of 32 with a value of 8 (matching\nthe default pg_num value of 8) using the optional --pg-num-min\n\u003cvalue\u003e argument at pool creation and pg_num_min value for\nexisting pools.\n\nChange-Id: Ie08fb367ec8b1803fcc6e8cd22dc8da43c90e5c4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/ba9f5c076376bcfeab1b2f4e7ae79ef91011deeb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/ba9f5c076376bcfeab1b2f4e7ae79ef91011deeb"}]},"branch":"refs/heads/master"},"e1cca3a5c080c9896a948dbc27245c015ac8c89f":{"kind":"REWORK","_number":21,"created":"2021-03-02 19:59:05.000000000","uploader":{"_account_id":18511,"name":"Brian Wickersham","email":"bkw86@bellsouth.net","username":"bw6938"},"ref":"refs/changes/11/777711/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-helm-infra","ref":"refs/changes/11/777711/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/11/777711/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/11/777711/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/11/777711/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-helm-infra refs/changes/11/777711/21"}}},"commit":{"parents":[{"commit":"cf7d665e793349f331b3eb76aea5c6d07c2b033a","subject":"[ceph-client] Separate pool quotas from pg_num calculations","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/cf7d665e793349f331b3eb76aea5c6d07c2b033a"}]}],"author":{"name":"bw6938","email":"brian.wickersham@att.com","date":"2021-02-26 03:43:06.000000000","tz":0},"committer":{"name":"bw6938","email":"brian.wickersham@att.com","date":"2021-03-02 19:58:58.000000000","tz":0},"subject":"[ceph-client] enhance logic to enable and disable the autoscaler","message":"[ceph-client] enhance logic to enable and disable the autoscaler\n\nThe autoscaler was introduced in the Nautilus release. This\nchange only sets the pg_num value for a pool if the autoscaler\nis disabled or the Ceph release is earlier than Nautilus.\n\nWhen pools are created with the autoscaler enabled, a pg_num_min\nvalue specifies the minimum value of pg_num that the autoscaler\nwill target. That default was recently changed from 8 to 32\nwhich severely limits the number of pools in a small cluster per\nhttps://github.com/rook/rook/issues/5091. This change overrides\nthe default pg_num_min value of 32 with a value of 8 (matching\nthe default pg_num value of 8) using the optional --pg-num-min\n\u003cvalue\u003e argument at pool creation and pg_num_min value for\nexisting pools.\n\nChange-Id: Ie08fb367ec8b1803fcc6e8cd22dc8da43c90e5c4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/e1cca3a5c080c9896a948dbc27245c015ac8c89f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/e1cca3a5c080c9896a948dbc27245c015ac8c89f"}]},"branch":"refs/heads/master"},"1e6f28f2d9047315bd04ff97792eeadede95d12b":{"kind":"REWORK","_number":22,"created":"2021-03-02 20:39:34.000000000","uploader":{"_account_id":18511,"name":"Brian Wickersham","email":"bkw86@bellsouth.net","username":"bw6938"},"ref":"refs/changes/11/777711/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-helm-infra","ref":"refs/changes/11/777711/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/11/777711/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/11/777711/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/11/777711/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-helm-infra refs/changes/11/777711/22"}}},"commit":{"parents":[{"commit":"cf7d665e793349f331b3eb76aea5c6d07c2b033a","subject":"[ceph-client] Separate pool quotas from pg_num calculations","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/cf7d665e793349f331b3eb76aea5c6d07c2b033a"}]}],"author":{"name":"bw6938","email":"brian.wickersham@att.com","date":"2021-02-26 03:43:06.000000000","tz":0},"committer":{"name":"bw6938","email":"brian.wickersham@att.com","date":"2021-03-02 20:39:25.000000000","tz":0},"subject":"[ceph-client] enhance logic to enable and disable the autoscaler","message":"[ceph-client] enhance logic to enable and disable the autoscaler\n\nThe autoscaler was introduced in the Nautilus release. This\nchange only sets the pg_num value for a pool if the autoscaler\nis disabled or the Ceph release is earlier than Nautilus.\n\nWhen pools are created with the autoscaler enabled, a pg_num_min\nvalue specifies the minimum value of pg_num that the autoscaler\nwill target. That default was recently changed from 8 to 32\nwhich severely limits the number of pools in a small cluster per\nhttps://github.com/rook/rook/issues/5091. This change overrides\nthe default pg_num_min value of 32 with a value of 8 (matching\nthe default pg_num value of 8) using the optional --pg-num-min\n\u003cvalue\u003e argument at pool creation and pg_num_min value for\nexisting pools.\n\nChange-Id: Ie08fb367ec8b1803fcc6e8cd22dc8da43c90e5c4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/1e6f28f2d9047315bd04ff97792eeadede95d12b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/1e6f28f2d9047315bd04ff97792eeadede95d12b"}]},"branch":"refs/heads/master"},"05841c40ab701fbf2500847d4568fb16f1c0a7cb":{"kind":"REWORK","_number":23,"created":"2021-03-02 22:33:33.000000000","uploader":{"_account_id":18511,"name":"Brian Wickersham","email":"bkw86@bellsouth.net","username":"bw6938"},"ref":"refs/changes/11/777711/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-helm-infra","ref":"refs/changes/11/777711/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/11/777711/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/11/777711/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/11/777711/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-helm-infra refs/changes/11/777711/23"}}},"commit":{"parents":[{"commit":"cf7d665e793349f331b3eb76aea5c6d07c2b033a","subject":"[ceph-client] Separate pool quotas from pg_num calculations","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/cf7d665e793349f331b3eb76aea5c6d07c2b033a"}]}],"author":{"name":"bw6938","email":"brian.wickersham@att.com","date":"2021-02-26 03:43:06.000000000","tz":0},"committer":{"name":"bw6938","email":"brian.wickersham@att.com","date":"2021-03-02 22:33:23.000000000","tz":0},"subject":"[ceph-client] enhance logic to enable and disable the autoscaler","message":"[ceph-client] enhance logic to enable and disable the autoscaler\n\nThe autoscaler was introduced in the Nautilus release. This\nchange only sets the pg_num value for a pool if the autoscaler\nis disabled or the Ceph release is earlier than Nautilus.\n\nWhen pools are created with the autoscaler enabled, a pg_num_min\nvalue specifies the minimum value of pg_num that the autoscaler\nwill target. That default was recently changed from 8 to 32\nwhich severely limits the number of pools in a small cluster per\nhttps://github.com/rook/rook/issues/5091. This change overrides\nthe default pg_num_min value of 32 with a value of 8 (matching\nthe default pg_num value of 8) using the optional --pg-num-min\n\u003cvalue\u003e argument at pool creation and pg_num_min value for\nexisting pools.\n\nChange-Id: Ie08fb367ec8b1803fcc6e8cd22dc8da43c90e5c4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/05841c40ab701fbf2500847d4568fb16f1c0a7cb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/05841c40ab701fbf2500847d4568fb16f1c0a7cb"}]},"branch":"refs/heads/master"},"6bae3d0b4138a004d0798019ef8a5f634b1a2d35":{"kind":"REWORK","_number":24,"created":"2021-03-03 04:05:51.000000000","uploader":{"_account_id":18511,"name":"Brian Wickersham","email":"bkw86@bellsouth.net","username":"bw6938"},"ref":"refs/changes/11/777711/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-helm-infra","ref":"refs/changes/11/777711/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/11/777711/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/11/777711/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/11/777711/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-helm-infra refs/changes/11/777711/24"}}},"commit":{"parents":[{"commit":"cf7d665e793349f331b3eb76aea5c6d07c2b033a","subject":"[ceph-client] Separate pool quotas from pg_num calculations","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/cf7d665e793349f331b3eb76aea5c6d07c2b033a"}]}],"author":{"name":"bw6938","email":"brian.wickersham@att.com","date":"2021-02-26 03:43:06.000000000","tz":0},"committer":{"name":"bw6938","email":"brian.wickersham@att.com","date":"2021-03-03 04:05:39.000000000","tz":0},"subject":"[ceph-client] enhance logic to enable and disable the autoscaler","message":"[ceph-client] enhance logic to enable and disable the autoscaler\n\nThe autoscaler was introduced in the Nautilus release. This\nchange only sets the pg_num value for a pool if the autoscaler\nis disabled or the Ceph release is earlier than Nautilus.\n\nWhen pools are created with the autoscaler enabled, a pg_num_min\nvalue specifies the minimum value of pg_num that the autoscaler\nwill target. That default was recently changed from 8 to 32\nwhich severely limits the number of pools in a small cluster per\nhttps://github.com/rook/rook/issues/5091. This change overrides\nthe default pg_num_min value of 32 with a value of 8 (matching\nthe default pg_num value of 8) using the optional --pg-num-min\n\u003cvalue\u003e argument at pool creation and pg_num_min value for\nexisting pools.\n\nChange-Id: Ie08fb367ec8b1803fcc6e8cd22dc8da43c90e5c4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/6bae3d0b4138a004d0798019ef8a5f634b1a2d35"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/6bae3d0b4138a004d0798019ef8a5f634b1a2d35"}]},"branch":"refs/heads/master"},"bb3ce70a1034b8815028468780438eddcca04119":{"kind":"TRIVIAL_REBASE","_number":25,"created":"2021-03-09 22:11:47.000000000","uploader":{"_account_id":18511,"name":"Brian Wickersham","email":"bkw86@bellsouth.net","username":"bw6938"},"ref":"refs/changes/11/777711/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-helm-infra","ref":"refs/changes/11/777711/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/11/777711/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/11/777711/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-helm-infra refs/changes/11/777711/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-helm-infra refs/changes/11/777711/25"}}},"commit":{"parents":[{"commit":"cda359ef1fadaa8771c88d8aef01effa28ffe50b","subject":"[Ceph] Add Ceph CSI plugin","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/cda359ef1fadaa8771c88d8aef01effa28ffe50b"}]}],"author":{"name":"bw6938","email":"brian.wickersham@att.com","date":"2021-02-26 03:43:06.000000000","tz":0},"committer":{"name":"Brian Wickersham","email":"brian.wickersham@att.com","date":"2021-03-09 22:11:47.000000000","tz":0},"subject":"[ceph-client] enhance logic to enable and disable the autoscaler","message":"[ceph-client] enhance logic to enable and disable the autoscaler\n\nThe autoscaler was introduced in the Nautilus release. This\nchange only sets the pg_num value for a pool if the autoscaler\nis disabled or the Ceph release is earlier than Nautilus.\n\nWhen pools are created with the autoscaler enabled, a pg_num_min\nvalue specifies the minimum value of pg_num that the autoscaler\nwill target. That default was recently changed from 8 to 32\nwhich severely limits the number of pools in a small cluster per\nhttps://github.com/rook/rook/issues/5091. This change overrides\nthe default pg_num_min value of 32 with a value of 8 (matching\nthe default pg_num value of 8) using the optional --pg-num-min\n\u003cvalue\u003e argument at pool creation and pg_num_min value for\nexisting pools.\n\nChange-Id: Ie08fb367ec8b1803fcc6e8cd22dc8da43c90e5c4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/bb3ce70a1034b8815028468780438eddcca04119"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm-infra/commit/bb3ce70a1034b8815028468780438eddcca04119"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":18250,"name":"Roy Tang","email":"roy.s.tang@att.com","username":"rt7380"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":18250,"name":"Roy Tang","email":"roy.s.tang@att.com","username":"rt7380"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
