)]}'
{"id":"openstack%2Fopenstack-helm~1003567","triplet_id":"openstack%2Fopenstack-helm~master~If3f398e7a0a6493d2ebb97e195410311f750452e","project":"openstack/openstack-helm","branch":"master","attention_set":{},"removed_from_attention_set":{"29974":{"account":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"last_update":"2026-09-03 15:54:48.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If3f398e7a0a6493d2ebb97e195410311f750452e","subject":"[ceph] Generate aes256k keys for CephX authentication","status":"MERGED","created":"2026-09-02 16:34:02.000000000","updated":"2026-09-03 15:56:32.000000000","submitted":"2026-09-03 15:54:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1003567","meta_rev_id":"998bb6727bc15a4d9a774b4ddee0c0f0c9694367","_number":1003567,"virtual_id_number":1003567,"owner":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21040,"name":"Alexey","email":"aodinokov@mirantis.com","username":"aodinokov"},{"value":0,"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},{"value":0,"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},{"value":0,"_account_id":34667,"name":"Dustin Spears","email":"ds443n@att.com","username":"ds443n"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-03 15:54:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32190,"name":"Frank Ritchie","email":"fr801x@att.com","username":"fr801x"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"recommended":{"_account_id":21040,"name":"Alexey","email":"aodinokov@mirantis.com","username":"aodinokov"},"all":[{"value":1,"date":"2026-09-02 20:35:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21040,"name":"Alexey","email":"aodinokov@mirantis.com","username":"aodinokov"},{"value":2,"date":"2026-09-02 20:42:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},{"value":2,"date":"2026-09-03 14:09:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},{"value":1,"date":"2026-09-02 20:36:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":34667,"name":"Dustin Spears","email":"ds443n@att.com","username":"ds443n"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-09-02 20:44:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32190,"name":"Frank Ritchie","email":"fr801x@att.com","username":"fr801x"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},"all":[{"value":0,"_account_id":21040,"name":"Alexey","email":"aodinokov@mirantis.com","username":"aodinokov"},{"value":0,"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},{"value":1,"date":"2026-09-03 14:09:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},{"value":0,"_account_id":34667,"name":"Dustin Spears","email":"ds443n@att.com","username":"ds443n"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32190,"name":"Frank Ritchie","email":"fr801x@att.com","username":"fr801x"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},{"_account_id":21040,"name":"Alexey","email":"aodinokov@mirantis.com","username":"aodinokov"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32190,"name":"Frank Ritchie","email":"fr801x@att.com","username":"fr801x"},{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},{"_account_id":34667,"name":"Dustin Spears","email":"ds443n@att.com","username":"ds443n"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-02 20:24:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-02 20:35:29.000000000","updated_by":{"_account_id":21040,"name":"Alexey","email":"aodinokov@mirantis.com","username":"aodinokov"},"reviewer":{"_account_id":21040,"name":"Alexey","email":"aodinokov@mirantis.com","username":"aodinokov"},"state":"REVIEWER"},{"updated":"2026-09-02 20:36:11.000000000","updated_by":{"_account_id":34667,"name":"Dustin Spears","email":"ds443n@att.com","username":"ds443n"},"reviewer":{"_account_id":34667,"name":"Dustin Spears","email":"ds443n@att.com","username":"ds443n"},"state":"REVIEWER"},{"updated":"2026-09-02 20:42:41.000000000","updated_by":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"reviewer":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"state":"REVIEWER"},{"updated":"2026-09-02 20:44:23.000000000","updated_by":{"_account_id":32190,"name":"Frank Ritchie","email":"fr801x@att.com","username":"fr801x"},"reviewer":{"_account_id":32190,"name":"Frank Ritchie","email":"fr801x@att.com","username":"fr801x"},"state":"REVIEWER"},{"updated":"2026-09-03 14:09:19.000000000","updated_by":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},"reviewer":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},"state":"REVIEWER"}],"messages":[{"id":"e83d6119f0f6fc73c8126dbb09bf2db0cc1231e7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"date":"2026-09-02 16:34:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f5b28105e35a5ede658a07a74e537316ae10554a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"date":"2026-09-02 17:50:36.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"84340cbda3ce23bcb9ff5a6fb494aad5bec801b0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-02 20:24:13.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1e0765d952124042a88a714cd9a82df1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/03194354af554404b058d23ab77bcfe7 : SUCCESS in 5m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7ff96b0654bb4814a36795a2645a5aea : SUCCESS in 5m 03s\n- openstack-helm-linter https://zuul.opendev.org/t/openstack/build/516cbe013e3b4a159d0dc4a3c4ccd71b : SUCCESS in 3m 40s\n- openstack-helm-mariadb-crd-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/9d2ee5db3a674b8fa0ad0dccc3d2bfcd : SUCCESS in 1h 38m 37s\n- openstack-helm-mariadb-operator-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/8f373880f5fb47abaf8d592bcd8c2461 : SUCCESS in 1h 39m 09s\n- openstack-helm-rabbitmq-crd-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/d32011510e1f49919247fe2428eab424 : SUCCESS in 47m 11s\n- openstack-helm-pre-commit https://zuul.opendev.org/t/openstack/build/0308f4abd2f143bca0dafba6e12f0f95 : SUCCESS in 3m 24s\n- openstack-helm-bandit https://zuul.opendev.org/t/openstack/build/3f259f447b1f49f4a0646bfc54c331d1 : SUCCESS in 3m 09s\n- openstack-helm-build-charts https://zuul.opendev.org/t/openstack/build/cffc4b173b0d46768a5713093807b0a1 : SUCCESS in 3m 59s\n- openstack-helm-cinder-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/53d91665f024491a847565e8c9ed937c : SUCCESS in 46m 54s\n- openstack-helm-compute-kit-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/dfc72f9d6aea4e71a7a3b96d53ad54ee : SUCCESS in 1h 14m 38s\n- openstack-helm-cinder-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/dff2186cd6224414b124cc6255db85d0 : SUCCESS in 44m 43s\n- openstack-helm-compute-kit-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/997dbfd1cc47444bba262f8081bae9a7 : SUCCESS in 1h 22m 59s\n- openstack-helm-cinder-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/2e9f2df2d24c4898a225d9f94f14c6a4 : SUCCESS in 47m 14s\n- openstack-helm-compute-kit-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/3534ee7f6baa49fbae90b5a23e8a329f : SUCCESS in 1h 13m 34s\n- openstack-helm-tls-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/852b889d5b054cf59a1d8d34c3c355ad : SUCCESS in 1h 45m 44s\n- openstack-helm-octavia-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/562ab891485c477a8e3a42c4246ec466 : SUCCESS in 1h 15m 05s\n- openstack-helm-logging https://zuul.opendev.org/t/openstack/build/edbd1996e4b14af39c85c018e49658ac : SUCCESS in 39m 54s\n- openstack-helm-monitoring https://zuul.opendev.org/t/openstack/build/476b1d3ea62947a1bd22d9e40e908450 : SUCCESS in 36m 51s","accounts_in_message":[],"_revision_number":2},{"id":"a93abd3dbeca02d02244c3386a32694e6c22cca5","author":{"_account_id":21040,"name":"Alexey","email":"aodinokov@mirantis.com","username":"aodinokov"},"date":"2026-09-02 20:35:29.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"863199a72ed970d45f2ae4f5d0e632f6180ef0a9","author":{"_account_id":34667,"name":"Dustin Spears","email":"ds443n@att.com","username":"ds443n"},"date":"2026-09-02 20:36:11.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"953d7382ce7222c129e772ff0341d149e962dfa8","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2026-09-02 20:42:41.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"9cd1fa5e7f19985a97e81fe5f794d647cfa69198","author":{"_account_id":32190,"name":"Frank Ritchie","email":"fr801x@att.com","username":"fr801x"},"date":"2026-09-02 20:44:23.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"92a4bb474cfc2b657ce44bb22b24d3e9cba94471","author":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},"date":"2026-09-03 14:09:19.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"2b5c2b7b28de05fa97e3d2b4ec542ffa536d2eab","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-03 14:09:31.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"b0d4ecaf068b4f7fd689543ad549801026591202","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-03 15:54:48.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9d3eb627808c4862b13e180334016100\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/abc20e5b57684fa3bad82e586f0b11d3 : SUCCESS in 5m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/03780aba86744c6cb48d0eaadd594837 : SUCCESS in 5m 18s\n- openstack-helm-linter https://zuul.opendev.org/t/openstack/build/2a1bccd0b5634175b2c0346a0d671768 : SUCCESS in 2m 55s\n- openstack-helm-cinder-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/9d3812bd987e4b2babbe89f46ab0cc0a : SUCCESS in 46m 06s\n- openstack-helm-compute-kit-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/0ea79864a324460996efe19664c6421e : SUCCESS in 1h 18m 18s\n- openstack-helm-logging https://zuul.opendev.org/t/openstack/build/e4a572a3739f4936a75358cff209855b : SUCCESS in 25m 24s\n- openstack-helm-monitoring https://zuul.opendev.org/t/openstack/build/a06dec72e2d0412789ff4071d5a2edd5 : SUCCESS in 20m 23s","accounts_in_message":[],"_revision_number":2},{"id":"63221827d8b4ac261ab8fa4542211b531d2b6754","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-03 15:54:48.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"998bb6727bc15a4d9a774b4ddee0c0f0c9694367","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-03 15:56:32.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/95716c109ca04ae0a07895aeb7e967f8\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/78a7d1d183e24adf85372a238982b0c5 : SUCCESS in 52s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c91ccef7cb074e54915193ba945a86a2 : SUCCESS in 59s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"83868b8a4bd2124de98523b4021cb729543fb1de","revisions":{"a00f63202ba8a464c42fb140fa71e2a79d7ef377":{"kind":"REWORK","_number":1,"created":"2026-09-02 16:34:02.000000000","uploader":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"ref":"refs/changes/67/1003567/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-helm","ref":"refs/changes/67/1003567/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-helm refs/changes/67/1003567/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-helm refs/changes/67/1003567/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-helm refs/changes/67/1003567/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-helm refs/changes/67/1003567/1"}}},"commit":{"parents":[{"commit":"bcf34e9d32fec5251c2100cbe1a7ebee59fc1f2b","subject":"Merge \"[ceph] Update Ceph to Tentacle 20.2.4 and Rook to 1.20.6\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm/commit/bcf34e9d32fec5251c2100cbe1a7ebee59fc1f2b"}]}],"author":{"name":"Stephen Taylor","email":"stephen.taylor.1@att.com","date":"2026-09-02 16:23:33.000000000","tz":-360},"committer":{"name":"Stephen Taylor","email":"stephen.taylor.1@att.com","date":"2026-09-02 16:34:01.000000000","tz":-360},"subject":"[ceph] Generate aes256k keys for CephX authentication","message":"[ceph] Generate aes256k keys for CephX authentication\n\nThe bootstrap keyring generator Python tool is removed and replaced\nwith ceph-authtool (already present in the ceph-daemon image) for\nCephX key generation in the keyring manager jobs. Key types are\nspecified as aes256k for compatibility with the secure key\nrequirement in Tentacle 20.2.4.\n\nChange-Id: If3f398e7a0a6493d2ebb97e195410311f750452e\nSigned-off-by: Stephen Taylor \u003cstephen.taylor.1@att.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm/commit/a00f63202ba8a464c42fb140fa71e2a79d7ef377"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm/commit/a00f63202ba8a464c42fb140fa71e2a79d7ef377"}]},"branch":"refs/heads/master"},"83868b8a4bd2124de98523b4021cb729543fb1de":{"kind":"REWORK","_number":2,"created":"2026-09-02 17:50:36.000000000","uploader":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"ref":"refs/changes/67/1003567/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-helm","ref":"refs/changes/67/1003567/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-helm refs/changes/67/1003567/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-helm refs/changes/67/1003567/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-helm refs/changes/67/1003567/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-helm refs/changes/67/1003567/2"}}},"commit":{"parents":[{"commit":"bcf34e9d32fec5251c2100cbe1a7ebee59fc1f2b","subject":"Merge \"[ceph] Update Ceph to Tentacle 20.2.4 and Rook to 1.20.6\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm/commit/bcf34e9d32fec5251c2100cbe1a7ebee59fc1f2b"}]}],"author":{"name":"Stephen Taylor","email":"stephen.taylor.1@att.com","date":"2026-09-02 16:23:33.000000000","tz":-360},"committer":{"name":"Stephen Taylor","email":"stephen.taylor.1@att.com","date":"2026-09-02 17:50:31.000000000","tz":-360},"subject":"[ceph] Generate aes256k keys for CephX authentication","message":"[ceph] Generate aes256k keys for CephX authentication\n\nThe bootstrap keyring generator Python tool is removed and replaced\nwith ceph-authtool (already present in the ceph-daemon image) for\nCephX key generation in the keyring manager jobs. Key types are\nspecified as aes256k for compatibility with the secure key\nrequirement in Tentacle 20.2.4.\n\nChange-Id: If3f398e7a0a6493d2ebb97e195410311f750452e\nSigned-off-by: Stephen Taylor \u003cstephen.taylor.1@att.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm/commit/83868b8a4bd2124de98523b4021cb729543fb1de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm/commit/83868b8a4bd2124de98523b4021cb729543fb1de"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
