)]}'
{"id":"openstack%2Fopenstack-helm~1003608","triplet_id":"openstack%2Fopenstack-helm~master~Id9837f86d53e0d61e0612a17577d1cb7559687c6","project":"openstack/openstack-helm","branch":"master","attention_set":{},"removed_from_attention_set":{"29974":{"account":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"last_update":"2026-09-03 18:00:36.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id9837f86d53e0d61e0612a17577d1cb7559687c6","subject":"[ceph] Update Ceph image references to Noble","status":"MERGED","created":"2026-09-02 20:57:27.000000000","updated":"2026-09-03 18:01:57.000000000","submitted":"2026-09-03 18:00:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1003608","meta_rev_id":"7787541f8e36f3ade033122506493b3ab3ae1953","_number":1003608,"virtual_id_number":1003608,"owner":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21040,"name":"Alexey","email":"aodinokov@mirantis.com","username":"aodinokov"},{"value":0,"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},{"value":0,"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},{"value":0,"_account_id":34667,"name":"Dustin Spears","email":"ds443n@att.com","username":"ds443n"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-03 18:00:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"recommended":{"_account_id":21040,"name":"Alexey","email":"aodinokov@mirantis.com","username":"aodinokov"},"all":[{"value":1,"date":"2026-09-03 14:04:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21040,"name":"Alexey","email":"aodinokov@mirantis.com","username":"aodinokov"},{"value":2,"date":"2026-09-03 16:07:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},{"value":2,"date":"2026-09-03 14:09:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},{"value":1,"date":"2026-09-03 14:04:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":34667,"name":"Dustin Spears","email":"ds443n@att.com","username":"ds443n"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"all":[{"value":0,"_account_id":21040,"name":"Alexey","email":"aodinokov@mirantis.com","username":"aodinokov"},{"value":1,"date":"2026-09-03 16:07:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},{"value":0,"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},{"value":0,"_account_id":34667,"name":"Dustin Spears","email":"ds443n@att.com","username":"ds443n"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},{"_account_id":21040,"name":"Alexey","email":"aodinokov@mirantis.com","username":"aodinokov"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},{"_account_id":34667,"name":"Dustin Spears","email":"ds443n@att.com","username":"ds443n"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-02 22:48:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-03 14:04:08.000000000","updated_by":{"_account_id":34667,"name":"Dustin Spears","email":"ds443n@att.com","username":"ds443n"},"reviewer":{"_account_id":34667,"name":"Dustin Spears","email":"ds443n@att.com","username":"ds443n"},"state":"REVIEWER"},{"updated":"2026-09-03 14:04:50.000000000","updated_by":{"_account_id":21040,"name":"Alexey","email":"aodinokov@mirantis.com","username":"aodinokov"},"reviewer":{"_account_id":21040,"name":"Alexey","email":"aodinokov@mirantis.com","username":"aodinokov"},"state":"REVIEWER"},{"updated":"2026-09-03 14:09:37.000000000","updated_by":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},"reviewer":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},"state":"REVIEWER"},{"updated":"2026-09-03 16:07:17.000000000","updated_by":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"reviewer":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"state":"REVIEWER"}],"messages":[{"id":"89d3398bbb622163a73b3a7ebcff5eaae9a225ae","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"date":"2026-09-02 20:57:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"69c3501f212d83fdd17192b25e7a939582e12b33","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-02 22:48:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3ae601288add44568aae133bac5f5024\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/766bd26e31e648fc825d96d20d87e822 : SUCCESS in 5m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e18037fac7ae4de5b3380fad6e48caea : SUCCESS in 4m 34s\n- openstack-helm-linter https://zuul.opendev.org/t/openstack/build/1636047772e8452295c333702fc32b13 : SUCCESS in 3m 50s\n- openstack-helm-mariadb-crd-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/4252c646bc0548928ebca5803fc29ac5 : SUCCESS in 1h 38m 25s\n- openstack-helm-mariadb-operator-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/c4efaa11619c476ea7fdb8482226838f : SUCCESS in 1h 03m 05s\n- openstack-helm-rabbitmq-crd-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/d5c8bc2b518247e1a8c8e0559bd6fac3 : SUCCESS in 1h 16m 56s\n- openstack-helm-rabbitmq-operator-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/26f3c6a44a9641c084f2c6a0a60b02cc : SUCCESS in 1h 13m 25s\n- openstack-helm-pre-commit https://zuul.opendev.org/t/openstack/build/210eba7c7f5e4e49b794c888ec53ab3a : SUCCESS in 3m 08s\n- openstack-helm-build-charts https://zuul.opendev.org/t/openstack/build/45f7d66133a84b9786e97fa35c931c96 : SUCCESS in 9m 14s\n- openstack-helm-cinder-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/48cdfeaa24254645bdc9aa28d2a957d3 : SUCCESS in 43m 47s\n- openstack-helm-compute-kit-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/34abab65894349d9a1ea524dc62d531f : SUCCESS in 1h 15m 42s\n- openstack-helm-cinder-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/04e88be3ffda4f7fa31deb53d79965f3 : SUCCESS in 43m 59s\n- openstack-helm-compute-kit-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/ae27c099dc674184b4eefa8f515b4076 : SUCCESS in 48m 42s\n- openstack-helm-cinder-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/bf75ded9fb1545dc976cfd3c00f3ecfd : SUCCESS in 49m 10s\n- openstack-helm-compute-kit-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/0ab64e0838fb45e199a1b71dedf178ad : SUCCESS in 50m 01s\n- openstack-helm-tls-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/18c837abade0459c9977b86c0a26e0fe : SUCCESS in 1h 46m 59s\n- openstack-helm-compute-kit-ovn-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/2a82b9f7457246b08152096c9a378074 : SUCCESS in 44m 36s\n- openstack-helm-compute-kit-dpdk-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/02f57a71f6e642b5814c7ddc77f529d7 : SUCCESS in 1h 05m 03s\n- openstack-helm-octavia-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/380b602624894e95bb6dfff8a265a344 : SUCCESS in 1h 24m 31s\n- openstack-helm-logging https://zuul.opendev.org/t/openstack/build/1b703b5d63f64f33b8bd0e8459be6216 : SUCCESS in 43m 18s\n- openstack-helm-monitoring https://zuul.opendev.org/t/openstack/build/8118f660794a48858b64b850493893ea : SUCCESS in 29m 51s","accounts_in_message":[],"_revision_number":1},{"id":"69f91fdd321fc2953c8577940e9483ca2f90cc86","author":{"_account_id":34667,"name":"Dustin Spears","email":"ds443n@att.com","username":"ds443n"},"date":"2026-09-03 14:04:08.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"e9fdd9c8bb6850d539772653ddb8e3b042c8715b","author":{"_account_id":21040,"name":"Alexey","email":"aodinokov@mirantis.com","username":"aodinokov"},"date":"2026-09-03 14:04:50.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"f7cf9463c86fa462d4319cf7d0122c9fdd256ee6","author":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},"date":"2026-09-03 14:09:37.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"1801eeae95d73e6fdf93ec22e53c3f38b28b5284","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2026-09-03 16:07:17.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5cac0ed996a3aac7d92f05718705af5dcd76ca20","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-03 16:07:29.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5afc7e944048d61c2ead49e39febdf7c8e43a6d0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-03 18:00:36.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a9086dbab7164c38a219a671e0241248\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0000ef61afcc455b865ca9a599f61e7a : SUCCESS in 2m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8cfe100248f448a7a039a45fcee10faf : SUCCESS in 3m 18s\n- openstack-helm-linter https://zuul.opendev.org/t/openstack/build/88a82bbcee514da1a43da800a7570a47 : SUCCESS in 3m 56s\n- openstack-helm-cinder-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/fd535481b9754eeda0a12c4b307c2d86 : SUCCESS in 35m 55s\n- openstack-helm-compute-kit-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/f477bbe616dc440eb2fceebb3857ba86 : SUCCESS in 1h 10m 37s\n- openstack-helm-logging https://zuul.opendev.org/t/openstack/build/2e1d9bca714846d3885d6be20ec1b04b : SUCCESS in 25m 54s\n- openstack-helm-monitoring https://zuul.opendev.org/t/openstack/build/dc7e64c8ef7148f4b4c65197e40be535 : SUCCESS in 37m 08s","accounts_in_message":[],"_revision_number":1},{"id":"a2eb05f9a922cef8e44a2b1f15a30fdc021d0745","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-03 18:00:36.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"7787541f8e36f3ade033122506493b3ab3ae1953","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-03 18:01:57.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/721d9f8687c347f980905af8c2f49067\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a258c511e212477ebd663d9092c70f3c : SUCCESS in 52s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/341fdeb920ab443986e9594d56a965a3 : SUCCESS in 53s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a34a2f9b9e590dfeb45998c25f4eb78272497e30","revisions":{"a34a2f9b9e590dfeb45998c25f4eb78272497e30":{"kind":"REWORK","_number":1,"created":"2026-09-02 20:57:27.000000000","uploader":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"ref":"refs/changes/08/1003608/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-helm","ref":"refs/changes/08/1003608/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-helm refs/changes/08/1003608/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-helm refs/changes/08/1003608/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-helm refs/changes/08/1003608/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-helm refs/changes/08/1003608/1"}}},"commit":{"parents":[{"commit":"0ddd9fe771963b00609d678d9d813f9bbeeacd7d","subject":"Merge \"Reconcile the RabbitMQ resources with the upstream operator\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm/commit/0ddd9fe771963b00609d678d9d813f9bbeeacd7d"}]}],"author":{"name":"Stephen Taylor","email":"stephen.taylor.1@att.com","date":"2026-09-02 20:56:10.000000000","tz":-360},"committer":{"name":"Stephen Taylor","email":"stephen.taylor.1@att.com","date":"2026-09-02 20:56:10.000000000","tz":-360},"subject":"[ceph] Update Ceph image references to Noble","message":"[ceph] Update Ceph image references to Noble\n\nAll Ceph image references are updated from Jammy to Noble. The\nCeph version remains unchanged at 20.2.4.\n\nChange-Id: Id9837f86d53e0d61e0612a17577d1cb7559687c6\nSigned-off-by: Stephen Taylor \u003cstephen.taylor.1@att.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm/commit/a34a2f9b9e590dfeb45998c25f4eb78272497e30"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm/commit/a34a2f9b9e590dfeb45998c25f4eb78272497e30"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":34667,"name":"Dustin Spears","email":"ds443n@att.com","username":"ds443n"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
