)]}'
{"id":"openstack%2Fopenstack-helm~1003988","triplet_id":"openstack%2Fopenstack-helm~master~I9dbc23782a82120fb702768fb853a970d7b363a3","project":"openstack/openstack-helm","branch":"master","attention_set":{},"removed_from_attention_set":{"21040":{"account":{"_account_id":21040,"name":"Alexey","email":"aodinokov@mirantis.com","username":"aodinokov"},"last_update":"2026-09-11 18:28:08.000000000","reason":"Change was submitted"},"3009":{"account":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"last_update":"2026-09-10 22:42:34.000000000","reason":"\u003cGERRIT_ACCOUNT_3009\u003e replied on the change","reason_account":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"}},"5890":{"account":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"last_update":"2026-09-11 18:28:08.000000000","reason":"Change was submitted"},"35943":{"account":{"_account_id":35943,"name":"Aleksei Chekunov","display_name":"Aleksei Chekunov","email":"chekunov87@gmail.com","username":"az7961"},"last_update":"2026-09-11 13:16:24.000000000","reason":"\u003cGERRIT_ACCOUNT_35943\u003e replied on the change","reason_account":{"_account_id":35943,"name":"Aleksei Chekunov","display_name":"Aleksei Chekunov","email":"chekunov87@gmail.com","username":"az7961"}},"34520":{"account":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},"last_update":"2026-09-10 23:21:23.000000000","reason":"\u003cGERRIT_ACCOUNT_34520\u003e replied on the change","reason_account":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"}}},"hashtags":[],"change_id":"I9dbc23782a82120fb702768fb853a970d7b363a3","subject":"Fixing permanent secrets for keystone","status":"MERGED","created":"2026-09-04 14:21:06.000000000","updated":"2026-09-11 18:29:15.000000000","submitted":"2026-09-11 18:28:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":14,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"1003988","meta_rev_id":"9675bfda5172c749cab1e11fa9744374d379b0d9","_number":1003988,"virtual_id_number":1003988,"owner":{"_account_id":21040,"name":"Alexey","email":"aodinokov@mirantis.com","username":"aodinokov"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},{"value":0,"_account_id":35943,"name":"Aleksei Chekunov","display_name":"Aleksei Chekunov","email":"chekunov87@gmail.com","username":"az7961"},{"value":0,"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-11 18:28:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"all":[{"value":2,"date":"2026-09-11 16:27:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},{"value":1,"date":"2026-09-11 14:29:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":35943,"name":"Aleksei Chekunov","display_name":"Aleksei Chekunov","email":"chekunov87@gmail.com","username":"az7961"},{"value":2,"date":"2026-09-11 14:29:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"all":[{"value":1,"date":"2026-09-11 17:05:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},{"value":0,"_account_id":35943,"name":"Aleksei Chekunov","display_name":"Aleksei Chekunov","email":"chekunov87@gmail.com","username":"az7961"},{"value":0,"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"}],"REVIEWER":[{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},{"_account_id":35943,"name":"Aleksei Chekunov","display_name":"Aleksei Chekunov","email":"chekunov87@gmail.com","username":"az7961"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-04 16:26:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-04 19:57:04.000000000","updated_by":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"reviewer":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"state":"CC"},{"updated":"2026-09-10 22:17:53.000000000","updated_by":{"_account_id":35943,"name":"Aleksei Chekunov","display_name":"Aleksei Chekunov","email":"chekunov87@gmail.com","username":"az7961"},"reviewer":{"_account_id":35943,"name":"Aleksei Chekunov","display_name":"Aleksei Chekunov","email":"chekunov87@gmail.com","username":"az7961"},"state":"REVIEWER"},{"updated":"2026-09-10 22:19:53.000000000","updated_by":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},"reviewer":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},"state":"REVIEWER"},{"updated":"2026-09-10 22:35:35.000000000","updated_by":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"reviewer":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"state":"CC"},{"updated":"2026-09-10 22:37:35.000000000","updated_by":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"reviewer":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"state":"REVIEWER"}],"messages":[{"id":"5c6ec5a5e5fb3d720cbccf74d682f2086b66ed36","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":21040,"name":"Alexey","email":"aodinokov@mirantis.com","username":"aodinokov"},"date":"2026-09-04 14:21:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7fac99a919360afe1bfc5259ce9a8458db632388","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":21040,"name":"Alexey","email":"aodinokov@mirantis.com","username":"aodinokov"},"date":"2026-09-04 14:21:37.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"45210fc795c99974b4006f8e002acf4f9837bb8c","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":21040,"name":"Alexey","email":"aodinokov@mirantis.com","username":"aodinokov"},"date":"2026-09-04 14:28:04.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"234fd01e716b6b5669182c26c9514870a52fb042","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-04 16:26:27.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/daea2d0726714065b30719788db2fc0d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c982f44b33842f2924439b286587e2c : SUCCESS in 3m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5a07d5cb14d24c81bf1e0e59a6235f86 : SUCCESS in 2m 11s\n- openstack-helm-linter https://zuul.opendev.org/t/openstack/build/c16dea5668c34495953b55c766b42603 : SUCCESS in 4m 22s\n- openstack-helm-mariadb-crd-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/29d030da6759483f851ad2daab7972bb : SUCCESS in 1h 42m 59s\n- openstack-helm-mariadb-operator-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/ac02bf0cc8584a8088ef7c8aac2836a4 : SUCCESS in 1h 23m 57s\n- openstack-helm-rabbitmq-crd-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/b80105a8cf9540509d1b23ece6a6bc3d : SUCCESS in 1h 15m 40s\n- openstack-helm-rabbitmq-operator-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/2e6a84c96e7b422699357a7c0769615a : FAILURE in 1h 43m 46s\n- openstack-helm-pre-commit https://zuul.opendev.org/t/openstack/build/3eacc4995e7b4d578bfd7dee5d0842a5 : SUCCESS in 2m 00s\n- openstack-helm-build-charts https://zuul.opendev.org/t/openstack/build/9eedef79cea641c89d46b0e6cb343987 : SUCCESS in 2m 47s\n- openstack-helm-cinder-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/41348b1bc2704118958c930b10715bab : SUCCESS in 50m 45s\n- openstack-helm-compute-kit-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/31c67190f3ef446ab26f7554fbf6b0b4 : SUCCESS in 1h 10m 55s\n- openstack-helm-cinder-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/f168f1d886be4839b3b99630d5a34a08 : SUCCESS in 44m 22s\n- openstack-helm-compute-kit-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/c9e50bf1c8dc4208af4ef5a59c7674a8 : SUCCESS in 1h 21m 57s\n- openstack-helm-cinder-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/a00b10fc2e2c4c60b319f700de48fe54 : SUCCESS in 38m 42s\n- openstack-helm-compute-kit-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/d9f9090da74441198c5565f4405ce173 : SUCCESS in 1h 23m 43s\n- openstack-helm-tls-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/ee0c3e79727a4a57a0e5fe85a9652155 : SUCCESS in 1h 49m 09s\n- openstack-helm-octavia-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/42c120441c7a4f4a9623c2d6cb6449fc : SUCCESS in 1h 14m 00s\n- openstack-helm-logging https://zuul.opendev.org/t/openstack/build/04e234a0f9d04ab9b06bd28a982d4a42 : SUCCESS in 38m 46s\n- openstack-helm-monitoring https://zuul.opendev.org/t/openstack/build/b9176e015af14d51804d6c83436df327 : SUCCESS in 37m 45s","accounts_in_message":[],"_revision_number":3},{"id":"56cc00e466271a5d132c48f29cea3ac78be6d350","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":21040,"name":"Alexey","email":"aodinokov@mirantis.com","username":"aodinokov"},"date":"2026-09-04 19:03:08.000000000","message":"Patch Set 4: Patch Set 3 was rebased\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"d0c82a7e24c42d0b08365267049401f7945b2f55","author":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"date":"2026-09-04 19:57:04.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"2776dd3a13134e15c7b7b36e13d5de61e4c2d48a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-04 21:15:39.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0676abcc331a4905a14e4caab968fe48\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/86149a88730f4abe9004174f807fc523 : SUCCESS in 3m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/aaf67239c8aa4642a1ec5ed7d0f46b2e : SUCCESS in 3m 37s\n- openstack-helm-linter https://zuul.opendev.org/t/openstack/build/0246910d43ba4237af98f3a1d1b13e61 : SUCCESS in 3m 28s\n- openstack-helm-mariadb-crd-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/7bd562b284544dda8dacc8efcdce8c19 : SUCCESS in 58m 24s\n- openstack-helm-mariadb-operator-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/8e9e6b4edb564f688c38d7ab8c12637f : SUCCESS in 59m 02s\n- openstack-helm-rabbitmq-crd-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/4e9ba6b0cbf44386afa3408a84d60138 : SUCCESS in 1h 18m 35s\n- openstack-helm-rabbitmq-operator-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/e6e0abe11cf94e549ffba6c5718a2bbb : SUCCESS in 1h 14m 49s\n- openstack-helm-pre-commit https://zuul.opendev.org/t/openstack/build/27909d3ad4c346cf9382b96d8bc283ef : SUCCESS in 3m 10s\n- openstack-helm-build-charts https://zuul.opendev.org/t/openstack/build/6f97a217cea64567be09f7f83c237528 : SUCCESS in 2m 37s\n- openstack-helm-cinder-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/43f515264f2149b9b2f56dc20522fc85 : SUCCESS in 38m 03s\n- openstack-helm-compute-kit-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/671ebac006184821b77ae0dff44249db : SUCCESS in 50m 45s\n- openstack-helm-cinder-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/d89e685a0ed84e97b400cdb0c49e9271 : SUCCESS in 41m 10s\n- openstack-helm-compute-kit-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/d476275af9464d398bbc423b02c8abb6 : SUCCESS in 45m 46s\n- openstack-helm-cinder-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/40b0273dd00d464697221aaedc78c77c : SUCCESS in 33m 41s\n- openstack-helm-compute-kit-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/d140a2b7559349f9943db679374f4a4e : SUCCESS in 1h 09m 24s\n- openstack-helm-tls-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/939009a6dfb449189d5b7aab10fc0cd3 : SUCCESS in 1h 55m 18s\n- openstack-helm-octavia-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/7a98131acc724235a652185cbccc4f7e : SUCCESS in 1h 24m 54s\n- openstack-helm-logging https://zuul.opendev.org/t/openstack/build/7e31673559e54ba9a50e7362b4ceb0a8 : SUCCESS in 36m 50s\n- openstack-helm-monitoring https://zuul.opendev.org/t/openstack/build/e26799a90b7143da86b835f5e6d4e559 : SUCCESS in 28m 09s","accounts_in_message":[],"_revision_number":4},{"id":"1ce259c11c6a77fe40ea8c83dc6ee26b7d4fc317","author":{"_account_id":21040,"name":"Alexey","email":"aodinokov@mirantis.com","username":"aodinokov"},"date":"2026-09-05 02:06:53.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"6295dde37e6593f9e113b67bd4faf3f8d70b1689","author":{"_account_id":21040,"name":"Alexey","email":"aodinokov@mirantis.com","username":"aodinokov"},"date":"2026-09-05 03:19:42.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"e7245e8426901f018903d60dcefa4a12e46064eb","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":21040,"name":"Alexey","email":"aodinokov@mirantis.com","username":"aodinokov"},"date":"2026-09-05 03:20:07.000000000","message":"Patch Set 5: Published edit on patch set 4.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":5},{"id":"732e398570d901261a57cc4ab1c0ae1a6947519d","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":21040,"name":"Alexey","email":"aodinokov@mirantis.com","username":"aodinokov"},"date":"2026-09-05 03:20:37.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"da01520d3e16fc96b504b36fe9f2027778194387","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-05 05:11:05.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cfd22289c9e246db8be0bee964a46b92\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/58606d11afd14e279693079d4db91fb7 : SUCCESS in 5m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8c3c8fa253234ea991be55c5616976a2 : SUCCESS in 3m 51s\n- openstack-helm-linter https://zuul.opendev.org/t/openstack/build/6beaa7a380314a1ca929b610ea55f31c : SUCCESS in 2m 00s\n- openstack-helm-mariadb-crd-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/9a7ea09a7246441f86e87435dddbe5dc : SUCCESS in 1h 44m 20s\n- openstack-helm-mariadb-operator-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/2ae6115f8ea24b1fbb21b5b247afe0eb : SUCCESS in 1h 24m 32s\n- openstack-helm-rabbitmq-crd-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/a68243a2f0fb4f19be8c455d9a7d6ede : SUCCESS in 53m 24s\n- openstack-helm-rabbitmq-operator-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/6d6e33ca18c148dfb2e881dd89d378ac : SUCCESS in 52m 51s\n- openstack-helm-pre-commit https://zuul.opendev.org/t/openstack/build/991e700da9594b50b51a34299ea0ea46 : SUCCESS in 2m 06s\n- openstack-helm-bandit https://zuul.opendev.org/t/openstack/build/9cb241ef51cc4cb4986c281c0aca84fb : SUCCESS in 3m 13s\n- openstack-helm-build-charts https://zuul.opendev.org/t/openstack/build/72b62df454694bec9ebea76922c939d5 : SUCCESS in 4m 17s\n- openstack-helm-cinder-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/7d79ceabc8b54ac6b5a6e425659ea69a : SUCCESS in 41m 49s\n- openstack-helm-compute-kit-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/399b1a2a5e024c69b793be27f3489ab6 : SUCCESS in 52m 34s\n- openstack-helm-cinder-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/0162d271a4f94ae299d305a486fc3349 : SUCCESS in 32m 39s\n- openstack-helm-compute-kit-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/3d179c3b86c94ef6b7d8b92c67be86d7 : SUCCESS in 1h 21m 11s\n- openstack-helm-cinder-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/72e7b13d96294ad0aec7f949e210ef2f : SUCCESS in 43m 57s\n- openstack-helm-compute-kit-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/f1650bb28a1f44b1b9f4e139b13ee9a1 : SUCCESS in 1h 07m 04s\n- openstack-helm-tls-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/ae934cc92b2d4c8ea2feea083d1170e1 : SUCCESS in 1h 04m 05s\n- openstack-helm-octavia-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/f6314313c7c2436e982dbe03e83b1f7e : SUCCESS in 1h 21m 00s\n- openstack-helm-logging https://zuul.opendev.org/t/openstack/build/0fb900ccf07e437fbdd94dea381d54f9 : SUCCESS in 43m 47s\n- openstack-helm-monitoring https://zuul.opendev.org/t/openstack/build/8ea56158f33a448b9901d3d665ce34d4 : SUCCESS in 24m 52s","accounts_in_message":[],"_revision_number":6},{"id":"35cf0e29481617d772f46f05695c0aedc4224bb5","author":{"_account_id":21040,"name":"Alexey","email":"aodinokov@mirantis.com","username":"aodinokov"},"date":"2026-09-08 22:56:49.000000000","message":"Patch Set 6:\n\n(1 comment)\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":6},{"id":"0b8a013471fce81131fa39134101b70829ce751b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":21040,"name":"Alexey","email":"aodinokov@mirantis.com","username":"aodinokov"},"date":"2026-09-09 20:00:07.000000000","message":"Patch Set 7: Published edit on patch set 6.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":7},{"id":"2069b8d4d46c5a9f91408d36e2ab826eedb7af4a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-09 22:37:08.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9a6b4186f7604c729e784fef8a481e92\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f1d9c08009ba4d83a2282eb16ca1a133 : SUCCESS in 4m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9d1d5c0050c642c0bdd67d396c7039bb : SUCCESS in 3m 23s\n- openstack-helm-linter https://zuul.opendev.org/t/openstack/build/4e7b6c5d076749859098da59386e12ea : SUCCESS in 4m 18s\n- openstack-helm-mariadb-crd-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/17333522f1154fc5b9be82f2f7dd9bd2 : FAILURE in 2h 31m 24s\n- openstack-helm-mariadb-operator-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/af150fba1deb40bebba2b1567811abc8 : FAILURE in 2h 29m 44s\n- openstack-helm-rabbitmq-crd-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/8e41face8e2742ac89938f8972ed62d4 : SUCCESS in 1h 16m 57s\n- openstack-helm-rabbitmq-operator-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/8fff4b61fca5447e953ce96ee2723793 : SUCCESS in 1h 11m 29s\n- openstack-helm-pre-commit https://zuul.opendev.org/t/openstack/build/394a3c6c9a5f4942a00269a0d095184b : SUCCESS in 3m 05s\n- openstack-helm-bandit https://zuul.opendev.org/t/openstack/build/8a892498638d4c9da6d3d6ba5489b730 : SUCCESS in 4m 26s\n- openstack-helm-build-charts https://zuul.opendev.org/t/openstack/build/218b07826c2c494fbb79f81a491e0bab : SUCCESS in 4m 00s\n- openstack-helm-cinder-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/cd8a81d3bf3d425790b9d75ecc3cee10 : SUCCESS in 47m 27s\n- openstack-helm-compute-kit-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/72f70087197247b7a53bc148d02fb0c1 : SUCCESS in 1h 07m 19s\n- openstack-helm-cinder-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/fccf98d08f3b40ddb02b7cb5deb293e8 : SUCCESS in 44m 37s\n- openstack-helm-compute-kit-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/e1cee00261a64056ae5a0c410072e876 : FAILURE in 49m 12s\n- openstack-helm-cinder-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/5d890f8ef3974772bef1d003b5a56c25 : SUCCESS in 29m 11s\n- openstack-helm-compute-kit-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/13deb1b32aff4aeba818f4387a549406 : SUCCESS in 1h 09m 57s\n- openstack-helm-tls-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/702fd1361e8a4c12885d788e21be4b23 : FAILURE in 2h 03m 27s\n- openstack-helm-octavia-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/f635869f84204188bc0c242d63a26472 : SUCCESS in 1h 15m 32s\n- openstack-helm-logging https://zuul.opendev.org/t/openstack/build/631de16ee417424091b81e78a08ba5c8 : SUCCESS in 39m 31s\n- openstack-helm-monitoring https://zuul.opendev.org/t/openstack/build/69bb27963bb445f0adfab96d5a8d09cc : SUCCESS in 30m 24s","accounts_in_message":[],"_revision_number":7},{"id":"e99f8059366f5706d3d8cf40991820409ef571de","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":21040,"name":"Alexey","email":"aodinokov@mirantis.com","username":"aodinokov"},"date":"2026-09-09 22:41:17.000000000","message":"Patch Set 8: Published edit on patch set 7.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":8},{"id":"c8dd85fc0d360a06a0eca75dbc25f70c754a4157","author":{"_account_id":21040,"name":"Alexey","email":"aodinokov@mirantis.com","username":"aodinokov"},"date":"2026-09-09 22:42:13.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"e588ae97971c8682c360010ff9d152c014fd8695","author":{"_account_id":21040,"name":"Alexey","email":"aodinokov@mirantis.com","username":"aodinokov"},"date":"2026-09-09 22:42:30.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"e5a95b9eda13f9514330c8963c9fcca9cf7e5c71","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":21040,"name":"Alexey","email":"aodinokov@mirantis.com","username":"aodinokov"},"date":"2026-09-09 23:07:11.000000000","message":"Patch Set 9: Published edit on patch set 8.","accounts_in_message":[],"_revision_number":9},{"id":"30d3e1cddb0a492bd7497413433843d6677f47e1","author":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"date":"2026-09-09 23:28:48.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"acc2060657071eea933500b5b9700e016572495b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-10 01:07:03.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0e3ce60092834f32bcd8af55777a9117\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/13de330fab734b6f9ce78fe8744d589f : SUCCESS in 3m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/01bfa50251c14d0cb97d7c414829dbb5 : SUCCESS in 4m 29s\n- openstack-helm-linter https://zuul.opendev.org/t/openstack/build/6eb4ac46f84847ae825fbf21dbfe40ef : SUCCESS in 2m 16s\n- openstack-helm-mariadb-crd-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/ecf30ef7579e4c87a12ef6d31b8b3def : SUCCESS in 1h 14m 41s\n- openstack-helm-mariadb-operator-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/8f2f0371a8f745508a4f358886981dad : SUCCESS in 1h 33m 41s\n- openstack-helm-rabbitmq-crd-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/feb45e1c5aec4e0f84ed633f907db5ca : SUCCESS in 1h 15m 16s\n- openstack-helm-rabbitmq-operator-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/61772fa72b9b40efaafdb42c16ebecdf : SUCCESS in 1h 21m 46s\n- openstack-helm-pre-commit https://zuul.opendev.org/t/openstack/build/d8b31af9f9ef423dae6f52535b160f58 : SUCCESS in 2m 12s\n- openstack-helm-bandit https://zuul.opendev.org/t/openstack/build/d08dac9d1adb4ded98bf46d26bd58f86 : SUCCESS in 2m 50s\n- openstack-helm-build-charts https://zuul.opendev.org/t/openstack/build/783f3e7d80da4ed8b18e1cb9a34fadb1 : SUCCESS in 2m 24s\n- openstack-helm-cinder-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/ae6fbbfbe8f842e98ae900034756548f : SUCCESS in 32m 09s\n- openstack-helm-compute-kit-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/6adbd672f3ff4f1f9afc2ba004fea0ba : SUCCESS in 1h 01m 33s\n- openstack-helm-cinder-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/f3f2846e226b42e8aef5186aaef42864 : SUCCESS in 39m 27s\n- openstack-helm-compute-kit-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/cf14ddc601dc4180a2840b178b00d456 : SUCCESS in 46m 57s\n- openstack-helm-cinder-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/5c1e250a369c4adf8ef292b7bd42ba0e : FAILURE in 17m 36s\n- openstack-helm-compute-kit-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/280c097d0f554ec385cdeae0db6c7abb : SUCCESS in 47m 06s\n- openstack-helm-tls-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/e83a7a8943d94529b08a66a8017b0bf9 : SUCCESS in 1h 22m 33s\n- openstack-helm-octavia-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/4faa6958ff924a9d931d9a66a7697ad0 : SUCCESS in 45m 51s\n- openstack-helm-logging https://zuul.opendev.org/t/openstack/build/8c8ac74b0e0240908a86b43aa45d7447 : SUCCESS in 37m 39s\n- openstack-helm-monitoring https://zuul.opendev.org/t/openstack/build/c9d9e1ab8f6d4310817fad0bc8449d01 : SUCCESS in 33m 21s","accounts_in_message":[],"_revision_number":9},{"id":"d47a5cd39fe5383a36a1637c26da8c0e6b2ec22e","author":{"_account_id":21040,"name":"Alexey","email":"aodinokov@mirantis.com","username":"aodinokov"},"date":"2026-09-10 03:15:24.000000000","message":"Patch Set 9:\n\n(2 comments)","accounts_in_message":[],"_revision_number":9},{"id":"d90de7b23f784ef90bba7c7010c0f463f964427d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":21040,"name":"Alexey","email":"aodinokov@mirantis.com","username":"aodinokov"},"date":"2026-09-10 03:16:41.000000000","message":"Patch Set 10: Commit message was updated.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":10},{"id":"0dab8d56f510a52bd57023d68652a5572ddfb0bf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-10 04:49:54.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8403215d667d47619d6ffa012cacdba6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4bfad0b4bb634612b644e28e28710092 : SUCCESS in 3m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cc2d699a21d244c98d9ad76edb82bf75 : SUCCESS in 4m 41s\n- openstack-helm-linter https://zuul.opendev.org/t/openstack/build/9897a3a077cd423aa50e8c3df3d6f160 : SUCCESS in 3m 29s\n- openstack-helm-mariadb-crd-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/56b733703f21421383f17958bb9727c2 : SUCCESS in 59m 38s\n- openstack-helm-mariadb-operator-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/2f0c06db518441c2aeebe51775a410d5 : SUCCESS in 1h 20m 12s\n- openstack-helm-rabbitmq-crd-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/215e7dd6654247b6a3f9b36285f8f313 : SUCCESS in 1h 17m 18s\n- openstack-helm-rabbitmq-operator-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/f0991aef86f040dda0b7c01cd29a9220 : SUCCESS in 1h 22m 17s\n- openstack-helm-pre-commit https://zuul.opendev.org/t/openstack/build/e73062a949274dc6aa1ef109f362aae4 : SUCCESS in 1m 58s\n- openstack-helm-bandit https://zuul.opendev.org/t/openstack/build/44161552574e4a2db63a0e2af6fa6adb : SUCCESS in 3m 07s\n- openstack-helm-build-charts https://zuul.opendev.org/t/openstack/build/24bd8ff3cac147d39b22503e3985bf3a : SUCCESS in 1m 52s\n- openstack-helm-cinder-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/39d4e38e03c9473e880fcce6a0236549 : SUCCESS in 25m 59s\n- openstack-helm-compute-kit-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/87dbff717df742629eaab5861a76c12f : SUCCESS in 47m 36s\n- openstack-helm-cinder-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/486d24f96fca46be80cbf94207da13c5 : SUCCESS in 24m 45s\n- openstack-helm-compute-kit-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/2e2545bdff7d4687b147a16f1ace22e9 : SUCCESS in 1h 15m 23s\n- openstack-helm-cinder-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/06d714cc3bc14186aa58783d3ef53882 : SUCCESS in 36m 06s\n- openstack-helm-compute-kit-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/7830d765cca749de925aada5ebbc5be4 : SUCCESS in 1h 16m 49s\n- openstack-helm-tls-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/b93797a972bc4c73a03e4471a1d521af : SUCCESS in 1h 00m 22s\n- openstack-helm-octavia-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/03716b753fd3443ba75947db9f4b4e0f : SUCCESS in 1h 17m 14s\n- openstack-helm-logging https://zuul.opendev.org/t/openstack/build/426d1985409d47ca96e49a7e30c3cc59 : SUCCESS in 33m 29s\n- openstack-helm-monitoring https://zuul.opendev.org/t/openstack/build/c4b30411146a4e9dbdbdaf71a0f1eeb2 : SUCCESS in 21m 02s","accounts_in_message":[],"_revision_number":10},{"id":"80473424dcfbcf585ef7a61bb398ca721006ba3d","author":{"_account_id":35943,"name":"Aleksei Chekunov","display_name":"Aleksei Chekunov","email":"chekunov87@gmail.com","username":"az7961"},"date":"2026-09-10 22:17:53.000000000","message":"Patch Set 10: Code-Review+1","accounts_in_message":[],"_revision_number":10},{"id":"27f004f5df6de2ccdca7675cd850e9edfb3359a8","author":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},"date":"2026-09-10 22:19:53.000000000","message":"Patch Set 10: Code-Review+2","accounts_in_message":[],"_revision_number":10},{"id":"7c1f8143b646cf16374ea9f32abc60a848e470ba","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2026-09-10 22:35:35.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"45fecdfda78238dba60348fa50d760e527556759","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2026-09-10 22:37:35.000000000","message":"Patch Set 10: Code-Review+2","accounts_in_message":[],"_revision_number":10},{"id":"0811d10747eeaca6d3368b051675af9dc17ac69a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":21040,"name":"Alexey","email":"aodinokov@mirantis.com","username":"aodinokov"},"date":"2026-09-10 22:41:35.000000000","message":"Patch Set 11: Published edit on patch set 10.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":11},{"id":"dea020cd8528882e413c7bada30b2b7d03cd5da0","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2026-09-10 22:42:34.000000000","message":"Patch Set 11: Code-Review+2","accounts_in_message":[],"_revision_number":11},{"id":"e0c6e9c206456a69f2e026d50b09061d00adfb44","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2026-09-10 22:43:04.000000000","message":"Patch Set 11: -Code-Review","accounts_in_message":[],"_revision_number":11},{"id":"83d060d92bcd4399dc31ea022186dcde129ed22f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":21040,"name":"Alexey","email":"aodinokov@mirantis.com","username":"aodinokov"},"date":"2026-09-10 23:20:38.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"ca7e2c5eaf090c926fa3e9d42ecfe08c2d742c93","author":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},"date":"2026-09-10 23:21:23.000000000","message":"Patch Set 12: Code-Review+2","accounts_in_message":[],"_revision_number":12},{"id":"681f4190a012a2f0a5241ed885d528c72e37a738","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-11 01:10:45.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/dda3d4035f874168bc5d86c57594e007\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/07a381c816d44d32b158abd83797d7d6 : SUCCESS in 3m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0198630f642c46a68e8e0099fc8391a2 : SUCCESS in 2m 13s\n- openstack-helm-linter https://zuul.opendev.org/t/openstack/build/57ac98eb1ba94dbdb1ba840783aaa839 : SUCCESS in 1m 52s\n- openstack-helm-mariadb-crd-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/3a3ed0a82bd74a6b9e939295aff74fc0 : SUCCESS in 58m 03s\n- openstack-helm-mariadb-operator-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/55e74d05a628460584c2f70e954c0b92 : SUCCESS in 1h 16m 15s\n- openstack-helm-rabbitmq-crd-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/478228cb10d245ff8e24379691ae0fca : SUCCESS in 42m 57s\n- openstack-helm-rabbitmq-operator-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/f9262178d065434397c5a01e5a43e7c0 : SUCCESS in 1h 18m 54s\n- openstack-helm-pre-commit https://zuul.opendev.org/t/openstack/build/3ee444b399824766811ff90ace418c75 : SUCCESS in 2m 42s\n- openstack-helm-bandit https://zuul.opendev.org/t/openstack/build/ac310a561b2344f791acebd2f04b56f8 : SUCCESS in 3m 18s\n- openstack-helm-build-charts https://zuul.opendev.org/t/openstack/build/de4124d0a66543ea8c8a85ee17cdfa00 : SUCCESS in 2m 45s\n- openstack-helm-cinder-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/adb5f8bdcaf743ffaee650a003544924 : SUCCESS in 27m 00s\n- openstack-helm-compute-kit-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/67b445ad32494fe9a1a9134e729ca3ce : SUCCESS in 1h 11m 17s\n- openstack-helm-cinder-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/4e1869769e4c44cd97d16e5ec256a54c : SUCCESS in 34m 12s\n- openstack-helm-compute-kit-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/25903074d0cb418bb4a62bd57e53b2ce : SUCCESS in 46m 06s\n- openstack-helm-cinder-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/eed470a0976048c99a1cffe70b1b2db8 : SUCCESS in 30m 09s\n- openstack-helm-compute-kit-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/02b2aa008ecb41b3b180076f70f08619 : SUCCESS in 46m 20s\n- openstack-helm-tls-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/fb2178e87669421780599e0c13c899d2 : SUCCESS in 1h 41m 56s\n- openstack-helm-octavia-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/cf308e2b045446fe80f55878621abccd : FAILURE in 1h 03m 38s\n- openstack-helm-logging https://zuul.opendev.org/t/openstack/build/7d189f1ee8824cd28c576a55b085c41a : SUCCESS in 39m 43s\n- openstack-helm-monitoring https://zuul.opendev.org/t/openstack/build/0072407ca9584ab1bcc1fc0e58aad1e1 : SUCCESS in 30m 25s","accounts_in_message":[],"_revision_number":12},{"id":"fdda972d12f2a99d1562e236449a69bcb78e4182","author":{"_account_id":21040,"name":"Alexey","email":"aodinokov@mirantis.com","username":"aodinokov"},"date":"2026-09-11 03:52:45.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"cb47ceff431e9fca8027ca7f7774b1608c154400","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-11 05:42:11.000000000","message":"Patch Set 12:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/eaf1980aabb6422c92002d714b6d8674\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7eba8fe8d8be4808b1811f09aa5aab8a : SUCCESS in 3m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/81425a3bb6a0405da549fca6c00ebf99 : SUCCESS in 2m 44s\n- openstack-helm-linter https://zuul.opendev.org/t/openstack/build/33af276d55f54d8094258a9d2fef22fa : SUCCESS in 3m 45s\n- openstack-helm-mariadb-crd-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/b5fe681aa69d42128374d18ec4f646ce : SUCCESS in 58m 30s\n- openstack-helm-mariadb-operator-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/bed4216f08864573b8e1b01fd1c1bff1 : SUCCESS in 1h 24m 45s\n- openstack-helm-rabbitmq-crd-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/a7a929d9fe034ec587bf45fcc6285d6a : SUCCESS in 57m 11s\n- openstack-helm-rabbitmq-operator-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/6957bcd15387426e84d2325b03d8c63e : SUCCESS in 1h 16m 47s\n- openstack-helm-pre-commit https://zuul.opendev.org/t/openstack/build/5236d97b889944fa96f81d9c4f45a52a : SUCCESS in 1m 50s\n- openstack-helm-bandit https://zuul.opendev.org/t/openstack/build/848c17e1dbc54358a9bde802973a1870 : SUCCESS in 3m 52s\n- openstack-helm-build-charts https://zuul.opendev.org/t/openstack/build/395abf87884044728c8e5c142aeabe82 : SUCCESS in 1m 52s\n- openstack-helm-cinder-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/f19c76027e804965aa1f881f782312b0 : SUCCESS in 26m 54s\n- openstack-helm-compute-kit-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/865f1a1ef02644a490dfae09b9ff111f : SUCCESS in 1h 11m 41s\n- openstack-helm-cinder-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/ff439d4ec7fd4745889678845f6789b4 : SUCCESS in 46m 02s\n- openstack-helm-compute-kit-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/798a925d322f430db0457f4e1544bb82 : SUCCESS in 1h 09m 01s\n- openstack-helm-cinder-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/36e180d94cfe4885a30b77bb8e64267a : SUCCESS in 43m 48s\n- openstack-helm-compute-kit-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/80aeb30d29354d05be5767d99226f340 : SUCCESS in 1h 12m 07s\n- openstack-helm-tls-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/2aa24ba9cbdf4443bcf3ac11933b14bc : SUCCESS in 1h 45m 01s\n- openstack-helm-octavia-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/7f338d64b6304f6ab504ef526f551eaf : FAILURE in 1h 03m 59s\n- openstack-helm-logging https://zuul.opendev.org/t/openstack/build/8351b5d98c3744c1b9716a705876eac5 : SUCCESS in 24m 18s\n- openstack-helm-monitoring https://zuul.opendev.org/t/openstack/build/a6a638f3159d4cf39d0d2040650b1933 : SUCCESS in 22m 16s","accounts_in_message":[],"_revision_number":12},{"id":"5ab8cc5a7233a964a150b5e4c6e1be488061a413","author":{"_account_id":21040,"name":"Alexey","email":"aodinokov@mirantis.com","username":"aodinokov"},"date":"2026-09-11 06:04:16.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"af7eee41792f7d59e33562d72801f203857c9e3f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-11 07:55:00.000000000","message":"Patch Set 12:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/36987195dd3a4f8fbc8df4e235236827\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/098dadd36aba46f1bcbd988e01344c59 : SUCCESS in 3m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/eef4c976a2564c17ac84dbbad6a81b58 : SUCCESS in 4m 04s\n- openstack-helm-linter https://zuul.opendev.org/t/openstack/build/9b3ff92ce7da42cda09f862699ca6604 : SUCCESS in 2m 56s\n- openstack-helm-mariadb-crd-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/0560bb5262a94117ab0b0042351cfbca : SUCCESS in 56m 29s\n- openstack-helm-mariadb-operator-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/5bb5ed1bf9a84dad8180a77c0422c2c8 : SUCCESS in 1h 34m 15s\n- openstack-helm-rabbitmq-crd-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/55a9900e9f3446bebfa7970423ecb7e3 : SUCCESS in 1h 12m 49s\n- openstack-helm-rabbitmq-operator-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/4d97c970ab6d4700813d1fc3727d6769 : SUCCESS in 1h 06m 28s\n- openstack-helm-pre-commit https://zuul.opendev.org/t/openstack/build/647803e90a0e43af97d5552c82ee9263 : SUCCESS in 2m 58s\n- openstack-helm-bandit https://zuul.opendev.org/t/openstack/build/05a55343e225416f908a011b8b8bad53 : SUCCESS in 3m 16s\n- openstack-helm-build-charts https://zuul.opendev.org/t/openstack/build/04569aa8f0b3484a9fa68f26325667a5 : SUCCESS in 1m 48s\n- openstack-helm-cinder-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/0a2888796bdc4e889f57741a569155be : SUCCESS in 45m 38s\n- openstack-helm-compute-kit-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/230bdee302b946ca879459935d6b7237 : SUCCESS in 1h 18m 41s\n- openstack-helm-cinder-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/f50d057c632c4c95bb87d7547a988e72 : SUCCESS in 25m 09s\n- openstack-helm-compute-kit-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/1f5bdb4f326d47ccb7b37757b91a4d39 : SUCCESS in 1h 20m 05s\n- openstack-helm-cinder-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/c8b88fc28ef3496498d9e4514e694a52 : SUCCESS in 43m 45s\n- openstack-helm-compute-kit-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/02c7910747be4cada76a8400fd7f2f1e : SUCCESS in 43m 12s\n- openstack-helm-tls-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/451cdd5442e44273910928d373765e37 : SUCCESS in 1h 43m 37s\n- openstack-helm-octavia-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/c5515433e03f49f4905fd677bf51e0cf : FAILURE in 1h 11m 19s\n- openstack-helm-logging https://zuul.opendev.org/t/openstack/build/8061ba8e89ba41d8bae71fb83d14de7a : SUCCESS in 37m 43s\n- openstack-helm-monitoring https://zuul.opendev.org/t/openstack/build/239c81a8488041f2843f77df09be3679 : SUCCESS in 26m 32s","accounts_in_message":[],"_revision_number":12},{"id":"68f43d61fa38e8c121b96612c7f43935bf7d861c","author":{"_account_id":35943,"name":"Aleksei Chekunov","display_name":"Aleksei Chekunov","email":"chekunov87@gmail.com","username":"az7961"},"date":"2026-09-11 13:16:24.000000000","message":"Patch Set 12: Code-Review+1","accounts_in_message":[],"_revision_number":12},{"id":"c0b676ea6bf681a8c2b1982fe3429af294d369d3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":21040,"name":"Alexey","email":"aodinokov@mirantis.com","username":"aodinokov"},"date":"2026-09-11 14:29:17.000000000","message":"Patch Set 13: Patch Set 12 was rebased\n\nCopied Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":13},{"id":"a6abc1430621f142716c79766003266280e4043c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-11 16:17:45.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fc990c5d737e4233a84da9a17590b48d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6df91702d9cf4013a79b7c6398a6dfc8 : SUCCESS in 3m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/957543747c854311aaa078da134a8ae0 : SUCCESS in 3m 36s\n- openstack-helm-linter https://zuul.opendev.org/t/openstack/build/533be5108ad04bf789ab70d7abb9476c : SUCCESS in 2m 02s\n- openstack-helm-mariadb-crd-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/e6c91ebb637b4cca869b9f6da521cf45 : SUCCESS in 1h 42m 18s\n- openstack-helm-mariadb-operator-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/5de72f21ec914d8b91b44770918b809c : SUCCESS in 1h 34m 20s\n- openstack-helm-rabbitmq-crd-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/55169eb6058b471e8818a01be2b03241 : SUCCESS in 1h 18m 03s\n- openstack-helm-rabbitmq-operator-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/f9c16b63816f4ccabbe6e3017b0921b6 : SUCCESS in 1h 18m 24s\n- openstack-helm-pre-commit https://zuul.opendev.org/t/openstack/build/2bf2e1633301460cb07791281f49b242 : SUCCESS in 2m 32s\n- openstack-helm-bandit https://zuul.opendev.org/t/openstack/build/cdf39e416fe44f6cb275125898b7e400 : SUCCESS in 3m 23s\n- openstack-helm-build-charts https://zuul.opendev.org/t/openstack/build/d6809f91a395464f85ad770c6c5d3e05 : SUCCESS in 3m 29s\n- openstack-helm-cinder-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/7bd4f91f6685473e8c90b57071fdbd20 : SUCCESS in 30m 56s\n- openstack-helm-compute-kit-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/6ec78c16e52b4470b87b2c3805ab2214 : SUCCESS in 52m 54s\n- openstack-helm-cinder-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/a6894557ca634df5a2b51bbd4d7cf0a0 : SUCCESS in 44m 56s\n- openstack-helm-compute-kit-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/98c8bb56027d42deb8c4cb4935cb94a6 : SUCCESS in 46m 40s\n- openstack-helm-cinder-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/3a4975c4defd41dc8c65b08ca6c5ba38 : SUCCESS in 45m 44s\n- openstack-helm-compute-kit-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/485f39d408324a1695d6f5fab17fb301 : SUCCESS in 1h 23m 07s\n- openstack-helm-tls-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/ab9c4ad8b3c447fd8ce3b4567ac0616b : SUCCESS in 1h 40m 21s\n- openstack-helm-octavia-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/4127db603c5c4d0fa1680f7abeed62ca : SUCCESS in 44m 43s\n- openstack-helm-logging https://zuul.opendev.org/t/openstack/build/0038675f335e44659b8b7444e12c875f : SUCCESS in 35m 35s\n- openstack-helm-monitoring https://zuul.opendev.org/t/openstack/build/0a29910f9d4e4b8cb763b1c3b43ad493 : SUCCESS in 29m 47s","accounts_in_message":[],"_revision_number":13},{"id":"d24c672fbec0473e9ef2ebce243fa9cc3da4f102","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2026-09-11 16:27:11.000000000","message":"Patch Set 13: Code-Review+2","accounts_in_message":[],"_revision_number":13},{"id":"2e392d0a65352c742402f5a17df25962ea88e932","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2026-09-11 17:05:41.000000000","message":"Patch Set 13: Workflow+1","accounts_in_message":[],"_revision_number":13},{"id":"fec2870f147755217fa1d7c314a64233282d82bc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-11 17:06:21.000000000","message":"Patch Set 13: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":13},{"id":"6b350af87a2ab7fb602be10d773dd979401b50e7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-11 18:28:08.000000000","message":"Patch Set 13: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9b191c5e54fb48cab7de5ecce3f16345\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/21c9df12230b48fa88a1214d27a41764 : SUCCESS in 4m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d2f3dfa7d25e4b59821b9c16dc074276 : SUCCESS in 3m 34s\n- openstack-helm-linter https://zuul.opendev.org/t/openstack/build/d36e44ee7e4749a6b75dab446c08dba6 : SUCCESS in 3m 08s\n- openstack-helm-cinder-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/801f26fcaac94c6aaccafde22ed475c2 : SUCCESS in 39m 18s\n- openstack-helm-compute-kit-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/4678e179f03a484c8ccaec0552432431 : SUCCESS in 1h 16m 00s\n- openstack-helm-logging https://zuul.opendev.org/t/openstack/build/de26ddfaa1d54938b395cb231f2986e8 : SUCCESS in 26m 16s\n- openstack-helm-monitoring https://zuul.opendev.org/t/openstack/build/9486406cf3f04410b7bb310ccd375a5c : SUCCESS in 22m 11s","accounts_in_message":[],"_revision_number":13},{"id":"57c1bdcd7bab011f637d4f157fc6ecf1ad95330c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-11 18:28:08.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":13},{"id":"9675bfda5172c749cab1e11fa9744374d379b0d9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-11 18:29:15.000000000","message":"Patch Set 13:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/85dab58ba75c43dfb09043e8c3469538\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/124019de227f494b860ea7c00d196ca5 : SUCCESS in 49s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/05fd4a969d504b69b9d7f1df4486fe9c : SUCCESS in 47s","accounts_in_message":[],"_revision_number":13}],"current_revision_number":13,"current_revision":"506b0bcb2746964cc571846e0651adf50237dc0e","revisions":{"8f25dec36cdf6ca7ee2c649ebd57f55b416de4b5":{"kind":"REWORK","_number":1,"created":"2026-09-04 14:21:06.000000000","uploader":{"_account_id":21040,"name":"Alexey","email":"aodinokov@mirantis.com","username":"aodinokov"},"ref":"refs/changes/88/1003988/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-helm","ref":"refs/changes/88/1003988/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-helm refs/changes/88/1003988/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-helm refs/changes/88/1003988/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-helm refs/changes/88/1003988/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-helm refs/changes/88/1003988/1"}}},"commit":{"parents":[{"commit":"c73f68ac866aeedbec2784c16e6a45b215664112","subject":"Merge \"ovn: fix liveness probe to check as ovs user, not root\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm/commit/c73f68ac866aeedbec2784c16e6a45b215664112"}]}],"author":{"name":"Alexey","email":"aodinokov@mirantis.com","date":"2026-09-04 14:21:06.000000000","tz":0},"committer":{"name":"Alexey","email":"aodinokov@mirantis.com","date":"2026-09-04 14:21:06.000000000","tz":0},"subject":"DNM/WIP Fixing permanent secrets for keystone","message":"DNM/WIP Fixing permanent secrets for keystone\n\nChange-Id: I9dbc23782a82120fb702768fb853a970d7b363a3\nSigned-off-by: Alexey Odinokov \u003caodinokov@mirantis.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm/commit/8f25dec36cdf6ca7ee2c649ebd57f55b416de4b5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm/commit/8f25dec36cdf6ca7ee2c649ebd57f55b416de4b5"}]},"branch":"refs/heads/master","conflicts":{"no_base_reason":"NO_MERGE_PERFORMED","contains_conflicts":false}},"8d1161504a62f530aa48670476984a5f28a231fe":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-09-04 14:21:37.000000000","uploader":{"_account_id":21040,"name":"Alexey","email":"aodinokov@mirantis.com","username":"aodinokov"},"ref":"refs/changes/88/1003988/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-helm","ref":"refs/changes/88/1003988/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-helm refs/changes/88/1003988/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-helm refs/changes/88/1003988/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-helm refs/changes/88/1003988/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-helm refs/changes/88/1003988/2"}}},"commit":{"parents":[{"commit":"3bc47c1e93f78e4df2e1040d553777068eb33d92","subject":"feat: add extraObjects to every chart","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm/commit/3bc47c1e93f78e4df2e1040d553777068eb33d92"}]}],"author":{"name":"Alexey","email":"aodinokov@mirantis.com","date":"2026-09-04 14:21:06.000000000","tz":0},"committer":{"name":"Alexey","email":"aodinokov@mirantis.com","date":"2026-09-04 14:21:37.000000000","tz":0},"subject":"DNM/WIP Fixing permanent secrets for keystone","message":"DNM/WIP Fixing permanent secrets for keystone\n\nChange-Id: I9dbc23782a82120fb702768fb853a970d7b363a3\nSigned-off-by: Alexey Odinokov \u003caodinokov@mirantis.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm/commit/8d1161504a62f530aa48670476984a5f28a231fe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm/commit/8d1161504a62f530aa48670476984a5f28a231fe"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"c73f68ac866aeedbec2784c16e6a45b215664112","ours":"8f25dec36cdf6ca7ee2c649ebd57f55b416de4b5","theirs":"3bc47c1e93f78e4df2e1040d553777068eb33d92","merge_strategy":"recursive","contains_conflicts":false}},"02aecf041258f599a01f65f3ba037b37697cf6a8":{"kind":"REWORK","_number":3,"created":"2026-09-04 14:28:04.000000000","uploader":{"_account_id":21040,"name":"Alexey","email":"aodinokov@mirantis.com","username":"aodinokov"},"ref":"refs/changes/88/1003988/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-helm","ref":"refs/changes/88/1003988/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-helm refs/changes/88/1003988/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-helm refs/changes/88/1003988/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-helm refs/changes/88/1003988/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-helm refs/changes/88/1003988/3"}}},"commit":{"parents":[{"commit":"3bc47c1e93f78e4df2e1040d553777068eb33d92","subject":"feat: add extraObjects to every chart","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm/commit/3bc47c1e93f78e4df2e1040d553777068eb33d92"}]}],"author":{"name":"Alexey","email":"aodinokov@mirantis.com","date":"2026-09-04 14:21:06.000000000","tz":0},"committer":{"name":"Alexey","email":"aodinokov@mirantis.com","date":"2026-09-04 14:28:04.000000000","tz":0},"subject":"DNM/WIP Fixing permanent secrets for keystone","message":"DNM/WIP Fixing permanent secrets for keystone\n\nChange-Id: I9dbc23782a82120fb702768fb853a970d7b363a3\nSigned-off-by: Alexey Odinokov \u003caodinokov@mirantis.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm/commit/02aecf041258f599a01f65f3ba037b37697cf6a8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm/commit/02aecf041258f599a01f65f3ba037b37697cf6a8"}]},"branch":"refs/heads/master"},"b1c3e47df7e921fd411dcf24d42aaf94e87b9e80":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2026-09-04 19:03:08.000000000","uploader":{"_account_id":21040,"name":"Alexey","email":"aodinokov@mirantis.com","username":"aodinokov"},"ref":"refs/changes/88/1003988/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-helm","ref":"refs/changes/88/1003988/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-helm refs/changes/88/1003988/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-helm refs/changes/88/1003988/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-helm refs/changes/88/1003988/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-helm refs/changes/88/1003988/4"}}},"commit":{"parents":[{"commit":"c73f68ac866aeedbec2784c16e6a45b215664112","subject":"Merge \"ovn: fix liveness probe to check as ovs user, not root\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm/commit/c73f68ac866aeedbec2784c16e6a45b215664112"}]}],"author":{"name":"Alexey","email":"aodinokov@mirantis.com","date":"2026-09-04 14:21:06.000000000","tz":0},"committer":{"name":"Alexey","email":"aodinokov@mirantis.com","date":"2026-09-04 19:03:08.000000000","tz":0},"subject":"DNM/WIP Fixing permanent secrets for keystone","message":"DNM/WIP Fixing permanent secrets for keystone\n\nChange-Id: I9dbc23782a82120fb702768fb853a970d7b363a3\nSigned-off-by: Alexey Odinokov \u003caodinokov@mirantis.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm/commit/b1c3e47df7e921fd411dcf24d42aaf94e87b9e80"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm/commit/b1c3e47df7e921fd411dcf24d42aaf94e87b9e80"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"3bc47c1e93f78e4df2e1040d553777068eb33d92","ours":"02aecf041258f599a01f65f3ba037b37697cf6a8","theirs":"c73f68ac866aeedbec2784c16e6a45b215664112","merge_strategy":"recursive","contains_conflicts":false}},"2a42a897d87786cf4a46c2bcc22f41a7059f0682":{"kind":"REWORK","_number":5,"created":"2026-09-05 03:20:07.000000000","uploader":{"_account_id":21040,"name":"Alexey","email":"aodinokov@mirantis.com","username":"aodinokov"},"ref":"refs/changes/88/1003988/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-helm","ref":"refs/changes/88/1003988/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-helm refs/changes/88/1003988/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-helm refs/changes/88/1003988/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-helm refs/changes/88/1003988/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-helm refs/changes/88/1003988/5"}}},"commit":{"parents":[{"commit":"c73f68ac866aeedbec2784c16e6a45b215664112","subject":"Merge \"ovn: fix liveness probe to check as ovs user, not root\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm/commit/c73f68ac866aeedbec2784c16e6a45b215664112"}]}],"author":{"name":"Alexey","email":"aodinokov@mirantis.com","date":"2026-09-04 14:21:06.000000000","tz":0},"committer":{"name":"Alexey","email":"aodinokov@mirantis.com","date":"2026-09-05 03:16:33.000000000","tz":0},"subject":"DNM/WIP Fixing permanent secrets for keystone","message":"DNM/WIP Fixing permanent secrets for keystone\n\nChange-Id: I9dbc23782a82120fb702768fb853a970d7b363a3\nSigned-off-by: Alexey Odinokov \u003caodinokov@mirantis.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm/commit/2a42a897d87786cf4a46c2bcc22f41a7059f0682"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm/commit/2a42a897d87786cf4a46c2bcc22f41a7059f0682"}]},"branch":"refs/heads/master"},"57a629fe05a1296d2c941d6fcf0b23720c4b08c7":{"kind":"REWORK","_number":6,"created":"2026-09-05 03:20:37.000000000","uploader":{"_account_id":21040,"name":"Alexey","email":"aodinokov@mirantis.com","username":"aodinokov"},"ref":"refs/changes/88/1003988/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-helm","ref":"refs/changes/88/1003988/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-helm refs/changes/88/1003988/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-helm refs/changes/88/1003988/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-helm refs/changes/88/1003988/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-helm refs/changes/88/1003988/6"}}},"commit":{"parents":[{"commit":"c73f68ac866aeedbec2784c16e6a45b215664112","subject":"Merge \"ovn: fix liveness probe to check as ovs user, not root\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm/commit/c73f68ac866aeedbec2784c16e6a45b215664112"}]}],"author":{"name":"Alexey","email":"aodinokov@mirantis.com","date":"2026-09-04 14:21:06.000000000","tz":0},"committer":{"name":"Alexey","email":"aodinokov@mirantis.com","date":"2026-09-05 03:20:36.000000000","tz":0},"subject":"DNM/WIP Fixing permanent secrets for keystone","message":"DNM/WIP Fixing permanent secrets for keystone\n\nChange-Id: I9dbc23782a82120fb702768fb853a970d7b363a3\nSigned-off-by: Alexey Odinokov \u003caodinokov@mirantis.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm/commit/57a629fe05a1296d2c941d6fcf0b23720c4b08c7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm/commit/57a629fe05a1296d2c941d6fcf0b23720c4b08c7"}]},"branch":"refs/heads/master"},"d0dbe0d009668fbcf16062bc877e328f31810295":{"kind":"REWORK","_number":7,"created":"2026-09-09 20:00:07.000000000","uploader":{"_account_id":21040,"name":"Alexey","email":"aodinokov@mirantis.com","username":"aodinokov"},"ref":"refs/changes/88/1003988/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-helm","ref":"refs/changes/88/1003988/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-helm refs/changes/88/1003988/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-helm refs/changes/88/1003988/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-helm refs/changes/88/1003988/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-helm refs/changes/88/1003988/7"}}},"commit":{"parents":[{"commit":"c73f68ac866aeedbec2784c16e6a45b215664112","subject":"Merge \"ovn: fix liveness probe to check as ovs user, not root\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm/commit/c73f68ac866aeedbec2784c16e6a45b215664112"}]}],"author":{"name":"Alexey","email":"aodinokov@mirantis.com","date":"2026-09-04 14:21:06.000000000","tz":0},"committer":{"name":"Alexey","email":"aodinokov@mirantis.com","date":"2026-09-09 20:00:07.000000000","tz":0},"subject":"DNM/WIP Fixing permanent secrets for keystone","message":"DNM/WIP Fixing permanent secrets for keystone\n\nChange-Id: I9dbc23782a82120fb702768fb853a970d7b363a3\nSigned-off-by: Alexey Odinokov \u003caodinokov@mirantis.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm/commit/d0dbe0d009668fbcf16062bc877e328f31810295"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm/commit/d0dbe0d009668fbcf16062bc877e328f31810295"}]},"branch":"refs/heads/master"},"7e90dce24e731bfe9571312c7e4b3044981399f7":{"kind":"REWORK","_number":8,"created":"2026-09-09 22:41:17.000000000","uploader":{"_account_id":21040,"name":"Alexey","email":"aodinokov@mirantis.com","username":"aodinokov"},"ref":"refs/changes/88/1003988/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-helm","ref":"refs/changes/88/1003988/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-helm refs/changes/88/1003988/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-helm refs/changes/88/1003988/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-helm refs/changes/88/1003988/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-helm refs/changes/88/1003988/8"}}},"commit":{"parents":[{"commit":"c73f68ac866aeedbec2784c16e6a45b215664112","subject":"Merge \"ovn: fix liveness probe to check as ovs user, not root\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm/commit/c73f68ac866aeedbec2784c16e6a45b215664112"}]}],"author":{"name":"Alexey","email":"aodinokov@mirantis.com","date":"2026-09-04 14:21:06.000000000","tz":0},"committer":{"name":"Alexey","email":"aodinokov@mirantis.com","date":"2026-09-09 20:39:48.000000000","tz":0},"subject":"DNM/WIP Fixing permanent secrets for keystone","message":"DNM/WIP Fixing permanent secrets for keystone\n\nChange-Id: I9dbc23782a82120fb702768fb853a970d7b363a3\nSigned-off-by: Alexey Odinokov \u003caodinokov@mirantis.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm/commit/7e90dce24e731bfe9571312c7e4b3044981399f7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm/commit/7e90dce24e731bfe9571312c7e4b3044981399f7"}]},"branch":"refs/heads/master"},"258a9ad21d52c13589153ed1c3c1c525495c37c8":{"kind":"REWORK","_number":9,"created":"2026-09-09 23:07:11.000000000","uploader":{"_account_id":21040,"name":"Alexey","email":"aodinokov@mirantis.com","username":"aodinokov"},"ref":"refs/changes/88/1003988/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-helm","ref":"refs/changes/88/1003988/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-helm refs/changes/88/1003988/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-helm refs/changes/88/1003988/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-helm refs/changes/88/1003988/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-helm refs/changes/88/1003988/9"}}},"commit":{"parents":[{"commit":"c73f68ac866aeedbec2784c16e6a45b215664112","subject":"Merge \"ovn: fix liveness probe to check as ovs user, not root\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm/commit/c73f68ac866aeedbec2784c16e6a45b215664112"}]}],"author":{"name":"Alexey","email":"aodinokov@mirantis.com","date":"2026-09-04 14:21:06.000000000","tz":0},"committer":{"name":"Alexey","email":"aodinokov@mirantis.com","date":"2026-09-09 23:07:11.000000000","tz":0},"subject":"DNM/WIP Fixing permanent secrets for keystone","message":"DNM/WIP Fixing permanent secrets for keystone\n\nChange-Id: I9dbc23782a82120fb702768fb853a970d7b363a3\nSigned-off-by: Alexey Odinokov \u003caodinokov@mirantis.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm/commit/258a9ad21d52c13589153ed1c3c1c525495c37c8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm/commit/258a9ad21d52c13589153ed1c3c1c525495c37c8"}]},"branch":"refs/heads/master"},"29fdc4ff53ae096bd1358d1ab73033a1ec11a530":{"kind":"NO_CODE_CHANGE","_number":10,"created":"2026-09-10 03:16:41.000000000","uploader":{"_account_id":21040,"name":"Alexey","email":"aodinokov@mirantis.com","username":"aodinokov"},"ref":"refs/changes/88/1003988/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-helm","ref":"refs/changes/88/1003988/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-helm refs/changes/88/1003988/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-helm refs/changes/88/1003988/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-helm refs/changes/88/1003988/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-helm refs/changes/88/1003988/10"}}},"commit":{"parents":[{"commit":"c73f68ac866aeedbec2784c16e6a45b215664112","subject":"Merge \"ovn: fix liveness probe to check as ovs user, not root\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm/commit/c73f68ac866aeedbec2784c16e6a45b215664112"}]}],"author":{"name":"Alexey","email":"aodinokov@mirantis.com","date":"2026-09-04 14:21:06.000000000","tz":0},"committer":{"name":"Alexey","email":"aodinokov@mirantis.com","date":"2026-09-10 03:16:41.000000000","tz":0},"subject":"Fixing permanent secrets for keystone","message":"Fixing permanent secrets for keystone\n\nMaking secrets completely dynamically created\n\nChange-Id: I9dbc23782a82120fb702768fb853a970d7b363a3\nSigned-off-by: Alexey Odinokov \u003caodinokov@mirantis.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm/commit/29fdc4ff53ae096bd1358d1ab73033a1ec11a530"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm/commit/29fdc4ff53ae096bd1358d1ab73033a1ec11a530"}]},"branch":"refs/heads/master","description":"Edit commit message"},"53cf508e14b86aa291f3df3c1fd2963a1fdf2471":{"kind":"REWORK","_number":11,"created":"2026-09-10 22:41:35.000000000","uploader":{"_account_id":21040,"name":"Alexey","email":"aodinokov@mirantis.com","username":"aodinokov"},"ref":"refs/changes/88/1003988/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-helm","ref":"refs/changes/88/1003988/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-helm refs/changes/88/1003988/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-helm refs/changes/88/1003988/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-helm refs/changes/88/1003988/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-helm refs/changes/88/1003988/11"}}},"commit":{"parents":[{"commit":"c73f68ac866aeedbec2784c16e6a45b215664112","subject":"Merge \"ovn: fix liveness probe to check as ovs user, not root\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm/commit/c73f68ac866aeedbec2784c16e6a45b215664112"}]}],"author":{"name":"Alexey","email":"aodinokov@mirantis.com","date":"2026-09-04 14:21:06.000000000","tz":0},"committer":{"name":"Alexey","email":"aodinokov@mirantis.com","date":"2026-09-10 22:41:35.000000000","tz":0},"subject":"Fixing permanent secrets for keystone","message":"Fixing permanent secrets for keystone\n\nMaking secrets completely dynamically created\n\nChange-Id: I9dbc23782a82120fb702768fb853a970d7b363a3\nSigned-off-by: Alexey Odinokov \u003caodinokov@mirantis.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm/commit/53cf508e14b86aa291f3df3c1fd2963a1fdf2471"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm/commit/53cf508e14b86aa291f3df3c1fd2963a1fdf2471"}]},"branch":"refs/heads/master"},"5c3a4d50cb13fe6d4d9a3f2e163b314e90a70398":{"kind":"REWORK","_number":12,"created":"2026-09-10 23:20:38.000000000","uploader":{"_account_id":21040,"name":"Alexey","email":"aodinokov@mirantis.com","username":"aodinokov"},"ref":"refs/changes/88/1003988/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-helm","ref":"refs/changes/88/1003988/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-helm refs/changes/88/1003988/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-helm refs/changes/88/1003988/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-helm refs/changes/88/1003988/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-helm refs/changes/88/1003988/12"}}},"commit":{"parents":[{"commit":"c73f68ac866aeedbec2784c16e6a45b215664112","subject":"Merge \"ovn: fix liveness probe to check as ovs user, not root\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm/commit/c73f68ac866aeedbec2784c16e6a45b215664112"}]}],"author":{"name":"Alexey","email":"aodinokov@mirantis.com","date":"2026-09-04 14:21:06.000000000","tz":0},"committer":{"name":"Alexey Odinokov","email":"aodinokov@mirantis.com","date":"2026-09-10 23:20:26.000000000","tz":-300},"subject":"Fixing permanent secrets for keystone","message":"Fixing permanent secrets for keystone\n\nMaking secrets completely dynamically created\n\nChange-Id: I9dbc23782a82120fb702768fb853a970d7b363a3\nSigned-off-by: Alexey Odinokov \u003caodinokov@mirantis.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm/commit/5c3a4d50cb13fe6d4d9a3f2e163b314e90a70398"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm/commit/5c3a4d50cb13fe6d4d9a3f2e163b314e90a70398"}]},"branch":"refs/heads/master"},"506b0bcb2746964cc571846e0651adf50237dc0e":{"kind":"TRIVIAL_REBASE","_number":13,"created":"2026-09-11 14:29:17.000000000","uploader":{"_account_id":21040,"name":"Alexey","email":"aodinokov@mirantis.com","username":"aodinokov"},"ref":"refs/changes/88/1003988/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-helm","ref":"refs/changes/88/1003988/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-helm refs/changes/88/1003988/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-helm refs/changes/88/1003988/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-helm refs/changes/88/1003988/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-helm refs/changes/88/1003988/13"}}},"commit":{"parents":[{"commit":"7745db9b124fabb5e95d075e5f9df573a92b3db2","subject":"ci: fix the octavia image and kubeadm join flakes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm/commit/7745db9b124fabb5e95d075e5f9df573a92b3db2"}]}],"author":{"name":"Alexey","email":"aodinokov@mirantis.com","date":"2026-09-04 14:21:06.000000000","tz":0},"committer":{"name":"Alexey","email":"aodinokov@mirantis.com","date":"2026-09-11 14:29:17.000000000","tz":0},"subject":"Fixing permanent secrets for keystone","message":"Fixing permanent secrets for keystone\n\nMaking secrets completely dynamically created\n\nChange-Id: I9dbc23782a82120fb702768fb853a970d7b363a3\nSigned-off-by: Alexey Odinokov \u003caodinokov@mirantis.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm/commit/506b0bcb2746964cc571846e0651adf50237dc0e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm/commit/506b0bcb2746964cc571846e0651adf50237dc0e"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"c73f68ac866aeedbec2784c16e6a45b215664112","ours":"5c3a4d50cb13fe6d4d9a3f2e163b314e90a70398","theirs":"7745db9b124fabb5e95d075e5f9df573a92b3db2","merge_strategy":"recursive","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":35943,"name":"Aleksei Chekunov","display_name":"Aleksei Chekunov","email":"chekunov87@gmail.com","username":"az7961"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
