)]}'
{"id":"openstack%2Fopenstack-helm~1004032","triplet_id":"openstack%2Fopenstack-helm~master~I5f7ce68ab99ccc152c86a9fb09c6661ec33b2a00","project":"openstack/openstack-helm","branch":"master","attention_set":{},"removed_from_attention_set":{"21040":{"account":{"_account_id":21040,"name":"Alexey","email":"aodinokov@mirantis.com","username":"aodinokov"},"last_update":"2026-09-10 15:07:26.000000000","reason":"Change was marked work in progress"},"29974":{"account":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"last_update":"2026-09-10 15:07:26.000000000","reason":"Change was marked work in progress"},"34520":{"account":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},"last_update":"2026-09-10 15:07:26.000000000","reason":"Change was marked work in progress"}},"hashtags":[],"change_id":"I5f7ce68ab99ccc152c86a9fb09c6661ec33b2a00","subject":"[ceph] Support both aes and aes256k CephX auth keys","status":"NEW","created":"2026-09-04 19:48:16.000000000","updated":"2026-09-24 18:07:33.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"work_in_progress":true,"has_review_started":true,"meta_rev_id":"523cd8ffee0f379f9aacde55177db1378796f23a","_number":1004032,"virtual_id_number":1004032,"owner":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":21040,"name":"Alexey","email":"aodinokov@mirantis.com","username":"aodinokov"},{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-09-24 18:07:33.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":21040,"name":"Alexey","email":"aodinokov@mirantis.com","username":"aodinokov"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":21040,"name":"Alexey","email":"aodinokov@mirantis.com","username":"aodinokov"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},{"_account_id":21040,"name":"Alexey","email":"aodinokov@mirantis.com","username":"aodinokov"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"}]},"pending_reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"reviewer_updates":[{"updated":"2026-09-04 21:23:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-04 21:31:26.000000000","updated_by":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},"reviewer":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},"state":"REVIEWER"},{"updated":"2026-09-08 14:42:37.000000000","updated_by":{"_account_id":21040,"name":"Alexey","email":"aodinokov@mirantis.com","username":"aodinokov"},"reviewer":{"_account_id":21040,"name":"Alexey","email":"aodinokov@mirantis.com","username":"aodinokov"},"state":"REVIEWER"},{"updated":"2026-09-09 21:37:54.000000000","updated_by":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"reviewer":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"state":"REVIEWER"}],"messages":[{"id":"499c59e009b4d448d99ad7a742c62730afb12d04","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"date":"2026-09-04 19:48:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"37de06e7de756724a6b8a906ae685c64325a36d5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-04 21:23:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cf9a8bd9a9484e8db80bf2609e6d68c5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aee103bfae3345ed894011d3217e9223 : SUCCESS in 2m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e4819825f9e643bfa3b3cfeb01bfa6ac : SUCCESS in 2m 58s\n- openstack-helm-linter https://zuul.opendev.org/t/openstack/build/d4aa74c8c15d4dc38bd8f50817e85b14 : SUCCESS in 2m 19s\n- openstack-helm-mariadb-crd-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/78cd2ea717bc4b93ab2fa6922c34b7d0 : SUCCESS in 1h 10m 47s\n- openstack-helm-mariadb-operator-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/ad2a6ee48dfc4becb7cc7e3a170bf492 : SUCCESS in 1h 28m 44s\n- openstack-helm-rabbitmq-crd-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/eb818682145047a09a2b8d08082ae937 : SUCCESS in 50m 59s\n- openstack-helm-rabbitmq-operator-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/40da147b6ff84d85b38b2e7409a41ced : SUCCESS in 1h 19m 51s\n- openstack-helm-pre-commit https://zuul.opendev.org/t/openstack/build/5b23c496155643f8a25ffb1fc9c7a08e : SUCCESS in 2m 13s\n- openstack-helm-build-charts https://zuul.opendev.org/t/openstack/build/6e8a836fbc1e49b19553b58093014413 : SUCCESS in 1m 51s\n- openstack-helm-cinder-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/a69ebfc6f8dd428cb7509b23e47536b9 : SUCCESS in 49m 10s\n- openstack-helm-compute-kit-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/de37e04807884573b096ade2fe8f6cf6 : SUCCESS in 1h 13m 15s\n- openstack-helm-cinder-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/5a7d6344dfed45f68d3af2fae68e5727 : SUCCESS in 45m 37s\n- openstack-helm-compute-kit-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/dfd5303ae3ac43f5a4ca27b43f0ac398 : SUCCESS in 50m 54s\n- openstack-helm-cinder-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/223737770ec04015b83c5c631ca34595 : SUCCESS in 30m 46s\n- openstack-helm-compute-kit-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/a583623cd7044b6e8ee7afcf80799413 : SUCCESS in 1h 24m 17s\n- openstack-helm-tls-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/6f5bc29ef17c4722bbd8017439292305 : SUCCESS in 1h 06m 44s\n- openstack-helm-octavia-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/9732d494e12e4bb2a8e32f6aa0871f24 : SUCCESS in 43m 21s\n- openstack-helm-logging https://zuul.opendev.org/t/openstack/build/28c631e64810405ea1df4f5999d077f1 : SUCCESS in 30m 48s\n- openstack-helm-monitoring https://zuul.opendev.org/t/openstack/build/d5a5e67e849448d5a9ee917334c96f68 : SUCCESS in 28m 48s","accounts_in_message":[],"_revision_number":1},{"id":"2fc77f26f0bf5346d51cb2dfed88bb6e6084f7e1","author":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},"date":"2026-09-04 21:31:26.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e73ec066b0fed5b46443b15119d80e0435c2bcc0","author":{"_account_id":21040,"name":"Alexey","email":"aodinokov@mirantis.com","username":"aodinokov"},"date":"2026-09-08 14:42:37.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"23066fe6ba309ab5edce2c669e02a03ac53cc762","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"date":"2026-09-08 16:42:38.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"f6670ada06e94b8e8017cdc1bb1cf350b6413f15","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-08 18:41:09.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d2369468917249de8195f4763ba63e1c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f43464db959146deae4e16601bbf3c26 : SUCCESS in 3m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7a734bf357814efb9400648296c9550e : SUCCESS in 2m 16s\n- openstack-helm-linter https://zuul.opendev.org/t/openstack/build/e865aab675b14d72883d3e894ca101c0 : SUCCESS in 3m 28s\n- openstack-helm-mariadb-crd-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/7c6358c497f544259d25477a796bd607 : SUCCESS in 1h 14m 28s\n- openstack-helm-mariadb-operator-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/96ca9254286c44e59e13e5b40f563a7d : FAILURE in 1h 54m 13s\n- openstack-helm-rabbitmq-crd-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/5625a5521651431d8b5839a4b5d2d5ce : SUCCESS in 52m 44s\n- openstack-helm-rabbitmq-operator-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/0f617a3157ee4c70997e2a14947be675 : SUCCESS in 47m 28s\n- openstack-helm-pre-commit https://zuul.opendev.org/t/openstack/build/8388e40a10394e89989da0ca06ebafad : SUCCESS in 2m 59s\n- openstack-helm-build-charts https://zuul.opendev.org/t/openstack/build/0b4709b6ffc94bba8b7d074e36324f9c : SUCCESS in 2m 56s\n- openstack-helm-cinder-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/ee276db239b44af190aba64c82c71ffa : SUCCESS in 29m 53s\n- openstack-helm-compute-kit-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/62ce3676564e4ad1a2390959aa0d65cf : SUCCESS in 1h 48m 03s\n- openstack-helm-cinder-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/a5b199cbea7a4f738efdb758022cbe6d : SUCCESS in 42m 59s\n- openstack-helm-compute-kit-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/785d1434d8664812bd8e9bab5d735b6a : SUCCESS in 1h 14m 53s\n- openstack-helm-cinder-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/17500155b4d742858e58e0b9cb88b61f : SUCCESS in 57m 03s\n- openstack-helm-compute-kit-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/194eb3fc8b194132a36b94a2a0c9d3b3 : SUCCESS in 1h 15m 29s\n- openstack-helm-tls-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/d7d6d0c59a1c4659a4472661f390a17f : SUCCESS in 1h 19m 28s\n- openstack-helm-octavia-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/881e6005340d4dec9c7c340f4da549d7 : SUCCESS in 1h 07m 28s\n- openstack-helm-logging https://zuul.opendev.org/t/openstack/build/8b9eb3dfe068492caae7c57ffefec651 : SUCCESS in 28m 28s\n- openstack-helm-monitoring https://zuul.opendev.org/t/openstack/build/e8fd8fcc906f455399114559ac58c16b : SUCCESS in 35m 54s","accounts_in_message":[],"_revision_number":2},{"id":"1801a545d3c8070d0d44d5d8dc192c717ab78d7b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"date":"2026-09-08 18:45:13.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"835246e65aa54803fbc18755005b018542303862","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-08 20:35:53.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c48bb478d5d840e1a6e9fb972d5d18ef\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/294c3e8abd95439fbcbca6432724b9be : SUCCESS in 5m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/586064c8df6f458799d976c6eb3ae6c0 : SUCCESS in 3m 18s\n- openstack-helm-linter https://zuul.opendev.org/t/openstack/build/e2c3808410cb488984fbc8484bd5ecae : SUCCESS in 3m 18s\n- openstack-helm-mariadb-crd-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/dac86f28dcbd4e5283dfdee74429b765 : SUCCESS in 1h 45m 09s\n- openstack-helm-mariadb-operator-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/550d3dc4406745388bd0742b011c4fa1 : SUCCESS in 1h 01m 31s\n- openstack-helm-rabbitmq-crd-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/fd7a17c2739a454bb42f194e0beb06cc : SUCCESS in 59m 25s\n- openstack-helm-rabbitmq-operator-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/b43c84f2bad548cb911951ac109d6c78 : SUCCESS in 46m 36s\n- openstack-helm-pre-commit https://zuul.opendev.org/t/openstack/build/a302bf26437044f7ade3fb68976ce1f1 : SUCCESS in 3m 30s\n- openstack-helm-build-charts https://zuul.opendev.org/t/openstack/build/52615524357f4977802e3ce7e0c7c750 : SUCCESS in 2m 09s\n- openstack-helm-cinder-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/bb4e12076170495a8046e4d36d482146 : SUCCESS in 1h 11m 03s\n- openstack-helm-compute-kit-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/a131d43c6a2a4665999c4dd246254886 : SUCCESS in 44m 21s\n- openstack-helm-cinder-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/00204bc87e3d47c9afd487f7df138ee0 : SUCCESS in 44m 18s\n- openstack-helm-compute-kit-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/ab824ed5215c4d398be91e9a11522453 : SUCCESS in 48m 05s\n- openstack-helm-cinder-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/f520472d5c9c45ceb2fdd99365a01bf6 : SUCCESS in 28m 33s\n- openstack-helm-compute-kit-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/0bdf4e062e1e4a67884852b09112a48c : SUCCESS in 1h 20m 17s\n- openstack-helm-tls-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/cf93c32e01434a759ebda42346fbd6e6 : SUCCESS in 1h 38m 56s\n- openstack-helm-octavia-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/0268bcb2c2ac46febbbd9e8002ecce77 : FAILURE in 1h 43m 25s\n- openstack-helm-logging https://zuul.opendev.org/t/openstack/build/972950fa59374c9cb9e3a3d4e198409f : SUCCESS in 38m 24s\n- openstack-helm-monitoring https://zuul.opendev.org/t/openstack/build/3a8259b74fc5412a9b60dab367a50898 : SUCCESS in 58m 07s","accounts_in_message":[],"_revision_number":3},{"id":"7b0d67aa6dbb023c1e09cb018d2ed9bd0b93d9d4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"date":"2026-09-08 20:41:37.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"b2d7860e27a68a235577332b038385d036ecd47b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"date":"2026-09-08 21:49:36.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"f6f6b95ee6894dcb4758bdde83716b1a75070a66","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-08 23:38:00.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cf9dde4170db473fba145a22b592a450\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/63fe4675c9324eac86e5fb6bdabb1787 : SUCCESS in 3m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a58a4ba21be14736b3c3ba00831945f7 : SUCCESS in 2m 34s\n- openstack-helm-linter https://zuul.opendev.org/t/openstack/build/b9ef01878c2f4af69066ce92339dc760 : SUCCESS in 2m 21s\n- openstack-helm-mariadb-crd-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/7b60de9f99c941d8b77bfae50fa7f9d4 : SUCCESS in 1h 03m 54s\n- openstack-helm-mariadb-operator-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/db6d77b38a9449eaa2acb89928ff49ef : SUCCESS in 1h 01m 11s\n- openstack-helm-rabbitmq-crd-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/f51568f636a947be874a4ad3bfa11a8d : SUCCESS in 1h 02m 19s\n- openstack-helm-rabbitmq-operator-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/d39794b9c1494862b811b997e5bd61fe : SUCCESS in 1h 23m 28s\n- openstack-helm-pre-commit https://zuul.opendev.org/t/openstack/build/437f3ff8601f40c7a038a1a73698957e : SUCCESS in 3m 15s\n- openstack-helm-build-charts https://zuul.opendev.org/t/openstack/build/de7fa062b6254eb5a9097db6d07e97fe : SUCCESS in 2m 07s\n- openstack-helm-cinder-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/7a55891e9fc34ac58a035311410c1d8f : SUCCESS in 40m 34s\n- openstack-helm-compute-kit-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/22216964f192455180ee4f1b7c474be7 : SUCCESS in 1h 14m 13s\n- openstack-helm-cinder-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/21d1d09ddacd463c94beb753a7f7e779 : SUCCESS in 45m 15s\n- openstack-helm-compute-kit-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/64b63d194bd84b93895fd03b3228b25b : SUCCESS in 48m 21s\n- openstack-helm-cinder-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/37addcfaba6a420dbf7d6ee4ee0119bc : SUCCESS in 26m 27s\n- openstack-helm-compute-kit-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/288c00474c0b403188bc505fc46e9b41 : SUCCESS in 1h 17m 38s\n- openstack-helm-tls-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/70a97f54e6e143668cf8f7b0b51c1ee6 : SUCCESS in 1h 20m 04s\n- openstack-helm-octavia-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/836b6de2d7f247068b6bb172631b8795 : SUCCESS in 1h 05m 20s\n- openstack-helm-logging https://zuul.opendev.org/t/openstack/build/20fb81bbc72a45148dc966456ac85847 : SUCCESS in 35m 08s\n- openstack-helm-monitoring https://zuul.opendev.org/t/openstack/build/4d07e5c5ce0945ca93570dfd378f6327 : SUCCESS in 20m 51s","accounts_in_message":[],"_revision_number":5},{"id":"aa63bc1ac0c5ef56a5d605e30db85d92db6302fe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"date":"2026-09-09 12:41:42.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":6},{"id":"c568fd39537ba45d1389520e0c90e0de3d7a861e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-09 15:36:46.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c4feac406142461abaca83519b7ca561\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/256390ff41234746889d12445427aa70 : SUCCESS in 3m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8ea67704109446b5b7a160723fdea738 : SUCCESS in 2m 09s\n- openstack-helm-linter https://zuul.opendev.org/t/openstack/build/4555a1cae16f44829f122a101c6c31f5 : SUCCESS in 2m 38s\n- openstack-helm-mariadb-crd-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/2225ff26a63c4d8695808a618efaf869 : FAILURE in 1h 55m 52s\n- openstack-helm-mariadb-operator-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/d1732ae1b316491880bb87d3be544d06 : FAILURE in 2h 42m 05s\n- openstack-helm-rabbitmq-crd-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/1efc698f50ec4fba857d62413036d380 : SUCCESS in 1h 22m 05s\n- openstack-helm-rabbitmq-operator-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/228b924fe8534d58abb9d4ad7fd23f5a : SUCCESS in 1h 15m 08s\n- openstack-helm-pre-commit https://zuul.opendev.org/t/openstack/build/9145316eded2415ab3545a696d531156 : SUCCESS in 2m 18s\n- openstack-helm-build-charts https://zuul.opendev.org/t/openstack/build/5fa26fb64d99484e9385a9a24bccb699 : SUCCESS in 3m 48s\n- openstack-helm-cinder-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/51adbb0fdd47400ea3c4dcf69bd5b783 : SUCCESS in 43m 12s\n- openstack-helm-compute-kit-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/4327234a49a2475eaefcf6bd3160a18b : SUCCESS in 1h 10m 48s\n- openstack-helm-cinder-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/44648c405bc344b99d0de2577d504d6a : FAILURE in 21m 21s\n- openstack-helm-compute-kit-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/1466faf022f048909112b3fbeeb3dc5a : SUCCESS in 56m 20s\n- openstack-helm-cinder-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/6483d4000125469286d324fe54cc01e2 : SUCCESS in 32m 53s\n- openstack-helm-compute-kit-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/88740db27aeb47e9b218b9165061d741 : SUCCESS in 1h 08m 48s\n- openstack-helm-tls-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/4de39880f7c94b88ab55a1ed918eee9e : FAILURE in 2h 47m 18s\n- openstack-helm-octavia-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/f33120fcba164d2fb65735b1ad706182 : SUCCESS in 1h 21m 03s\n- openstack-helm-logging https://zuul.opendev.org/t/openstack/build/245cf3986e0d4e9aa1e877730e3736af : SUCCESS in 44m 42s\n- openstack-helm-monitoring https://zuul.opendev.org/t/openstack/build/9770f6fbc9614f1486a00f1bb18c7ff4 : SUCCESS in 32m 23s","accounts_in_message":[],"_revision_number":6},{"id":"50cef26394938c5e9ebdb14e262691ae685de0a8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"real_author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2026-09-09 19:47:51.000000000","message":"Patch Set 7: Patch Set 6 was rebased on behalf of \u003cGERRIT_ACCOUNT_29974\u003e\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"}],"_revision_number":7},{"id":"d3133d1dcbbf44ac1b2c8f26e4c2be15891763de","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-09 21:33:56.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f4130cddf42741bfb8a2b4cb75a47163\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1db1d61e1f4f428299fbd6fbf451d494 : SUCCESS in 4m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2a71f253354a4f1487c77142e7febae9 : SUCCESS in 4m 26s\n- openstack-helm-linter https://zuul.opendev.org/t/openstack/build/84121768cbc24138886fb1ed7337a51a : SUCCESS in 3m 35s\n- openstack-helm-mariadb-crd-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/19687104be0641a9a46b084cead01577 : SUCCESS in 1h 38m 56s\n- openstack-helm-mariadb-operator-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/d5da24a88220433e99086124fd5ec47e : SUCCESS in 1h 38m 02s\n- openstack-helm-rabbitmq-crd-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/c0688a219913435aa857ab1f0aa0686a : SUCCESS in 1h 21m 43s\n- openstack-helm-rabbitmq-operator-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/1e5eb2188aec473fbc1eebea875e4aaf : SUCCESS in 50m 52s\n- openstack-helm-pre-commit https://zuul.opendev.org/t/openstack/build/fa4a3c5cb3b64c3ba3a89708bcc5bc85 : SUCCESS in 2m 47s\n- openstack-helm-build-charts https://zuul.opendev.org/t/openstack/build/017dd74b767144eb95e65057282c2dc9 : SUCCESS in 2m 11s\n- openstack-helm-cinder-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/a78a1a7c315b439aa46a8745a280915f : SUCCESS in 38m 40s\n- openstack-helm-compute-kit-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/ed0ae51764514bca8f016a4fe18080cc : SUCCESS in 1h 15m 56s\n- openstack-helm-cinder-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/72e007cbb7784b47a6701fef61195586 : SUCCESS in 57m 30s\n- openstack-helm-compute-kit-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/7adf5f229d534de19ec3746253454dd1 : SUCCESS in 50m 05s\n- openstack-helm-cinder-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/387dffaf745847c9b7e8ff179f52926d : SUCCESS in 44m 11s\n- openstack-helm-compute-kit-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/0b57af9fdf5d47a0acff26a9b0279470 : SUCCESS in 1h 09m 18s\n- openstack-helm-tls-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/46642f318fd3403cb9e588f802fa0158 : SUCCESS in 1h 04m 44s\n- openstack-helm-octavia-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/2d2dd29fd0484be4813edf8b7606b7a8 : SUCCESS in 1h 06m 43s\n- openstack-helm-logging https://zuul.opendev.org/t/openstack/build/ee862bdfddd24986ac17ef7cd4308647 : SUCCESS in 38m 43s\n- openstack-helm-monitoring https://zuul.opendev.org/t/openstack/build/db990343510d48299549d82c996c8789 : SUCCESS in 20m 54s","accounts_in_message":[],"_revision_number":7},{"id":"f81ab34206904a30cf5a99f2aa7c0601962cf58a","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2026-09-09 21:37:54.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"e4f4ca69156e57d83948a21bb41876cc5d61cf67","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"date":"2026-09-10 15:07:26.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":7},{"id":"a26573158969050fd0996c1f7d256a2e63ec1a3f","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"date":"2026-09-10 15:21:59.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":8},{"id":"61abfeff3c7a8a1bae057b41082057742ad3509a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-10 17:03:02.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0daed7e868564ad9957406c0e43ea3e7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e3fba771cac4a85a813865cab1da007 : SUCCESS in 2m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e38e135c354241f08a1f585b9e0bfb39 : SUCCESS in 2m 35s\n- openstack-helm-linter https://zuul.opendev.org/t/openstack/build/5801ace44cbd4b7d9ca7dad10063238d : SUCCESS in 3m 46s\n- openstack-helm-mariadb-crd-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/62230c4302624a9a9b603ee8d6a2ba73 : SUCCESS in 56m 21s\n- openstack-helm-mariadb-operator-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/d31d8fa86db6403ebb486cfac4dd9d6a : SUCCESS in 1h 35m 04s\n- openstack-helm-rabbitmq-crd-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/7335e01d373b4d6591676b14441ac45f : SUCCESS in 1h 16m 01s\n- openstack-helm-rabbitmq-operator-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/4868d7c4b5ec4e72924600d8e249992e : SUCCESS in 1h 17m 30s\n- openstack-helm-pre-commit https://zuul.opendev.org/t/openstack/build/141280500abc474db18a1a04552f602e : SUCCESS in 4m 27s\n- openstack-helm-build-charts https://zuul.opendev.org/t/openstack/build/c5374dc0178542c8abd1c4b4c702a0ae : SUCCESS in 1m 42s\n- openstack-helm-cinder-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/6bede2ddaa5e41bf9340a63608d5dc72 : SUCCESS in 28m 45s\n- openstack-helm-compute-kit-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/fc24d73e8e49403b87add9613d2e5b7a : SUCCESS in 1h 24m 19s\n- openstack-helm-cinder-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/d0e8f1ee96af4142911ea6a632fac076 : SUCCESS in 26m 31s\n- openstack-helm-compute-kit-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/55740dce218648099e557d4a5b68859f : SUCCESS in 1h 16m 58s\n- openstack-helm-cinder-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/f0e1a1bda3164f8684d39cc30acd83bc : SUCCESS in 46m 35s\n- openstack-helm-compute-kit-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/e085715b6fb642f7a78f0985266338af : SUCCESS in 1h 15m 00s\n- openstack-helm-tls-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/8ee333ecdb66416a98e5870865e6003e : FAILURE in 1h 34m 01s\n- openstack-helm-octavia-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/a130f4878fff4acfa16a9d86432a5079 : FAILURE in 1h 01m 11s\n- openstack-helm-logging https://zuul.opendev.org/t/openstack/build/e5fbe6fd847b46f396782ca04ed3577e : SUCCESS in 40m 31s\n- openstack-helm-monitoring https://zuul.opendev.org/t/openstack/build/376cda74dc2048f48319c288459a914b : SUCCESS in 29m 31s","accounts_in_message":[],"_revision_number":8},{"id":"fd6ac86be9c049bbf35b74cb4c1583c8f1dbd253","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"date":"2026-09-10 20:11:34.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":9},{"id":"3f0680c4afcd97e29c98f889f0cb6f3b2d209f5f","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"date":"2026-09-10 20:21:49.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"85668e2b987b6f566a791512dd33eec2b3120e8b","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"date":"2026-09-10 20:56:45.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"191d7c6332348e0e88a7f942435bf08f92b2e94a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-10 22:45:56.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/08cad6d40e8645b59807fd7ac2bb28d7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/226daa5c2069408886492a90a386c0eb : SUCCESS in 3m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0f93ac73d7b641a292e7c9df7f7889f3 : SUCCESS in 3m 33s\n- openstack-helm-linter https://zuul.opendev.org/t/openstack/build/88e55c6d8a6f4195bd2cb8e0291d725c : SUCCESS in 2m 30s\n- openstack-helm-mariadb-crd-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/dd38b1225fd44247a758d945460996f2 : SUCCESS in 1h 32m 11s\n- openstack-helm-mariadb-operator-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/32286af8d7054524b93d2e06791d351a : SUCCESS in 1h 43m 04s\n- openstack-helm-rabbitmq-crd-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/eeae9a3a34374c858a6ce34ec4de09ac : SUCCESS in 1h 15m 56s\n- openstack-helm-rabbitmq-operator-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/1438d393c6724dc0a3d5fd5bd450f993 : SUCCESS in 1h 21m 51s\n- openstack-helm-pre-commit https://zuul.opendev.org/t/openstack/build/8622e1f0b6ee46ad82f72296eef43fd5 : SUCCESS in 2m 47s\n- openstack-helm-build-charts https://zuul.opendev.org/t/openstack/build/c953a307c96f4890a32e0163d6fbd9c5 : SUCCESS in 4m 54s\n- openstack-helm-cinder-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/01ddc80722d04be1b703e8a12fd27c92 : SUCCESS in 29m 42s\n- openstack-helm-compute-kit-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/c3a098294df54982b99ebdf1778ee59e : SUCCESS in 1h 16m 19s\n- openstack-helm-cinder-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/70fed4060bba4d599edbfacfdc37664d : SUCCESS in 37m 20s\n- openstack-helm-compute-kit-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/21327d405889416ba0a4a5471932924c : SUCCESS in 1h 13m 31s\n- openstack-helm-cinder-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/9b1d5e45d5b9403f8636bf4e12af1db4 : SUCCESS in 30m 45s\n- openstack-helm-compute-kit-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/a0ca20fd1a734befbdc8714b7e62aa1b : SUCCESS in 1h 04m 46s\n- openstack-helm-tls-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/c3f11b1a694045faae7303a8367daa61 : SUCCESS in 1h 36m 14s\n- openstack-helm-octavia-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/bb4d3195c9eb426186d9f3eef2458591 : FAILURE in 1h 28m 54s\n- openstack-helm-logging https://zuul.opendev.org/t/openstack/build/17323e6a70fa41cdad300f238e7607fc : SUCCESS in 39m 14s\n- openstack-helm-monitoring https://zuul.opendev.org/t/openstack/build/aad2df75941a435f8b6dd89c46b2c500 : SUCCESS in 29m 46s","accounts_in_message":[],"_revision_number":11},{"id":"87fd23c9fbe65450b151fcc0b78fb260a90339e5","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"date":"2026-09-11 15:06:24.000000000","message":"Uploaded patch set 12.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":12},{"id":"ffab81ddaf29ad1fdf904a52a66b00a0109a32c7","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"date":"2026-09-11 15:06:48.000000000","message":"Patch Set 13: Patch Set 12 was rebased","accounts_in_message":[],"_revision_number":13},{"id":"06ab5c3b321b5557219d18ba6e440908f2062fb3","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"date":"2026-09-11 16:13:12.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"2ab24de99184c1f0470f12ec81700a403cea4360","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"date":"2026-09-11 16:20:14.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"6324fbf162f9fd6071dc6c58ac9e41423c3cf3bf","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"date":"2026-09-11 16:36:23.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"09a11ea5a2b73466a647b4be2a02022002413af9","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"date":"2026-09-11 18:14:35.000000000","message":"Uploaded patch set 17.","accounts_in_message":[],"_revision_number":17},{"id":"c67e958b1a4e88f79ab2fa777130a2b94cf13174","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-11 19:54:19.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d603dd4d24ed4278ad5537d840455326\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b951e44090f42c7ad88fbab8ce5d116 : SUCCESS in 2m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d5d97010386548f1bed6502d02282dca : SUCCESS in 2m 09s\n- openstack-helm-linter https://zuul.opendev.org/t/openstack/build/dc25851cd39a4941a9e753541d7f5ff7 : SUCCESS in 1m 58s\n- openstack-helm-mariadb-crd-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/4509931614b8406489548da21c0019a5 : SUCCESS in 58m 03s\n- openstack-helm-mariadb-operator-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/f0ea228c0b0f491d93b6816af3688dba : SUCCESS in 54m 54s\n- openstack-helm-rabbitmq-crd-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/fa0857078e9b4403850263a8b4a2d7bd : SUCCESS in 43m 51s\n- openstack-helm-rabbitmq-operator-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/eb46a1f3d3cc4fdb8633f5212a1f543c : SUCCESS in 1h 16m 29s\n- openstack-helm-pre-commit https://zuul.opendev.org/t/openstack/build/a35e53bd3d014858adac887a8ebabbf9 : SUCCESS in 2m 54s\n- openstack-helm-build-charts https://zuul.opendev.org/t/openstack/build/0de2dce2de7448b786517173dfb60bfc : SUCCESS in 37m 55s\n- openstack-helm-cinder-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/699da3dd8e5c45d89f279a2fd69ab3da : SUCCESS in 38m 30s\n- openstack-helm-compute-kit-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/92bb4f3822834342bd0a29d6b22659eb : SUCCESS in 52m 38s\n- openstack-helm-cinder-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/9e81cac1a1504ad8aed15479aa9b836b : SUCCESS in 31m 30s\n- openstack-helm-compute-kit-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/84b2bf70d4c14e2ebbfb60a912bf2bdf : SUCCESS in 46m 32s\n- openstack-helm-cinder-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/bd2ff737895b42fbac6919cb5310fa2e : SUCCESS in 43m 14s\n- openstack-helm-compute-kit-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/91a261d81931423484f7470b63bfd079 : SUCCESS in 1h 12m 07s\n- openstack-helm-tls-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/9f87daf699a642548cf2b916910efca3 : SUCCESS in 1h 35m 34s\n- openstack-helm-octavia-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/fbd84d6a921c49e4a64cc83f0164817c : SUCCESS in 1h 17m 40s\n- openstack-helm-logging https://zuul.opendev.org/t/openstack/build/fc45e2203a6d4a0f8e1d3297b98f6b91 : SUCCESS in 39m 35s\n- openstack-helm-monitoring https://zuul.opendev.org/t/openstack/build/77fed92c601543d081ec9b5356d8b3cf : SUCCESS in 20m 52s","accounts_in_message":[],"_revision_number":17},{"id":"acd75f7f7fecda17cb6406e1edd311ee2aa7622b","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"date":"2026-09-11 20:52:48.000000000","message":"Uploaded patch set 18.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":18},{"id":"f4de797cbe10fe00819bfb69bda8db009fadb575","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-11 22:44:27.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d1479a62db2441b38e24dd3529081091\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d7d42416bf7342899ba20954aaf7ec06 : SUCCESS in 5m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bdddda05d7624aedb8770b12dcd5e0da : SUCCESS in 2m 18s\n- openstack-helm-linter https://zuul.opendev.org/t/openstack/build/51ed3e664f93499faa8a6b59b038b377 : SUCCESS in 3m 28s\n- openstack-helm-mariadb-crd-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/f1e6526b83a24678a5777f73cfe7ee0a : SUCCESS in 1h 35m 02s\n- openstack-helm-mariadb-operator-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/822806663d114da3aa8582ce882bdaee : SUCCESS in 57m 20s\n- openstack-helm-rabbitmq-crd-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/3ef1db7298ee4fb78fbef9e89d35afdd : SUCCESS in 44m 16s\n- openstack-helm-rabbitmq-operator-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/ce13f4d0786b4c2aaf3bac8330d05e0f : SUCCESS in 1h 14m 44s\n- openstack-helm-pre-commit https://zuul.opendev.org/t/openstack/build/12be0cd552084112948a25f14dcdecfb : SUCCESS in 1m 59s\n- openstack-helm-build-charts https://zuul.opendev.org/t/openstack/build/b5e85fe46f85463e98b47b443c55d803 : SUCCESS in 34m 17s\n- openstack-helm-cinder-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/d5bcdf6b67824500bf93bc371b011422 : SUCCESS in 26m 31s\n- openstack-helm-compute-kit-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/a1b050cbd0ae4bdc8bdd6c341be13e29 : SUCCESS in 1h 13m 59s\n- openstack-helm-cinder-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/91b53b7e0fcd4cb6916574d31f1581db : SUCCESS in 48m 08s\n- openstack-helm-compute-kit-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/defee8d868f04b8685b3ab9444fe5d18 : SUCCESS in 1h 16m 51s\n- openstack-helm-cinder-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/1be2f076db72448ebba0a27b333547e7 : SUCCESS in 37m 43s\n- openstack-helm-compute-kit-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/36fd7e57d599466681a318762e5b0238 : SUCCESS in 1h 20m 53s\n- openstack-helm-tls-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/b19f021864344c8281ddd1e863a5fe0e : SUCCESS in 1h 45m 34s\n- openstack-helm-octavia-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/f5b3e44d63894989918f9e6067f12441 : SUCCESS in 1h 18m 26s\n- openstack-helm-logging https://zuul.opendev.org/t/openstack/build/d62cc1d07ed2491c9d5b5fe140aa601d : SUCCESS in 24m 44s\n- openstack-helm-monitoring https://zuul.opendev.org/t/openstack/build/95417c8af4424d2fb451991f7e062400 : SUCCESS in 28m 05s","accounts_in_message":[],"_revision_number":18},{"id":"6a08b8a834b08604ad9a1fb8fa73525e6f48229b","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"date":"2026-09-14 18:24:07.000000000","message":"Uploaded patch set 19.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":19},{"id":"10d69d1e09d3943aefcbfcffa71d2959678fb8e6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-14 19:09:39.000000000","message":"Patch Set 19: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/39186154bf4f4c708f3e56137f2edd77\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef95d333703f4f68a377dddcd20fd0ff : SUCCESS in 2m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9b4c99205d3045b7bb2c4d6a45116ddb : SUCCESS in 3m 25s\n- openstack-helm-linter https://zuul.opendev.org/t/openstack/build/bd05e18a8e234b5294c4caa4aa0dab3c : SUCCESS in 2m 59s\n- openstack-helm-mariadb-crd-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/809290f4fb3049b3badeeefec145d824 : FAILURE in 8m 39s\n- openstack-helm-mariadb-operator-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/391752af56674fb2a05c837afae96f42 : FAILURE in 11m 12s\n- openstack-helm-rabbitmq-crd-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/337474c19cd54436a77b3c0e0bbdb7f6 : FAILURE in 7m 21s\n- openstack-helm-rabbitmq-operator-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/9d76cbf2e14e4bff8079a663749c0b22 : FAILURE in 11m 25s\n- openstack-helm-pre-commit https://zuul.opendev.org/t/openstack/build/61861583979e4c82b0843fbdbec46bda : SUCCESS in 2m 47s\n- openstack-helm-build-charts https://zuul.opendev.org/t/openstack/build/b73633e0c7914b36998d2e13c9232ec7 : SUCCESS in 43m 38s\n- openstack-helm-cinder-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/bcab41677b9b4594a0776782bcec243e : FAILURE in 10m 38s\n- openstack-helm-compute-kit-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/c4793f75e189476da3fde53cd9dc4010 : FAILURE in 12m 24s\n- openstack-helm-cinder-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/ebd578e574fd45a2a781d5d16cf9bab5 : FAILURE in 12m 42s\n- openstack-helm-compute-kit-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/43381fefa324474fb6427303daf292d6 : FAILURE in 13m 41s\n- openstack-helm-cinder-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/a350c4168cd8441b9c6872f288f0fd75 : FAILURE in 12m 22s\n- openstack-helm-compute-kit-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/d2c178717502499d9a69a02cfe53d389 : FAILURE in 11m 52s\n- openstack-helm-tls-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/92dd01a5c01e4ddb8e1bed16e0d42afd : FAILURE in 9m 24s\n- openstack-helm-octavia-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/3eb8485283db4c66bc52cfe4dfe9206f : FAILURE in 11m 27s\n- openstack-helm-logging https://zuul.opendev.org/t/openstack/build/d0e3c43d604740f6a80c21e605c4819d : FAILURE in 10m 05s\n- openstack-helm-monitoring https://zuul.opendev.org/t/openstack/build/10dd743fd2d540789af042d2e8b63bd3 : FAILURE in 9m 46s","accounts_in_message":[],"_revision_number":19},{"id":"d160531306368eb5b207714c97329aec80af44ff","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"date":"2026-09-15 14:29:49.000000000","message":"Uploaded patch set 20.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":20},{"id":"57fca278079624faaaaab4058fbdc18e1f614847","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-15 16:16:19.000000000","message":"Patch Set 20: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/98cd7e97426642b4bc1198a90d824bf9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/677b5c12acc9499b8058c702fc054c0b : SUCCESS in 3m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7a2e5a3df30b44ec81645da7c7bdc40a : SUCCESS in 3m 44s\n- openstack-helm-linter https://zuul.opendev.org/t/openstack/build/336b646275714aa9982f729d1e6ea6a3 : SUCCESS in 3m 14s\n- openstack-helm-mariadb-crd-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/3f65f26290b944b281f383d2fd2b885e : SUCCESS in 1h 40m 03s\n- openstack-helm-mariadb-operator-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/023cfd7b93ed4a05a284771a6f1981b1 : SUCCESS in 1h 25m 26s\n- openstack-helm-rabbitmq-crd-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/fdb75739b5a94cc996497b8c26dcdf84 : SUCCESS in 44m 43s\n- openstack-helm-rabbitmq-operator-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/cff7cbc792d14e3da05f9ebf099dc08f : SUCCESS in 1h 02m 53s\n- openstack-helm-pre-commit https://zuul.opendev.org/t/openstack/build/f869b27624fd4372a9990767c6d26923 : SUCCESS in 2m 50s\n- openstack-helm-build-charts https://zuul.opendev.org/t/openstack/build/ef4fa4061ddc49ae848ae27f3cadc536 : SUCCESS in 17m 25s\n- openstack-helm-cinder-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/b712b88b9e7847758b59a95c3b594728 : SUCCESS in 25m 53s\n- openstack-helm-compute-kit-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/1086a589ee684b38b85b868620cf8af8 : SUCCESS in 1h 02m 52s\n- openstack-helm-cinder-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/d11c43932a7a4bc1a9cf85a714462bc2 : SUCCESS in 35m 38s\n- openstack-helm-compute-kit-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/65f5aa4b5f6547b69550f5cad3fb7a33 : SUCCESS in 1h 15m 19s\n- openstack-helm-cinder-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/42725600d4894515828232ad6296dcfa : SUCCESS in 44m 39s\n- openstack-helm-compute-kit-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/e8e48b4a8f9143708861fb29ceb5d959 : SUCCESS in 47m 07s\n- openstack-helm-tls-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/70d452e7868c4501bf81947112738814 : SUCCESS in 1h 30m 47s\n- openstack-helm-octavia-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/2ac7d9e238f64e92b9ecfdd1107ede0d : SUCCESS in 1h 18m 46s\n- openstack-helm-logging https://zuul.opendev.org/t/openstack/build/276d52edf94645b290d2a1d18e3715ea : SUCCESS in 25m 56s\n- openstack-helm-monitoring https://zuul.opendev.org/t/openstack/build/e10d3e497e634acfb437ce991f26b4d3 : SUCCESS in 26m 26s","accounts_in_message":[],"_revision_number":20},{"id":"fab1e1b5a111cbf5a1e365872458273c9cd7927e","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"date":"2026-09-15 17:46:05.000000000","message":"Uploaded patch set 21.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":21},{"id":"7907f3e79e17bf505c38bc9a322d2c4801e8042f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-15 19:26:04.000000000","message":"Patch Set 21: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/43b3ced242484561bb29206b928b75ba\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9cce8ef9418241d7a47cca7b1c489cdb : SUCCESS in 3m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e8f2842d0a454b1cab2bdb7d6c0c1193 : SUCCESS in 3m 38s\n- openstack-helm-linter https://zuul.opendev.org/t/openstack/build/f9acf6afa937417ca13fa7b1967de19a : SUCCESS in 3m 28s\n- openstack-helm-mariadb-crd-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/d37f798bcba34a8ab817d31432334119 : SUCCESS in 1h 01m 01s\n- openstack-helm-mariadb-operator-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/7f53d424f71143f4ad1f7973e4cb7d8a : SUCCESS in 54m 53s\n- openstack-helm-rabbitmq-crd-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/4c9dfefbef0c46479bacf3bfc06c89bf : SUCCESS in 1h 17m 05s\n- openstack-helm-rabbitmq-operator-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/7ad28781c1b748ca833b376e2fd8b3e5 : SUCCESS in 1h 03m 28s\n- openstack-helm-pre-commit https://zuul.opendev.org/t/openstack/build/beb1cf8df16d4378b759b0694eb9b843 : SUCCESS in 2m 52s\n- openstack-helm-build-charts https://zuul.opendev.org/t/openstack/build/cd30ef30e6f04516a2d35d6eef8c5376 : SUCCESS in 32m 30s\n- openstack-helm-cinder-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/9e133dd660804b82b6d27d1fca0e9098 : SUCCESS in 26m 11s\n- openstack-helm-compute-kit-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/5d89a504372e445cbd80c01dddbe3374 : SUCCESS in 1h 13m 08s\n- openstack-helm-cinder-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/b3a84acc3de44be58d962ffcb56be480 : SUCCESS in 47m 35s\n- openstack-helm-compute-kit-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/d73d7225da454ae6b5beb6ae76d53966 : SUCCESS in 1h 22m 48s\n- openstack-helm-cinder-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/3b57b4b441a14965bb98f5bedb103162 : SUCCESS in 43m 49s\n- openstack-helm-compute-kit-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/60cad0aa0fa047db9f9d4ae7987a8a53 : SUCCESS in 48m 21s\n- openstack-helm-tls-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/4050aff2495e47c58acff6e41e62ee94 : SUCCESS in 1h 22m 03s\n- openstack-helm-octavia-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/e72e20aa727143d68b5482cf438bf321 : SUCCESS in 43m 55s\n- openstack-helm-logging https://zuul.opendev.org/t/openstack/build/9d2c7fce459b44518a54282f09c66245 : FAILURE in 25m 29s\n- openstack-helm-monitoring https://zuul.opendev.org/t/openstack/build/82ec7e0136d54d0992dd8ae62eadc62d : SUCCESS in 21m 14s","accounts_in_message":[],"_revision_number":21},{"id":"35d2c88790e2416f7c7680723c5948450f4119f3","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"date":"2026-09-15 20:19:30.000000000","message":"Uploaded patch set 22.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":22},{"id":"47818c7861ee103baf2ae46bb9766c70ad16dac5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-15 22:18:52.000000000","message":"Patch Set 22: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/91ce97c3cb0247e2958436edc9bf0b6d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d04a143c63ad4507b14a29c0e8f2657d : SUCCESS in 3m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/16a66b3ccd1a409c9940eab5f75e94f5 : SUCCESS in 3m 28s\n- openstack-helm-linter https://zuul.opendev.org/t/openstack/build/af6eb87887fd43eeb8b00fd40793a29b : SUCCESS in 3m 18s\n- openstack-helm-mariadb-crd-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/ac0a9c7c6c1c4229a63c14b133e6173e : SUCCESS in 1h 36m 58s\n- openstack-helm-mariadb-operator-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/d3eef84f12e24b2f96971520e14500fa : SUCCESS in 1h 36m 05s\n- openstack-helm-rabbitmq-crd-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/0df206b0e9db4351b0ec44b15f1db9f2 : SUCCESS in 1h 11m 17s\n- openstack-helm-rabbitmq-operator-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/50c72bd17a4d40688ea5c098c8bb2ac6 : SUCCESS in 1h 08m 53s\n- openstack-helm-pre-commit https://zuul.opendev.org/t/openstack/build/2ecf516204d84294b7f7583f8e781c9e : SUCCESS in 1m 51s\n- openstack-helm-build-charts https://zuul.opendev.org/t/openstack/build/64c55076b0f44e96a3f6d3472898c5be : SUCCESS in 48m 28s\n- openstack-helm-cinder-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/e18436db0fc04164a450ff6a71ea8c81 : SUCCESS in 42m 42s\n- openstack-helm-compute-kit-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/1e4023ab73554f6bb93836fde1166beb : SUCCESS in 1h 12m 04s\n- openstack-helm-cinder-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/58d92bf72e05437ba926c932753d27b1 : SUCCESS in 28m 12s\n- openstack-helm-compute-kit-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/b5c6ce865ecb408cafe2525bbfd45470 : SUCCESS in 1h 09m 09s\n- openstack-helm-cinder-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/9c4e33307a3e4d8594b98f49a382223f : SUCCESS in 46m 49s\n- openstack-helm-compute-kit-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/8366befae4584d98818c7c54e2cde177 : SUCCESS in 47m 54s\n- openstack-helm-tls-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/3715316e422344759ae207e1a45d5fc8 : SUCCESS in 1h 44m 17s\n- openstack-helm-octavia-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/38ab246dd5ce45e7871c91fb28fd8612 : SUCCESS in 1h 04m 24s\n- openstack-helm-logging https://zuul.opendev.org/t/openstack/build/29ddfc2a5c0e4591b97a0e0304df37ef : SUCCESS in 37m 03s\n- openstack-helm-monitoring https://zuul.opendev.org/t/openstack/build/e70f1c354abc40f58c8d55ee2d32fa90 : SUCCESS in 32m 11s","accounts_in_message":[],"_revision_number":22},{"id":"5984868fc8ce6ce77849c1e2581cfe1ea63303fb","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"date":"2026-09-16 12:17:46.000000000","message":"Uploaded patch set 23.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":23},{"id":"bab18feae9b9d0f5e8c3e0d53203191f072f329d","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"date":"2026-09-16 14:03:41.000000000","message":"Uploaded patch set 24.","accounts_in_message":[],"_revision_number":24},{"id":"8495edd7b4b6a618b192da50c1327ac43caf7aae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-16 15:57:22.000000000","message":"Patch Set 24: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/452cb55cd8774f578f78636b7a8e0528\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e7583611ef9442c1b40d6f616dc081f9 : SUCCESS in 4m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7cc9bc03c3da42eaa6c33459ca66d0f3 : SUCCESS in 5m 48s\n- openstack-helm-linter https://zuul.opendev.org/t/openstack/build/767c2801937644a6a153772adaa02d1f : SUCCESS in 4m 53s\n- openstack-helm-mariadb-crd-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/216efd3190e740cd93ea72d335647a3f : SUCCESS in 1h 43m 14s\n- openstack-helm-mariadb-operator-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/336368c576cb4929b80aef541d9cdf2b : SUCCESS in 1h 30m 24s\n- openstack-helm-rabbitmq-crd-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/cc34de8664de4f1482506b0b9733b9b7 : SUCCESS in 1h 15m 14s\n- openstack-helm-rabbitmq-operator-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/f943868b9b5d42eaa2cbcf4bf2bf2d68 : SUCCESS in 1h 11m 57s\n- openstack-helm-pre-commit https://zuul.opendev.org/t/openstack/build/bb26fbe8773c467f9688d119e80d2572 : SUCCESS in 3m 14s\n- openstack-helm-build-charts https://zuul.opendev.org/t/openstack/build/8da347c04c6f43e1a74ce8140f8baffe : SUCCESS in 48m 19s\n- openstack-helm-cinder-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/05853ad6a64547e7b41a3335eeab182a : SUCCESS in 47m 47s\n- openstack-helm-compute-kit-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/b51f42a48812434a9f552d9c33a86e8f : SUCCESS in 1h 17m 01s\n- openstack-helm-cinder-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/105f40f4c16f4ddfa80400e21ad2ddad : SUCCESS in 47m 31s\n- openstack-helm-compute-kit-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/e85ed78ecb994568af8c6a5c750987d1 : SUCCESS in 1h 19m 04s\n- openstack-helm-cinder-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/5d00a3bf9510474caf63fdc737f1ef9e : SUCCESS in 35m 12s\n- openstack-helm-compute-kit-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/918af31b15234a7bb9e00d5b1d21d4b2 : SUCCESS in 1h 16m 43s\n- openstack-helm-tls-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/44625f7aa9f9432dace2550eca8e2cee : FAILURE in 1h 12m 43s\n- openstack-helm-octavia-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/2f2c16989d5b4afe9704c11b647c3a72 : SUCCESS in 1h 26m 10s\n- openstack-helm-logging https://zuul.opendev.org/t/openstack/build/151f189d1f114d4e8c5e1848ee1e075a : SUCCESS in 36m 04s\n- openstack-helm-monitoring https://zuul.opendev.org/t/openstack/build/02f375afca9b43fdaaf605620853f90d : SUCCESS in 30m 18s","accounts_in_message":[],"_revision_number":24},{"id":"a309290037b716da8b7807574084379515b95bbf","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"date":"2026-09-16 16:15:11.000000000","message":"Uploaded patch set 25.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":25},{"id":"a4d97e84dab60b5d23505b6b41a8eb0d188d3f82","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-16 17:50:49.000000000","message":"Patch Set 25: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6650c1dfbefd46d296be31d30e75a723\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b2e3a36480c46a9997d17cd0c6a760c : SUCCESS in 3m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6dc5b329d89b466a8be9ed1dc3c3202b : SUCCESS in 4m 46s\n- openstack-helm-linter https://zuul.opendev.org/t/openstack/build/0c88985ccf4f4f2382ce78423641b705 : SUCCESS in 3m 28s\n- openstack-helm-mariadb-crd-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/0c965a278d7349d1a1f90fed75950b0e : SUCCESS in 1h 04m 02s\n- openstack-helm-mariadb-operator-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/f6b609e211854b4794d9d2587d8732b3 : SUCCESS in 58m 06s\n- openstack-helm-rabbitmq-crd-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/489f8d658d13468db3fdc1713531bf86 : SUCCESS in 50m 15s\n- openstack-helm-rabbitmq-operator-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/ca13edd4e5ac46e9ae2125f10236f73f : SUCCESS in 54m 27s\n- openstack-helm-pre-commit https://zuul.opendev.org/t/openstack/build/82f5832d05624abdb1085e5f595b4b77 : SUCCESS in 2m 52s\n- openstack-helm-build-charts https://zuul.opendev.org/t/openstack/build/c59b618e2ba849ada62323a42b9e059d : SUCCESS in 14m 29s\n- openstack-helm-cinder-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/59c23b1ba2a04ef7875fe1e9b8cd8dde : SUCCESS in 36m 17s\n- openstack-helm-compute-kit-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/0d927ab3d16b4f59b7745c43ce776712 : SUCCESS in 1h 13m 29s\n- openstack-helm-cinder-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/391728a594db48c88b548ed11b2d82ca : SUCCESS in 46m 45s\n- openstack-helm-compute-kit-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/4b1436fa8f9343bd9ccc95016472bd93 : SUCCESS in 50m 54s\n- openstack-helm-cinder-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/09a3b5df455b4779a6f599d162d457f1 : SUCCESS in 1h 05m 49s\n- openstack-helm-compute-kit-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/55b2034b549a4360a1a66eff0a482357 : SUCCESS in 52m 21s\n- openstack-helm-tls-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/242ff6560f5046ac95a5fad510c3e39e : SUCCESS in 1h 22m 37s\n- openstack-helm-octavia-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/11681dbbd3924cdd9bb07c75aec20681 : SUCCESS in 1h 19m 53s\n- openstack-helm-logging https://zuul.opendev.org/t/openstack/build/47d81adda977444692fad9e97a5f4451 : SUCCESS in 27m 22s\n- openstack-helm-monitoring https://zuul.opendev.org/t/openstack/build/5a9e331b6038425c803d222c25a2866c : SUCCESS in 31m 24s","accounts_in_message":[],"_revision_number":25},{"id":"163a1067ca75200b8d5b48806f76f825d3c08ed7","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"date":"2026-09-16 21:42:36.000000000","message":"Uploaded patch set 26.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":26},{"id":"462c045d70908f2dbce797cb293ce17599a5c21f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-16 23:24:00.000000000","message":"Patch Set 26: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b238bc0f474840a5a262bf6f9c922d32\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/06252bed33664c73bff712328c4d8fab : SUCCESS in 2m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b9d5bb0853084b319875d983f84df50b : SUCCESS in 2m 13s\n- openstack-helm-linter https://zuul.opendev.org/t/openstack/build/d7cd9bd9729442f0bb0bfa1c12712922 : SUCCESS in 3m 11s\n- openstack-helm-mariadb-crd-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/3bb87ff994224e60bace24f61f5b3ec7 : SUCCESS in 1h 34m 55s\n- openstack-helm-mariadb-operator-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/02f162050a474e28b8d8d9e7d3a78abd : SUCCESS in 1h 02m 25s\n- openstack-helm-rabbitmq-crd-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/c8a296e5f6334052b75fbf0c46268554 : SUCCESS in 1h 17m 54s\n- openstack-helm-rabbitmq-operator-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/2f6792bd68064818a6114882d0a24361 : SUCCESS in 1h 12m 03s\n- openstack-helm-pre-commit https://zuul.opendev.org/t/openstack/build/53b94c0571b847a49fbb89527feb938b : SUCCESS in 2m 48s\n- openstack-helm-build-charts https://zuul.opendev.org/t/openstack/build/50fbecf929d84bd2895e2007c48acbd0 : SUCCESS in 45m 06s\n- openstack-helm-cinder-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/c86fd73cc1c34de1a44de3d31af3241f : SUCCESS in 44m 01s\n- openstack-helm-compute-kit-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/bf54db4c503a4b35a5c43250f6b6a501 : SUCCESS in 1h 12m 07s\n- openstack-helm-cinder-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/0ef08ecdc45a43c7943290cc4b5832a2 : SUCCESS in 28m 57s\n- openstack-helm-compute-kit-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/1053c5e39652433e9e4720ec77447673 : SUCCESS in 1h 15m 49s\n- openstack-helm-cinder-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/ae0e34304bfe46869b30e570f8dce26b : SUCCESS in 38m 28s\n- openstack-helm-compute-kit-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/ebe700e6c6ea4c1890512b23f6958afc : SUCCESS in 53m 29s\n- openstack-helm-tls-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/5c64b50de02e43e4b7980c6efd06194f : SUCCESS in 1h 29m 33s\n- openstack-helm-octavia-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/d5c87dd7f79c4e15a7adfc29219b8297 : SUCCESS in 47m 44s\n- openstack-helm-logging https://zuul.opendev.org/t/openstack/build/b28162d49f6242e3839d8f0a1210167c : SUCCESS in 34m 31s\n- openstack-helm-monitoring https://zuul.opendev.org/t/openstack/build/9d3b5bcc51394d78b7f7e24dcb52932d : SUCCESS in 31m 18s","accounts_in_message":[],"_revision_number":26},{"id":"af04bc02a172ac5fe20b35dea67c41143fc78493","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"date":"2026-09-22 18:39:29.000000000","message":"Uploaded patch set 27.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":27},{"id":"38fda5eacb839dde3026aac443f49e32b2113f45","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-22 20:19:19.000000000","message":"Patch Set 27: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/acf17d15d3e041c6bcad59581caf64fc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3470f7dfb3644e71ad066ece605a2383 : SUCCESS in 4m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/783ff8252c02431489c09542aef29eff : SUCCESS in 3m 19s\n- openstack-helm-linter https://zuul.opendev.org/t/openstack/build/4b99e086e0404ff39381f35f212dd6a5 : SUCCESS in 2m 44s\n- openstack-helm-mariadb-crd-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/3c28e2d2d5a044c3906454aa5003bf13 : SUCCESS in 1h 07m 32s\n- openstack-helm-mariadb-operator-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/60aa7191af3e4b1c83dc2415cc1d8ede : SUCCESS in 1h 10m 35s\n- openstack-helm-rabbitmq-crd-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/a7e73fbcc42d4f7a9fee9515cc4f8ef2 : SUCCESS in 1h 23m 13s\n- openstack-helm-rabbitmq-operator-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/054d2940beb44197bfb2dd7ffe2aaf6a : SUCCESS in 1h 23m 56s\n- openstack-helm-pre-commit https://zuul.opendev.org/t/openstack/build/f6926f41741f47a6971421fb7403b916 : SUCCESS in 2m 54s\n- openstack-helm-build-charts https://zuul.opendev.org/t/openstack/build/d556399592cc462baf4a55e838cd58ad : SUCCESS in 42m 36s\n- openstack-helm-cinder-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/4d4189d74d3741ca9be7d549f1a597fa : SUCCESS in 41m 23s\n- openstack-helm-compute-kit-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/0d50373bee944b8298b612d71f2fa58e : SUCCESS in 1h 17m 08s\n- openstack-helm-cinder-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/08405b6fa9ae4be4b94ce428b0fec14c : SUCCESS in 45m 36s\n- openstack-helm-compute-kit-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/a95c9249bf894a60bc05730fa7952c4b : SUCCESS in 1h 18m 42s\n- openstack-helm-cinder-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/acd73a3ad3c7413f924e3de98eae958b : SUCCESS in 46m 52s\n- openstack-helm-compute-kit-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/4e8580b4928243e4b49852ffcd8c89ff : SUCCESS in 1h 08m 51s\n- openstack-helm-tls-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/ecf50916fd9d4c509b50b47daddefaf0 : SUCCESS in 1h 34m 27s\n- openstack-helm-octavia-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/69abbf39e8d641feb9a5bd020f74a84d : SUCCESS in 1h 20m 31s\n- openstack-helm-logging https://zuul.opendev.org/t/openstack/build/99943819ccfb4b008b9f7b4b5bd18fc3 : SUCCESS in 40m 06s\n- openstack-helm-monitoring https://zuul.opendev.org/t/openstack/build/8a76a7c82914472389d29a90808406bc : SUCCESS in 29m 36s","accounts_in_message":[],"_revision_number":27},{"id":"f50a9eea3357c1eef4f0c89e68eba54e0616f69a","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"date":"2026-09-24 15:38:17.000000000","message":"Uploaded patch set 28.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":28},{"id":"523cd8ffee0f379f9aacde55177db1378796f23a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-24 18:07:33.000000000","message":"Patch Set 28: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0b18387bbc0f4cba85837f0f1ba78ee4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dbadaee33d2c4d89bd8ef88092c4abca : SUCCESS in 2m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/db3e0afbf44b4b7ea025384047b2e0aa : SUCCESS in 3m 38s\n- openstack-helm-linter https://zuul.opendev.org/t/openstack/build/bcec3ef0afe34c91b320aadcc3817484 : SUCCESS in 3m 32s\n- openstack-helm-pre-commit https://zuul.opendev.org/t/openstack/build/40f5b439780f431a99d5dc17d8809f67 : SUCCESS in 3m 54s\n- openstack-helm-build-charts https://zuul.opendev.org/t/openstack/build/0afb45312a62465998aec88e480858ea : SUCCESS in 21m 47s\n- openstack-helm-compute-kit-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/4b9ca3420e4b487092931f86e9468b64 : SUCCESS in 1h 40m 20s\n- openstack-helm-compute-kit-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/835a256c1e484a4d9f196c1c7738968f : SUCCESS in 1h 28m 49s\n- openstack-helm-compute-kit-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/93e29553742f4079858e9edee09109eb : SUCCESS in 1h 51m 52s\n- openstack-helm-tls-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/88ed8290e52042d085d67949e8cda59a : SUCCESS in 1h 44m 30s\n- openstack-helm-crd-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/ad35a9e9643b4219af7393080771a212 : SUCCESS in 1h 26m 53s\n- openstack-helm-operators-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/20d6a5084a0d42228ade67b79b53e046 : SUCCESS in 1h 28m 59s\n- openstack-helm-logging https://zuul.opendev.org/t/openstack/build/0e176a4e694a4ef69e6028358b6dd23b : SUCCESS in 39m 10s\n- openstack-helm-monitoring https://zuul.opendev.org/t/openstack/build/3662ec61de2f4193aae29aafc9ced033 : SUCCESS in 28m 42s","accounts_in_message":[],"_revision_number":28}],"current_revision_number":28,"current_revision":"e543cc15c9fd0d7755dc8c8adda93a18e6d99507","revisions":{"32294464d4c4633d3f204adc177f00879f53d692":{"kind":"REWORK","_number":1,"created":"2026-09-04 19:48:16.000000000","uploader":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"ref":"refs/changes/32/1004032/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-helm","ref":"refs/changes/32/1004032/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-helm refs/changes/32/1004032/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-helm refs/changes/32/1004032/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-helm refs/changes/32/1004032/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-helm refs/changes/32/1004032/1"}}},"commit":{"parents":[{"commit":"c73f68ac866aeedbec2784c16e6a45b215664112","subject":"Merge \"ovn: fix liveness probe to check as ovs user, not root\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm/commit/c73f68ac866aeedbec2784c16e6a45b215664112"}]}],"author":{"name":"Stephen Taylor","email":"stephen.taylor.1@att.com","date":"2026-09-04 19:44:57.000000000","tz":-360},"committer":{"name":"Stephen Taylor","email":"stephen.taylor.1@att.com","date":"2026-09-04 19:44:57.000000000","tz":-360},"subject":"[ceph-mon] Revert client.admin key to aes for compatibility","message":"[ceph-mon] Revert client.admin key to aes for compatibility\n\nThe aes256k key type is only compatible with 7.x Linux kernels.\nThe client admin key is reverted back to aes for compatibility\nwith older kernels. Otherwise, Ceph CSI is unable to map RBDs.\n\nChange-Id: I5f7ce68ab99ccc152c86a9fb09c6661ec33b2a00\nSigned-off-by: Stephen Taylor \u003cstaylor@mirantis.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm/commit/32294464d4c4633d3f204adc177f00879f53d692"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm/commit/32294464d4c4633d3f204adc177f00879f53d692"}]},"branch":"refs/heads/master"},"db26f07d67236b0ee184fa06b088847dbe6ea829":{"kind":"REWORK","_number":2,"created":"2026-09-08 16:42:38.000000000","uploader":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"ref":"refs/changes/32/1004032/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-helm","ref":"refs/changes/32/1004032/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-helm refs/changes/32/1004032/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-helm refs/changes/32/1004032/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-helm refs/changes/32/1004032/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-helm refs/changes/32/1004032/2"}}},"commit":{"parents":[{"commit":"c73f68ac866aeedbec2784c16e6a45b215664112","subject":"Merge \"ovn: fix liveness probe to check as ovs user, not root\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm/commit/c73f68ac866aeedbec2784c16e6a45b215664112"}]}],"author":{"name":"Stephen Taylor","email":"stephen.taylor.1@att.com","date":"2026-09-04 19:44:57.000000000","tz":-360},"committer":{"name":"Stephen Taylor","email":"stephen.taylor.1@att.com","date":"2026-09-08 16:42:10.000000000","tz":-360},"subject":"[ceph-mon] Revert client.admin key to aes for compatibility","message":"[ceph-mon] Revert client.admin key to aes for compatibility\n\nThe aes256k key type is only compatible with 7.x Linux kernels.\nThe client admin key is reverted back to aes for compatibility\nwith older kernels. Otherwise, Ceph CSI is unable to map RBDs.\n\nChange-Id: I5f7ce68ab99ccc152c86a9fb09c6661ec33b2a00\nSigned-off-by: Stephen Taylor \u003cstaylor@mirantis.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm/commit/db26f07d67236b0ee184fa06b088847dbe6ea829"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm/commit/db26f07d67236b0ee184fa06b088847dbe6ea829"}]},"branch":"refs/heads/master"},"7d895d7f4c942307769b6b62352639f815ac5ab1":{"kind":"REWORK","_number":3,"created":"2026-09-08 18:45:13.000000000","uploader":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"ref":"refs/changes/32/1004032/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-helm","ref":"refs/changes/32/1004032/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-helm refs/changes/32/1004032/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-helm refs/changes/32/1004032/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-helm refs/changes/32/1004032/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-helm refs/changes/32/1004032/3"}}},"commit":{"parents":[{"commit":"c73f68ac866aeedbec2784c16e6a45b215664112","subject":"Merge \"ovn: fix liveness probe to check as ovs user, not root\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm/commit/c73f68ac866aeedbec2784c16e6a45b215664112"}]}],"author":{"name":"Stephen Taylor","email":"stephen.taylor.1@att.com","date":"2026-09-04 19:44:57.000000000","tz":-360},"committer":{"name":"Stephen Taylor","email":"stephen.taylor.1@att.com","date":"2026-09-08 18:45:07.000000000","tz":-360},"subject":"[ceph-mon] Revert client.admin key to aes for compatibility","message":"[ceph-mon] Revert client.admin key to aes for compatibility\n\nThe aes256k key type is only compatible with 7.x Linux kernels.\nThe client admin key is reverted back to aes for compatibility\nwith older kernels. Otherwise, Ceph CSI is unable to map RBDs.\n\nChange-Id: I5f7ce68ab99ccc152c86a9fb09c6661ec33b2a00\nSigned-off-by: Stephen Taylor \u003cstaylor@mirantis.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm/commit/7d895d7f4c942307769b6b62352639f815ac5ab1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm/commit/7d895d7f4c942307769b6b62352639f815ac5ab1"}]},"branch":"refs/heads/master"},"c7d6ff387195f3f34abcd76960dd304590e6e569":{"kind":"REWORK","_number":4,"created":"2026-09-08 20:41:37.000000000","uploader":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"ref":"refs/changes/32/1004032/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-helm","ref":"refs/changes/32/1004032/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-helm refs/changes/32/1004032/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-helm refs/changes/32/1004032/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-helm refs/changes/32/1004032/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-helm refs/changes/32/1004032/4"}}},"commit":{"parents":[{"commit":"c73f68ac866aeedbec2784c16e6a45b215664112","subject":"Merge \"ovn: fix liveness probe to check as ovs user, not root\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm/commit/c73f68ac866aeedbec2784c16e6a45b215664112"}]}],"author":{"name":"Stephen Taylor","email":"stephen.taylor.1@att.com","date":"2026-09-04 19:44:57.000000000","tz":-360},"committer":{"name":"Stephen Taylor","email":"stephen.taylor.1@att.com","date":"2026-09-08 20:40:43.000000000","tz":-360},"subject":"[ceph-mon] Revert client.admin key to aes for compatibility","message":"[ceph-mon] Revert client.admin key to aes for compatibility\n\nThe aes256k key type is only compatible with 7.x Linux kernels.\nThe client admin key is reverted back to aes for compatibility\nwith older kernels. Otherwise, Ceph CSI is unable to map RBDs.\n\nChange-Id: I5f7ce68ab99ccc152c86a9fb09c6661ec33b2a00\nSigned-off-by: Stephen Taylor \u003cstaylor@mirantis.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm/commit/c7d6ff387195f3f34abcd76960dd304590e6e569"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm/commit/c7d6ff387195f3f34abcd76960dd304590e6e569"}]},"branch":"refs/heads/master"},"42f8998aebe32b6b219ffff59bbafe27318337f5":{"kind":"REWORK","_number":5,"created":"2026-09-08 21:49:36.000000000","uploader":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"ref":"refs/changes/32/1004032/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-helm","ref":"refs/changes/32/1004032/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-helm refs/changes/32/1004032/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-helm refs/changes/32/1004032/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-helm refs/changes/32/1004032/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-helm refs/changes/32/1004032/5"}}},"commit":{"parents":[{"commit":"c73f68ac866aeedbec2784c16e6a45b215664112","subject":"Merge \"ovn: fix liveness probe to check as ovs user, not root\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm/commit/c73f68ac866aeedbec2784c16e6a45b215664112"}]}],"author":{"name":"Stephen Taylor","email":"stephen.taylor.1@att.com","date":"2026-09-04 19:44:57.000000000","tz":-360},"committer":{"name":"Stephen Taylor","email":"stephen.taylor.1@att.com","date":"2026-09-08 21:49:15.000000000","tz":-360},"subject":"[ceph-mon] Revert client.admin key to aes for compatibility","message":"[ceph-mon] Revert client.admin key to aes for compatibility\n\nThe aes256k key type is only compatible with 7.x Linux kernels.\nThe client admin key is reverted back to aes for compatibility\nwith older kernels. Otherwise, Ceph CSI is unable to map RBDs.\n\nChange-Id: I5f7ce68ab99ccc152c86a9fb09c6661ec33b2a00\nSigned-off-by: Stephen Taylor \u003cstaylor@mirantis.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm/commit/42f8998aebe32b6b219ffff59bbafe27318337f5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm/commit/42f8998aebe32b6b219ffff59bbafe27318337f5"}]},"branch":"refs/heads/master"},"e7f934c0340d1e9c8cd7baaf67c7fae8681dbc2c":{"kind":"REWORK","_number":6,"created":"2026-09-09 12:41:42.000000000","uploader":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"ref":"refs/changes/32/1004032/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-helm","ref":"refs/changes/32/1004032/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-helm refs/changes/32/1004032/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-helm refs/changes/32/1004032/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-helm refs/changes/32/1004032/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-helm refs/changes/32/1004032/6"}}},"commit":{"parents":[{"commit":"c73f68ac866aeedbec2784c16e6a45b215664112","subject":"Merge \"ovn: fix liveness probe to check as ovs user, not root\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm/commit/c73f68ac866aeedbec2784c16e6a45b215664112"}]}],"author":{"name":"Stephen Taylor","email":"stephen.taylor.1@att.com","date":"2026-09-04 19:44:57.000000000","tz":-360},"committer":{"name":"Stephen Taylor","email":"stephen.taylor.1@att.com","date":"2026-09-09 12:41:19.000000000","tz":-360},"subject":"[ceph-mon] Revert client.admin key to aes for compatibility","message":"[ceph-mon] Revert client.admin key to aes for compatibility\n\nThe aes256k key type is only compatible with 7.x Linux kernels.\nThe client admin key is reverted back to aes for compatibility\nwith older kernels. Otherwise, Ceph CSI is unable to map RBDs.\n\nChange-Id: I5f7ce68ab99ccc152c86a9fb09c6661ec33b2a00\nSigned-off-by: Stephen Taylor \u003cstaylor@mirantis.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm/commit/e7f934c0340d1e9c8cd7baaf67c7fae8681dbc2c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm/commit/e7f934c0340d1e9c8cd7baaf67c7fae8681dbc2c"}]},"branch":"refs/heads/master"},"c04deabff148dd4cc49294865698e2ad56c14d9b":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2026-09-09 19:47:51.000000000","uploader":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"real_uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/32/1004032/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-helm","ref":"refs/changes/32/1004032/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-helm refs/changes/32/1004032/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-helm refs/changes/32/1004032/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-helm refs/changes/32/1004032/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-helm refs/changes/32/1004032/7"}}},"commit":{"parents":[{"commit":"dc13ee9889dfbe7dc7ab796b019b13c9f93624aa","subject":"deploy: keep the OpenStack CephX keys on the aes cipher","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm/commit/dc13ee9889dfbe7dc7ab796b019b13c9f93624aa"}]}],"author":{"name":"Stephen Taylor","email":"stephen.taylor.1@att.com","date":"2026-09-04 19:44:57.000000000","tz":-360},"committer":{"name":"Stephen Taylor","email":"stephen.taylor.1@att.com","date":"2026-09-09 19:47:51.000000000","tz":0},"subject":"[ceph-mon] Revert client.admin key to aes for compatibility","message":"[ceph-mon] Revert client.admin key to aes for compatibility\n\nThe aes256k key type is only compatible with 7.x Linux kernels.\nThe client admin key is reverted back to aes for compatibility\nwith older kernels. Otherwise, Ceph CSI is unable to map RBDs.\n\nChange-Id: I5f7ce68ab99ccc152c86a9fb09c6661ec33b2a00\nSigned-off-by: Stephen Taylor \u003cstaylor@mirantis.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm/commit/c04deabff148dd4cc49294865698e2ad56c14d9b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm/commit/c04deabff148dd4cc49294865698e2ad56c14d9b"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"c73f68ac866aeedbec2784c16e6a45b215664112","ours":"e7f934c0340d1e9c8cd7baaf67c7fae8681dbc2c","theirs":"dc13ee9889dfbe7dc7ab796b019b13c9f93624aa","merge_strategy":"recursive","contains_conflicts":false}},"ba22d9147d234eb73870c6452b53b4610350d7ef":{"kind":"REWORK","_number":8,"created":"2026-09-10 15:21:59.000000000","uploader":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"ref":"refs/changes/32/1004032/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-helm","ref":"refs/changes/32/1004032/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-helm refs/changes/32/1004032/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-helm refs/changes/32/1004032/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-helm refs/changes/32/1004032/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-helm refs/changes/32/1004032/8"}}},"commit":{"parents":[{"commit":"c73f68ac866aeedbec2784c16e6a45b215664112","subject":"Merge \"ovn: fix liveness probe to check as ovs user, not root\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm/commit/c73f68ac866aeedbec2784c16e6a45b215664112"}]}],"author":{"name":"Stephen Taylor","email":"stephen.taylor.1@att.com","date":"2026-09-04 19:44:57.000000000","tz":-360},"committer":{"name":"Stephen Taylor","email":"stephen.taylor.1@att.com","date":"2026-09-10 15:18:00.000000000","tz":-360},"subject":"[ceph] Support both aes and aes256k CephX auth keys","message":"[ceph] Support both aes and aes256k CephX auth keys\n\nThe ceph-mon start script logic has been expanded significantly\nto support both aes and aes256k CephX auth keys for the mon key\nand the bootstrap keys. Key types are now defined in values.yaml,\nand the start script will attempt to rotate keys if existing keys\ndon\u0027t match the expected key types.\n\nChange-Id: I5f7ce68ab99ccc152c86a9fb09c6661ec33b2a00\nSigned-off-by: Stephen Taylor \u003cstaylor@mirantis.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm/commit/ba22d9147d234eb73870c6452b53b4610350d7ef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm/commit/ba22d9147d234eb73870c6452b53b4610350d7ef"}]},"branch":"refs/heads/master"},"847e2d7a3174aa3a6b9bd83ae12f851c039d3a11":{"kind":"REWORK","_number":9,"created":"2026-09-10 20:11:34.000000000","uploader":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"ref":"refs/changes/32/1004032/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-helm","ref":"refs/changes/32/1004032/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-helm refs/changes/32/1004032/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-helm refs/changes/32/1004032/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-helm refs/changes/32/1004032/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-helm refs/changes/32/1004032/9"}}},"commit":{"parents":[{"commit":"c73f68ac866aeedbec2784c16e6a45b215664112","subject":"Merge \"ovn: fix liveness probe to check as ovs user, not root\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm/commit/c73f68ac866aeedbec2784c16e6a45b215664112"}]}],"author":{"name":"Stephen Taylor","email":"stephen.taylor.1@att.com","date":"2026-09-04 19:44:57.000000000","tz":-360},"committer":{"name":"Stephen Taylor","email":"stephen.taylor.1@att.com","date":"2026-09-10 20:05:55.000000000","tz":-360},"subject":"[ceph] Support both aes and aes256k CephX auth keys","message":"[ceph] Support both aes and aes256k CephX auth keys\n\nThe ceph-mon start script logic has been expanded significantly\nto support both aes and aes256k CephX auth keys for the mon key\nand the bootstrap keys. Key types are now defined in values.yaml,\nand the start script will attempt to rotate keys if existing keys\ndon\u0027t match the expected key types.\n\nSimilar logic as been added to the ceph-osd osd-init scripts and\nthe ceph-rgw start script to attempt to rotate keys to the desired\nkey type if specified. A desired key type option has also been\nadded to the ceph-osd values since the ceph-osd type can be defined\nexplicitly in some cases. The ceph-rgw key type is inferred from\nthe ceph-mon config, so that config is queried to determine if\nkey rotate is appropriate for RGW.\n\nChange-Id: I5f7ce68ab99ccc152c86a9fb09c6661ec33b2a00\nSigned-off-by: Stephen Taylor \u003cstaylor@mirantis.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm/commit/847e2d7a3174aa3a6b9bd83ae12f851c039d3a11"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm/commit/847e2d7a3174aa3a6b9bd83ae12f851c039d3a11"}]},"branch":"refs/heads/master"},"c4db1c29da78417bbf846d636b76e39c7b95fc4c":{"kind":"REWORK","_number":10,"created":"2026-09-10 20:21:49.000000000","uploader":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"ref":"refs/changes/32/1004032/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-helm","ref":"refs/changes/32/1004032/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-helm refs/changes/32/1004032/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-helm refs/changes/32/1004032/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-helm refs/changes/32/1004032/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-helm refs/changes/32/1004032/10"}}},"commit":{"parents":[{"commit":"c73f68ac866aeedbec2784c16e6a45b215664112","subject":"Merge \"ovn: fix liveness probe to check as ovs user, not root\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm/commit/c73f68ac866aeedbec2784c16e6a45b215664112"}]}],"author":{"name":"Stephen Taylor","email":"stephen.taylor.1@att.com","date":"2026-09-04 19:44:57.000000000","tz":-360},"committer":{"name":"Stephen Taylor","email":"stephen.taylor.1@att.com","date":"2026-09-10 20:21:43.000000000","tz":-360},"subject":"[ceph] Support both aes and aes256k CephX auth keys","message":"[ceph] Support both aes and aes256k CephX auth keys\n\nThe ceph-mon start script logic has been expanded significantly\nto support both aes and aes256k CephX auth keys for the mon key\nand the bootstrap keys. Key types are now defined in values.yaml,\nand the start script will attempt to rotate keys if existing keys\ndon\u0027t match the expected key types.\n\nSimilar logic as been added to the ceph-osd osd-init scripts and\nthe ceph-rgw start script to attempt to rotate keys to the desired\nkey type if specified. A desired key type option has also been\nadded to the ceph-osd values since the ceph-osd type can be defined\nexplicitly in some cases. The ceph-rgw key type is inferred from\nthe ceph-mon config, so that config is queried to determine if\nkey rotate is appropriate for RGW.\n\nChange-Id: I5f7ce68ab99ccc152c86a9fb09c6661ec33b2a00\nSigned-off-by: Stephen Taylor \u003cstaylor@mirantis.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm/commit/c4db1c29da78417bbf846d636b76e39c7b95fc4c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm/commit/c4db1c29da78417bbf846d636b76e39c7b95fc4c"}]},"branch":"refs/heads/master"},"e50ce1ff0a514307b5831a165aec82b5c958d812":{"kind":"REWORK","_number":11,"created":"2026-09-10 20:56:45.000000000","uploader":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"ref":"refs/changes/32/1004032/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-helm","ref":"refs/changes/32/1004032/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-helm refs/changes/32/1004032/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-helm refs/changes/32/1004032/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-helm refs/changes/32/1004032/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-helm refs/changes/32/1004032/11"}}},"commit":{"parents":[{"commit":"c73f68ac866aeedbec2784c16e6a45b215664112","subject":"Merge \"ovn: fix liveness probe to check as ovs user, not root\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm/commit/c73f68ac866aeedbec2784c16e6a45b215664112"}]}],"author":{"name":"Stephen Taylor","email":"stephen.taylor.1@att.com","date":"2026-09-04 19:44:57.000000000","tz":-360},"committer":{"name":"Stephen Taylor","email":"stephen.taylor.1@att.com","date":"2026-09-10 20:54:36.000000000","tz":-360},"subject":"[ceph] Support both aes and aes256k CephX auth keys","message":"[ceph] Support both aes and aes256k CephX auth keys\n\nThe ceph-mon start script logic has been expanded significantly\nto support both aes and aes256k CephX auth keys for the mon key\nand the bootstrap keys. Key types are now defined in values.yaml,\nand the start script will attempt to rotate keys if existing keys\ndon\u0027t match the expected key types.\n\nSimilar logic as been added to the mgr start script in ceph-mon,\nthe ceph-osd osd init scripts, and the ceph-rgw start script to\nattempt to rotate keys to the desired key type if specified.\n\nA desired key type option has also been added to the ceph-osd\nvalues since the ceph-osd type can be defined explicitly in some\ncases.\n\nThe ceph-rgw key type is inferred from the ceph-mon config, so that\nconfig is queried to determine if key rotate is appropriate for\nRGW.\n\nChange-Id: I5f7ce68ab99ccc152c86a9fb09c6661ec33b2a00\nSigned-off-by: Stephen Taylor \u003cstaylor@mirantis.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm/commit/e50ce1ff0a514307b5831a165aec82b5c958d812"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm/commit/e50ce1ff0a514307b5831a165aec82b5c958d812"}]},"branch":"refs/heads/master"},"7a68d1af21833c778acbc2cc09bb7c67bab21abf":{"kind":"REWORK","_number":12,"created":"2026-09-11 15:06:24.000000000","uploader":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"ref":"refs/changes/32/1004032/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-helm","ref":"refs/changes/32/1004032/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-helm refs/changes/32/1004032/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-helm refs/changes/32/1004032/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-helm refs/changes/32/1004032/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-helm refs/changes/32/1004032/12"}}},"commit":{"parents":[{"commit":"c73f68ac866aeedbec2784c16e6a45b215664112","subject":"Merge \"ovn: fix liveness probe to check as ovs user, not root\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm/commit/c73f68ac866aeedbec2784c16e6a45b215664112"}]}],"author":{"name":"Stephen Taylor","email":"stephen.taylor.1@att.com","date":"2026-09-04 19:44:57.000000000","tz":-360},"committer":{"name":"Stephen Taylor","email":"stephen.taylor.1@att.com","date":"2026-09-11 14:59:02.000000000","tz":-360},"subject":"[ceph] Support both aes and aes256k CephX auth keys","message":"[ceph] Support both aes and aes256k CephX auth keys\n\nA new script has been added to the helm-toolkit chart to\nfacilitate CephX key rotation. Support for key rotate is required\nin the ceph-mon, ceph-osd, and ceph-rgw charts in order to comply\nwith secure (aes256k) key requirements in Ceph Tentacle 20.2.4.\n\nThe ceph-mon start script has been modified to support both aes and\naes256k CephX auth keys for the mon key and the bootstrap keys. Key\ntypes are now defined in values.yaml, and the start script attempts\nto rotate keys if existing keys don\u0027t match the expected key types.\n\nSimilar logic as been added to the mgr start script in ceph-mon,\nthe ceph-osd osd init scripts, and the ceph-rgw start script to\nattempt to rotate keys to the desired key type if specified.\n\nA desired key type option has also been added to the ceph-osd\nvalues since the ceph-osd type can be defined explicitly in some\ncases.\n\nThe ceph-mgr and ceph-rgw key types are inferred from the monmap\nconfig defined in ceph-mon values, so that config is queried to\ndetermine if key rotate is necessary for those components.\n\nChange-Id: I5f7ce68ab99ccc152c86a9fb09c6661ec33b2a00\nSigned-off-by: Stephen Taylor \u003cstaylor@mirantis.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm/commit/7a68d1af21833c778acbc2cc09bb7c67bab21abf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm/commit/7a68d1af21833c778acbc2cc09bb7c67bab21abf"}]},"branch":"refs/heads/master"},"9b242cd3ba6156e78fa42e96f572db65a21ce91e":{"kind":"TRIVIAL_REBASE","_number":13,"created":"2026-09-11 15:06:48.000000000","uploader":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"ref":"refs/changes/32/1004032/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-helm","ref":"refs/changes/32/1004032/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-helm refs/changes/32/1004032/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-helm refs/changes/32/1004032/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-helm refs/changes/32/1004032/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-helm refs/changes/32/1004032/13"}}},"commit":{"parents":[{"commit":"85bb718fdd7bf555c58eae1fdc981b13d56fe742","subject":"Merge \"masakari: set kubernetes driver as default for host-monitor masakari-host-monitor changed to deployment Task: 55600 Story: 2011878\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm/commit/85bb718fdd7bf555c58eae1fdc981b13d56fe742"}]}],"author":{"name":"Stephen Taylor","email":"stephen.taylor.1@att.com","date":"2026-09-04 19:44:57.000000000","tz":-360},"committer":{"name":"Stephen Taylor","email":"stephen.taylor.1@att.com","date":"2026-09-11 15:06:48.000000000","tz":0},"subject":"[ceph] Support both aes and aes256k CephX auth keys","message":"[ceph] Support both aes and aes256k CephX auth keys\n\nA new script has been added to the helm-toolkit chart to\nfacilitate CephX key rotation. Support for key rotate is required\nin the ceph-mon, ceph-osd, and ceph-rgw charts in order to comply\nwith secure (aes256k) key requirements in Ceph Tentacle 20.2.4.\n\nThe ceph-mon start script has been modified to support both aes and\naes256k CephX auth keys for the mon key and the bootstrap keys. Key\ntypes are now defined in values.yaml, and the start script attempts\nto rotate keys if existing keys don\u0027t match the expected key types.\n\nSimilar logic as been added to the mgr start script in ceph-mon,\nthe ceph-osd osd init scripts, and the ceph-rgw start script to\nattempt to rotate keys to the desired key type if specified.\n\nA desired key type option has also been added to the ceph-osd\nvalues since the ceph-osd type can be defined explicitly in some\ncases.\n\nThe ceph-mgr and ceph-rgw key types are inferred from the monmap\nconfig defined in ceph-mon values, so that config is queried to\ndetermine if key rotate is necessary for those components.\n\nChange-Id: I5f7ce68ab99ccc152c86a9fb09c6661ec33b2a00\nSigned-off-by: Stephen Taylor \u003cstaylor@mirantis.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm/commit/9b242cd3ba6156e78fa42e96f572db65a21ce91e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm/commit/9b242cd3ba6156e78fa42e96f572db65a21ce91e"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"c73f68ac866aeedbec2784c16e6a45b215664112","ours":"7a68d1af21833c778acbc2cc09bb7c67bab21abf","theirs":"85bb718fdd7bf555c58eae1fdc981b13d56fe742","merge_strategy":"recursive","contains_conflicts":false}},"84b6e645306d4075b84b0b7e50a9702f95130fff":{"kind":"REWORK","_number":14,"created":"2026-09-11 16:13:12.000000000","uploader":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"ref":"refs/changes/32/1004032/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-helm","ref":"refs/changes/32/1004032/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-helm refs/changes/32/1004032/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-helm refs/changes/32/1004032/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-helm refs/changes/32/1004032/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-helm refs/changes/32/1004032/14"}}},"commit":{"parents":[{"commit":"85bb718fdd7bf555c58eae1fdc981b13d56fe742","subject":"Merge \"masakari: set kubernetes driver as default for host-monitor masakari-host-monitor changed to deployment Task: 55600 Story: 2011878\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm/commit/85bb718fdd7bf555c58eae1fdc981b13d56fe742"}]}],"author":{"name":"Stephen Taylor","email":"stephen.taylor.1@att.com","date":"2026-09-04 19:44:57.000000000","tz":-360},"committer":{"name":"Stephen Taylor","email":"stephen.taylor.1@att.com","date":"2026-09-11 16:13:05.000000000","tz":-360},"subject":"[ceph] Support both aes and aes256k CephX auth keys","message":"[ceph] Support both aes and aes256k CephX auth keys\n\nA new script has been added to the helm-toolkit chart to\nfacilitate CephX key rotation. Support for key rotate is required\nin the ceph-mon, ceph-osd, and ceph-rgw charts in order to comply\nwith secure (aes256k) key requirements in Ceph Tentacle 20.2.4.\n\nThe ceph-mon start script has been modified to support both aes and\naes256k CephX auth keys for the mon key and the bootstrap keys. Key\ntypes are now defined in values.yaml, and the start script attempts\nto rotate keys if existing keys don\u0027t match the expected key types.\n\nSimilar logic as been added to the mgr start script in ceph-mon,\nthe ceph-osd osd init scripts, and the ceph-rgw start script to\nattempt to rotate keys to the desired key type if specified.\n\nA desired key type option has also been added to the ceph-osd\nvalues since the ceph-osd type can be defined explicitly in some\ncases.\n\nThe ceph-mgr and ceph-rgw key types are inferred from the monmap\nconfig defined in ceph-mon values, so that config is queried to\ndetermine if key rotate is necessary for those components.\n\nChange-Id: I5f7ce68ab99ccc152c86a9fb09c6661ec33b2a00\nSigned-off-by: Stephen Taylor \u003cstaylor@mirantis.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm/commit/84b6e645306d4075b84b0b7e50a9702f95130fff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm/commit/84b6e645306d4075b84b0b7e50a9702f95130fff"}]},"branch":"refs/heads/master"},"d4c463a1242554294a8fe09f274fd4c3eb3ea8ad":{"kind":"REWORK","_number":15,"created":"2026-09-11 16:20:14.000000000","uploader":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"ref":"refs/changes/32/1004032/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-helm","ref":"refs/changes/32/1004032/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-helm refs/changes/32/1004032/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-helm refs/changes/32/1004032/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-helm refs/changes/32/1004032/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-helm refs/changes/32/1004032/15"}}},"commit":{"parents":[{"commit":"85bb718fdd7bf555c58eae1fdc981b13d56fe742","subject":"Merge \"masakari: set kubernetes driver as default for host-monitor masakari-host-monitor changed to deployment Task: 55600 Story: 2011878\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm/commit/85bb718fdd7bf555c58eae1fdc981b13d56fe742"}]}],"author":{"name":"Stephen Taylor","email":"stephen.taylor.1@att.com","date":"2026-09-04 19:44:57.000000000","tz":-360},"committer":{"name":"Stephen Taylor","email":"stephen.taylor.1@att.com","date":"2026-09-11 16:20:09.000000000","tz":-360},"subject":"[ceph] Support both aes and aes256k CephX auth keys","message":"[ceph] Support both aes and aes256k CephX auth keys\n\nA new script has been added to the helm-toolkit chart to\nfacilitate CephX key rotation. Support for key rotate is required\nin the ceph-mon, ceph-osd, and ceph-rgw charts in order to comply\nwith secure (aes256k) key requirements in Ceph Tentacle 20.2.4.\n\nThe ceph-mon start script has been modified to support both aes and\naes256k CephX auth keys for the mon key and the bootstrap keys. Key\ntypes are now defined in values.yaml, and the start script attempts\nto rotate keys if existing keys don\u0027t match the expected key types.\n\nSimilar logic as been added to the mgr start script in ceph-mon,\nthe ceph-osd osd init scripts, and the ceph-rgw start script to\nattempt to rotate keys to the desired key type if specified.\n\nA desired key type option has also been added to the ceph-osd\nvalues since the ceph-osd type can be defined explicitly in some\ncases.\n\nThe ceph-mgr and ceph-rgw key types are inferred from the monmap\nconfig defined in ceph-mon values, so that config is queried to\ndetermine if key rotate is necessary for those components.\n\nChange-Id: I5f7ce68ab99ccc152c86a9fb09c6661ec33b2a00\nSigned-off-by: Stephen Taylor \u003cstaylor@mirantis.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm/commit/d4c463a1242554294a8fe09f274fd4c3eb3ea8ad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm/commit/d4c463a1242554294a8fe09f274fd4c3eb3ea8ad"}]},"branch":"refs/heads/master"},"bcbce0a18049c2e0399bdc410f2475e6077dae8c":{"kind":"REWORK","_number":16,"created":"2026-09-11 16:36:23.000000000","uploader":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"ref":"refs/changes/32/1004032/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-helm","ref":"refs/changes/32/1004032/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-helm refs/changes/32/1004032/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-helm refs/changes/32/1004032/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-helm refs/changes/32/1004032/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-helm refs/changes/32/1004032/16"}}},"commit":{"parents":[{"commit":"85bb718fdd7bf555c58eae1fdc981b13d56fe742","subject":"Merge \"masakari: set kubernetes driver as default for host-monitor masakari-host-monitor changed to deployment Task: 55600 Story: 2011878\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm/commit/85bb718fdd7bf555c58eae1fdc981b13d56fe742"}]}],"author":{"name":"Stephen Taylor","email":"stephen.taylor.1@att.com","date":"2026-09-04 19:44:57.000000000","tz":-360},"committer":{"name":"Stephen Taylor","email":"stephen.taylor.1@att.com","date":"2026-09-11 16:36:18.000000000","tz":-360},"subject":"[ceph] Support both aes and aes256k CephX auth keys","message":"[ceph] Support both aes and aes256k CephX auth keys\n\nA new script has been added to the helm-toolkit chart to\nfacilitate CephX key rotation. Support for key rotate is required\nin the ceph-mon, ceph-osd, and ceph-rgw charts in order to comply\nwith secure (aes256k) key requirements in Ceph Tentacle 20.2.4.\n\nThe ceph-mon start script has been modified to support both aes and\naes256k CephX auth keys for the mon key and the bootstrap keys. Key\ntypes are now defined in values.yaml, and the start script attempts\nto rotate keys if existing keys don\u0027t match the expected key types.\n\nSimilar logic as been added to the mgr start script in ceph-mon,\nthe ceph-osd osd init scripts, and the ceph-rgw start script to\nattempt to rotate keys to the desired key type if specified.\n\nA desired key type option has also been added to the ceph-osd\nvalues since the ceph-osd type can be defined explicitly in some\ncases.\n\nThe ceph-mgr and ceph-rgw key types are inferred from the monmap\nconfig defined in ceph-mon values, so that config is queried to\ndetermine if key rotate is necessary for those components.\n\nChange-Id: I5f7ce68ab99ccc152c86a9fb09c6661ec33b2a00\nSigned-off-by: Stephen Taylor \u003cstaylor@mirantis.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm/commit/bcbce0a18049c2e0399bdc410f2475e6077dae8c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm/commit/bcbce0a18049c2e0399bdc410f2475e6077dae8c"}]},"branch":"refs/heads/master"},"a4f5241721cb4825dae438ed118abc1aba6c8a89":{"kind":"REWORK","_number":17,"created":"2026-09-11 18:14:35.000000000","uploader":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"ref":"refs/changes/32/1004032/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-helm","ref":"refs/changes/32/1004032/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-helm refs/changes/32/1004032/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-helm refs/changes/32/1004032/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-helm refs/changes/32/1004032/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-helm refs/changes/32/1004032/17"}}},"commit":{"parents":[{"commit":"85bb718fdd7bf555c58eae1fdc981b13d56fe742","subject":"Merge \"masakari: set kubernetes driver as default for host-monitor masakari-host-monitor changed to deployment Task: 55600 Story: 2011878\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm/commit/85bb718fdd7bf555c58eae1fdc981b13d56fe742"}]}],"author":{"name":"Stephen Taylor","email":"stephen.taylor.1@att.com","date":"2026-09-04 19:44:57.000000000","tz":-360},"committer":{"name":"Stephen Taylor","email":"stephen.taylor.1@att.com","date":"2026-09-11 18:14:29.000000000","tz":-360},"subject":"[ceph] Support both aes and aes256k CephX auth keys","message":"[ceph] Support both aes and aes256k CephX auth keys\n\nA new script has been added to the helm-toolkit chart to\nfacilitate CephX key rotation. Support for key rotate is required\nin the ceph-mon, ceph-osd, and ceph-rgw charts in order to comply\nwith secure (aes256k) key requirements in Ceph Tentacle 20.2.4.\n\nThe ceph-mon start script has been modified to support both aes and\naes256k CephX auth keys for the mon key and the bootstrap keys. Key\ntypes are now defined in values.yaml, and the start script attempts\nto rotate keys if existing keys don\u0027t match the expected key types.\n\nSimilar logic as been added to the mgr start script in ceph-mon,\nthe ceph-osd osd init scripts, and the ceph-rgw start script to\nattempt to rotate keys to the desired key type if specified.\n\nA desired key type option has also been added to the ceph-osd\nvalues since the ceph-osd type can be defined explicitly in some\ncases.\n\nThe ceph-mgr and ceph-rgw key types are inferred from the monmap\nconfig defined in ceph-mon values, so that config is queried to\ndetermine if key rotate is necessary for those components.\n\nChange-Id: I5f7ce68ab99ccc152c86a9fb09c6661ec33b2a00\nSigned-off-by: Stephen Taylor \u003cstaylor@mirantis.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm/commit/a4f5241721cb4825dae438ed118abc1aba6c8a89"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm/commit/a4f5241721cb4825dae438ed118abc1aba6c8a89"}]},"branch":"refs/heads/master"},"dbfaf3cb011400b55a33a23dd1820e2202943624":{"kind":"REWORK","_number":18,"created":"2026-09-11 20:52:48.000000000","uploader":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"ref":"refs/changes/32/1004032/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-helm","ref":"refs/changes/32/1004032/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-helm refs/changes/32/1004032/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-helm refs/changes/32/1004032/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-helm refs/changes/32/1004032/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-helm refs/changes/32/1004032/18"}}},"commit":{"parents":[{"commit":"85bb718fdd7bf555c58eae1fdc981b13d56fe742","subject":"Merge \"masakari: set kubernetes driver as default for host-monitor masakari-host-monitor changed to deployment Task: 55600 Story: 2011878\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm/commit/85bb718fdd7bf555c58eae1fdc981b13d56fe742"}]}],"author":{"name":"Stephen Taylor","email":"stephen.taylor.1@att.com","date":"2026-09-04 19:44:57.000000000","tz":-360},"committer":{"name":"Stephen Taylor","email":"stephen.taylor.1@att.com","date":"2026-09-11 20:52:43.000000000","tz":-360},"subject":"[ceph] Support both aes and aes256k CephX auth keys","message":"[ceph] Support both aes and aes256k CephX auth keys\n\nA new script has been added to the helm-toolkit chart to\nfacilitate CephX key rotation. Support for key rotate is required\nin the ceph-mon, ceph-osd, and ceph-rgw charts in order to comply\nwith secure (aes256k) key requirements in Ceph Tentacle 20.2.4.\n\nThe ceph-mon start script has been modified to support both aes and\naes256k CephX auth keys for the mon key and the bootstrap keys. Key\ntypes are now defined in values.yaml, and the start script attempts\nto rotate keys if existing keys don\u0027t match the expected key types.\n\nSimilar logic as been added to the mgr start script in ceph-mon,\nthe ceph-osd osd init scripts, and the ceph-rgw start script to\nattempt to rotate keys to the desired key type if specified.\n\nA desired key type option has also been added to the ceph-osd\nvalues since the ceph-osd type can be defined explicitly in some\ncases.\n\nThe ceph-mgr and ceph-rgw key types are inferred from the monmap\nconfig defined in ceph-mon values, so that config is queried to\ndetermine if key rotate is necessary for those components.\n\nChange-Id: I5f7ce68ab99ccc152c86a9fb09c6661ec33b2a00\nSigned-off-by: Stephen Taylor \u003cstaylor@mirantis.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm/commit/dbfaf3cb011400b55a33a23dd1820e2202943624"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm/commit/dbfaf3cb011400b55a33a23dd1820e2202943624"}]},"branch":"refs/heads/master"},"0f9db114da7ff18e3ecbb125bdce817b460976fa":{"kind":"REWORK","_number":19,"created":"2026-09-14 18:24:07.000000000","uploader":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"ref":"refs/changes/32/1004032/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-helm","ref":"refs/changes/32/1004032/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-helm refs/changes/32/1004032/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-helm refs/changes/32/1004032/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-helm refs/changes/32/1004032/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-helm refs/changes/32/1004032/19"}}},"commit":{"parents":[{"commit":"85bb718fdd7bf555c58eae1fdc981b13d56fe742","subject":"Merge \"masakari: set kubernetes driver as default for host-monitor masakari-host-monitor changed to deployment Task: 55600 Story: 2011878\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm/commit/85bb718fdd7bf555c58eae1fdc981b13d56fe742"}]}],"author":{"name":"Stephen Taylor","email":"stephen.taylor.1@att.com","date":"2026-09-04 19:44:57.000000000","tz":-360},"committer":{"name":"Stephen Taylor","email":"stephen.taylor.1@att.com","date":"2026-09-14 18:23:47.000000000","tz":-360},"subject":"[ceph] Support both aes and aes256k CephX auth keys","message":"[ceph] Support both aes and aes256k CephX auth keys\n\nA new script has been added to the helm-toolkit chart to\nfacilitate CephX key rotation. Support for key rotate is required\nin the ceph-mon, ceph-osd, and ceph-rgw charts in order to comply\nwith secure (aes256k) key requirements in Ceph Tentacle 20.2.4.\n\nThe ceph-mon start script has been modified to support both aes and\naes256k CephX auth keys for the mon key and the bootstrap keys. Key\ntypes are now defined in values.yaml, and the start script attempts\nto rotate keys if existing keys don\u0027t match the expected key types.\n\nSimilar logic as been added to the mgr start script in ceph-mon,\nthe ceph-osd osd init scripts, and the ceph-rgw start script to\nattempt to rotate keys to the desired key type if specified.\n\nA desired key type option has also been added to the ceph-osd\nvalues since the ceph-osd type can be defined explicitly in some\ncases.\n\nThe ceph-mgr and ceph-rgw key types are inferred from the monmap\nconfig defined in ceph-mon values, so that config is queried to\ndetermine if key rotate is necessary for those components.\n\nChange-Id: I5f7ce68ab99ccc152c86a9fb09c6661ec33b2a00\nSigned-off-by: Stephen Taylor \u003cstaylor@mirantis.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm/commit/0f9db114da7ff18e3ecbb125bdce817b460976fa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm/commit/0f9db114da7ff18e3ecbb125bdce817b460976fa"}]},"branch":"refs/heads/master"},"4fe57674046c028b90ff71b5a029d10554405a88":{"kind":"REWORK","_number":20,"created":"2026-09-15 14:29:49.000000000","uploader":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"ref":"refs/changes/32/1004032/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-helm","ref":"refs/changes/32/1004032/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-helm refs/changes/32/1004032/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-helm refs/changes/32/1004032/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-helm refs/changes/32/1004032/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-helm refs/changes/32/1004032/20"}}},"commit":{"parents":[{"commit":"85bb718fdd7bf555c58eae1fdc981b13d56fe742","subject":"Merge \"masakari: set kubernetes driver as default for host-monitor masakari-host-monitor changed to deployment Task: 55600 Story: 2011878\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm/commit/85bb718fdd7bf555c58eae1fdc981b13d56fe742"}]}],"author":{"name":"Stephen Taylor","email":"stephen.taylor.1@att.com","date":"2026-09-04 19:44:57.000000000","tz":-360},"committer":{"name":"Stephen Taylor","email":"stephen.taylor.1@att.com","date":"2026-09-15 14:29:25.000000000","tz":-360},"subject":"[ceph] Support both aes and aes256k CephX auth keys","message":"[ceph] Support both aes and aes256k CephX auth keys\n\nA new script has been added to the helm-toolkit chart to\nfacilitate CephX key rotation. Support for key rotate is required\nin the ceph-mon, ceph-osd, and ceph-rgw charts in order to comply\nwith secure (aes256k) key requirements in Ceph Tentacle 20.2.4.\n\nThe ceph-mon start script has been modified to support both aes and\naes256k CephX auth keys for the mon key and the bootstrap keys. Key\ntypes are now defined in values.yaml, and the start script attempts\nto rotate keys if existing keys don\u0027t match the expected key types.\n\nSimilar logic as been added to the mgr start script in ceph-mon,\nthe ceph-osd osd init scripts, and the ceph-rgw start script to\nattempt to rotate keys to the desired key type if specified.\n\nA desired key type option has also been added to the ceph-osd\nvalues since the ceph-osd type can be defined explicitly in some\ncases.\n\nThe ceph-mgr and ceph-rgw key types are inferred from the monmap\nconfig defined in ceph-mon values, so that config is queried to\ndetermine if key rotate is necessary for those components.\n\nChange-Id: I5f7ce68ab99ccc152c86a9fb09c6661ec33b2a00\nSigned-off-by: Stephen Taylor \u003cstaylor@mirantis.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm/commit/4fe57674046c028b90ff71b5a029d10554405a88"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm/commit/4fe57674046c028b90ff71b5a029d10554405a88"}]},"branch":"refs/heads/master"},"34c1c5ed4de2653b80ca81ccb36f61649445de63":{"kind":"REWORK","_number":21,"created":"2026-09-15 17:46:05.000000000","uploader":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"ref":"refs/changes/32/1004032/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-helm","ref":"refs/changes/32/1004032/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-helm refs/changes/32/1004032/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-helm refs/changes/32/1004032/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-helm refs/changes/32/1004032/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-helm refs/changes/32/1004032/21"}}},"commit":{"parents":[{"commit":"85bb718fdd7bf555c58eae1fdc981b13d56fe742","subject":"Merge \"masakari: set kubernetes driver as default for host-monitor masakari-host-monitor changed to deployment Task: 55600 Story: 2011878\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm/commit/85bb718fdd7bf555c58eae1fdc981b13d56fe742"}]}],"author":{"name":"Stephen Taylor","email":"stephen.taylor.1@att.com","date":"2026-09-04 19:44:57.000000000","tz":-360},"committer":{"name":"Stephen Taylor","email":"stephen.taylor.1@att.com","date":"2026-09-15 17:45:58.000000000","tz":-360},"subject":"[ceph] Support both aes and aes256k CephX auth keys","message":"[ceph] Support both aes and aes256k CephX auth keys\n\nA new script has been added to the helm-toolkit chart to\nfacilitate CephX key rotation. Support for key rotate is required\nin the ceph-mon, ceph-osd, and ceph-rgw charts in order to comply\nwith secure (aes256k) key requirements in Ceph Tentacle 20.2.4.\n\nThe ceph-mon start script has been modified to support both aes and\naes256k CephX auth keys for the mon key and the bootstrap keys. Key\ntypes are now defined in values.yaml, and the start script attempts\nto rotate keys if existing keys don\u0027t match the expected key types.\n\nSimilar logic as been added to the mgr start script in ceph-mon,\nthe ceph-osd osd init scripts, and the ceph-rgw start script to\nattempt to rotate keys to the desired key type if specified.\n\nA desired key type option has also been added to the ceph-osd\nvalues since the ceph-osd type can be defined explicitly in some\ncases.\n\nThe ceph-mgr and ceph-rgw key types are inferred from the monmap\nconfig defined in ceph-mon values, so that config is queried to\ndetermine if key rotate is necessary for those components.\n\nChange-Id: I5f7ce68ab99ccc152c86a9fb09c6661ec33b2a00\nSigned-off-by: Stephen Taylor \u003cstaylor@mirantis.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm/commit/34c1c5ed4de2653b80ca81ccb36f61649445de63"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm/commit/34c1c5ed4de2653b80ca81ccb36f61649445de63"}]},"branch":"refs/heads/master"},"44f71d4cb1b891676f12ff3e81d9486156fdcc3d":{"kind":"REWORK","_number":22,"created":"2026-09-15 20:19:30.000000000","uploader":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"ref":"refs/changes/32/1004032/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-helm","ref":"refs/changes/32/1004032/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-helm refs/changes/32/1004032/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-helm refs/changes/32/1004032/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-helm refs/changes/32/1004032/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-helm refs/changes/32/1004032/22"}}},"commit":{"parents":[{"commit":"85bb718fdd7bf555c58eae1fdc981b13d56fe742","subject":"Merge \"masakari: set kubernetes driver as default for host-monitor masakari-host-monitor changed to deployment Task: 55600 Story: 2011878\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm/commit/85bb718fdd7bf555c58eae1fdc981b13d56fe742"}]}],"author":{"name":"Stephen Taylor","email":"stephen.taylor.1@att.com","date":"2026-09-04 19:44:57.000000000","tz":-360},"committer":{"name":"Stephen Taylor","email":"stephen.taylor.1@att.com","date":"2026-09-15 20:19:25.000000000","tz":-360},"subject":"[ceph] Support both aes and aes256k CephX auth keys","message":"[ceph] Support both aes and aes256k CephX auth keys\n\nA new script has been added to the helm-toolkit chart to\nfacilitate CephX key rotation. Support for key rotate is required\nin the ceph-mon, ceph-osd, and ceph-rgw charts in order to comply\nwith secure (aes256k) key requirements in Ceph Tentacle 20.2.4.\n\nThe ceph-mon start script has been modified to support both aes and\naes256k CephX auth keys for the mon key and the bootstrap keys. Key\ntypes are now defined in values.yaml, and the start script attempts\nto rotate keys if existing keys don\u0027t match the expected key types.\n\nSimilar logic as been added to the mgr start script in ceph-mon,\nthe ceph-osd osd init scripts, and the ceph-rgw start script to\nattempt to rotate keys to the desired key type if specified.\n\nA desired key type option has also been added to the ceph-osd\nvalues since the ceph-osd type can be defined explicitly in some\ncases.\n\nThe ceph-mgr and ceph-rgw key types are inferred from the monmap\nconfig defined in ceph-mon values, so that config is queried to\ndetermine if key rotate is necessary for those components.\n\nChange-Id: I5f7ce68ab99ccc152c86a9fb09c6661ec33b2a00\nSigned-off-by: Stephen Taylor \u003cstaylor@mirantis.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm/commit/44f71d4cb1b891676f12ff3e81d9486156fdcc3d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm/commit/44f71d4cb1b891676f12ff3e81d9486156fdcc3d"}]},"branch":"refs/heads/master"},"acb3f6c120a785fd2330572d548d1f849ae4790f":{"kind":"REWORK","_number":23,"created":"2026-09-16 12:17:46.000000000","uploader":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"ref":"refs/changes/32/1004032/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-helm","ref":"refs/changes/32/1004032/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-helm refs/changes/32/1004032/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-helm refs/changes/32/1004032/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-helm refs/changes/32/1004032/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-helm refs/changes/32/1004032/23"}}},"commit":{"parents":[{"commit":"85bb718fdd7bf555c58eae1fdc981b13d56fe742","subject":"Merge \"masakari: set kubernetes driver as default for host-monitor masakari-host-monitor changed to deployment Task: 55600 Story: 2011878\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm/commit/85bb718fdd7bf555c58eae1fdc981b13d56fe742"}]}],"author":{"name":"Stephen Taylor","email":"stephen.taylor.1@att.com","date":"2026-09-04 19:44:57.000000000","tz":-360},"committer":{"name":"Stephen Taylor","email":"stephen.taylor.1@att.com","date":"2026-09-16 12:17:19.000000000","tz":-360},"subject":"[ceph] Support both aes and aes256k CephX auth keys","message":"[ceph] Support both aes and aes256k CephX auth keys\n\nA new script has been added to the helm-toolkit chart to\nfacilitate CephX key rotation. Support for key rotate is required\nin the ceph-mon, ceph-osd, and ceph-rgw charts in order to comply\nwith secure (aes256k) key requirements in Ceph Tentacle 20.2.4.\n\nThe ceph-mon start script has been modified to support both aes and\naes256k CephX auth keys for the mon key and the bootstrap keys. Key\ntypes are now defined in values.yaml, and the start script attempts\nto rotate keys if existing keys don\u0027t match the expected key types.\n\nSimilar logic as been added to the mgr start script in ceph-mon,\nthe ceph-osd osd init scripts, and the ceph-rgw start script to\nattempt to rotate keys to the desired key type if specified.\n\nA desired key type option has also been added to the ceph-osd\nvalues since the ceph-osd type can be defined explicitly in some\ncases.\n\nThe ceph-mgr and ceph-rgw key types are inferred from the monmap\nconfig defined in ceph-mon values, so that config is queried to\ndetermine if key rotate is necessary for those components.\n\nChange-Id: I5f7ce68ab99ccc152c86a9fb09c6661ec33b2a00\nSigned-off-by: Stephen Taylor \u003cstaylor@mirantis.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm/commit/acb3f6c120a785fd2330572d548d1f849ae4790f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm/commit/acb3f6c120a785fd2330572d548d1f849ae4790f"}]},"branch":"refs/heads/master"},"e4b7a3596b0cbad8017c34cbf2a1fc1bcae3c8dc":{"kind":"REWORK","_number":24,"created":"2026-09-16 14:03:41.000000000","uploader":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"ref":"refs/changes/32/1004032/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-helm","ref":"refs/changes/32/1004032/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-helm refs/changes/32/1004032/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-helm refs/changes/32/1004032/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-helm refs/changes/32/1004032/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-helm refs/changes/32/1004032/24"}}},"commit":{"parents":[{"commit":"85bb718fdd7bf555c58eae1fdc981b13d56fe742","subject":"Merge \"masakari: set kubernetes driver as default for host-monitor masakari-host-monitor changed to deployment Task: 55600 Story: 2011878\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm/commit/85bb718fdd7bf555c58eae1fdc981b13d56fe742"}]}],"author":{"name":"Stephen Taylor","email":"stephen.taylor.1@att.com","date":"2026-09-04 19:44:57.000000000","tz":-360},"committer":{"name":"Stephen Taylor","email":"stephen.taylor.1@att.com","date":"2026-09-16 14:03:37.000000000","tz":-360},"subject":"[ceph] Support both aes and aes256k CephX auth keys","message":"[ceph] Support both aes and aes256k CephX auth keys\n\nA new script has been added to the helm-toolkit chart to\nfacilitate CephX key rotation. Support for key rotate is required\nin the ceph-mon, ceph-osd, and ceph-rgw charts in order to comply\nwith secure (aes256k) key requirements in Ceph Tentacle 20.2.4.\n\nThe ceph-mon start script has been modified to support both aes and\naes256k CephX auth keys for the mon key and the bootstrap keys. Key\ntypes are now defined in values.yaml, and the start script attempts\nto rotate keys if existing keys don\u0027t match the expected key types.\n\nSimilar logic as been added to the mgr start script in ceph-mon,\nthe ceph-osd osd init scripts, and the ceph-rgw start script to\nattempt to rotate keys to the desired key type if specified.\n\nA desired key type option has also been added to the ceph-osd\nvalues since the ceph-osd type can be defined explicitly in some\ncases.\n\nThe ceph-mgr and ceph-rgw key types are inferred from the monmap\nconfig defined in ceph-mon values, so that config is queried to\ndetermine if key rotate is necessary for those components.\n\nChange-Id: I5f7ce68ab99ccc152c86a9fb09c6661ec33b2a00\nSigned-off-by: Stephen Taylor \u003cstaylor@mirantis.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm/commit/e4b7a3596b0cbad8017c34cbf2a1fc1bcae3c8dc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm/commit/e4b7a3596b0cbad8017c34cbf2a1fc1bcae3c8dc"}]},"branch":"refs/heads/master"},"a22cfa42f91fd7cc81debb771e65487c13d921fa":{"kind":"REWORK","_number":25,"created":"2026-09-16 16:15:11.000000000","uploader":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"ref":"refs/changes/32/1004032/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-helm","ref":"refs/changes/32/1004032/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-helm refs/changes/32/1004032/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-helm refs/changes/32/1004032/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-helm refs/changes/32/1004032/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-helm refs/changes/32/1004032/25"}}},"commit":{"parents":[{"commit":"85bb718fdd7bf555c58eae1fdc981b13d56fe742","subject":"Merge \"masakari: set kubernetes driver as default for host-monitor masakari-host-monitor changed to deployment Task: 55600 Story: 2011878\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm/commit/85bb718fdd7bf555c58eae1fdc981b13d56fe742"}]}],"author":{"name":"Stephen Taylor","email":"stephen.taylor.1@att.com","date":"2026-09-04 19:44:57.000000000","tz":-360},"committer":{"name":"Stephen Taylor","email":"stephen.taylor.1@att.com","date":"2026-09-16 16:15:04.000000000","tz":-360},"subject":"[ceph] Support both aes and aes256k CephX auth keys","message":"[ceph] Support both aes and aes256k CephX auth keys\n\nA new script has been added to the helm-toolkit chart to\nfacilitate CephX key rotation. Support for key rotate is required\nin the ceph-mon, ceph-osd, and ceph-rgw charts in order to comply\nwith secure (aes256k) key requirements in Ceph Tentacle 20.2.4.\n\nThe ceph-mon start script has been modified to support both aes and\naes256k CephX auth keys for the mon key and the bootstrap keys. Key\ntypes are now defined in values.yaml, and the start script attempts\nto rotate keys if existing keys don\u0027t match the expected key types.\n\nSimilar logic as been added to the mgr start script in ceph-mon,\nthe ceph-osd osd init scripts, and the ceph-rgw start script to\nattempt to rotate keys to the desired key type if specified.\n\nA desired key type option has also been added to the ceph-osd\nvalues since the ceph-osd type can be defined explicitly in some\ncases.\n\nThe ceph-mgr and ceph-rgw key types are inferred from the monmap\nconfig defined in ceph-mon values, so that config is queried to\ndetermine if key rotate is necessary for those components.\n\nChange-Id: I5f7ce68ab99ccc152c86a9fb09c6661ec33b2a00\nSigned-off-by: Stephen Taylor \u003cstaylor@mirantis.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm/commit/a22cfa42f91fd7cc81debb771e65487c13d921fa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm/commit/a22cfa42f91fd7cc81debb771e65487c13d921fa"}]},"branch":"refs/heads/master"},"3d05b9c8e5f4ca8dcd67edc7e31161a063ae625d":{"kind":"REWORK","_number":26,"created":"2026-09-16 21:42:36.000000000","uploader":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"ref":"refs/changes/32/1004032/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-helm","ref":"refs/changes/32/1004032/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-helm refs/changes/32/1004032/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-helm refs/changes/32/1004032/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-helm refs/changes/32/1004032/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-helm refs/changes/32/1004032/26"}}},"commit":{"parents":[{"commit":"85bb718fdd7bf555c58eae1fdc981b13d56fe742","subject":"Merge \"masakari: set kubernetes driver as default for host-monitor masakari-host-monitor changed to deployment Task: 55600 Story: 2011878\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm/commit/85bb718fdd7bf555c58eae1fdc981b13d56fe742"}]}],"author":{"name":"Stephen Taylor","email":"stephen.taylor.1@att.com","date":"2026-09-04 19:44:57.000000000","tz":-360},"committer":{"name":"Stephen Taylor","email":"stephen.taylor.1@att.com","date":"2026-09-16 21:42:31.000000000","tz":-360},"subject":"[ceph] Support both aes and aes256k CephX auth keys","message":"[ceph] Support both aes and aes256k CephX auth keys\n\nA new script has been added to the helm-toolkit chart to\nfacilitate CephX key rotation. Support for key rotate is required\nin the ceph-mon, ceph-osd, and ceph-rgw charts in order to comply\nwith secure (aes256k) key requirements in Ceph Tentacle 20.2.4.\n\nThe ceph-mon start script has been modified to support both aes and\naes256k CephX auth keys for the mon key and the bootstrap keys. Key\ntypes are now defined in values.yaml, and the start script attempts\nto rotate keys if existing keys don\u0027t match the expected key types.\n\nSimilar logic as been added to the mgr start script in ceph-mon,\nthe ceph-osd osd init scripts, and the ceph-rgw start script to\nattempt to rotate keys to the desired key type if specified.\n\nA desired key type option has also been added to the ceph-osd\nvalues since the ceph-osd type can be defined explicitly in some\ncases.\n\nThe ceph-mgr and ceph-rgw key types are inferred from the monmap\nconfig defined in ceph-mon values, so that config is queried to\ndetermine if key rotate is necessary for those components.\n\nChange-Id: I5f7ce68ab99ccc152c86a9fb09c6661ec33b2a00\nSigned-off-by: Stephen Taylor \u003cstaylor@mirantis.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm/commit/3d05b9c8e5f4ca8dcd67edc7e31161a063ae625d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm/commit/3d05b9c8e5f4ca8dcd67edc7e31161a063ae625d"}]},"branch":"refs/heads/master"},"784e1a9ebb47c7f3352bf56b8c2080c437b99730":{"kind":"REWORK","_number":27,"created":"2026-09-22 18:39:29.000000000","uploader":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"ref":"refs/changes/32/1004032/27","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-helm","ref":"refs/changes/32/1004032/27","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-helm refs/changes/32/1004032/27 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-helm refs/changes/32/1004032/27 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-helm refs/changes/32/1004032/27 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-helm refs/changes/32/1004032/27"}}},"commit":{"parents":[{"commit":"7a13fefe85e45290ff11d2dc5dd8ce7275cb9852","subject":"Merge \"Spec: declarative Keystone identity management\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm/commit/7a13fefe85e45290ff11d2dc5dd8ce7275cb9852"}]}],"author":{"name":"Stephen Taylor","email":"stephen.taylor.1@att.com","date":"2026-09-04 19:44:57.000000000","tz":-360},"committer":{"name":"Stephen Taylor","email":"stephen.taylor.1@att.com","date":"2026-09-22 18:39:05.000000000","tz":-360},"subject":"[ceph] Support both aes and aes256k CephX auth keys","message":"[ceph] Support both aes and aes256k CephX auth keys\n\nA new script has been added to the helm-toolkit chart to\nfacilitate CephX key rotation. Support for key rotate is required\nin the ceph-mon, ceph-osd, and ceph-rgw charts in order to comply\nwith secure (aes256k) key requirements in Ceph Tentacle 20.2.4.\n\nThe ceph-mon start script has been modified to support both aes and\naes256k CephX auth keys for the mon key and the bootstrap keys. Key\ntypes are now defined in values.yaml, and the start script attempts\nto rotate keys if existing keys don\u0027t match the expected key types.\n\nSimilar logic as been added to the mgr start script in ceph-mon,\nthe ceph-osd osd init scripts, and the ceph-rgw start script to\nattempt to rotate keys to the desired key type if specified.\n\nA desired key type option has also been added to the ceph-osd\nvalues since the ceph-osd type can be defined explicitly in some\ncases.\n\nThe ceph-mgr and ceph-rgw key types are inferred from the monmap\nconfig defined in ceph-mon values, so that config is queried to\ndetermine if key rotate is necessary for those components.\n\nChange-Id: I5f7ce68ab99ccc152c86a9fb09c6661ec33b2a00\nSigned-off-by: Stephen Taylor \u003cstaylor@mirantis.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm/commit/784e1a9ebb47c7f3352bf56b8c2080c437b99730"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm/commit/784e1a9ebb47c7f3352bf56b8c2080c437b99730"}]},"branch":"refs/heads/master"},"e543cc15c9fd0d7755dc8c8adda93a18e6d99507":{"kind":"REWORK","_number":28,"created":"2026-09-24 15:38:17.000000000","uploader":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"ref":"refs/changes/32/1004032/28","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-helm","ref":"refs/changes/32/1004032/28","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-helm refs/changes/32/1004032/28 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-helm refs/changes/32/1004032/28 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-helm refs/changes/32/1004032/28 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-helm refs/changes/32/1004032/28"}}},"commit":{"parents":[{"commit":"f0429a776cb4d5e51cef8ab39a28287220cb3b36","subject":"Merge \"Stop losing charts when post events are superseded\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm/commit/f0429a776cb4d5e51cef8ab39a28287220cb3b36"}]}],"author":{"name":"Stephen Taylor","email":"stephen.taylor.1@att.com","date":"2026-09-04 19:44:57.000000000","tz":-360},"committer":{"name":"Stephen Taylor","email":"stephen.taylor.1@att.com","date":"2026-09-24 15:37:58.000000000","tz":-360},"subject":"[ceph] Support both aes and aes256k CephX auth keys","message":"[ceph] Support both aes and aes256k CephX auth keys\n\nA new script has been added to the helm-toolkit chart to\nfacilitate CephX key rotation. Support for key rotate is required\nin the ceph-mon, ceph-osd, and ceph-rgw charts in order to comply\nwith secure (aes256k) key requirements in Ceph Tentacle 20.2.4.\n\nThe ceph-mon start script has been modified to support both aes and\naes256k CephX auth keys for the mon key and the bootstrap keys. Key\ntypes are now defined in values.yaml, and the start script attempts\nto rotate keys if existing keys don\u0027t match the expected key types.\n\nSimilar logic as been added to the mgr start script in ceph-mon,\nthe ceph-osd osd init scripts, and the ceph-rgw start script to\nattempt to rotate keys to the desired key type if specified.\n\nA desired key type option has also been added to the ceph-osd\nvalues since the ceph-osd type can be defined explicitly in some\ncases.\n\nThe ceph-mgr and ceph-rgw key types are inferred from the monmap\nconfig defined in ceph-mon values, so that config is queried to\ndetermine if key rotate is necessary for those components.\n\nChange-Id: I5f7ce68ab99ccc152c86a9fb09c6661ec33b2a00\nSigned-off-by: Stephen Taylor \u003cstaylor@mirantis.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm/commit/e543cc15c9fd0d7755dc8c8adda93a18e6d99507"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm/commit/e543cc15c9fd0d7755dc8c8adda93a18e6d99507"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
