)]}'
{"id":"openstack%2Fopenstack-helm~1004246","triplet_id":"openstack%2Fopenstack-helm~master~Id77d281368629e6dd8093dac666cc66bc6732d1d","project":"openstack/openstack-helm","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-09-08 19:58:11.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id77d281368629e6dd8093dac666cc66bc6732d1d","subject":"tox: Drop redundant passenv","status":"MERGED","created":"2026-09-05 18:35:49.000000000","updated":"2026-09-08 20:00:17.000000000","submitted":"2026-09-08 19:58:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1004246","meta_rev_id":"b74595e626f3c3130a67a29d41638fb2c64e06b0","_number":1004246,"virtual_id_number":1004246,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},{"value":0,"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-08 19:58:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"all":[{"value":2,"date":"2026-09-08 17:56:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},{"value":2,"date":"2026-09-08 18:58:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},"all":[{"value":0,"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},{"value":1,"date":"2026-09-08 18:58:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-05 20:43:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-08 17:56:01.000000000","updated_by":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"reviewer":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"state":"REVIEWER"},{"updated":"2026-09-08 18:58:45.000000000","updated_by":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},"reviewer":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},"state":"REVIEWER"}],"messages":[{"id":"faccf4eae71846e4ccf42df7f6843cdc8977ce64","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-05 18:35:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"986760ffe057c0c314eeec4de0925b18a24a54b6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-05 20:43:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/181f8712725c423fb513b67d11b8e24c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e47ab00962184875be2218b372a809b6 : SUCCESS in 3m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9588b4aa6fc84c028345942d5ec5da10 : SUCCESS in 3m 37s\n- openstack-helm-linter https://zuul.opendev.org/t/openstack/build/4f41805119384a069a72b97709a3c4e6 : SUCCESS in 3m 17s\n- openstack-helm-mariadb-crd-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/037139662f1a41be80d01273cce63751 : SUCCESS in 1h 25m 15s\n- openstack-helm-mariadb-operator-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/840715b10c9a49d3a5e967d3852be1e7 : SUCCESS in 1h 57m 24s\n- openstack-helm-rabbitmq-crd-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/28d6f3024d124b5a9243ef42fbaaa3b3 : SUCCESS in 58m 29s\n- openstack-helm-rabbitmq-operator-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/ebf32b62cd974d8ba5919764c2d52a40 : SUCCESS in 1h 27m 37s\n- openstack-helm-pre-commit https://zuul.opendev.org/t/openstack/build/922a1fee37854891806ca9d70b48062a : SUCCESS in 4m 30s\n- openstack-helm-build-charts https://zuul.opendev.org/t/openstack/build/6a7c37d07e9040b2a1e0bbd8839340af : SUCCESS in 2m 36s\n- openstack-helm-cinder-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/97ab8b1beba249f1b339b5cf7443fd3b : SUCCESS in 43m 23s\n- openstack-helm-compute-kit-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/34b606203d7747a4bf91069bf92ba334 : SUCCESS in 1h 32m 05s\n- openstack-helm-cinder-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/e550080b15b64acb9a93e454c155ddcb : SUCCESS in 42m 02s\n- openstack-helm-compute-kit-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/175cf72b7b0544fd87dfbaef3576845c : SUCCESS in 1h 14m 19s\n- openstack-helm-cinder-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/b682f3955d2f4aaaaa62269575025fb2 : SUCCESS in 41m 31s\n- openstack-helm-compute-kit-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/dbf60829689a43a4a50b274f4f8be9d6 : SUCCESS in 1h 15m 40s\n- openstack-helm-tls-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/ad44bc27794349bbb79295beb28714e8 : SUCCESS in 1h 03m 37s\n- openstack-helm-octavia-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/d9e2acaaf6c446b5bf21f1e1f7a71aa2 : SUCCESS in 1h 29m 51s\n- openstack-helm-logging https://zuul.opendev.org/t/openstack/build/55c564b939924b749aa31dc864a26543 : SUCCESS in 36m 25s\n- openstack-helm-monitoring https://zuul.opendev.org/t/openstack/build/01f4acb77c644631b53b4683faf68281 : SUCCESS in 29m 29s","accounts_in_message":[],"_revision_number":1},{"id":"39a44e6ce06b1809df9be6d9b2b823b78b3fa5e0","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2026-09-08 17:56:01.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ac92697d2acb6126342170ca285d4e115e10aa16","author":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},"date":"2026-09-08 18:58:45.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ff1175a2b5fd9e9d5a9516d6aa3472e77f974a06","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-08 18:59:03.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a1451df8873ab58aab7a8014e57e063e98f16356","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-08 19:58:11.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/62353b27eb8e4037aba2fdfa6b185ce1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5922cb2519004862bbec72e2c598dd95 : SUCCESS in 4m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ac0e084b514f4a52b2a78aab59a8f923 : SUCCESS in 2m 31s\n- openstack-helm-linter https://zuul.opendev.org/t/openstack/build/9ce5aea539c44a16adb65dc088322e6f : SUCCESS in 2m 45s\n- openstack-helm-cinder-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/49ed2db5dbf443be9f15fe825a3b3062 : SUCCESS in 32m 13s\n- openstack-helm-compute-kit-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/1a098179620641a1b0bf938de5426550 : SUCCESS in 58m 41s\n- openstack-helm-logging https://zuul.opendev.org/t/openstack/build/c9387fa779ce48cb9d233614f522192f : SUCCESS in 23m 00s\n- openstack-helm-monitoring https://zuul.opendev.org/t/openstack/build/2823ca66a22b42fea90896a9644f2128 : SUCCESS in 30m 06s","accounts_in_message":[],"_revision_number":1},{"id":"19b24de318ce0dec472c6aaaa3bf4cd74ba09c81","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-08 19:58:11.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"b74595e626f3c3130a67a29d41638fb2c64e06b0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-08 20:00:17.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cfbfd96c231b46aea188f0f0f0bfe157\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ef09ccc44a548aa8c7e03c767139869 : SUCCESS in 49s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ab659eea67c64f26a17bed05bf0a98de : SUCCESS in 1m 47s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ed9009f08c210b49d8ca50e7cf1c36851ce32965","revisions":{"ed9009f08c210b49d8ca50e7cf1c36851ce32965":{"kind":"REWORK","_number":1,"created":"2026-09-05 18:35:49.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/46/1004246/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-helm","ref":"refs/changes/46/1004246/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-helm refs/changes/46/1004246/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-helm refs/changes/46/1004246/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-helm refs/changes/46/1004246/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-helm refs/changes/46/1004246/1"}}},"commit":{"parents":[{"commit":"7c81bea7cb7b6792b26807ac217099e0f50d520b","subject":"Merge \"Declarative RabbitMQ topology management\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm/commit/7c81bea7cb7b6792b26807ac217099e0f50d520b"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-05 18:35:24.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-05 18:35:24.000000000","tz":540},"subject":"tox: Drop redundant passenv","message":"tox: Drop redundant passenv\n\nSince 4.0.0[1], tox passes basic environment variables automatically\nand we don\u0027t have to add these explicitly.\n\ncf. https://tox.wiki/en/latest/reference/config.html#passenv\n\nSome variables are always passed through to ensure the basic\nfunctionality of standard library functions or tooling like pip.\nThis is also not case sensitive on all platforms except Windows.\n\n[1] https://github.com/tox-dev/tox/commit/afb3b395dd\n\nAlso remove the VIRTUAL_ENV environment because the same is already\npassed by tox.\n\nChange-Id: Id77d281368629e6dd8093dac666cc66bc6732d1d\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm/commit/ed9009f08c210b49d8ca50e7cf1c36851ce32965"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm/commit/ed9009f08c210b49d8ca50e7cf1c36851ce32965"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
