)]}'
{"id":"openstack%2Fopenstack-helm~1004800","triplet_id":"openstack%2Fopenstack-helm~master~I3f222984cca42c29df5907421a0d382c2960cb80","project":"openstack/openstack-helm","branch":"master","attention_set":{},"removed_from_attention_set":{"37387":{"account":{"_account_id":37387,"name":"Alberto Molina Coballes","display_name":"Alberto Molina Coballes","email":"a.molinacoballes@fairbanks.nl","username":"a.molinacoballes","status":"Cloud Engineer at fairbanks.nl"},"last_update":"2026-09-10 16:03:04.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I3f222984cca42c29df5907421a0d382c2960cb80","subject":"masakari: set kubernetes driver as default for host-monitor masakari-host-monitor changed to deployment Task: 55600 Story: 2011878","status":"MERGED","created":"2026-09-09 11:01:50.000000000","updated":"2026-09-10 16:04:57.000000000","submitted":"2026-09-10 16:03:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1004800","meta_rev_id":"a436474e1c4dfc9434b117dc59ccc73b6454511f","_number":1004800,"virtual_id_number":1004800,"owner":{"_account_id":37387,"name":"Alberto Molina Coballes","display_name":"Alberto Molina Coballes","email":"a.molinacoballes@fairbanks.nl","username":"a.molinacoballes","status":"Cloud Engineer at fairbanks.nl"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},{"value":0,"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-10 16:03:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"all":[{"value":2,"date":"2026-09-10 01:30:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},{"value":2,"date":"2026-09-10 14:33:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},"all":[{"value":0,"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},{"value":1,"date":"2026-09-10 14:33:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-09 13:48:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-09 22:32:45.000000000","updated_by":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"reviewer":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"state":"CC"},{"updated":"2026-09-10 01:30:06.000000000","updated_by":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"reviewer":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"state":"REVIEWER"},{"updated":"2026-09-10 14:33:01.000000000","updated_by":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},"reviewer":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},"state":"REVIEWER"}],"messages":[{"id":"501199d7e6ca61fe26107e883e05bc3c21b55fe8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37387,"name":"Alberto Molina Coballes","display_name":"Alberto Molina Coballes","email":"a.molinacoballes@fairbanks.nl","username":"a.molinacoballes","status":"Cloud Engineer at fairbanks.nl"},"date":"2026-09-09 11:01:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9eedb25b79f39cca5b4b58b5bb3c2697e8e3c568","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-09 13:48:58.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a0a31d1ed30d4e17a45be38b8a899b3b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b1d8bc052d54142b75b5e3963a07087 : SUCCESS in 2m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5029e7e234114bb5a0cfab860db2c6e9 : SUCCESS in 4m 03s\n- openstack-helm-linter https://zuul.opendev.org/t/openstack/build/b8bb6c1df3f24dda8b26ed9a4eadec0d : SUCCESS in 3m 23s\n- openstack-helm-mariadb-crd-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/eb9a4c6005fa42c9948537757f03e9f5 : FAILURE in 2h 20m 20s\n- openstack-helm-mariadb-operator-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/7d98bc40b09048c7a18232ef273b2d23 : FAILURE in 2h 37m 01s\n- openstack-helm-rabbitmq-crd-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/89ec7549284548488ad679b26b11ca8b : SUCCESS in 42m 48s\n- openstack-helm-rabbitmq-operator-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/38f0682c2b8b456b86ff0f9e4392f5e2 : SUCCESS in 46m 24s\n- openstack-helm-pre-commit https://zuul.opendev.org/t/openstack/build/71ac3e9bf94441bd92f52be27152aef4 : SUCCESS in 4m 21s\n- openstack-helm-build-charts https://zuul.opendev.org/t/openstack/build/cbb57568b69140a1b2a930a73a044f51 : SUCCESS in 2m 52s\n- openstack-helm-cinder-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/fcb2f0bd1a194b28ba7eade04c8f334d : SUCCESS in 43m 13s\n- openstack-helm-compute-kit-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/05f3db593f9843a1a79fa06cc3ccf91c : SUCCESS in 1h 17m 04s\n- openstack-helm-cinder-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/b151b1aec4f44ce482952b7b633bc8e9 : FAILURE in 21m 18s\n- openstack-helm-compute-kit-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/c348c38b20944c0980482084ca232592 : SUCCESS in 1h 19m 44s\n- openstack-helm-cinder-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/be947ec754b74bad9acc12827fac7622 : SUCCESS in 47m 46s\n- openstack-helm-compute-kit-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/a4147b647e394ac0a7be1e40e69e43f4 : SUCCESS in 1h 15m 06s\n- openstack-helm-tls-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/5bcb607b860748fdbeff90686bf34486 : FAILURE in 2h 40m 54s\n- openstack-helm-octavia-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/ba827d4c37104919943453105b12598d : SUCCESS in 43m 03s\n- openstack-helm-logging https://zuul.opendev.org/t/openstack/build/97b0c312ac4e4b20acec50b8ca603221 : FAILURE in 7m 04s\n- openstack-helm-monitoring https://zuul.opendev.org/t/openstack/build/ffee5d1b7379450bb2a59d6e5df0e4da : SUCCESS in 26m 37s","accounts_in_message":[],"_revision_number":1},{"id":"2b4cf325674723b65f21f8121081e9c49c2d658e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37387,"name":"Alberto Molina Coballes","display_name":"Alberto Molina Coballes","email":"a.molinacoballes@fairbanks.nl","username":"a.molinacoballes","status":"Cloud Engineer at fairbanks.nl"},"real_author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2026-09-09 20:16:32.000000000","message":"Patch Set 2: Patch Set 1 was rebased on behalf of \u003cGERRIT_ACCOUNT_37387\u003e\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[{"_account_id":37387,"name":"Alberto Molina Coballes","display_name":"Alberto Molina Coballes","email":"a.molinacoballes@fairbanks.nl","username":"a.molinacoballes","status":"Cloud Engineer at fairbanks.nl"}],"_revision_number":2},{"id":"25bc0d424e77099b43ca1eb2d2bda7a123d1f6e3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-09 22:03:17.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b738edbcc1e7453cbac9da315c6851b6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b4ab130975e549329b05d2502e4b9ed2 : SUCCESS in 5m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e4e43130f6884b3aa9fe3196e1c0ce33 : SUCCESS in 3m 24s\n- openstack-helm-linter https://zuul.opendev.org/t/openstack/build/dcb495115fe54b4ea46a014f3e5c096b : SUCCESS in 4m 27s\n- openstack-helm-mariadb-crd-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/803d5e356e0645c9a9a8443dfbfe7d61 : SUCCESS in 1h 04m 34s\n- openstack-helm-mariadb-operator-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/bdf2a00ef5504011931969791f37d287 : SUCCESS in 1h 33m 36s\n- openstack-helm-rabbitmq-crd-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/743570e1d11843e39bf3e46a24cd09a9 : SUCCESS in 1h 06m 12s\n- openstack-helm-rabbitmq-operator-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/465ffe5a81584d599bd073be1d326c73 : SUCCESS in 1h 17m 57s\n- openstack-helm-pre-commit https://zuul.opendev.org/t/openstack/build/fdac0af825924869953a1a6ed38fc56c : SUCCESS in 3m 19s\n- openstack-helm-build-charts https://zuul.opendev.org/t/openstack/build/031f8be626bc4732aebf388d9c7b505d : SUCCESS in 2m 10s\n- openstack-helm-cinder-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/1f5db15abc1e4d39b51f837db80bd200 : SUCCESS in 36m 48s\n- openstack-helm-compute-kit-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/c04c38192d144edc8bbcb64b24edd0ce : SUCCESS in 50m 17s\n- openstack-helm-cinder-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/45afd00799084dc79e515837bf77ac70 : SUCCESS in 51m 26s\n- openstack-helm-compute-kit-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/ebcbde999068450d8b81f8412a29a62d : SUCCESS in 1h 15m 11s\n- openstack-helm-cinder-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/3dc6a9bbbbdc4c2b8d50ca56633d5c63 : SUCCESS in 56m 00s\n- openstack-helm-compute-kit-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/a29ce08824fc4532b94e8c739d1ceeac : SUCCESS in 50m 54s\n- openstack-helm-tls-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/115c506bb9b34a41b375b8d622f5b51c : FAILURE in 1h 41m 57s\n- openstack-helm-octavia-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/638ece876ce94a4a8e40b9959e2b81f7 : SUCCESS in 1h 15m 47s\n- openstack-helm-logging https://zuul.opendev.org/t/openstack/build/d524b824932f4072b601d64af734d1f3 : SUCCESS in 22m 36s\n- openstack-helm-monitoring https://zuul.opendev.org/t/openstack/build/3c650b6ed7a74a27be14df7f5fe519dc : SUCCESS in 29m 02s","accounts_in_message":[],"_revision_number":2},{"id":"ccce946b126f9c01abb6934ad565eae5af542f37","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2026-09-09 22:32:45.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d93c083bfe76b8d649208491f83fd432b54a0249","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-10 00:28:45.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cce3ab0b26d145459db5f24fad8319a8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/268738ac8c7a4e649735f22f856455cb : SUCCESS in 4m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1721396f29224ee8813b099fa6b9fc0d : SUCCESS in 4m 28s\n- openstack-helm-linter https://zuul.opendev.org/t/openstack/build/1e0566bee0874eea9071377a80293b36 : SUCCESS in 3m 19s\n- openstack-helm-mariadb-crd-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/698661b25e90437381c813765c308709 : SUCCESS in 1h 34m 57s\n- openstack-helm-mariadb-operator-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/23fbd502aab54159a66c5391b90aa56d : SUCCESS in 1h 36m 53s\n- openstack-helm-rabbitmq-crd-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/7001b0c8181e49a988c5eb768d1302cf : SUCCESS in 1h 16m 24s\n- openstack-helm-rabbitmq-operator-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/360067a9f7b24108b64fecbef2c377ce : SUCCESS in 47m 34s\n- openstack-helm-pre-commit https://zuul.opendev.org/t/openstack/build/fabc0b38dffd4d87889c9437973128d2 : SUCCESS in 4m 08s\n- openstack-helm-build-charts https://zuul.opendev.org/t/openstack/build/435bfb8aadac440bb13751158e65414f : SUCCESS in 1m 46s\n- openstack-helm-cinder-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/e6d6b9070f7947d5aae26c0b0f825c75 : SUCCESS in 27m 38s\n- openstack-helm-compute-kit-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/75aa6134b5ce4d968024b78ad56d32f9 : SUCCESS in 1h 03m 13s\n- openstack-helm-cinder-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/dffa1a69b59346e6a72e7cc17462f4ee : SUCCESS in 42m 49s\n- openstack-helm-compute-kit-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/1879853e8dd0442fbfc1d44aa7459a25 : SUCCESS in 1h 15m 55s\n- openstack-helm-cinder-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/55f1eda9a1a44adeb365c7688cf29fad : SUCCESS in 28m 27s\n- openstack-helm-compute-kit-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/87a68d31f44b4d8eafe5fe830a0e0d4b : SUCCESS in 45m 18s\n- openstack-helm-tls-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/6823c978cdea4cfca84fc1909e75fff8 : SUCCESS in 1h 50m 37s\n- openstack-helm-octavia-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/1493fd2b8777498a9400e9616ab11f52 : SUCCESS in 1h 23m 44s\n- openstack-helm-logging https://zuul.opendev.org/t/openstack/build/268d82f1986d4f878f20aafe2a3462a2 : SUCCESS in 39m 08s\n- openstack-helm-monitoring https://zuul.opendev.org/t/openstack/build/bc4745e2a968444eab1adf420f0a094a : SUCCESS in 21m 18s","accounts_in_message":[],"_revision_number":2},{"id":"69be879e6d43ad93b4a2d0a7ea2c97eefd548e55","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2026-09-10 01:30:06.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"0038ead8d9a1185b121a48dde50d0bb659bdaa73","author":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},"date":"2026-09-10 14:33:01.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"62bf3df1aaf8d2c90a572d5965698c4cdbfbd069","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-10 14:33:14.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"1f76af36727beb486c1abbb90ada3ce3c5f52364","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-10 16:03:04.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c1747c9db5d74425804fc5454b457d88\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/75a0148e19a14696aadb47103b7a23a7 : SUCCESS in 2m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a8fb2275ce3b45f7b266acdf4c149497 : SUCCESS in 3m 07s\n- openstack-helm-linter https://zuul.opendev.org/t/openstack/build/66ffc19d693940ddaf118586fac5bd7a : SUCCESS in 3m 49s\n- openstack-helm-cinder-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/afc850dec3e14225ac5b5c63adea0fa5 : SUCCESS in 30m 01s\n- openstack-helm-compute-kit-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/10a58d25ee04447d8adade7635a82a79 : SUCCESS in 1h 24m 33s\n- openstack-helm-logging https://zuul.opendev.org/t/openstack/build/8d852ad62384478e85f83a9e95a52ab9 : SUCCESS in 39m 09s\n- openstack-helm-monitoring https://zuul.opendev.org/t/openstack/build/54fc93554abd45249fdf56d0df9dd97a : SUCCESS in 31m 17s","accounts_in_message":[],"_revision_number":2},{"id":"a0dbcf80e4feeefc9004b188d4c97142e6339a86","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-10 16:03:04.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"a436474e1c4dfc9434b117dc59ccc73b6454511f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-10 16:04:57.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/43418e2f6baf4e76bd09efaf727c78ac\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/12be44ae79a9470b9718171909244da4 : SUCCESS in 50s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"aa9a6bc72452d1bbc5399309f707bf8a48e9c702","revisions":{"f6bb8959280860a56af48796d8271c1007076b82":{"kind":"REWORK","_number":1,"created":"2026-09-09 11:01:50.000000000","uploader":{"_account_id":37387,"name":"Alberto Molina Coballes","display_name":"Alberto Molina Coballes","email":"a.molinacoballes@fairbanks.nl","username":"a.molinacoballes","status":"Cloud Engineer at fairbanks.nl"},"ref":"refs/changes/00/1004800/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-helm","ref":"refs/changes/00/1004800/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-helm refs/changes/00/1004800/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-helm refs/changes/00/1004800/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-helm refs/changes/00/1004800/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-helm refs/changes/00/1004800/1"}}},"commit":{"parents":[{"commit":"aa72dac1a2cc570d49d17cb4578ee340c4621844","subject":"Merge \"glance: point the liveness probe at the uwsgi stats port\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm/commit/aa72dac1a2cc570d49d17cb4578ee340c4621844"}]}],"author":{"name":"Alberto Molina Coballes","email":"a.molinacoballes@fairbanks.nl","date":"2026-09-09 10:59:56.000000000","tz":120},"committer":{"name":"Alberto Molina Coballes","email":"a.molinacoballes@fairbanks.nl","date":"2026-09-09 11:00:49.000000000","tz":120},"subject":"masakari: set kubernetes driver as default for host-monitor masakari-host-monitor changed to deployment Task: 55600 Story: 2011878","message":"masakari: set kubernetes driver as default for host-monitor\nmasakari-host-monitor changed to deployment\nTask: 55600\nStory: 2011878\n\nChange-Id: I3f222984cca42c29df5907421a0d382c2960cb80\nSigned-off-by: Alberto Molina Coballes \u003ca.molinacoballes@fairbanks.nl\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm/commit/f6bb8959280860a56af48796d8271c1007076b82"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm/commit/f6bb8959280860a56af48796d8271c1007076b82"}]},"branch":"refs/heads/master"},"aa9a6bc72452d1bbc5399309f707bf8a48e9c702":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-09-09 20:16:32.000000000","uploader":{"_account_id":37387,"name":"Alberto Molina Coballes","display_name":"Alberto Molina Coballes","email":"a.molinacoballes@fairbanks.nl","username":"a.molinacoballes","status":"Cloud Engineer at fairbanks.nl"},"real_uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/00/1004800/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-helm","ref":"refs/changes/00/1004800/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-helm refs/changes/00/1004800/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-helm refs/changes/00/1004800/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-helm refs/changes/00/1004800/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-helm refs/changes/00/1004800/2"}}},"commit":{"parents":[{"commit":"033a007b398b096431e1da2add0b4aebc78b0900","subject":"Merge \"deploy: keep the OpenStack CephX keys on the aes cipher\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm/commit/033a007b398b096431e1da2add0b4aebc78b0900"}]}],"author":{"name":"Alberto Molina Coballes","email":"a.molinacoballes@fairbanks.nl","date":"2026-09-09 10:59:56.000000000","tz":120},"committer":{"name":"Alberto Molina Coballes","email":"a.molinacoballes@fairbanks.nl","date":"2026-09-09 20:16:32.000000000","tz":0},"subject":"masakari: set kubernetes driver as default for host-monitor masakari-host-monitor changed to deployment Task: 55600 Story: 2011878","message":"masakari: set kubernetes driver as default for host-monitor\nmasakari-host-monitor changed to deployment\nTask: 55600\nStory: 2011878\n\nChange-Id: I3f222984cca42c29df5907421a0d382c2960cb80\nSigned-off-by: Alberto Molina Coballes \u003ca.molinacoballes@fairbanks.nl\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm/commit/aa9a6bc72452d1bbc5399309f707bf8a48e9c702"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm/commit/aa9a6bc72452d1bbc5399309f707bf8a48e9c702"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"aa72dac1a2cc570d49d17cb4578ee340c4621844","ours":"f6bb8959280860a56af48796d8271c1007076b82","theirs":"033a007b398b096431e1da2add0b4aebc78b0900","merge_strategy":"recursive","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
