)]}'
{"id":"openstack%2Fopenstack-helm~1006556","triplet_id":"openstack%2Fopenstack-helm~master~I0267a9358b52ac00f1d3a4a4a59466a1ea304524","project":"openstack/openstack-helm","branch":"master","attention_set":{},"removed_from_attention_set":{"3009":{"account":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"last_update":"2026-09-21 18:46:04.000000000","reason":"\u003cGERRIT_ACCOUNT_3009\u003e replied on the change","reason_account":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"}}},"hashtags":[],"change_id":"I0267a9358b52ac00f1d3a4a4a59466a1ea304524","subject":"neutron: Noop change to rebuild and republish the chart","status":"MERGED","created":"2026-09-21 14:37:33.000000000","updated":"2026-09-21 19:54:18.000000000","submitted":"2026-09-21 19:53:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1006556","meta_rev_id":"ad1c0c87a7dacbf63b197c7b4c17f39ee96d9f94","_number":1006556,"virtual_id_number":1006556,"owner":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},{"value":0,"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-21 19:53:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"all":[{"value":0,"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},{"value":2,"date":"2026-09-21 15:27:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"all":[{"value":1,"date":"2026-09-21 18:46:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},{"value":1,"date":"2026-09-21 15:27:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-21 15:27:51.000000000","updated_by":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"reviewer":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"state":"REVIEWER"},{"updated":"2026-09-21 16:26:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"83d57adc660a9e46e14f7110dca25e512aa05d35","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2026-09-21 14:37:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f1895473ac7d76bc6af3b5075412834a7cf7906d","author":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"date":"2026-09-21 15:27:51.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a9f679b969acd1f97375f53eb7b1443b26713496","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 16:26:44.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3820316aad344ab3965942ac0fe56e69\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/456e30a52bcf4f0b8a94df957502c95f : SUCCESS in 2m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b28431e3c32a4f10bf6f22684e77c08a : SUCCESS in 3m 45s\n- openstack-helm-linter https://zuul.opendev.org/t/openstack/build/4e4050864f554c609668ff1d2d2615d4 : SUCCESS in 2m 13s\n- openstack-helm-mariadb-crd-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/220339dbba6d428982287d06624f019c : SUCCESS in 1h 38m 37s\n- openstack-helm-mariadb-operator-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/42531dadb9344e29a3df4604243e0c24 : SUCCESS in 1h 29m 00s\n- openstack-helm-rabbitmq-crd-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/374ce3b39f9e41dbb640e7e750a6cd31 : FAILURE in 41m 14s\n- openstack-helm-rabbitmq-operator-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/41359721e66f4673930b0a79092edadf : FAILURE in 9m 20s\n- openstack-helm-pre-commit https://zuul.opendev.org/t/openstack/build/1ded61bb552e46d485eab7a1231ced47 : SUCCESS in 2m 40s\n- openstack-helm-build-charts https://zuul.opendev.org/t/openstack/build/e69c5e4504a64de9a3a769193eb06a82 : FAILURE in 3m 30s\n- openstack-helm-cinder-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/045fd443618b43c0b6f433b1a7045be4 : SUCCESS in 46m 52s\n- openstack-helm-compute-kit-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/51766ce4a9f648fab5f3d4d07be2fbb8 : FAILURE in 8m 45s\n- openstack-helm-cinder-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/a4e593be25f04b05959779e3d8da96d4 : SUCCESS in 47m 58s\n- openstack-helm-compute-kit-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/ebb631bc2b884060b14654a8f23bcf50 : SUCCESS in 1h 04m 37s\n- openstack-helm-cinder-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/440ff041806f47cb927fab2540d5212f : FAILURE in 11m 59s\n- openstack-helm-compute-kit-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/20eb3471fbda4f14add2f60f61c00adc : FAILURE in 6m 09s\n- openstack-helm-tls-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/a20c388735b74f55bb3ee8949ee83633 : SUCCESS in 1h 46m 22s\n- openstack-helm-compute-kit-dpdk-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/406b1d113b4044129a7e688b1d27c6c9 : SUCCESS in 57m 53s\n- openstack-helm-octavia-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/e661fa69ebbe429bb40713bf426bbec5 : FAILURE in 7m 30s\n- openstack-helm-logging https://zuul.opendev.org/t/openstack/build/83e390f050a149d2bc9b4a8f34a3b4da : FAILURE in 11m 12s\n- openstack-helm-monitoring https://zuul.opendev.org/t/openstack/build/8575adcd32bd4b26a893a72311b0ef9f : SUCCESS in 25m 16s","accounts_in_message":[],"_revision_number":1},{"id":"5427e4c41409ebfa557179e0c35d50e977822622","author":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"date":"2026-09-21 16:49:22.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bddb35d680d88acf6e4cf9ca7d744aa7213b8ee9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 18:43:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6952eb60dfce4f65b7c94d7d502f68bc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eda07da8b13a437d88540fc014b73a46 : SUCCESS in 3m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/118f001e722f409b8f4b17d867887b25 : SUCCESS in 3m 08s\n- openstack-helm-linter https://zuul.opendev.org/t/openstack/build/b5813a4c4df84275ac1c70aff99c46cb : SUCCESS in 3m 57s\n- openstack-helm-mariadb-crd-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/45a2170796334521a51afd2232bcd6af : SUCCESS in 1h 35m 58s\n- openstack-helm-mariadb-operator-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/72a8dcf59a9540aca7a4433ad3335737 : SUCCESS in 1h 40m 59s\n- openstack-helm-rabbitmq-crd-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/8f3d5bc4e9b147c79886237c0632bba8 : SUCCESS in 1h 14m 04s\n- openstack-helm-rabbitmq-operator-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/1bc6baba9e5448cb93d26246c0977237 : SUCCESS in 1h 09m 15s\n- openstack-helm-pre-commit https://zuul.opendev.org/t/openstack/build/d5d7eb5e57764d9a931ba73c4ea1398d : SUCCESS in 3m 17s\n- openstack-helm-build-charts https://zuul.opendev.org/t/openstack/build/42e94b2673844a63b2f3d956c37b254e : SUCCESS in 4m 15s\n- openstack-helm-cinder-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/1504ddcf747347faa737fdfc159fc320 : SUCCESS in 45m 19s\n- openstack-helm-compute-kit-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/8670cc614d944064bc02b3962f353adc : SUCCESS in 42m 59s\n- openstack-helm-cinder-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/d1e842568dce482391da58967a9b43fe : SUCCESS in 45m 17s\n- openstack-helm-compute-kit-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/16a446850414402fa0ba6fd064b7f686 : SUCCESS in 1h 19m 02s\n- openstack-helm-cinder-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/12046b1394774f0a8ad56dae35df496a : SUCCESS in 44m 08s\n- openstack-helm-compute-kit-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/54886dd3b90a4c01b61f98909081923e : SUCCESS in 1h 20m 45s\n- openstack-helm-tls-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/4ccd175476fa4350a6db63e748cf18fb : SUCCESS in 59m 56s\n- openstack-helm-compute-kit-dpdk-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/d43cc4ff01a4469295386a6aa64741a0 : SUCCESS in 57m 54s\n- openstack-helm-octavia-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/0b578447958641eeb5b1b7bd51c8fe14 : SUCCESS in 1h 16m 03s\n- openstack-helm-logging https://zuul.opendev.org/t/openstack/build/25fed827297140bbb9c4a10331a60b69 : SUCCESS in 36m 12s\n- openstack-helm-monitoring https://zuul.opendev.org/t/openstack/build/4d982855ca4f48aa858e396e1125c313 : SUCCESS in 27m 55s","accounts_in_message":[],"_revision_number":1},{"id":"04796c3e743369b064535251df584a76774af74f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 18:43:57.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b2545b66dc09fb21618468e26c519e3d3ec1b3db","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2026-09-21 18:46:04.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8fd039fc5431dae5e01a054cd5b70703178c4536","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 19:53:03.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0ed18b6db7f048f883e1fe9e3c3c3fb6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/86a1d8a5dd5e42808b405490b42d97d7 : SUCCESS in 3m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0434a4f4ce8442f79eb148b5c890c708 : SUCCESS in 3m 04s\n- openstack-helm-linter https://zuul.opendev.org/t/openstack/build/ef9b4b05bf874ddbbec7af9bd0fb52e1 : SUCCESS in 3m 07s\n- openstack-helm-cinder-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/4d7fbb40fef442ee97041dee11cac0c1 : SUCCESS in 37m 23s\n- openstack-helm-compute-kit-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/d74b32cc1193465fb0eeea8a9ac42833 : SUCCESS in 1h 06m 34s\n- openstack-helm-logging https://zuul.opendev.org/t/openstack/build/173c0f61282444a6a0e21f35afc94e5f : SUCCESS in 35m 51s\n- openstack-helm-monitoring https://zuul.opendev.org/t/openstack/build/2d53f2a6d601409db2a3a26238015449 : SUCCESS in 29m 03s","accounts_in_message":[],"_revision_number":1},{"id":"bf632173552292a68c5c2af187d9c1d00e9df587","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 19:53:03.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"ad1c0c87a7dacbf63b197c7b4c17f39ee96d9f94","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 19:54:18.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5a0fc5261fb546769b40c32cb5bc271b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa8544e3a03348f6bf073d3c73cafc3b : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"76eae2d5ad88cfa1d5ba5eadec57bb28aebb7cb1","revisions":{"76eae2d5ad88cfa1d5ba5eadec57bb28aebb7cb1":{"kind":"REWORK","_number":1,"created":"2026-09-21 14:37:33.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/56/1006556/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-helm","ref":"refs/changes/56/1006556/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-helm refs/changes/56/1006556/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-helm refs/changes/56/1006556/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-helm refs/changes/56/1006556/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-helm refs/changes/56/1006556/1"}}},"commit":{"parents":[{"commit":"0105ad1a628747465d760e13b9107cbc6f6c64e6","subject":"Merge \"Fix the Selenium dashboard tests, stop ignoring them\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm/commit/0105ad1a628747465d760e13b9107cbc6f6c64e6"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2026-09-21 14:37:10.000000000","tz":-300},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2026-09-21 14:37:10.000000000","tz":-300},"subject":"neutron: Noop change to rebuild and republish the chart","message":"neutron: Noop change to rebuild and republish the chart\n\nSigned-off-by: Vladimir Kozhukalov \u003ckozhukalov@gmail.com\u003e\nChange-Id: I0267a9358b52ac00f1d3a4a4a59466a1ea304524\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm/commit/76eae2d5ad88cfa1d5ba5eadec57bb28aebb7cb1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm/commit/76eae2d5ad88cfa1d5ba5eadec57bb28aebb7cb1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
