)]}'
{"id":"openstack%2Fopenstack-helm~1007705","triplet_id":"openstack%2Fopenstack-helm~master~I7b3ca2047dfc2d63bc5359551a2af900b98f1b47","project":"openstack/openstack-helm","branch":"master","topic":"story/2011883","attention_set":{"36741":{"account":{"_account_id":36741,"name":"Eunkyung Kim","display_name":"eunkyung","email":"ek121.kim@samsung.com","username":"eunkyung"},"last_update":"2026-09-30 15:36:25.000000000","reason":"\u003cGERRIT_ACCOUNT_3009\u003e replied on the change","reason_account":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"}}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I7b3ca2047dfc2d63bc5359551a2af900b98f1b47","subject":"Rally: Add CronJob to run tests periodically","status":"NEW","created":"2026-09-28 12:12:27.000000000","updated":"2026-09-30 15:36:25.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"c226796b90593e92ed238ee86de35397b69da1bc","_number":1007705,"virtual_id_number":1007705,"owner":{"_account_id":36741,"name":"Eunkyung Kim","display_name":"eunkyung","email":"ek121.kim@samsung.com","username":"eunkyung"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2026-09-28 14:04:25.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-28 14:04:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-30 15:36:25.000000000","updated_by":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"reviewer":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"state":"CC"}],"messages":[{"id":"20cbb6d720808a94e74783c95804884cadae5b8e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36741,"name":"Eunkyung Kim","display_name":"eunkyung","email":"ek121.kim@samsung.com","username":"eunkyung"},"date":"2026-09-28 12:12:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3d53d09592728e870be64b2ee8773c021f7c8dc6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-28 14:04:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0816d3dc46624c2f80cd3e35203b2ac4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e09456ef24c644f58abcc3ed2865e71f : SUCCESS in 3m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2403549e410e4e788966a9fef3d0148d : SUCCESS in 2m 19s\n- openstack-helm-linter https://zuul.opendev.org/t/openstack/build/36bcde704c9341d4adce8d60b2caa33b : SUCCESS in 4m 27s\n- openstack-helm-pre-commit https://zuul.opendev.org/t/openstack/build/ce0dcddff6ae44e5b3e85adfac029f40 : SUCCESS in 2m 39s\n- openstack-helm-build-charts https://zuul.opendev.org/t/openstack/build/a5c42d69df6d4ac0ae3cdfbbd52319c7 : SUCCESS in 3m 54s\n- openstack-helm-compute-kit-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/8c4bc5c2c0404a92ad2e509d616a2f0b : SUCCESS in 1h 44m 15s\n- openstack-helm-compute-kit-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/34343d645a66411095ca5779ae20f0c0 : SUCCESS in 1h 43m 24s\n- openstack-helm-compute-kit-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/e92e18368ab445c7b68e6e40190c5241 : SUCCESS in 59m 40s\n- openstack-helm-tls-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/cae924f6bf774290af3a229b9919096c : SUCCESS in 1h 46m 32s\n- openstack-helm-crd-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/b449c56355624cbbb45a21e05f9b6a6e : SUCCESS in 1h 04m 16s\n- openstack-helm-operators-2026-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/dc1028189e024ed29f833daa8c3e80f5 : SUCCESS in 1h 06m 06s\n- openstack-helm-logging https://zuul.opendev.org/t/openstack/build/c78270dd30104b26a575fedb8716c593 : SUCCESS in 40m 38s\n- openstack-helm-monitoring https://zuul.opendev.org/t/openstack/build/a20c971d6abc4b62adb3eaf6e4b3dd80 : SUCCESS in 30m 01s","accounts_in_message":[],"_revision_number":1},{"id":"c226796b90593e92ed238ee86de35397b69da1bc","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2026-09-30 15:36:25.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1ba2a4577e6c4e0027a24e9ccba27adaea58dd25","revisions":{"1ba2a4577e6c4e0027a24e9ccba27adaea58dd25":{"kind":"REWORK","_number":1,"created":"2026-09-28 12:12:27.000000000","uploader":{"_account_id":36741,"name":"Eunkyung Kim","display_name":"eunkyung","email":"ek121.kim@samsung.com","username":"eunkyung"},"ref":"refs/changes/05/1007705/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-helm","ref":"refs/changes/05/1007705/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-helm refs/changes/05/1007705/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-helm refs/changes/05/1007705/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-helm refs/changes/05/1007705/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-helm refs/changes/05/1007705/1"}}},"commit":{"parents":[{"commit":"a606f21e37e41e71fb5c3100f3809cb851343ec8","subject":"rabbitmq: adjust helm test script to not require `partitions` key","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm/commit/a606f21e37e41e71fb5c3100f3809cb851343ec8"}]}],"author":{"name":"Eunkyung99","email":"ek121.kim@samsung.com","date":"2026-09-28 12:11:54.000000000","tz":540},"committer":{"name":"Eunkyung99","email":"ek121.kim@samsung.com","date":"2026-09-28 12:11:54.000000000","tz":540},"subject":"Rally: Add CronJob to run tests periodically","message":"Rally: Add CronJob to run tests periodically\n\nAdd an optional CronJob to run rally tasks periodically.\nIt is disabled by default and can be enabled with\nmanifests.cron_job_run_task. The schedule, retry count\nand history limits are configured under jobs.cron_run_task,\nand reports are stored in a separate PVC configured under cron_pvc\nand created when manifests.pvc_rally_cron is enabled.\nNote that rally stores the results of every run in its database.\n\nStory: 2011883\nTask: 55686\nChange-Id: I7b3ca2047dfc2d63bc5359551a2af900b98f1b47\nSigned-off-by: eunkyung \u003cek121.kim@samsung.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm/commit/1ba2a4577e6c4e0027a24e9ccba27adaea58dd25"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm/commit/1ba2a4577e6c4e0027a24e9ccba27adaea58dd25"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
