)]}'
{"id":"openstack%2Fopenstack-helm~688644","triplet_id":"openstack%2Fopenstack-helm~master~I61379a5e00de4311c02c3f64cbe7c7345a9b3569","project":"openstack/openstack-helm","branch":"master","topic":"upstream","hashtags":[],"change_id":"I61379a5e00de4311c02c3f64cbe7c7345a9b3569","subject":"Neutron: Add support for disabling Readiness/Liveness probes","status":"ABANDONED","created":"2019-10-15 07:50:56.000000000","updated":"2019-11-14 07:29:57.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"6edf44711302d971b44300bc46b0868698aed1c6","_number":688644,"virtual_id_number":688644,"owner":{"_account_id":28570,"name":"zhipeng liu","email":"zhipengs.liu@intel.com","username":"zhipengl"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":29397,"name":"Kudaka Poorna Rajesh","email":"rajesh.kudaka@imaginea.com","username":"rajesh.kudaka"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2019-10-15 08:58:42.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":29397,"name":"Kudaka Poorna Rajesh","email":"rajesh.kudaka@imaginea.com","username":"rajesh.kudaka"},"all":[{"value":-1,"date":"2019-10-16 08:59:27.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":29397,"name":"Kudaka Poorna Rajesh","email":"rajesh.kudaka@imaginea.com","username":"rajesh.kudaka"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":29397,"name":"Kudaka Poorna Rajesh","email":"rajesh.kudaka@imaginea.com","username":"rajesh.kudaka"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29397,"name":"Kudaka Poorna Rajesh","email":"rajesh.kudaka@imaginea.com","username":"rajesh.kudaka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-15 08:58:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-10-16 08:59:27.000000000","updated_by":{"_account_id":29397,"name":"Kudaka Poorna Rajesh","email":"rajesh.kudaka@imaginea.com","username":"rajesh.kudaka"},"reviewer":{"_account_id":29397,"name":"Kudaka Poorna Rajesh","email":"rajesh.kudaka@imaginea.com","username":"rajesh.kudaka"},"state":"REVIEWER"}],"messages":[{"id":"b8c48573a1df0163eabbf80d6a72a429a3b83b7e","author":{"_account_id":28570,"name":"zhipeng liu","email":"zhipengs.liu@intel.com","username":"zhipengl"},"date":"2019-10-15 07:50:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"690d0ee017735874fa3462728d2cfc8dc3a4df0d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-15 08:58:42.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3fa7e17680c54a9c9e32cb5e4eebc891 : SUCCESS in 5m 56s\n- openstack-helm-lint https://zuul.opendev.org/t/openstack/build/6b3b65517b1c46819143e97a397b1f2d : SUCCESS in 2m 25s\n- openstack-helm-cinder https://zuul.opendev.org/t/openstack/build/01d07bc8cef64e5993c8b7b6559eab69 : SUCCESS in 33m 15s\n- openstack-helm-cinder-pike-ubuntu_xenial https://zuul.opendev.org/t/openstack/build/d9057ae3c492474180b899488b49d9c1 : SUCCESS in 35m 15s\n- openstack-helm-cinder-queens-ubuntu_xenial https://zuul.opendev.org/t/openstack/build/d56f172770c64c209becc906d51c97bd : SUCCESS in 39m 10s\n- openstack-helm-cinder-rocky-ubuntu_xenial https://zuul.opendev.org/t/openstack/build/f19b03ce654d4510a72b5117bff35a62 : SUCCESS in 41m 24s\n- openstack-helm-cinder-rocky-ubuntu_bionic https://zuul.opendev.org/t/openstack/build/f8789d3301fd4b0daf2009b429d964a9 : SUCCESS in 36m 07s\n- openstack-helm-cinder-rocky-opensuse_15 https://zuul.opendev.org/t/openstack/build/53a81cfd86954291923231546f890663 : SUCCESS in 36m 03s\n- openstack-helm-cinder-stein-ubuntu_bionic https://zuul.opendev.org/t/openstack/build/c90de70becff492d93216464f4cbcdd2 : SUCCESS in 38m 42s\n- openstack-helm-compute-kit https://zuul.opendev.org/t/openstack/build/a3842175734f4bea82439b0053d6f829 : SUCCESS in 36m 36s\n- openstack-helm-compute-kit-pike-ubuntu_xenial https://zuul.opendev.org/t/openstack/build/e6869eaa555d4f6d9be14c71960c7a92 : SUCCESS in 44m 14s\n- openstack-helm-compute-kit-queens-ubuntu_xenial https://zuul.opendev.org/t/openstack/build/8c6bc76f168444258affe4cad8ababa3 : SUCCESS in 39m 59s\n- openstack-helm-compute-kit-rocky-ubuntu_xenial https://zuul.opendev.org/t/openstack/build/dc2108f4ea564283b2b73b1e140b7eee : SUCCESS in 42m 53s\n- openstack-helm-compute-kit-rocky-ubuntu_bionic https://zuul.opendev.org/t/openstack/build/b4626cdae69c4f839d98a5ee93f16300 : SUCCESS in 44m 38s\n- openstack-helm-compute-kit-rocky-opensuse_15 https://zuul.opendev.org/t/openstack/build/0ae93af5842f4069bd35ef11b8e77e7b : FAILURE in 57m 34s\n- openstack-helm-compute-kit-stein-ubuntu_bionic https://zuul.opendev.org/t/openstack/build/a243233317774437b07b33d97872462b : SUCCESS in 47m 44s\n- openstack-helm-horizon https://zuul.opendev.org/t/openstack/build/f57feaab45a448f9842026b0fd582aaf : SUCCESS in 22m 22s\n- openstack-helm-apparmor https://zuul.opendev.org/t/openstack/build/b06f6255e769456fa5f0fbbbc59ad6d3 : SUCCESS in 39m 16s (non-voting)\n- openstack-helm-netpol-compute-kit https://zuul.opendev.org/t/openstack/build/b5ee3ccca92640e4aa2fff255820fbad : SUCCESS in 44m 16s (non-voting)\n- openstack-helm-netpol-cinder https://zuul.opendev.org/t/openstack/build/16e29ec991484c8cbeab5880ec218abd : SUCCESS in 34m 49s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ecb4473ed8fdcf6b59b3249caef67ff1d36dea16","author":{"_account_id":29397,"name":"Kudaka Poorna Rajesh","email":"rajesh.kudaka@imaginea.com","username":"rajesh.kudaka"},"date":"2019-10-16 08:59:27.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)\n\nSame should apply to all the files in this PS.","accounts_in_message":[],"_revision_number":1},{"id":"889fe4ee4a00a83c079749790293aee3d238d9f5","author":{"_account_id":28570,"name":"zhipeng liu","email":"zhipengs.liu@intel.com","username":"zhipengl"},"date":"2019-11-14 07:29:57.000000000","message":"Abandoned\n\nThanks Rajesh!\nNo need anymore","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"29eee9ffa6e4997965d11e5efd1005e2dc089843","revisions":{"29eee9ffa6e4997965d11e5efd1005e2dc089843":{"kind":"REWORK","_number":1,"created":"2019-10-15 07:50:56.000000000","uploader":{"_account_id":28570,"name":"zhipeng liu","email":"zhipengs.liu@intel.com","username":"zhipengl"},"ref":"refs/changes/44/688644/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-helm","ref":"refs/changes/44/688644/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-helm refs/changes/44/688644/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-helm refs/changes/44/688644/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-helm refs/changes/44/688644/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-helm refs/changes/44/688644/1"}}},"commit":{"parents":[{"commit":"d88e3f80055ccaf22bff9f48c76ce173479456ce","subject":"Merge \"[Glance] add job-metadefs-load\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm/commit/d88e3f80055ccaf22bff9f48c76ce173479456ce"}]}],"author":{"name":"zhipengl","email":"zhipengs.liu@intel.com","date":"2019-10-15 16:38:26.000000000","tz":480},"committer":{"name":"zhipengl","email":"zhipengs.liu@intel.com","date":"2019-10-15 16:38:26.000000000","tz":480},"subject":"Neutron: Add support for disabling Readiness/Liveness probes","message":"Neutron: Add support for disabling Readiness/Liveness probes\n\nWith the introduction of Readiness/Liveness probes in\nIb99ceaabbad1d1e0faf34cc74314da9aa688fa0a, some probes are failing and\npreventing successful armada manifest applies.\n\nAdd support to disable the probes.\n\nChange-Id: I61379a5e00de4311c02c3f64cbe7c7345a9b3569\nSigned-off-by: Zhipeng Liu \u003czhipengs.liu@intel.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm/commit/29eee9ffa6e4997965d11e5efd1005e2dc089843"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-helm/commit/29eee9ffa6e4997965d11e5efd1005e2dc089843"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
