)]}'
{"id":"openstack%2Fopenstack-manuals~1006577","triplet_id":"openstack%2Fopenstack-manuals~master~I1980e5cdd168c10a71e64e7011a6048f94347e4b","project":"openstack/openstack-manuals","branch":"master","topic":"www-hibiscus-final","attention_set":{"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2026-09-21 15:50:28.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{"15993":{"account":{"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},"last_update":"2026-09-23 20:48:07.000000000","reason":"\u003cGERRIT_ACCOUNT_15993\u003e replied on the change","reason_account":{"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"}}},"hashtags":[],"change_id":"I1980e5cdd168c10a71e64e7011a6048f94347e4b","subject":"[www] Setup 2027.1 Indri and add project data to Hibiscus","status":"NEW","created":"2026-09-21 15:32:41.000000000","updated":"2026-09-23 20:48:07.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"dc73d9bd44b3a96d7710ccb69a0616233a559b21","_number":1006577,"virtual_id_number":1006577,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-09-22 08:58:44.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},"recommended":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"all":[{"value":2,"date":"2026-09-23 20:48:07.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-09-23 17:01:35.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-21 15:50:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-23 17:01:35.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2026-09-23 17:01:35.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},"state":"REVIEWER"}],"messages":[{"id":"a56fa0b9400bf42a6161b6deafd64c20994b849b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-09-21 15:32:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b8685faa9f0577008fe6de2ad2b99819d42f46b0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 15:50:28.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3336509637014df6a49d8cbf647fcbe7\n\n- build-tox-manuals-publishlang https://zuul.opendev.org/t/openstack/build/bcb2a7394ca24c29b44d21b8220c1169 : SUCCESS in 10m 54s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/6b0fd81b6f3a43d59a93d9dbcbb787dd : SUCCESS in 5m 55s\n- build-tox-manuals-publishdocs https://zuul.opendev.org/t/openstack/build/0e9134cdde9e444fbd86442c656cf904 : FAILURE in 5m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d3d14c776d6b4262b46f58c33379b93c : SUCCESS in 6m 11s","accounts_in_message":[],"_revision_number":1},{"id":"6f4b0c6ff69ced4c3b9d3c987256fe263a5ab40c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-09-22 08:51:08.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"00bbf684d45f404fd846742806b3fff25a1cf5fd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-22 08:58:44.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7701ddfd0b0b4c8485a008bac51a20d8\n\n- build-tox-manuals-publishlang https://zuul.opendev.org/t/openstack/build/7b474e6b30054aa7bb19243e3d80bbb2 : SUCCESS in 5m 15s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/fa43ee6159b84aafa615ef573f17ba15 : SUCCESS in 3m 05s\n- build-tox-manuals-publishdocs https://zuul.opendev.org/t/openstack/build/42a14b55848c47f3875cd369ef07fe6a : SUCCESS in 6m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/85f0cd9226bc4382b2049bd0455393f7 : SUCCESS in 4m 07s","accounts_in_message":[],"_revision_number":2},{"id":"ffe513c7ed714c03dfeb02a49efd850920f7300c","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-23 17:01:35.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"dc73d9bd44b3a96d7710ccb69a0616233a559b21","author":{"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},"date":"2026-09-23 20:48:07.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"4691526f0e74fffa61abc38cee04a1e6649f340a","revisions":{"4a92afad0b46b235795cf2c1dff30b4e31a81397":{"kind":"REWORK","_number":1,"created":"2026-09-21 15:32:41.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/77/1006577/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-manuals","ref":"refs/changes/77/1006577/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-manuals refs/changes/77/1006577/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-manuals refs/changes/77/1006577/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-manuals refs/changes/77/1006577/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-manuals refs/changes/77/1006577/1"}}},"commit":{"parents":[{"commit":"22ce59f07f6b521a642ff3a5bf6edbca2bea221f","subject":"[glossary] Add 2027.1 Indri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-manuals/commit/22ce59f07f6b521a642ff3a5bf6edbca2bea221f"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2026-09-21 15:28:36.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2026-09-21 15:32:08.000000000","tz":120},"subject":"[www] Setup 2027.1 Indri and add project data to Hibiscus","message":"[www] Setup 2027.1 Indri and add project data to Hibiscus\n\nThis change adds\n* default pages for next release, 2027.1 Indri\n* 2026.2 Hibiscus project data so that index pages will be generated\n\nIt still keeps 2026.2 Hibiscus in development phase. A future change\nwill transition it to \u0027maintained\u0027 at release time.\n\nThe patch content was setup by following the www page release\nguideline [1].\n\n[1] https://docs.openstack.org/doc-contrib-guide/release/taskdetail.html#update-www-pages-for-end-of-release\n\nChange-Id: I1980e5cdd168c10a71e64e7011a6048f94347e4b\nSigned-off-by: Elod Illes \u003celod.illes@est.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-manuals/commit/4a92afad0b46b235795cf2c1dff30b4e31a81397"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-manuals/commit/4a92afad0b46b235795cf2c1dff30b4e31a81397"}]},"branch":"refs/heads/master"},"4691526f0e74fffa61abc38cee04a1e6649f340a":{"kind":"REWORK","_number":2,"created":"2026-09-22 08:51:08.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/77/1006577/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-manuals","ref":"refs/changes/77/1006577/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-manuals refs/changes/77/1006577/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-manuals refs/changes/77/1006577/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-manuals refs/changes/77/1006577/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-manuals refs/changes/77/1006577/2"}}},"commit":{"parents":[{"commit":"22ce59f07f6b521a642ff3a5bf6edbca2bea221f","subject":"[glossary] Add 2027.1 Indri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-manuals/commit/22ce59f07f6b521a642ff3a5bf6edbca2bea221f"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2026-09-21 15:28:36.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2026-09-22 08:50:56.000000000","tz":120},"subject":"[www] Setup 2027.1 Indri and add project data to Hibiscus","message":"[www] Setup 2027.1 Indri and add project data to Hibiscus\n\nThis change adds\n* default pages for next release, 2027.1 Indri\n* 2026.2 Hibiscus project data so that index pages will be generated\n\nIt still keeps 2026.2 Hibiscus in development phase. A future change\nwill transition it to \u0027maintained\u0027 at release time.\n\nThe patch content was setup by following the www page release\nguideline [1].\n\n[1] https://docs.openstack.org/doc-contrib-guide/release/taskdetail.html#update-www-pages-for-end-of-release\n\nChange-Id: I1980e5cdd168c10a71e64e7011a6048f94347e4b\nSigned-off-by: Elod Illes \u003celod.illes@est.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-manuals/commit/4691526f0e74fffa61abc38cee04a1e6649f340a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-manuals/commit/4691526f0e74fffa61abc38cee04a1e6649f340a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"}},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
