)]}'
{"id":"openstack%2Fopenstack-manuals~245163","triplet_id":"openstack%2Fopenstack-manuals~master~I6b52e700463906e2705465d4c2cc01ca45aad49e","project":"openstack/openstack-manuals","branch":"master","topic":"bp/config-ref-rst","hashtags":[],"change_id":"I6b52e700463906e2705465d4c2cc01ca45aad49e","subject":"Un-comment config options for consistency","status":"MERGED","created":"2015-11-13 14:06:47.000000000","updated":"2015-11-13 19:51:28.000000000","submitted":"2015-11-13 19:51:26.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"8fbd2058d29317d293e10b9deebba432b0270f19","_number":245163,"virtual_id_number":245163,"owner":{"_account_id":6635,"name":"John Davidge","email":"john.davidge@rackspace.com","username":"john-davidge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":0,"_account_id":10497,"name":"KATO Tomoyuki","email":"kato.tomoyuki@jp.fujitsu.com","username":"to222"},{"value":0,"_account_id":17361,"name":"Shashirekha Gundur","email":"shashirekha.j.gundur@intel.com","username":"shashirekha.j.gundur"},{"value":2,"date":"2015-11-13 19:51:26.000000000","_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":6772,"name":"Stephen Gordon","email":"sgordon@redhat.com","username":"sgordon"},{"value":0,"_account_id":17832,"name":"Linette Williams","email":"stemke@us.ibm.com","username":"lwilliams"},{"value":0,"_account_id":6635,"name":"John Davidge","email":"john.davidge@rackspace.com","username":"john-davidge"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"recommended":{"_account_id":17361,"name":"Shashirekha Gundur","email":"shashirekha.j.gundur@intel.com","username":"shashirekha.j.gundur"},"all":[{"value":0,"_account_id":10497,"name":"KATO Tomoyuki","email":"kato.tomoyuki@jp.fujitsu.com","username":"to222"},{"value":1,"date":"2015-11-13 15:39:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17361,"name":"Shashirekha Gundur","email":"shashirekha.j.gundur@intel.com","username":"shashirekha.j.gundur"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":2,"date":"2015-11-13 15:57:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":2,"date":"2015-11-13 19:29:02.000000000","_account_id":6772,"name":"Stephen Gordon","email":"sgordon@redhat.com","username":"sgordon"},{"value":1,"date":"2015-11-13 15:40:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17832,"name":"Linette Williams","email":"stemke@us.ibm.com","username":"lwilliams"},{"value":0,"_account_id":6635,"name":"John Davidge","email":"john.davidge@rackspace.com","username":"john-davidge"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6772,"name":"Stephen Gordon","email":"sgordon@redhat.com","username":"sgordon"},"all":[{"value":0,"_account_id":10497,"name":"KATO Tomoyuki","email":"kato.tomoyuki@jp.fujitsu.com","username":"to222"},{"value":0,"_account_id":17361,"name":"Shashirekha Gundur","email":"shashirekha.j.gundur@intel.com","username":"shashirekha.j.gundur"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":1,"date":"2015-11-13 19:29:02.000000000","_account_id":6772,"name":"Stephen Gordon","email":"sgordon@redhat.com","username":"sgordon"},{"value":0,"_account_id":17832,"name":"Linette Williams","email":"stemke@us.ibm.com","username":"lwilliams"},{"value":0,"_account_id":6635,"name":"John Davidge","email":"john.davidge@rackspace.com","username":"john-davidge"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":6635,"name":"John Davidge","email":"john.davidge@rackspace.com","username":"john-davidge"},{"_account_id":6772,"name":"Stephen Gordon","email":"sgordon@redhat.com","username":"sgordon"},{"_account_id":10497,"name":"KATO Tomoyuki","email":"kato.tomoyuki@jp.fujitsu.com","username":"to222"},{"_account_id":17361,"name":"Shashirekha Gundur","email":"shashirekha.j.gundur@intel.com","username":"shashirekha.j.gundur"},{"_account_id":17832,"name":"Linette Williams","email":"stemke@us.ibm.com","username":"lwilliams"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2015-11-13 14:37:16.000000000","updated_by":{"_account_id":10497,"name":"KATO Tomoyuki","email":"kato.tomoyuki@jp.fujitsu.com","username":"to222"},"reviewer":{"_account_id":10497,"name":"KATO Tomoyuki","email":"kato.tomoyuki@jp.fujitsu.com","username":"to222"},"state":"REVIEWER"},{"updated":"2015-11-13 15:39:40.000000000","updated_by":{"_account_id":17361,"name":"Shashirekha Gundur","email":"shashirekha.j.gundur@intel.com","username":"shashirekha.j.gundur"},"reviewer":{"_account_id":17361,"name":"Shashirekha Gundur","email":"shashirekha.j.gundur@intel.com","username":"shashirekha.j.gundur"},"state":"REVIEWER"},{"updated":"2015-11-13 15:40:25.000000000","updated_by":{"_account_id":17832,"name":"Linette Williams","email":"stemke@us.ibm.com","username":"lwilliams"},"reviewer":{"_account_id":17832,"name":"Linette Williams","email":"stemke@us.ibm.com","username":"lwilliams"},"state":"REVIEWER"},{"updated":"2015-11-13 15:57:40.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2015-11-13 19:29:02.000000000","updated_by":{"_account_id":6772,"name":"Stephen Gordon","email":"sgordon@redhat.com","username":"sgordon"},"reviewer":{"_account_id":6772,"name":"Stephen Gordon","email":"sgordon@redhat.com","username":"sgordon"},"state":"REVIEWER"},{"updated":"2015-11-13 19:51:26.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"52dfed80690314802b173f32c0ca3ce914fb9009","author":{"_account_id":6635,"name":"John Davidge","email":"john.davidge@rackspace.com","username":"john-davidge"},"date":"2015-11-13 14:06:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0875c7d93ae63e3c75d8a287de69a2c57958e374","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-11-13 14:17:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-openstack-manuals-tox-checkniceness http://logs.openstack.org/63/245163/1/check/gate-openstack-manuals-tox-checkniceness/8dcb8b8/ : SUCCESS in 4m 01s\n- gate-openstack-manuals-tox-checkdeletions http://logs.openstack.org/63/245163/1/check/gate-openstack-manuals-tox-checkdeletions/8b00d4d/ : SUCCESS in 4m 24s\n- gate-openstack-manuals-tox-doc-publish-checkbuild http://docs-draft.openstack.org/63/245163/1/check/gate-openstack-manuals-tox-doc-publish-checkbuild/b3981ad//publish-docs/ : SUCCESS in 5m 20s","accounts_in_message":[],"_revision_number":1},{"id":"e59478ddd693b7823106f99ec4df638651c5f08c","author":{"_account_id":10497,"name":"KATO Tomoyuki","email":"kato.tomoyuki@jp.fujitsu.com","username":"to222"},"date":"2015-11-13 14:37:16.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)\n\nCloud you also update the same point in the \"doc/common/section_config_format.xml\" file?\n\nAlso, please associate the bug report.","accounts_in_message":[],"_revision_number":1},{"id":"fd522a488cb7e4c895769bd473e64e1ccfa20117","author":{"_account_id":6635,"name":"John Davidge","email":"john.davidge@rackspace.com","username":"john-davidge"},"date":"2015-11-13 14:50:10.000000000","message":"Patch Set 1:\n\nAh, I hadn\u0027t noticed a bug has been filed for this. Thanks.","accounts_in_message":[],"_revision_number":1},{"id":"89599e150d0824574b1edb412967876309ca19b9","author":{"_account_id":6635,"name":"John Davidge","email":"john.davidge@rackspace.com","username":"john-davidge"},"date":"2015-11-13 14:54:48.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"dffc377a8adc6eeaae7bc1d562df8cd8e95b755d","author":{"_account_id":6635,"name":"John Davidge","email":"john.davidge@rackspace.com","username":"john-davidge"},"date":"2015-11-13 14:54:52.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c2d7eab5916442b1778cbcf5bfcaaa10dbe87595","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-11-13 15:30:25.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-openstack-manuals-tox-checklinks http://logs.openstack.org/63/245163/2/check/gate-openstack-manuals-tox-checklinks/54553e1/ : SUCCESS in 2m 31s (non-voting)\n- gate-openstack-manuals-tox-checkniceness http://logs.openstack.org/63/245163/2/check/gate-openstack-manuals-tox-checkniceness/aa6b396/ : SUCCESS in 3m 33s\n- gate-openstack-manuals-tox-checksyntax http://logs.openstack.org/63/245163/2/check/gate-openstack-manuals-tox-checksyntax/47ed020/ : SUCCESS in 2m 16s\n- gate-openstack-manuals-tox-checkdeletions http://logs.openstack.org/63/245163/2/check/gate-openstack-manuals-tox-checkdeletions/2084afc/ : SUCCESS in 2m 57s\n- gate-openstack-manuals-tox-doc-publish-checkbuild http://docs-draft.openstack.org/63/245163/2/check/gate-openstack-manuals-tox-doc-publish-checkbuild/64a3317//publish-docs/ : SUCCESS in 22m 03s","accounts_in_message":[],"_revision_number":2},{"id":"9f109df33e5261fc8caaca09d2dadfcabc46f7fa","author":{"_account_id":17361,"name":"Shashirekha Gundur","email":"shashirekha.j.gundur@intel.com","username":"shashirekha.j.gundur"},"date":"2015-11-13 15:39:40.000000000","message":"Patch Set 2: Code-Review+1\n\nlooks good!","accounts_in_message":[],"_revision_number":2},{"id":"455b942f3230b3639f6fbf053354dfa54171cd16","author":{"_account_id":17832,"name":"Linette Williams","email":"stemke@us.ibm.com","username":"lwilliams"},"date":"2015-11-13 15:40:25.000000000","message":"Patch Set 2: Code-Review+1\n\ngood catch.","accounts_in_message":[],"_revision_number":2},{"id":"75457ecd9bcc9218426e1f809edf3e1134c84622","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2015-11-13 15:57:40.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"08429e3dfae91ca3ccd0783de2ba77678e1f21ae","author":{"_account_id":6772,"name":"Stephen Gordon","email":"sgordon@redhat.com","username":"sgordon"},"date":"2015-11-13 19:29:02.000000000","message":"Patch Set 2: Workflow+1 Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"173249b6e6c3bb2c0263d215bb9aa332352ccc77","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-11-13 19:29:08.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":2},{"id":"f234a5e955726499c2f0b040b8e809b8fae8b776","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-11-13 19:51:26.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- gate-openstack-manuals-tox-checkniceness http://logs.openstack.org/63/245163/2/gate/gate-openstack-manuals-tox-checkniceness/befec05/ : SUCCESS in 3m 08s\n- gate-openstack-manuals-tox-checksyntax http://logs.openstack.org/63/245163/2/gate/gate-openstack-manuals-tox-checksyntax/6072d7f/ : SUCCESS in 2m 25s\n- gate-openstack-manuals-tox-checkdeletions http://logs.openstack.org/63/245163/2/gate/gate-openstack-manuals-tox-checkdeletions/7e6519f/ : SUCCESS in 3m 00s\n- gate-openstack-manuals-tox-doc-publish-checkbuild http://docs-draft.openstack.org/63/245163/2/gate/gate-openstack-manuals-tox-doc-publish-checkbuild/b4f4ff0//publish-docs/ : SUCCESS in 21m 30s","accounts_in_message":[],"_revision_number":2},{"id":"5bf66f8570902f70d09d30d9c4d300b9f9b1f356","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-11-13 19:51:28.000000000","message":"Change has been successfully merged into the git repository.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"c400d9b2f06bf821b0438b57950f0de4bab201a4","revisions":{"c1cf95df8168be16d8066521d8a2886dbe3793f0":{"kind":"REWORK","_number":1,"created":"2015-11-13 14:06:47.000000000","uploader":{"_account_id":6635,"name":"John Davidge","email":"john.davidge@rackspace.com","username":"john-davidge"},"ref":"refs/changes/63/245163/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-manuals","ref":"refs/changes/63/245163/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-manuals refs/changes/63/245163/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-manuals refs/changes/63/245163/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-manuals refs/changes/63/245163/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-manuals refs/changes/63/245163/1"}}},"commit":{"parents":[{"commit":"e135ca8442befda40ba396ef94952a2062b67d34","subject":"Merge \"small changes to shared_file_systems file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-manuals/commit/e135ca8442befda40ba396ef94952a2062b67d34"}]}],"author":{"name":"John Davidge","email":"jodavidg@cisco.com","date":"2015-11-13 14:00:54.000000000","tz":0},"committer":{"name":"John Davidge","email":"jodavidg@cisco.com","date":"2015-11-13 14:00:54.000000000","tz":0},"subject":"Un-comment config options for consistency","message":"Un-comment config options for consistency\n\nTwo of the config options in this doc are commented out.\n\nUn-comment them to match with the rest of the document,\nand to improve readability.\n\nChange-Id: I6b52e700463906e2705465d4c2cc01ca45aad49e\nPartially-Implements: blueprint config-ref-rst\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-manuals/commit/c1cf95df8168be16d8066521d8a2886dbe3793f0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-manuals/commit/c1cf95df8168be16d8066521d8a2886dbe3793f0"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"e135ca8442befda40ba396ef94952a2062b67d34","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"c400d9b2f06bf821b0438b57950f0de4bab201a4":{"kind":"REWORK","_number":2,"created":"2015-11-13 14:54:48.000000000","uploader":{"_account_id":6635,"name":"John Davidge","email":"john.davidge@rackspace.com","username":"john-davidge"},"ref":"refs/changes/63/245163/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-manuals","ref":"refs/changes/63/245163/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-manuals refs/changes/63/245163/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-manuals refs/changes/63/245163/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-manuals refs/changes/63/245163/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-manuals refs/changes/63/245163/2"}}},"commit":{"parents":[{"commit":"e135ca8442befda40ba396ef94952a2062b67d34","subject":"Merge \"small changes to shared_file_systems file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-manuals/commit/e135ca8442befda40ba396ef94952a2062b67d34"}]}],"author":{"name":"John Davidge","email":"jodavidg@cisco.com","date":"2015-11-13 14:00:54.000000000","tz":0},"committer":{"name":"John Davidge","email":"jodavidg@cisco.com","date":"2015-11-13 14:53:44.000000000","tz":0},"subject":"Un-comment config options for consistency","message":"Un-comment config options for consistency\n\nTwo of the config options in this section are commented out.\n\nUn-comment them to match with the rest of the document,\nand to improve readability.\n\nChange-Id: I6b52e700463906e2705465d4c2cc01ca45aad49e\nCloses-Bug: #1514930\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-manuals/commit/c400d9b2f06bf821b0438b57950f0de4bab201a4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-manuals/commit/c400d9b2f06bf821b0438b57950f0de4bab201a4"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"e135ca8442befda40ba396ef94952a2062b67d34","is_merged_in_target_branch":true}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
