)]}'
{"id":"openstack%2Fopenstack-manuals~421758","triplet_id":"openstack%2Fopenstack-manuals~stable%2Fnewton~If42ce1ffeeb2976581008b1ca45af9d09464927d","project":"openstack/openstack-manuals","branch":"stable/newton","topic":"zanata/translations","hashtags":[],"change_id":"If42ce1ffeeb2976581008b1ca45af9d09464927d","subject":"Imported Translations from Zanata","status":"MERGED","created":"2017-01-18 09:16:37.000000000","updated":"2017-01-18 09:44:10.000000000","submitted":"2017-01-18 09:44:10.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"b4320a2f76428d84cb13bfc476ec7bf604094ccc","_number":421758,"virtual_id_number":421758,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":2,"date":"2017-01-18 09:44:10.000000000","post_submit":true,"_account_id":3,"name":"Jenkins","username":"jenkins"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":2,"date":"2017-01-18 09:18:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":1,"date":"2017-01-18 09:18:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2017-01-18 09:18:38.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2017-01-18 09:44:10.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"99531ab104758cab7713fdba20eb828f1f550d2d","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2017-01-18 09:16:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"dbca9792ed9efdcc246b136fd7c5b7cd0f987483","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2017-01-18 09:18:38.000000000","message":"Patch Set 1: Workflow+1 Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"fa65d864be6367d1afa66ced2e2df5a0c766b540","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-01-18 09:27:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-openstack-manuals-tox-checkniceness-ubuntu-xenial http://logs.openstack.org/58/421758/1/check/gate-openstack-manuals-tox-checkniceness-ubuntu-xenial/93ec25d/ : SUCCESS in 2m 10s\n- gate-openstack-manuals-tox-doc-publish-checkbuild http://docs-draft.openstack.org/58/421758/1/check/gate-openstack-manuals-tox-doc-publish-checkbuild/74b1eee//publish-docs/ : SUCCESS in 2m 23s\n- gate-openstack-manuals-tox-checklang-ubuntu-xenial http://logs.openstack.org/58/421758/1/check/gate-openstack-manuals-tox-checklang-ubuntu-xenial/63a3c4d/ : SUCCESS in 10m 04s","accounts_in_message":[],"_revision_number":1},{"id":"9ef6f8198dcebcbebe78eadf6b4ac3b26019ac52","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-01-18 09:28:00.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":1},{"id":"1d2dfb95e3677b8a5655889fb6595bcf8a81d612","date":"2017-01-18 09:44:10.000000000","message":"Change has been successfully merged into the git repository by Jenkins","accounts_in_message":[],"_revision_number":1},{"id":"f4eb837103788a52505cea1ccd6df64b6cfcf8cc","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-01-18 09:44:10.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- gate-openstack-manuals-tox-checkniceness-ubuntu-xenial http://logs.openstack.org/58/421758/1/gate/gate-openstack-manuals-tox-checkniceness-ubuntu-xenial/3dfd726/ : SUCCESS in 3m 37s\n- gate-openstack-manuals-tox-doc-publish-checkbuild http://docs-draft.openstack.org/58/421758/1/gate/gate-openstack-manuals-tox-doc-publish-checkbuild/1649551//publish-docs/ : SUCCESS in 3m 20s\n- gate-openstack-manuals-tox-checklang-ubuntu-xenial http://logs.openstack.org/58/421758/1/gate/gate-openstack-manuals-tox-checklang-ubuntu-xenial/1883785/ : SUCCESS in 15m 03s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7bb0a9e62cb11e46e74fe02c53718ad617406958","revisions":{"7bb0a9e62cb11e46e74fe02c53718ad617406958":{"kind":"REWORK","_number":1,"created":"2017-01-18 09:16:37.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/58/421758/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-manuals","ref":"refs/changes/58/421758/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-manuals refs/changes/58/421758/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-manuals refs/changes/58/421758/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-manuals refs/changes/58/421758/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-manuals refs/changes/58/421758/1"}}},"commit":{"parents":[{"commit":"4d07874216f8c069b736f5ce0ee8987192f3d780","subject":"Tintri driver - Updating a missing config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-manuals/commit/4d07874216f8c069b736f5ce0ee8987192f3d780"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2017-01-18 09:16:36.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2017-01-18 09:16:36.000000000","tz":0},"subject":"Imported Translations from Zanata","message":"Imported Translations from Zanata\n\nFor more information about this automatic import see:\nhttp://docs.openstack.org/developer/i18n/reviewing-translation-import.html\n\nChange-Id: If42ce1ffeeb2976581008b1ca45af9d09464927d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-manuals/commit/7bb0a9e62cb11e46e74fe02c53718ad617406958"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-manuals/commit/7bb0a9e62cb11e46e74fe02c53718ad617406958"}]},"parents_data":[{"branch_name":"refs/heads/stable/newton","commit_id":"4d07874216f8c069b736f5ce0ee8987192f3d780","is_merged_in_target_branch":true,"change_id":"Ic6ce9fa849ff958f4eed0e74cb89606ae1308d61","change_number":421559,"patch_set_number":1,"change_status":"MERGED"}],"branch":"refs/heads/stable/newton"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
