)]}'
{"id":"openstack%2Fopenstack-manuals~728386","triplet_id":"openstack%2Fopenstack-manuals~master~Ie1efade0dd39de4d68c16177b6abf484cd1eaaca","project":"openstack/openstack-manuals","branch":"master","hashtags":[],"change_id":"Ie1efade0dd39de4d68c16177b6abf484cd1eaaca","subject":"www: Regenerate sitemap","status":"MERGED","created":"2020-05-15 08:49:55.000000000","updated":"2020-05-20 10:54:37.000000000","submitted":"2020-05-20 10:22:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"728386-1589970127922-3b3ce844","meta_rev_id":"f24e2e61cb29af233c71a2c33fa718ada479f736","_number":728386,"virtual_id_number":728386,"owner":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-20 10:22:07.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2020-05-20 09:57:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-05-20 10:09:32.000000000","_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-05-20 10:09:32.000000000","_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-20 09:57:53.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2020-05-20 10:09:32.000000000","updated_by":{"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"},"reviewer":{"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"},"state":"REVIEWER"},{"updated":"2020-05-20 10:22:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9de74506e2fd8be4b06330dac62bde1c2633f8a5","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-05-15 08:49:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ac01969dead707403f07aa8b1eed49624150ad82","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-15 09:02:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/987ade188e62444283f2912e6426241c : SUCCESS in 6m 35s\n- build-tox-manuals-publishdocs https://zuul.opendev.org/t/openstack/build/6b466b28bb1f4286bc1bc9c0c6f4bea4 : SUCCESS in 10m 51s","accounts_in_message":[],"_revision_number":1},{"id":"8d175fc6323a933abda22e0dbe9a7a61084c164e","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-05-20 09:57:53.000000000","message":"Patch Set 1: Code-Review+2\n\nThis won\u0027t even load in my browser, so I\u0027m going to just trust that it\u0027s correct","accounts_in_message":[],"_revision_number":1},{"id":"eb85f71b53df1bf334d62e3ddb2451476af2f3a7","author":{"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"},"date":"2020-05-20 10:09:32.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nAll good here :)","accounts_in_message":[],"_revision_number":1},{"id":"0e9464cd9fb74a5ace86166b620a69dd8d0272ae","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-20 10:09:47.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8fbed5b7626d27f059b09ca0d14aa15d1234abdf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-20 10:22:07.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"fc95c656e28b23d38bc415f152e682b7cc64c669","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-20 10:22:07.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/70f4073539a8406c8ecc024438f3deeb : SUCCESS in 5m 32s\n- build-tox-manuals-publishdocs https://zuul.opendev.org/t/openstack/build/c673beeb6b4b49a79eab2d98b4ab934c : SUCCESS in 7m 25s","accounts_in_message":[],"_revision_number":1},{"id":"f24e2e61cb29af233c71a2c33fa718ada479f736","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-20 10:54:37.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-manuals https://zuul.opendev.org/t/openstack/build/472e876d013a40c1b0bbb0a2e0f9d9d1 : SUCCESS in 32m 01s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"be0d6360c2f713c069a79f01f07c6d5dff405cb0","revisions":{"be0d6360c2f713c069a79f01f07c6d5dff405cb0":{"kind":"REWORK","_number":1,"created":"2020-05-15 08:49:55.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/86/728386/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-manuals","ref":"refs/changes/86/728386/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-manuals refs/changes/86/728386/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-manuals refs/changes/86/728386/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-manuals refs/changes/86/728386/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-manuals refs/changes/86/728386/1"}}},"commit":{"parents":[{"commit":"0c714e51507007a0245f67ec1c1c04aecb46e7b9","subject":"Merge \"www: Mark Ussuri as released\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-manuals/commit/0c714e51507007a0245f67ec1c1c04aecb46e7b9"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-05-15 08:49:20.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-05-15 08:49:20.000000000","tz":120},"subject":"www: Regenerate sitemap","message":"www: Regenerate sitemap\n\nRegenerate sitemap after Ussuri release.\n\nChange-Id: Ie1efade0dd39de4d68c16177b6abf484cd1eaaca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-manuals/commit/be0d6360c2f713c069a79f01f07c6d5dff405cb0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-manuals/commit/be0d6360c2f713c069a79f01f07c6d5dff405cb0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
