)]}'
{"id":"openstack%2Fopenstack-tempest-skiplist~775493","triplet_id":"openstack%2Fopenstack-tempest-skiplist~master~I18f217da244466ff5a17d9ba715e89ba0a0d830b","project":"openstack/openstack-tempest-skiplist","branch":"master","topic":"Id0bfd4a0dfa57d8c4412cac1be0f4eb9dcc138bd","hashtags":[],"change_id":"I18f217da244466ff5a17d9ba715e89ba0a0d830b","subject":"Revert \"Add NetworkSecGroupTest in skiplist\"","status":"MERGED","created":"2021-02-15 09:25:02.000000000","updated":"2021-02-22 12:16:17.000000000","submitted":"2021-02-22 12:14:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"revert_of":768722,"submission_id":"775493-Id0bfd4a0dfa57d8c4412cac1be0f4eb9dcc138bd","meta_rev_id":"67149499eca867851a4dcd4041c806b0ef2cbb78","_number":775493,"virtual_id_number":775493,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-22 12:14:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"all":[{"value":2,"date":"2021-02-22 10:59:02.000000000","_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":2,"date":"2021-02-22 11:12:44.000000000","_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"all":[{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":1,"date":"2021-02-22 11:13:04.000000000","_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"}],"reviewers":{"REVIEWER":[{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-15 09:25:02.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},"state":"REVIEWER"},{"updated":"2021-02-15 09:25:02.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"state":"REVIEWER"},{"updated":"2021-02-15 09:25:02.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"state":"REVIEWER"},{"updated":"2021-02-15 09:25:02.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-22 10:59:02.000000000","updated_by":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"reviewer":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"state":"REVIEWER"}],"messages":[{"id":"d32f5d22f46a5f24b2445fc0a3e481551b57f1a9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-02-15 09:25:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cd72fb225cc4613808598e210f71cd8d673529ec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-15 09:25:22.000000000","message":"Patch Set 1: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"fdbea6ed226c9d2e3b0e3c36a6aff84c4bc58410","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-02-15 10:29:34.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"32191dcab0884da8482303c46a8e321cf702ccf6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-15 10:39:39.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a224de3ee5e04d51b20cdf79c6de86b4 : SUCCESS in 5m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/050af6c4afac4f199037efcd179a5b75 : SUCCESS in 2m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/03a5866e031d471eb8c44360dc544325 : SUCCESS in 3m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e0ac57925ed54be2a04f3b0636a6354a : SUCCESS in 3m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b28fabce8b754d96986e7c3d2cc05056 : SUCCESS in 3m 10s (non-voting)\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f2f307513343465293f8450e064cd652 : SUCCESS in 3m 45s\n- openstack-tempest-skiplist-validate https://zuul.opendev.org/t/openstack/build/cd3c74cd200349c6a4a67c8b14f3fdd1 : SUCCESS in 2m 38s","accounts_in_message":[],"_revision_number":2},{"id":"7cd8fbec3721cda2f1d024fe3c229ee4b2f73113","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-02-17 08:26:56.000000000","message":"Patch Set 2:\n\nTest patch https://review.rdoproject.org/r/32006","accounts_in_message":[],"_revision_number":2},{"id":"0ada68d84b670a0a45d4cda50eacdd61f83c36ce","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-02-17 09:23:04.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"be6a461aad024eda6c63a4868ccbcabe5cefb092","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-17 09:43:43.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a9c4c9b9b1d14225bba3559c423dca79 : SUCCESS in 6m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ece0f64f0c934bb39b378f2d1d8f4017 : SUCCESS in 4m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f4d6ebc5e73b47919b41df30d08c5ce5 : SUCCESS in 4m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c43e8ba55f9340a6b298cb0edf13a695 : SUCCESS in 3m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d02b0ce5285945cf8c082de95e6a74fe : SUCCESS in 4m 35s (non-voting)\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/993b5e7c5ab749bfa92f1038af6e387b : SUCCESS in 4m 59s\n- openstack-tempest-skiplist-validate https://zuul.opendev.org/t/openstack/build/7a1906e421e4422381b4c06ba6ff3f5a : SUCCESS in 3m 03s","accounts_in_message":[],"_revision_number":3},{"id":"917bb099e2146314f255048f36cdf79bcca86444","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2021-02-22 10:59:02.000000000","message":"Patch Set 3: Code-Review+2\n\nis there a test for this one like https://review.rdoproject.org/r/#/c/32006/","accounts_in_message":[],"_revision_number":3},{"id":"a5adf908bcee29222418dcbd799d8f6e903b896c","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2021-02-22 10:59:43.000000000","message":"Patch Set 3:\n\n\u003e Patch Set 3: Code-Review+2\n\u003e \n\u003e is there a test for this one like https://review.rdoproject.org/r/#/c/32006/\n\noh i see it depends on this :D i saw it referenced on the other patch ack","accounts_in_message":[],"_revision_number":3},{"id":"3c477c66c245e2a101c58b1149276fe9a0686297","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2021-02-22 11:12:44.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"955a982620e8ef3c0132a9f3e057afe85acfd734","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2021-02-22 11:13:04.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"6f0b8d39e1f33c2efd9c29ba2d30de6d1fe751da","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-22 11:57:40.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"cc1e09e9146346850e83348cdc68ee3cfa38b966","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-22 12:14:55.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c1dd0736982d4dc49f1daccc3e2e9266 : SUCCESS in 5m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/406350d3d565498fbec5d16b442c55c4 : SUCCESS in 3m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/65ccfedb1ac249c39acb6746510be6fc : SUCCESS in 4m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/13747137b9af42e2a1aebb1a95816be2 : SUCCESS in 3m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a6f0edbc52db49dc9de269dc73078c2a : SUCCESS in 4m 14s\n- openstack-tempest-skiplist-validate https://zuul.opendev.org/t/openstack/build/9d0db13daaa0442e875551ec4d855d63 : SUCCESS in 3m 31s","accounts_in_message":[],"_revision_number":3},{"id":"2ea58980bd68adc4e4e39dcf9e781f69ab95ffa7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-22 12:14:56.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"67149499eca867851a4dcd4041c806b0ef2cbb78","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-22 12:16:17.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f5ddd63de4e4f7485003703848b310c : SUCCESS in 55s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"41cbe159e1fa89bb75c98020fcf44ecbb36d316b","revisions":{"1be9aa8fe36a13ad4afe6ad977d54005a0515893":{"kind":"REWORK","_number":1,"created":"2021-02-15 09:25:02.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/93/775493/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-tempest-skiplist","ref":"refs/changes/93/775493/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-tempest-skiplist refs/changes/93/775493/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-tempest-skiplist refs/changes/93/775493/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-tempest-skiplist refs/changes/93/775493/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-tempest-skiplist refs/changes/93/775493/1"}}},"commit":{"parents":[{"commit":"7c50ced28ef83e162008334ef36f1c1121140ddf","subject":"Add NetworkSecGroupTest in skiplist","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-tempest-skiplist/commit/7c50ced28ef83e162008334ef36f1c1121140ddf"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-02-15 09:25:02.000000000","tz":0},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-02-15 09:25:02.000000000","tz":0},"subject":"Revert \"Add NetworkSecGroupTest in skiplist\"","message":"Revert \"Add NetworkSecGroupTest in skiplist\"\n\nThis reverts commit 7c50ced28ef83e162008334ef36f1c1121140ddf.\n\nReason for revert: New OVN is now available so bug should be fixed.\n\nChange-Id: I18f217da244466ff5a17d9ba715e89ba0a0d830b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-tempest-skiplist/commit/1be9aa8fe36a13ad4afe6ad977d54005a0515893"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-tempest-skiplist/commit/1be9aa8fe36a13ad4afe6ad977d54005a0515893"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"7c50ced28ef83e162008334ef36f1c1121140ddf","is_merged_in_target_branch":true,"change_id":"I46eed8e0e1639f75b642ce76e45dd7a6af82065a","change_number":768722,"patch_set_number":3,"change_status":"MERGED"}],"branch":"refs/heads/master"},"46e433acece00cdfed2b7cb839063ac583422e0c":{"kind":"REWORK","_number":2,"created":"2021-02-15 10:29:34.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/93/775493/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-tempest-skiplist","ref":"refs/changes/93/775493/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-tempest-skiplist refs/changes/93/775493/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-tempest-skiplist refs/changes/93/775493/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-tempest-skiplist refs/changes/93/775493/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-tempest-skiplist refs/changes/93/775493/2"}}},"commit":{"parents":[{"commit":"5682e5d9544461005f7794810132e7b1fd7dc305","subject":"Adding jobs that are failing on TrafficOperationsScenarioTest","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-tempest-skiplist/commit/5682e5d9544461005f7794810132e7b1fd7dc305"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-02-15 10:27:50.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-02-15 10:27:50.000000000","tz":60},"subject":"Revert \"Add NetworkSecGroupTest in skiplist\"","message":"Revert \"Add NetworkSecGroupTest in skiplist\"\n\nThis reverts commits 7c50ced28ef83e162008334ef36f1c1121140ddf\nand 37830be390b4d43c5608003e4a587e5a789cfc8d.\n\nReason for revert: New OVN is now available so bug should be fixed.\n\nChange-Id: I18f217da244466ff5a17d9ba715e89ba0a0d830b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-tempest-skiplist/commit/46e433acece00cdfed2b7cb839063ac583422e0c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-tempest-skiplist/commit/46e433acece00cdfed2b7cb839063ac583422e0c"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"5682e5d9544461005f7794810132e7b1fd7dc305","is_merged_in_target_branch":true,"change_id":"If04bfdebc2507d7dfcb4ab5a70632a8304e86177","change_number":774868,"patch_set_number":1,"change_status":"MERGED"}],"branch":"refs/heads/master"},"41cbe159e1fa89bb75c98020fcf44ecbb36d316b":{"kind":"REWORK","_number":3,"created":"2021-02-17 09:23:04.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/93/775493/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-tempest-skiplist","ref":"refs/changes/93/775493/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-tempest-skiplist refs/changes/93/775493/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-tempest-skiplist refs/changes/93/775493/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-tempest-skiplist refs/changes/93/775493/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-tempest-skiplist refs/changes/93/775493/3"}}},"commit":{"parents":[{"commit":"e39b2e6c844155257e09814579a0feb5f7f210d8","subject":"Revert \"Add temptest test failing on scenario002 to skiplist\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-tempest-skiplist/commit/e39b2e6c844155257e09814579a0feb5f7f210d8"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-02-15 10:27:50.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-02-17 09:22:17.000000000","tz":60},"subject":"Revert \"Add NetworkSecGroupTest in skiplist\"","message":"Revert \"Add NetworkSecGroupTest in skiplist\"\n\nThis reverts commits 7c50ced28ef83e162008334ef36f1c1121140ddf\nand 37830be390b4d43c5608003e4a587e5a789cfc8d.\n\nReason for revert: New OVN is now available so bug should be fixed.\n\nChange-Id: I18f217da244466ff5a17d9ba715e89ba0a0d830b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-tempest-skiplist/commit/41cbe159e1fa89bb75c98020fcf44ecbb36d316b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-tempest-skiplist/commit/41cbe159e1fa89bb75c98020fcf44ecbb36d316b"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"e39b2e6c844155257e09814579a0feb5f7f210d8","is_merged_in_target_branch":true,"change_id":"I5d25b3b55c174ce256b6ac094eba2c18f9cda34a","change_number":775582,"patch_set_number":4,"change_status":"MERGED"}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
