)]}'
{"id":"openstack%2Fopenstack-tempest-skiplist~825506","triplet_id":"openstack%2Fopenstack-tempest-skiplist~master~I2728489db634a620c4c908d0f21f825b09cb4e3f","project":"openstack/openstack-tempest-skiplist","branch":"master","topic":"components","attention_set":{},"removed_from_attention_set":{"8449":{"account":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"last_update":"2022-01-28 16:06:30.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I2728489db634a620c4c908d0f21f825b09cb4e3f","subject":"Partial Revert \"Adding CentOS 9 jobs to existing skip list\"","status":"MERGED","created":"2022-01-20 13:55:24.000000000","updated":"2022-01-28 16:09:00.000000000","submitted":"2022-01-28 16:06:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":12,"unresolved_comment_count":8,"has_review_started":true,"submission_id":"825506-components","meta_rev_id":"b99b87350a8feb3df4e11cccc4cb6171c2a83d73","_number":825506,"virtual_id_number":825506,"owner":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"value":0,"_account_id":32458,"name":"Ananya","display_name":"Ananya Banerjee","email":"anbanerj@redhat.com","username":"frenzyfriday"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-01-28 16:06:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"all":[{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":2,"date":"2022-01-28 13:20:59.000000000","_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"value":1,"date":"2022-01-28 13:30:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32458,"name":"Ananya","display_name":"Ananya Banerjee","email":"anbanerj@redhat.com","username":"frenzyfriday"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"all":[{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":1,"date":"2022-01-28 15:55:37.000000000","_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"value":0,"_account_id":32458,"name":"Ananya","display_name":"Ananya Banerjee","email":"anbanerj@redhat.com","username":"frenzyfriday"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"_account_id":32458,"name":"Ananya","display_name":"Ananya Banerjee","email":"anbanerj@redhat.com","username":"frenzyfriday"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-20 14:15:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-28 13:20:59.000000000","updated_by":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"reviewer":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"state":"REVIEWER"},{"updated":"2022-01-28 13:30:14.000000000","updated_by":{"_account_id":32458,"name":"Ananya","display_name":"Ananya Banerjee","email":"anbanerj@redhat.com","username":"frenzyfriday"},"reviewer":{"_account_id":32458,"name":"Ananya","display_name":"Ananya Banerjee","email":"anbanerj@redhat.com","username":"frenzyfriday"},"state":"REVIEWER"},{"updated":"2022-01-28 15:55:37.000000000","updated_by":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"reviewer":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"state":"REVIEWER"}],"messages":[{"id":"18e378623bbc25bc8ad8dd4eb6662f215206ebf7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2022-01-20 13:55:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8dcf00bc4234984990a295d52c6ac68de05a3fe8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-20 14:15:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/656012a6ffae4df1bb102965402b5dd2 : SUCCESS in 7m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9bc1dd525f694856b63fc872751de8b6 : SUCCESS in 2m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/731711ac46b54656a97665daf9bf678b : SUCCESS in 3m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/327211a7e0924fbeb0980ebd04b8d1a0 : SUCCESS in 3m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2522a13f803843b5a3299fabe3016ff3 : SUCCESS in 3m 47s (non-voting)\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/33ef4f6eaa4644b0a2c9c70e9e1b8218 : SUCCESS in 3m 26s\n- openstack-tempest-skiplist-validate https://zuul.opendev.org/t/openstack/build/dec75393a3ca4610a404fe139aa8b7b2 : SUCCESS in 2m 46s","accounts_in_message":[],"_revision_number":1},{"id":"9b270b2f63be956adb05dae6e6163af7a7dbb7c4","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2022-01-21 10:07:35.000000000","message":"Patch Set 1: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"2cbb0e38e27ac0859ec9a39abde5aacbb3c28fab","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2022-01-25 09:17:38.000000000","message":"Patch Set 1: -Code-Review\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"db5034f95057839146500a821622fa29229a5dea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2022-01-25 09:33:00.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"e4088b18df31c2b62922eb303cc1a18ebbcc7031","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-25 09:43:54.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eda4b9aca8c14e5fa18bd49745ca0e2d : SUCCESS in 5m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/00ae73e7ff0942cf86fa1b9cff37fbb0 : SUCCESS in 3m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/de9280d728d2484684091175cc510992 : SUCCESS in 3m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e02bd4fd3fa64550ab2fa7120d6c27aa : SUCCESS in 4m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a4973a57bf78461faeb882cf63ee8513 : SUCCESS in 4m 30s (non-voting)\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5aaf4ff6ed8748c6847d9e9b829fe1e9 : SUCCESS in 3m 36s\n- openstack-tempest-skiplist-validate https://zuul.opendev.org/t/openstack/build/e18530dcdf1f467bb65825d1ecf3428c : SUCCESS in 2m 32s","accounts_in_message":[],"_revision_number":2},{"id":"24d35115c6f69205feea6df1625def157627994f","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2022-01-26 10:18:45.000000000","message":"Topic set to revert_skips_ovb_c9","accounts_in_message":[],"_revision_number":2},{"id":"5d4747a8c1656bc7915fc5ed18b8bab8e679ab0a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2022-01-26 10:41:56.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"4802738a4f2a527156a4ed596e42903c256b2b41","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-26 10:53:53.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e1d97a630b45456eb5a6abf8afbc43b9 : SUCCESS in 6m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7c8fa26d72824036a0713e503791dcdb : SUCCESS in 3m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/000345a592b1497795e0de1ea3fe8d76 : SUCCESS in 3m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/62038014ed9e437ea80492abfa643acf : SUCCESS in 3m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d44cf8a6a1214847ab031d38c8986b5b : SUCCESS in 3m 16s (non-voting)\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f81dca6dbbf4408db0b882e346f39a6e : SUCCESS in 3m 14s\n- openstack-tempest-skiplist-validate https://zuul.opendev.org/t/openstack/build/baaf34ffc31d4726aa2266d08a45e1fe : SUCCESS in 2m 31s","accounts_in_message":[],"_revision_number":3},{"id":"26948b1ab21530db415bf4873588f374938e9ca5","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2022-01-26 15:38:29.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"79161380278bc5f98cc410a5044b4436aadc6144","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2022-01-26 16:03:23.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"37fa40da15530f105614ecb2dfe7e8e6b67a1575","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-26 16:16:26.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/96884679e4f34a6ca6b97b83abfe3bb5 : SUCCESS in 6m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5a7b98b20c58429e954a945ee1cccfd0 : SUCCESS in 5m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4576ca31e6fe4575a7e53ac36c7f78a7 : SUCCESS in 5m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f812a218451b45dcbf3e739684385438 : SUCCESS in 5m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2ed35948311d41d494082ec56b31b7df : SUCCESS in 4m 35s (non-voting)\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8e2e21e15d944f7ba0727b21587c35df : SUCCESS in 5m 27s\n- openstack-tempest-skiplist-validate https://zuul.opendev.org/t/openstack/build/64dc9b9622cd4ea88acf98ddb31039a9 : SUCCESS in 4m 03s","accounts_in_message":[],"_revision_number":4},{"id":"47d1b14a495d622e1f4eed7675329564459f8b00","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2022-01-28 13:14:54.000000000","message":"Patch Set 4:\n\n(5 comments)","accounts_in_message":[],"_revision_number":4},{"id":"b4e6e70ef47c1fd1e670e8df827ce1f91347ef7a","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2022-01-28 13:20:59.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"793d657d0e40e485d806103fc932f3cd0f381c9b","author":{"_account_id":32458,"name":"Ananya","display_name":"Ananya Banerjee","email":"anbanerj@redhat.com","username":"frenzyfriday"},"date":"2022-01-28 13:30:14.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"2a1b15ea13e3ed60d4ba058e9db347210165653a","author":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"date":"2022-01-28 15:55:37.000000000","message":"Patch Set 4: Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"89041a6ac234436ced66e977f69a9b37ba50f61d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-28 15:56:25.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"61264284595cdba3c8c855eb33f74420edf2e9cf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-28 16:06:30.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/61ed2a98e16d4f578420ca6d8bf307ab : SUCCESS in 5m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c23f2966739543b79a3e82d96c91c8a8 : SUCCESS in 3m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f74a642576004753814783ec1f8c1a2f : SUCCESS in 4m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/30dd3d6b992f4be6a8a09bf7168b7a51 : SUCCESS in 4m 40s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d7e5689e459d41b3a88761f926d3ad0e : SUCCESS in 3m 53s\n- openstack-tempest-skiplist-validate https://zuul.opendev.org/t/openstack/build/3aca968cf92347e09e8a088c22c8937b : SUCCESS in 3m 17s","accounts_in_message":[],"_revision_number":4},{"id":"8736369eeee5905822c6ab42ea774b3dc4db622d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-28 16:06:30.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"b99b87350a8feb3df4e11cccc4cb6171c2a83d73","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-28 16:09:00.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e6e8fa30148d4f26a5fa3ad49172251a : SUCCESS in 1m 46s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"52416c9c4a19d141bb719ee9869591fd333653f7","revisions":{"6694e668c597f642423d02a9d9858d1a9e2e4d7a":{"kind":"REWORK","_number":1,"created":"2022-01-20 13:55:24.000000000","uploader":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"ref":"refs/changes/06/825506/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-tempest-skiplist","ref":"refs/changes/06/825506/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-tempest-skiplist refs/changes/06/825506/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-tempest-skiplist refs/changes/06/825506/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-tempest-skiplist refs/changes/06/825506/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-tempest-skiplist refs/changes/06/825506/1"}}},"commit":{"parents":[{"commit":"cbfba8be1f3bfdb65a5d533ceea044a468984aa4","subject":"Merge \"Revert \"Add fs1 centos 9 to tempest.api.compute tests skip for bug\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-tempest-skiplist/commit/cbfba8be1f3bfdb65a5d533ceea044a468984aa4"}]}],"author":{"name":"Marios Andreou","email":"marios@redhat.com","date":"2022-01-20 13:40:52.000000000","tz":120},"committer":{"name":"Marios Andreou","email":"marios@redhat.com","date":"2022-01-20 13:42:23.000000000","tz":120},"subject":"Partial Revert \"Adding CentOS 9 jobs to existing skip list\"","message":"Partial Revert \"Adding CentOS 9 jobs to existing skip list\"\n\nThis reverts commit 7b17b12be849792d4e00ff51b613d50523c7d81a.\nOnly partial revert as we keep skip for tripleo/+bug/1956233\nperiodic-tripleo-ci-centos-9-standalone-full-tempest-scenario-tempest-master\n\nRelated-Bug: 1957031\nChange-Id: I2728489db634a620c4c908d0f21f825b09cb4e3f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-tempest-skiplist/commit/6694e668c597f642423d02a9d9858d1a9e2e4d7a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-tempest-skiplist/commit/6694e668c597f642423d02a9d9858d1a9e2e4d7a"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"cbfba8be1f3bfdb65a5d533ceea044a468984aa4","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"c7280e11eb219bd108c3f075a8cec7cafa83fe55":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-01-25 09:33:00.000000000","uploader":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"ref":"refs/changes/06/825506/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-tempest-skiplist","ref":"refs/changes/06/825506/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-tempest-skiplist refs/changes/06/825506/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-tempest-skiplist refs/changes/06/825506/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-tempest-skiplist refs/changes/06/825506/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-tempest-skiplist refs/changes/06/825506/2"}}},"commit":{"parents":[{"commit":"b3f7cbb5cf5d4d6a8899323f1f25f5dbcafc2273","subject":"Revert \"Move test_trunk_subport_lifecycle to skip list\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-tempest-skiplist/commit/b3f7cbb5cf5d4d6a8899323f1f25f5dbcafc2273"}]}],"author":{"name":"Marios Andreou","email":"marios@redhat.com","date":"2022-01-20 13:40:52.000000000","tz":120},"committer":{"name":"Marios Andreou","email":"marios@redhat.com","date":"2022-01-25 09:33:00.000000000","tz":0},"subject":"Partial Revert \"Adding CentOS 9 jobs to existing skip list\"","message":"Partial Revert \"Adding CentOS 9 jobs to existing skip list\"\n\nThis reverts commit 7b17b12be849792d4e00ff51b613d50523c7d81a.\nOnly partial revert as we keep skip for tripleo/+bug/1956233\nperiodic-tripleo-ci-centos-9-standalone-full-tempest-scenario-tempest-master\n\nRelated-Bug: 1957031\nChange-Id: I2728489db634a620c4c908d0f21f825b09cb4e3f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-tempest-skiplist/commit/c7280e11eb219bd108c3f075a8cec7cafa83fe55"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-tempest-skiplist/commit/c7280e11eb219bd108c3f075a8cec7cafa83fe55"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"b3f7cbb5cf5d4d6a8899323f1f25f5dbcafc2273","is_merged_in_target_branch":true,"change_id":"Ifcadbf47ad3ec5cea82f83e6a8dd54dbad5e8af7","change_number":823480,"patch_set_number":3,"change_status":"MERGED"}],"branch":"refs/heads/master","description":"Rebase"},"f49d467a71a04b9a7ecc7514b34efb0357a1e732":{"kind":"REWORK","_number":3,"created":"2022-01-26 10:41:56.000000000","uploader":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"ref":"refs/changes/06/825506/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-tempest-skiplist","ref":"refs/changes/06/825506/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-tempest-skiplist refs/changes/06/825506/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-tempest-skiplist refs/changes/06/825506/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-tempest-skiplist refs/changes/06/825506/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-tempest-skiplist refs/changes/06/825506/3"}}},"commit":{"parents":[{"commit":"7168c281803b100038873e42d59cd3006bd3adc8","subject":"Revert centos9 OVB skips full-tempest-api \u0026 fs35","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-tempest-skiplist/commit/7168c281803b100038873e42d59cd3006bd3adc8"}]}],"author":{"name":"Marios Andreou","email":"marios@redhat.com","date":"2022-01-20 13:40:52.000000000","tz":120},"committer":{"name":"Marios Andreou","email":"marios@redhat.com","date":"2022-01-26 10:38:06.000000000","tz":120},"subject":"Partial Revert \"Adding CentOS 9 jobs to existing skip list\"","message":"Partial Revert \"Adding CentOS 9 jobs to existing skip list\"\n\nThis reverts commit 7b17b12be849792d4e00ff51b613d50523c7d81a.\nOnly partial revert until we get green runs for all in tests [1]\n\n[1] https://review.rdoproject.org/r/c/testproject/+/38357/1#message-f0e15374f8983202dfe76992219ee92b5ea500ce\nRelated-Bug: 1957031\nChange-Id: I2728489db634a620c4c908d0f21f825b09cb4e3f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-tempest-skiplist/commit/f49d467a71a04b9a7ecc7514b34efb0357a1e732"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-tempest-skiplist/commit/f49d467a71a04b9a7ecc7514b34efb0357a1e732"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"7168c281803b100038873e42d59cd3006bd3adc8","is_merged_in_target_branch":true,"change_id":"Ida71a0d9381e9de28312899357595a8f9786921b","change_number":825508,"patch_set_number":5,"change_status":"MERGED"}],"branch":"refs/heads/master"},"52416c9c4a19d141bb719ee9869591fd333653f7":{"kind":"REWORK","_number":4,"created":"2022-01-26 16:03:23.000000000","uploader":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"ref":"refs/changes/06/825506/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-tempest-skiplist","ref":"refs/changes/06/825506/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-tempest-skiplist refs/changes/06/825506/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-tempest-skiplist refs/changes/06/825506/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-tempest-skiplist refs/changes/06/825506/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-tempest-skiplist refs/changes/06/825506/4"}}},"commit":{"parents":[{"commit":"7168c281803b100038873e42d59cd3006bd3adc8","subject":"Revert centos9 OVB skips full-tempest-api \u0026 fs35","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-tempest-skiplist/commit/7168c281803b100038873e42d59cd3006bd3adc8"}]}],"author":{"name":"Marios Andreou","email":"marios@redhat.com","date":"2022-01-20 13:40:52.000000000","tz":120},"committer":{"name":"Marios Andreou","email":"marios@redhat.com","date":"2022-01-26 16:03:08.000000000","tz":120},"subject":"Partial Revert \"Adding CentOS 9 jobs to existing skip list\"","message":"Partial Revert \"Adding CentOS 9 jobs to existing skip list\"\n\nPartially reverts commit 7b17b12be849792d4e00ff51b613d50523c7d81a.\n\n[1] https://review.rdoproject.org/r/c/testproject/+/38357/1#message-f0e15374f8983202dfe76992219ee92b5ea500ce\nRelated-Bug: 1957031\n\nChange-Id: I2728489db634a620c4c908d0f21f825b09cb4e3f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-tempest-skiplist/commit/52416c9c4a19d141bb719ee9869591fd333653f7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-tempest-skiplist/commit/52416c9c4a19d141bb719ee9869591fd333653f7"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"7168c281803b100038873e42d59cd3006bd3adc8","is_merged_in_target_branch":true,"change_id":"Ida71a0d9381e9de28312899357595a8f9786921b","change_number":825508,"patch_set_number":5,"change_status":"MERGED"}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
