)]}'
{"id":"openstack%2Fopenstack-zuul-jobs~627554","triplet_id":"openstack%2Fopenstack-zuul-jobs~master~Id541bb066b553ea1d75c2832a4f18ef695bb8cc4","project":"openstack/openstack-zuul-jobs","branch":"master","topic":"forge-publish","hashtags":[],"change_id":"Id541bb066b553ea1d75c2832a4f18ef695bb8cc4","subject":"Remove bindep role for Puppet build job","status":"MERGED","created":"2018-12-27 19:56:10.000000000","updated":"2019-07-01 15:40:21.000000000","submitted":"2019-07-01 14:40:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"627554-1561992016153-a3bcdd19","meta_rev_id":"aefa4b6363b2b8077f8074f7a1570f5bc7efc38a","_number":627554,"virtual_id_number":627554,"owner":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":30610,"name":"Jeff Liu","email":"jeliu@redhat.com","username":"jeffliu"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2019-07-01 14:40:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"all":[{"value":2,"date":"2019-07-01 13:33:37.000000000","_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":30610,"name":"Jeff Liu","email":"jeliu@redhat.com","username":"jeffliu"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":1,"date":"2019-06-27 10:33:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-06-27 05:07:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"all":[{"value":1,"date":"2019-07-01 13:33:37.000000000","_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"date":"2019-07-01 15:40:21.000000000","post_submit":true,"_account_id":30610,"name":"Jeff Liu","email":"jeliu@redhat.com","username":"jeffliu"},{"value":1,"date":"2019-07-01 14:29:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30610,"name":"Jeff Liu","email":"jeliu@redhat.com","username":"jeffliu"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-27 05:07:50.000000000","updated_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"reviewer":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"state":"REVIEWER"},{"updated":"2019-06-27 10:33:55.000000000","updated_by":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"reviewer":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"state":"REVIEWER"},{"updated":"2019-07-01 13:33:37.000000000","updated_by":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"reviewer":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"state":"REVIEWER"},{"updated":"2019-07-01 14:29:13.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2019-07-01 14:40:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-07-01 15:40:21.000000000","updated_by":{"_account_id":30610,"name":"Jeff Liu","email":"jeliu@redhat.com","username":"jeffliu"},"reviewer":{"_account_id":30610,"name":"Jeff Liu","email":"jeliu@redhat.com","username":"jeffliu"},"state":"REVIEWER"}],"messages":[{"id":"40a67786cf17565a7510bcd91564094b70151aae","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2018-12-27 19:56:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"52e5ff98f5213a0b106c742363511f25911d4fd8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-12-27 20:01:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tox-docs http://logs.openstack.org/54/627554/1/check/tox-docs/c5e708b/html/ : SUCCESS in 2m 48s\n- openstack-zuul-jobs-linters http://logs.openstack.org/54/627554/1/check/openstack-zuul-jobs-linters/11cdee3/ : SUCCESS in 3m 45s","accounts_in_message":[],"_revision_number":1},{"id":"219a5cb2d276ffe5c8f390e782e26182fb0b4c1f","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2019-01-07 15:08:51.000000000","message":"Topic changed from remove-bindep-puppet to forge-publish","accounts_in_message":[],"_revision_number":1},{"id":"26f1ce5bd1decbb955dc7009c576c79d8368fff4","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2019-01-11 10:24:38.000000000","message":"Patch Set 1: Workflow-1\n\nSetting WIP so we don\u0027t accidentally merge this until [1] is merged since we might need to follow zuul policy on that change which can take some some weeks.","accounts_in_message":[],"_revision_number":1},{"id":"870a9ec8718f4a441f12cfc148dbc8c2d04fdca1","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2019-02-26 17:33:44.000000000","message":"Patch Set 1: -Workflow","accounts_in_message":[],"_revision_number":1},{"id":"8b7fad28fdde07569b4b8e1031f5385a00e357f4","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-06-26 12:06:31.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"17be046eef6cae974c084de0987b268c81fcf741","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-26 12:17:06.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tox-docs http://logs.openstack.org/54/627554/2/check/tox-docs/c57a26c/html/ : SUCCESS in 3m 10s\n- openstack-zuul-jobs-linters http://logs.openstack.org/54/627554/2/check/openstack-zuul-jobs-linters/6af1969/ : FAILURE in 4m 58s","accounts_in_message":[],"_revision_number":2},{"id":"35765c7f9de7123d4f223c3412234e0a2e292bbe","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-06-26 12:24:18.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"1413662089218760bb6d3b3bc4197b1dd0da9736","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-06-26 12:27:04.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"364345b1029a86b60d77b4f6179a2c00170a8b6a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-26 12:36:59.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tox-docs http://logs.openstack.org/54/627554/3/check/tox-docs/e715705/html/ : SUCCESS in 3m 33s\n- openstack-zuul-jobs-linters http://logs.openstack.org/54/627554/3/check/openstack-zuul-jobs-linters/0150e0c/ : SUCCESS in 5m 17s","accounts_in_message":[],"_revision_number":3},{"id":"f6fd54567219367f9b04e21698e4c702933f7344","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2019-06-27 05:07:50.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"1d691d134937fd1323d788660afd520e67154f13","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-06-27 10:33:55.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"3f1a92456cc7c6936d5d34b2cc02dd8258ea7cf1","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2019-07-01 13:33:37.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"f0a80d61cf32c90adf6e111b11cd8d01ccc0455b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-01 13:55:08.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"dd95c35871f0740ea6f860cee1f848ae177c6ec8","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-07-01 14:00:16.000000000","message":"Patch Set 3: Workflow-1\n\nwaiting for dependency to merge first - using review.openstack.org does not work anymore.","accounts_in_message":[],"_revision_number":3},{"id":"c4ac21ae7ff804e832e696aa359dc3a86d27f92b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-01 14:06:28.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tox-docs http://logs.openstack.org/54/627554/3/gate/tox-docs/8ba9015/html/ : SUCCESS in 3m 19s\n- openstack-zuul-jobs-linters http://logs.openstack.org/54/627554/3/gate/openstack-zuul-jobs-linters/85120ce/ : SUCCESS in 6m 18s","accounts_in_message":[],"_revision_number":3},{"id":"697225a4d17e0636844f8d0eb3408d0bb2987013","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-07-01 14:29:13.000000000","message":"Patch Set 3: Workflow+1\n\ndependency merged","accounts_in_message":[],"_revision_number":3},{"id":"ce9e1e192e72bcb1f6b948ec22cf544760a29893","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-01 14:29:24.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"f698794256b4787fff72e690ec1cc4ab7dc70248","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-01 14:40:15.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tox-docs http://logs.openstack.org/54/627554/3/gate/tox-docs/9ab3afb/html/ : SUCCESS in 3m 20s\n- openstack-zuul-jobs-linters http://logs.openstack.org/54/627554/3/gate/openstack-zuul-jobs-linters/88ac1c8/ : SUCCESS in 4m 54s","accounts_in_message":[],"_revision_number":3},{"id":"35f37c2de4ddde4a146cc54fe46ad96eff5ffc83","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-01 14:40:16.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"5c6d6d639f4f8de89b7de8b8cb2ca55698fd99c0","revisions":{"bd72c57539a6f5ff45eaf08c5022b22d8f755883":{"kind":"REWORK","_number":1,"created":"2018-12-27 19:56:10.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/54/627554/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-zuul-jobs","ref":"refs/changes/54/627554/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/54/627554/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/54/627554/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/54/627554/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/54/627554/1"}}},"commit":{"parents":[{"commit":"80efcb052689842b4653ccc1e11891748164747b","subject":"Merge \"Fix name of \u0027integrated-gate-py35\u0027 project template\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/80efcb052689842b4653ccc1e11891748164747b"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2018-12-27 19:53:32.000000000","tz":60},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2018-12-27 19:56:01.000000000","tz":60},"subject":"Remove bindep role for Puppet build job","message":"Remove bindep role for Puppet build job\n\nWhen [1] is merged we no longer have to rely\non the bindep file when puppet modules are built\nsince it doesn\u0027t need the puppet binary.\n\n[1] https://review.openstack.org/#/c/627534/\n\nDepends-On: https://review.openstack.org/#/c/627534/\nChange-Id: Id541bb066b553ea1d75c2832a4f18ef695bb8cc4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/bd72c57539a6f5ff45eaf08c5022b22d8f755883"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/bd72c57539a6f5ff45eaf08c5022b22d8f755883"}]},"branch":"refs/heads/master"},"d4186c4ec78fddea8aad6608ba76735c06369c3b":{"kind":"REWORK","_number":2,"created":"2019-06-26 12:06:31.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/54/627554/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-zuul-jobs","ref":"refs/changes/54/627554/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/54/627554/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/54/627554/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/54/627554/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/54/627554/2"}}},"commit":{"parents":[{"commit":"c081a8e19f446838836fff2488ff9211189fa0d8","subject":"Install PDK dependencies for puppet building","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/c081a8e19f446838836fff2488ff9211189fa0d8"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2018-12-27 19:53:32.000000000","tz":60},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-06-26 12:06:02.000000000","tz":120},"subject":"Remove bindep role for Puppet build job","message":"Remove bindep role for Puppet build job\n\nWhen [1] is merged we no longer have to rely\non the bindep file when puppet modules are built\nsince it doesn\u0027t need the puppet binary.\n\n[1] https://review.openstack.org/#/c/627534/\n\nDepends-On: https://review.openstack.org/627534\nChange-Id: Id541bb066b553ea1d75c2832a4f18ef695bb8cc4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/d4186c4ec78fddea8aad6608ba76735c06369c3b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/d4186c4ec78fddea8aad6608ba76735c06369c3b"}]},"branch":"refs/heads/master"},"5c6d6d639f4f8de89b7de8b8cb2ca55698fd99c0":{"kind":"NO_CHANGE","_number":3,"created":"2019-06-26 12:24:18.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/54/627554/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-zuul-jobs","ref":"refs/changes/54/627554/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/54/627554/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/54/627554/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/54/627554/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/54/627554/3"}}},"commit":{"parents":[{"commit":"5f3f86d40849240bc1d01d513f96e9af0960fbd0","subject":"Install PDK dependencies for puppet building","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/5f3f86d40849240bc1d01d513f96e9af0960fbd0"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2018-12-27 19:53:32.000000000","tz":60},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-06-26 12:23:57.000000000","tz":120},"subject":"Remove bindep role for Puppet build job","message":"Remove bindep role for Puppet build job\n\nWhen [1] is merged we no longer have to rely\non the bindep file when puppet modules are built\nsince it doesn\u0027t need the puppet binary.\n\n[1] https://review.openstack.org/#/c/627534/\n\nDepends-On: https://review.openstack.org/627534\nChange-Id: Id541bb066b553ea1d75c2832a4f18ef695bb8cc4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/5c6d6d639f4f8de89b7de8b8cb2ca55698fd99c0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/5c6d6d639f4f8de89b7de8b8cb2ca55698fd99c0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
