)]}'
{"id":"openstack%2Fopenstack-zuul-jobs~654571","triplet_id":"openstack%2Fopenstack-zuul-jobs~master~Ie2941e89e543db406db5a89666dfcb576e3e8f96","project":"openstack/openstack-zuul-jobs","branch":"master","topic":"periodic-py36","hashtags":[],"change_id":"Ie2941e89e543db406db5a89666dfcb576e3e8f96","subject":"Use py36 instead of py35 for periodic master jobs","status":"MERGED","created":"2019-04-22 16:59:31.000000000","updated":"2019-04-23 11:50:22.000000000","submitted":"2019-04-23 11:50:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"654571-1556020222510-81340386","meta_rev_id":"093278da74af5a6af99b2997e2bf16d6f7bf9be8","_number":654571,"virtual_id_number":654571,"owner":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":2,"date":"2019-04-23 11:50:22.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2019-04-22 19:08:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":2,"date":"2019-04-23 11:39:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2019-04-22 18:58:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":1,"date":"2019-04-23 11:39:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-22 18:58:49.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2019-04-22 19:08:40.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2019-04-23 11:39:17.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2019-04-23 11:50:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e8c6a827cd55f9de16a8a41c86c3e9cff4836158","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-04-22 16:59:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bfaf9464e67ec470500497a5b97e88a8156743c4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 16:59:47.000000000","message":"Patch Set 1: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/project-config on branch master.  The error was:\n\n  Job openstack-tox-py35-with-neutron-lib-master not defined\n\nThe error appears in the following project stanza:\n\n  project:\n      name: openstack/networking-odl\n      templates:\n        - periodic-jobs-with-neutron-lib-master\n        - translation-jobs-master-only\n        - publish-to-pypi\n      periodic:\n        jobs:\n          - openstack-tox-py35-with-neutron-lib-master:\n              required-projects:\n                - openstack/ceilometer\n                - openstack/neutron-fwaas\n                - openstack/neutron-lbaas\n                - openstack/networking-l2gw\n                - openstack/networking-sfc\n                - openstack/networking-bgpvpn\n\n  in \"openstack/project-config/zuul.d/projects.yaml@master\", line 3224, column 3","accounts_in_message":[],"_revision_number":1},{"id":"f737de8bd8bf18445fc9069a32d08c069a1edc8d","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-04-22 17:19:29.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ed903c8bae9be7d0bbfa3c763b0d3f29b9d9a47f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 17:41:44.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tox-docs http://logs.openstack.org/71/654571/2/check/tox-docs/67f78d5/html/ : SUCCESS in 3m 45s\n- openstack-infra-extra-integration-xenial http://logs.openstack.org/71/654571/2/check/openstack-infra-extra-integration-xenial/bd11fc4/ : SUCCESS in 2m 50s\n- openstack-infra-extra-integration-bionic http://logs.openstack.org/71/654571/2/check/openstack-infra-extra-integration-bionic/425f9e8/ : SUCCESS in 2m 34s\n- openstack-zuul-jobs-linters http://logs.openstack.org/71/654571/2/check/openstack-zuul-jobs-linters/4e4e32f/ : SUCCESS in 3m 56s","accounts_in_message":[],"_revision_number":2},{"id":"7a68f27409a179ba95479a06208d417320327ab9","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-04-22 18:58:49.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)\n\nthanks","accounts_in_message":[],"_revision_number":2},{"id":"fcb58366ee0ccabbd5392bc7f99d27a7fe874fa8","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2019-04-22 19:08:40.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"351d4cf6a8c53adcb7be140d5a07256df84ed8d7","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2019-04-23 11:39:17.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"e5e8b0bbe7a5e9163641e2230e4c2a7f117f1533","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-23 11:39:33.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"caf98143a796f5b1bde0206e6be8e98cfeef5cbd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-23 11:50:22.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"fdd2ad803b4c6888b2500b0eff016a6e21024f62","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-23 11:50:22.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tox-docs http://logs.openstack.org/71/654571/2/gate/tox-docs/787e19d/html/ : SUCCESS in 4m 31s\n- openstack-zuul-jobs-linters http://logs.openstack.org/71/654571/2/gate/openstack-zuul-jobs-linters/f822d10/ : SUCCESS in 5m 12s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"86925f88717c4c141707cdc502ccef378f9dead6","revisions":{"b56e0f9a9d007f670f220a5c0538bbf83d4f0c1d":{"kind":"REWORK","_number":1,"created":"2019-04-22 16:59:31.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/71/654571/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-zuul-jobs","ref":"refs/changes/71/654571/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/71/654571/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/71/654571/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/71/654571/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/71/654571/1"}}},"commit":{"parents":[{"commit":"1011fd9b83e3c26bd2c5f6522e5cb0ebf5b03ef6","subject":"Remove oslosphinx","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/1011fd9b83e3c26bd2c5f6522e5cb0ebf5b03ef6"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-04-22 16:58:07.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-04-22 16:58:07.000000000","tz":120},"subject":"Use py36 instead of py35 for periodic master jobs","message":"Use py36 instead of py35 for periodic master jobs\n\nOpenStack Train will test on py36 instead of py35, change the periodic\nmaster jobs periodic-jobs-with-X-master to use py36.\n\nChange-Id: Ie2941e89e543db406db5a89666dfcb576e3e8f96\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/b56e0f9a9d007f670f220a5c0538bbf83d4f0c1d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/b56e0f9a9d007f670f220a5c0538bbf83d4f0c1d"}]},"branch":"refs/heads/master"},"86925f88717c4c141707cdc502ccef378f9dead6":{"kind":"REWORK","_number":2,"created":"2019-04-22 17:19:29.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/71/654571/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-zuul-jobs","ref":"refs/changes/71/654571/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/71/654571/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/71/654571/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/71/654571/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/71/654571/2"}}},"commit":{"parents":[{"commit":"1011fd9b83e3c26bd2c5f6522e5cb0ebf5b03ef6","subject":"Remove oslosphinx","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/1011fd9b83e3c26bd2c5f6522e5cb0ebf5b03ef6"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-04-22 16:58:07.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-04-22 17:06:41.000000000","tz":120},"subject":"Use py36 instead of py35 for periodic master jobs","message":"Use py36 instead of py35 for periodic master jobs\n\nOpenStack Train will test on py36 instead of py35, change the periodic\nmaster jobs periodic-jobs-with-X-master to use py36.\n\nLeave openstack-tox-py35-with-neutron-lib-master and -with-ovsdbapp\nin for now, we first need to change a few users.\n\nChange-Id: Ie2941e89e543db406db5a89666dfcb576e3e8f96\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/86925f88717c4c141707cdc502ccef378f9dead6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/86925f88717c4c141707cdc502ccef378f9dead6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
