)]}'
{"id":"openstack%2Fopenstack-zuul-jobs~654580","triplet_id":"openstack%2Fopenstack-zuul-jobs~master~Id338db706d9bf3edeb3af31d5e5a665687a8e7b7","project":"openstack/openstack-zuul-jobs","branch":"master","topic":"periodic-py36","hashtags":[],"change_id":"Id338db706d9bf3edeb3af31d5e5a665687a8e7b7","subject":"Remove openstack-tox-py35-with-neutron-lib-master","status":"MERGED","created":"2019-04-22 17:19:29.000000000","updated":"2019-05-12 14:56:31.000000000","submitted":"2019-05-12 14:56:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"654580-1557672991977-a4b81bb3","meta_rev_id":"2d1d6ae1381f7c9b607eaf1806999944a06ba799","_number":654580,"virtual_id_number":654580,"owner":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},{"value":2,"date":"2019-05-12 14:56:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2019-05-03 16:35:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":2,"date":"2019-05-12 14:48:25.000000000","_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":1,"date":"2019-05-03 15:10:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":1,"date":"2019-04-24 13:50:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-04-24 20:26:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":1,"date":"2019-05-12 14:48:25.000000000","_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-24 13:50:12.000000000","updated_by":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"reviewer":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"state":"REVIEWER"},{"updated":"2019-04-24 20:26:35.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2019-05-03 15:10:17.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2019-05-03 16:35:06.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2019-05-12 14:48:25.000000000","updated_by":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"reviewer":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"state":"REVIEWER"},{"updated":"2019-05-12 14:56:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"08137761fa57fb7194a28855f31d7028d2c8443a","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-04-22 17:19:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"227694e5d99395f11c9d3ae0b0b36ec4d3eee334","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 17:19:57.000000000","message":"Patch Set 1: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/project-config on branch master.  The error was:\n\n  Job openstack-tox-py35-with-neutron-lib-master not defined\n\nThe error appears in the following project stanza:\n\n  project:\n      name: openstack/networking-odl\n      templates:\n        - periodic-jobs-with-neutron-lib-master\n        - translation-jobs-master-only\n        - publish-to-pypi\n      periodic:\n        jobs:\n          - openstack-tox-py35-with-neutron-lib-master:\n              required-projects:\n                - openstack/ceilometer\n                - openstack/neutron-fwaas\n                - openstack/neutron-lbaas\n                - openstack/networking-l2gw\n                - openstack/networking-sfc\n                - openstack/networking-bgpvpn\n\n  in \"openstack/project-config/zuul.d/projects.yaml@master\", line 3224, column 3","accounts_in_message":[],"_revision_number":1},{"id":"66b0eac7c5d3b7d09e161acdaaf8d03be3f14456","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-04-23 13:14:24.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"5f9645875eea579e480d3864b7d87e3b2c5eff81","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-23 13:16:10.000000000","message":"Patch Set 2: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/networking-odl on branch stable/ocata.  The error was:\n\n  Job openstack-tox-py35-with-neutron-lib-master not defined\n\nThe error appears in the following project stanza:\n\n  project:\n      templates:\n        - openstack-python-jobs-neutron\n        - openstack-python35-jobs-neutron\n        - release-notes-jobs\n        - periodic-stable-jobs-neutron\n        - publish-openstack-sphinx-docs\n        - check-requirements\n      check:\n        jobs:\n          - networking-odl-grenade-dsvm:\n              voting: false\n              branches: ^(?!(driverfixes|stable/(mitaka|newton))).*$\n              irrelevant-files:\n                - ^(test-|)requirements.txt$\n                - ^.*\\.rst$\n                - ^doc/.*$\n                - ^releasenotes/.*$\n                - ^setup.cfg$\n          - networking-odl-tempest-dsvm-boron-snapshot-v2driver:\n              voting: false\n              branches: ^(?!stable/newton).*$\n              irrelevant-files:\n                - ^(test-|)requirements.txt$\n                - ^.*\\.rst$\n                - ^doc/.*$\n                - ^releasenotes/.*$\n                - ^setup.cfg$\n          - networking-odl-tempest-dsvm-carbon-snapshot-vpnservice-v2driver:\n              voting: false\n              irrelevant-files:\n                - ^(test-|)requirements.txt$\n                - ^.*\\.rst$\n                - ^doc/.*$\n                - ^releasenotes/.*$\n                - ^setup.cfg$\n          - networking-odl-tempest-dsvm-multinode-carbon-snapshot:\n              voting: false\n              irrelevant-files:\n                - ^(test-|)requirements.txt$\n                - ^.*\\.rst$\n                - ^doc/.*$\n                - ^releasenotes/.*$\n                - ^setup.cfg$\n          - networking-odl-rally-dsvm-carbon-snapshot:\n              branches: ^(?!stable/newton).*$\n              irrelevant-files:\n                - ^.*\\.rst$\n                - ^doc/.*$\n                - ^releasenotes/.*$\n          - networking-odl-dsvm-fullstack-carbon-snapshot:\n              branches: ^(?!stable/(newton|ocata)).*$\n              irrelevant-files:\n                - ^.*\\.rst$\n                - ^doc/.*$\n                - ^releasenotes/.*$\n          - networking-odl-dsvm-functional-carbon-snapshot:\n              branches: ^(?!stable/newton).*$\n              irrelevant-files:\n                - ^.*\\.rst$\n                - ^doc/.*$\n                - ^releasenotes/.*$\n          - openstack-tox-pep8:\n              required-projects:\n                - openstack/ceilometer\n                - openstack/neutron-fwaas\n                - openstack/neutron-lbaas\n                - openstack/networking-l2gw\n                - openstack/networking-sfc\n                - openstack/networking-bgpvpn\n          - openstack-tox-py27:\n              required-projects:\n                - openstack/ceilometer\n                - openstack/neutron-fwaas\n                - openstack/neutron-lbaas\n                - openstack/networking-l2gw\n                - openstack/networking-sfc\n                - openstack/networking-bgpvpn\n          - openstack-tox-py35:\n              required-projects:\n                - openstack/ceilometer\n                - openstack/neutron-fwaas\n                - openstack/neutron-lbaas\n                - openstack/networking-l2gw\n                - openstack/networking-sfc\n                - openstack/networking-bgpvpn\n          - openstack-tox-cover:\n              required-projects:\n                - openstack/ceilometer\n                - openstack/neutron\n                - openstack/neutron-fwaas\n                - openstack/neutron-lbaas\n                - openstack/networking-l2gw\n                - openstack/networking-sfc\n                - openstack/networking-bgpvpn\n          - build-openstack-sphinx-docs:\n              required-projects:\n                - openstack/ceilometer\n                - openstack/neutron-fwaas\n                - openstack/neutron-lbaas\n                - openstack/networking-l2gw\n                - openstack/networking-sfc\n                - openstack/networking-bgpvpn\n      gate:\n        jobs:\n          - networking-odl-rally-dsvm-carbon-snapshot:\n              branches: ^(?!stable/newton).*$\n              irrelevant-files:\n                - ^.*\\.rst$\n                - ^doc/.*$\n                - ^releasenotes/.*$\n          - networking-odl-dsvm-fullstack-carbon-snapshot:\n              branches: ^(?!stable/(newton|ocata)).*$\n              irrelevant-files:\n                - ^.*\\.rst$\n                - ^doc/.*$\n                - ^releasenotes/.*$\n          - networking-odl-dsvm-functional-carbon-snapshot:\n              branches: ^(?!stable/newton).*$\n              irrelevant-files:\n                - ^.*\\.rst$\n                - ^doc/.*$\n                - ^releasenotes/.*$\n          - openstack-tox-pep8:\n              required-projects:\n                - openstack/ceilometer\n                - openstack/neutron-fwaas\n                - openstack/neutron-lbaas\n                - openstack/networking-l2gw\n                - openstack/networking-sfc\n                - openstack/networking-bgpvpn\n          - openstack-tox-py27:\n              required-projects:\n                - openstack/ceilometer\n                - openstack/neutron-fwaas\n                - openstack/neutron-lbaas\n                - openstack/networking-l2gw\n                - openstack/networking-sfc\n                - openstack/networking-bgpvpn\n          - openstack-tox-py35:\n              required-projects:\n                - openstack/ceilometer\n                - openstack/neutron-fwaas\n                - openstack/neutron-lbaas\n                - openstack/networking-l2gw\n                - openstack/networking-sfc\n                - openstack/networking-bgpvpn\n          - openstack-tox-cover:\n              required-projects:\n                - openstack/ceilometer\n                - openstack/neutron\n                - openstack/neutron-fwaas\n                - openstack/neutron-lbaas\n                - openstack/networking-l2gw\n                - openstack/networking-sfc\n                - openstack/networking-bgpvpn\n          - build-openstack-sphinx-docs:\n              required-projects:\n                - openstack/ceilometer\n                - openstack/neutron-fwaas\n                - openstack/neutron-lbaas\n                - openstack/networking-l2gw\n                - openstack/networking-sfc\n                - openstack/networking-bgpvpn\n      post:\n        jobs:\n          - publish-openstack-python-branch-tarball:\n              required-projects:\n                - openstack/ceilometer\n                - openstack/neutron-fwaas\n                - openstack/neutron-lbaas\n                - openstack/networking-l2gw\n                - openstack/networking-sfc\n                - openstack/networking-bgpvpn\n      periodic:\n        jobs:\n          - openstack-tox-py35-with-neutron-lib-master:\n              required-projects:\n                - openstack/ceilometer\n                - openstack/neutron-fwaas\n                - openstack/neutron-lbaas\n                - openstack/networking-l2gw\n                - openstack/networking-sfc\n                - openstack/networking-bgpvpn\n      periodic-stable:\n        jobs:\n          - build-openstack-sphinx-docs:\n              required-projects:\n                - openstack/ceilometer\n                - openstack/neutron-fwaas\n                - openstack/neutron-lbaas\n                - openstack/networking-l2gw\n                - openstack/networking-sfc\n                - openstack/networking-bgpvpn\n          - openstack-tox-py27:\n              required-projects:\n                - openstack/ceilometer\n                - openstack/neutron-fwaas\n                - openstack/neutron-lbaas\n                - openstack/networking-l2gw\n                - openstack/networking-sfc\n                - openstack/networking-bgpvpn\n\n  in \"openstack/networking-odl/.zuul.d/project.yaml@stable/ocata\", line 1, column 3","accounts_in_message":[],"_revision_number":2},{"id":"bd179604844260c0d5093bbbd0ae2b0f2255a995","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-04-23 13:32:45.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"c06a1efe7d769eb57268109711d64a5834ae7d45","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-23 13:33:15.000000000","message":"Patch Set 3: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/networking-odl on branch stable/rocky.  The error was:\n\n  Job openstack-tox-py35-with-neutron-lib-master not defined\n\nThe error appears in the following project stanza:\n\n  project:\n      templates:\n        - openstack-python-jobs-neutron\n        - openstack-python35-jobs-neutron\n        - release-notes-jobs\n        - periodic-stable-jobs-neutron\n        - publish-openstack-sphinx-docs\n        - check-requirements\n      check:\n        jobs:\n          - openstack-tox-lower-constraints:\n              required-projects:\n                - openstack/ceilometer\n                - openstack/neutron\n                - openstack/neutron-fwaas\n                - openstack/neutron-lbaas\n                - openstack/networking-l2gw\n                - openstack/networking-sfc\n                - openstack/networking-bgpvpn\n          - networking-odl-grenade:\n              voting: false\n              branches: ^(?!(driverfixes|stable/(mitaka|newton))).*$\n              irrelevant-files:\n                - ^(test-|)requirements.txt$\n                - ^.*\\.rst$\n                - ^doc/.*$\n                - ^releasenotes/.*$\n                - ^setup.cfg$\n          - networking-odl-tempest-oxygen\n          - networking-odl-tempest-fluorine\n          - networking-odl-tempest-oxygen-multinode:\n              voting: false\n              irrelevant-files:\n                - ^(test-|)requirements.txt$\n                - ^.*\\.rst$\n                - ^doc/.*$\n                - ^releasenotes/.*$\n                - ^setup.cfg$\n          - networking-odl-tempest-fluorine-multinode:\n              voting: false\n              irrelevant-files:\n                - ^(test-|)requirements.txt$\n                - ^.*\\.rst$\n                - ^doc/.*$\n                - ^releasenotes/.*$\n                - ^setup.cfg$\n          - networking-odl-rally-oxygen:\n              voting: false\n              branches: ^(?!stable/(newton|ocata|pike)).*$\n              irrelevant-files:\n                - ^.*\\.rst$\n                - ^doc/.*$\n                - ^releasenotes/.*$\n          - networking-odl-rally-fluorine:\n              voting: false\n              branches: ^(?!stable/(newton|ocata|pike)).*$\n              irrelevant-files:\n                - ^.*\\.rst$\n                - ^doc/.*$\n                - ^releasenotes/.*$\n          - networking-odl-functional-oxygen\n          - networking-odl-functional-fluorine\n          - openstack-tox-pep8:\n              required-projects:\n                - openstack/ceilometer\n                - openstack/neutron-fwaas\n                - openstack/neutron-lbaas\n                - openstack/networking-l2gw\n                - openstack/networking-sfc\n                - openstack/networking-bgpvpn\n          - openstack-tox-py27:\n              required-projects:\n                - openstack/ceilometer\n                - openstack/neutron-fwaas\n                - openstack/neutron-lbaas\n                - openstack/networking-l2gw\n                - openstack/networking-sfc\n                - openstack/networking-bgpvpn\n          - openstack-tox-py35:\n              required-projects:\n                - openstack/ceilometer\n                - openstack/neutron-fwaas\n                - openstack/neutron-lbaas\n                - openstack/networking-l2gw\n                - openstack/networking-sfc\n                - openstack/networking-bgpvpn\n          - openstack-tox-cover:\n              required-projects:\n                - openstack/ceilometer\n                - openstack/neutron\n                - openstack/neutron-fwaas\n                - openstack/neutron-lbaas\n                - openstack/networking-l2gw\n                - openstack/networking-sfc\n                - openstack/networking-bgpvpn\n          - build-openstack-sphinx-docs:\n              required-projects:\n                - openstack/ceilometer\n                - openstack/neutron-fwaas\n                - openstack/neutron-lbaas\n                - openstack/networking-l2gw\n                - openstack/networking-sfc\n                - openstack/networking-bgpvpn\n      gate:\n        jobs:\n          - openstack-tox-lower-constraints:\n              required-projects:\n                - openstack/ceilometer\n                - openstack/neutron\n                - openstack/neutron-fwaas\n                - openstack/neutron-lbaas\n                - openstack/networking-l2gw\n                - openstack/networking-sfc\n                - openstack/networking-bgpvpn\n          - networking-odl-functional-oxygen\n          - networking-odl-functional-fluorine\n          - openstack-tox-pep8:\n              required-projects:\n                - openstack/ceilometer\n                - openstack/neutron-fwaas\n                - openstack/neutron-lbaas\n                - openstack/networking-l2gw\n                - openstack/networking-sfc\n                - openstack/networking-bgpvpn\n          - openstack-tox-py27:\n              required-projects:\n                - openstack/ceilometer\n                - openstack/neutron-fwaas\n                - openstack/neutron-lbaas\n                - openstack/networking-l2gw\n                - openstack/networking-sfc\n                - openstack/networking-bgpvpn\n          - openstack-tox-py35:\n              required-projects:\n                - openstack/ceilometer\n                - openstack/neutron-fwaas\n                - openstack/neutron-lbaas\n                - openstack/networking-l2gw\n                - openstack/networking-sfc\n                - openstack/networking-bgpvpn\n          - openstack-tox-cover:\n              required-projects:\n                - openstack/ceilometer\n                - openstack/neutron\n                - openstack/neutron-fwaas\n                - openstack/neutron-lbaas\n                - openstack/networking-l2gw\n                - openstack/networking-sfc\n                - openstack/networking-bgpvpn\n          - build-openstack-sphinx-docs:\n              required-projects:\n                - openstack/ceilometer\n                - openstack/neutron-fwaas\n                - openstack/neutron-lbaas\n                - openstack/networking-l2gw\n                - openstack/networking-sfc\n                - openstack/networking-bgpvpn\n      post:\n        jobs:\n          - publish-openstack-python-branch-tarball:\n              required-projects:\n                - openstack/ceilometer\n                - openstack/neutron-fwaas\n                - openstack/neutron-lbaas\n                - openstack/networking-l2gw\n                - openstack/networking-sfc\n                - openstack/networking-bgpvpn\n      periodic:\n        jobs:\n          - openstack-tox-py35-with-neutron-lib-master:\n              required-projects:\n                - openstack/ceilometer\n                - openstack/neutron-fwaas\n                - openstack/neutron-lbaas\n                - openstack/networking-l2gw\n                - openstack/networking-sfc\n                - openstack/networking-bgpvpn\n\n  in \"openstack/networking-odl/.zuul.d/project.yaml@stable/rocky\", line 1, column 3","accounts_in_message":[],"_revision_number":3},{"id":"f28edc5b5ae82b0a7e39b1afdd1a85ab2f1ba7d8","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-04-23 13:34:59.000000000","message":"Uploaded patch set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"59d22647bb1241b7a2588ea07ba25fa2aff8dec7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-23 13:35:20.000000000","message":"Patch Set 4: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/networking-bagpipe on branch stable/ocata.  The error\nwas:\n\n  Job openstack-tox-py35-with-neutron-lib-master not defined\n\nThe error appears in the following project stanza:\n\n  project:\n      templates:\n        - openstack-python-jobs-neutron\n        - openstack-python35-jobs-neutron\n        - publish-openstack-sphinx-docs\n        - release-notes-jobs\n        - check-requirements\n        - periodic-stable-jobs-neutron\n      check:\n        jobs:\n          - build-openstack-sphinx-docs:\n              required-projects:\n                - openstack/networking-bgpvpn\n                - openstack/networking-sfc\n          - openstack-tox-pep8:\n              required-projects:\n                - openstack/networking-bgpvpn\n                - openstack/networking-sfc\n          - openstack-tox-py27:\n              required-projects:\n                - openstack/networking-bgpvpn\n                - openstack/networking-sfc\n          - openstack-tox-py35:\n              required-projects:\n                - openstack/networking-bgpvpn\n                - openstack/networking-sfc\n          - openstack-tox-cover:\n              required-projects:\n                - openstack/neutron\n                - openstack/networking-bgpvpn\n                - openstack/networking-sfc\n          - legacy-tempest-dsvm-networking-bagpipe:\n              irrelevant-files:\n                - ^(test-|)requirements.txt$\n                - ^setup.cfg$\n          - legacy-tempest-dsvm-networking-bgpvpn-bagpipe:\n              irrelevant-files:\n                - ^(test-|)requirements.txt$\n                - ^setup.cfg$\n      gate:\n        jobs:\n          - build-openstack-sphinx-docs:\n              required-projects:\n                - openstack/networking-bgpvpn\n                - openstack/networking-sfc\n          - openstack-tox-pep8:\n              required-projects:\n                - openstack/networking-bgpvpn\n                - openstack/networking-sfc\n          - openstack-tox-py27:\n              required-projects:\n                - openstack/networking-bgpvpn\n                - openstack/networking-sfc\n          - openstack-tox-py35:\n              required-projects:\n                - openstack/networking-bgpvpn\n                - openstack/networking-sfc\n          - openstack-tox-cover:\n              required-projects:\n                - openstack/neutron\n                - openstack/networking-bgpvpn\n                - openstack/networking-sfc\n          - legacy-tempest-dsvm-networking-bagpipe:\n              irrelevant-files:\n                - ^(test-|)requirements.txt$\n                - ^setup.cfg$\n          - legacy-tempest-dsvm-networking-bgpvpn-bagpipe:\n              irrelevant-files:\n                - ^(test-|)requirements.txt$\n                - ^setup.cfg$\n      post:\n        jobs:\n          - publish-openstack-python-branch-tarball:\n              required-projects:\n                - openstack/neutron\n                - openstack/networking-bgpvpn\n                - openstack/networking-sfc\n      periodic:\n        jobs:\n          - openstack-tox-py35-with-neutron-lib-master:\n              required-projects:\n                - openstack/networking-bgpvpn\n                - openstack/networking-sfc\n      periodic-stable:\n        jobs:\n          - build-openstack-sphinx-docs:\n              required-projects:\n                - openstack/networking-bgpvpn\n                - openstack/networking-sfc\n          - openstack-tox-py27:\n              required-projects:\n                - openstack/networking-bgpvpn\n                - openstack/networking-sfc\n\n  in \"openstack/networking-bagpipe/.zuul.yaml@stable/ocata\", line 1, column 3","accounts_in_message":[],"_revision_number":4},{"id":"68e26293f53622762a3513d76ae046be423d0ae2","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-04-23 13:49:06.000000000","message":"Uploaded patch set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"2b1dd4c0ce7e3225d8ce0725089f9f0a5dd3fc49","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-23 14:08:03.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tox-docs http://logs.openstack.org/80/654580/5/check/tox-docs/f8f7817/html/ : SUCCESS in 4m 13s\n- openstack-infra-extra-integration-xenial http://logs.openstack.org/80/654580/5/check/openstack-infra-extra-integration-xenial/bd1e481/ : SUCCESS in 2m 28s\n- openstack-infra-extra-integration-bionic http://logs.openstack.org/80/654580/5/check/openstack-infra-extra-integration-bionic/a7214d2/ : SUCCESS in 4m 39s\n- openstack-zuul-jobs-linters http://logs.openstack.org/80/654580/5/check/openstack-zuul-jobs-linters/86a4f41/ : SUCCESS in 5m 58s","accounts_in_message":[],"_revision_number":5},{"id":"39399138105108c38cfef48517015bf06855952e","author":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"date":"2019-04-24 13:50:12.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"78db3f760a34bc90bca0167b45806b27fcdfbf4e","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-04-24 20:26:35.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"71b1c56e24374f28d927bbe6d086efdd9b1c8a74","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2019-05-03 15:10:17.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"7ea670617652e95192197c4aa575b1f119dcc1e2","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2019-05-03 16:35:06.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"57e7a8fd72e01f6fd2af2d588a9a8507fde9d60d","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2019-05-12 14:48:25.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"1d73927a90eea46dbcf1c0eba40f5179644325c0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-12 14:48:40.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"e860c813380026f22439f0f6fc00c33b2832e7bf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-12 14:56:31.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tox-docs http://logs.openstack.org/80/654580/5/gate/tox-docs/6bfafde/html/ : SUCCESS in 3m 13s\n- openstack-zuul-jobs-linters http://logs.openstack.org/80/654580/5/gate/openstack-zuul-jobs-linters/65b7c18/ : SUCCESS in 5m 17s","accounts_in_message":[],"_revision_number":5},{"id":"e6b79f5bd074ac5383a4fbc5feda119227cdae26","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-12 14:56:31.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"38e75a3d4973ad92dffd292993c8339a72e577b2","revisions":{"c019d91dd33cc02b14251a0b568548b4c65f52a0":{"kind":"REWORK","_number":1,"created":"2019-04-22 17:19:29.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/80/654580/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-zuul-jobs","ref":"refs/changes/80/654580/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/80/654580/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/80/654580/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/80/654580/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/80/654580/1"}}},"commit":{"parents":[{"commit":"86925f88717c4c141707cdc502ccef378f9dead6","subject":"Use py36 instead of py35 for periodic master jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/86925f88717c4c141707cdc502ccef378f9dead6"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-04-22 17:12:46.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-04-22 17:16:08.000000000","tz":120},"subject":"Remove periodic-py35 jobs","message":"Remove periodic-py35 jobs\n\nThese jobs are not needed anymore, we use instead now the py36 variants.\nRemove the jobs.\n\nDepends-On: https://review.opendev.org/654573\nDepends-On: https://review.opendev.org/654574\nDepends-On: https://review.opendev.org/654575\nDepends-On: https://review.opendev.org/654576\nDepends-On: https://review.opendev.org/654578\n\nChange-Id: Id338db706d9bf3edeb3af31d5e5a665687a8e7b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/c019d91dd33cc02b14251a0b568548b4c65f52a0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/c019d91dd33cc02b14251a0b568548b4c65f52a0"}]},"branch":"refs/heads/master"},"96018e05940a5a1843957e8663fc899eea16e1cc":{"kind":"REWORK","_number":2,"created":"2019-04-23 13:14:24.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/80/654580/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-zuul-jobs","ref":"refs/changes/80/654580/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/80/654580/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/80/654580/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/80/654580/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/80/654580/2"}}},"commit":{"parents":[{"commit":"86925f88717c4c141707cdc502ccef378f9dead6","subject":"Use py36 instead of py35 for periodic master jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/86925f88717c4c141707cdc502ccef378f9dead6"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-04-22 17:12:46.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-04-23 13:12:44.000000000","tz":120},"subject":"Remove openstack-tox-py35-with-neutron-lib-master","message":"Remove openstack-tox-py35-with-neutron-lib-master\n\nThe job is not needed anymore, we use instead now the py36 variant.\nRemove the job.\n\nDepends-On: https://review.opendev.org/654574\n\nChange-Id: Id338db706d9bf3edeb3af31d5e5a665687a8e7b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/96018e05940a5a1843957e8663fc899eea16e1cc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/96018e05940a5a1843957e8663fc899eea16e1cc"}]},"branch":"refs/heads/master"},"16e0023d514a238155dbdcab7985e9aa27d31aeb":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2019-04-23 13:32:45.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/80/654580/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-zuul-jobs","ref":"refs/changes/80/654580/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/80/654580/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/80/654580/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/80/654580/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/80/654580/3"}}},"commit":{"parents":[{"commit":"86925f88717c4c141707cdc502ccef378f9dead6","subject":"Use py36 instead of py35 for periodic master jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/86925f88717c4c141707cdc502ccef378f9dead6"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-04-22 17:12:46.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-04-23 13:31:05.000000000","tz":120},"subject":"Remove openstack-tox-py35-with-neutron-lib-master","message":"Remove openstack-tox-py35-with-neutron-lib-master\n\nThe job is not needed anymore, we use instead now the py36 variant.\nRemove the job.\n\nDepends-On: https://review.opendev.org/654574\nDepends-On: https://review.opendev.org/655146\nDepends-On: https://review.opendev.org/655149\nDepends-On: https://review.opendev.org/655150\n\nChange-Id: Id338db706d9bf3edeb3af31d5e5a665687a8e7b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/16e0023d514a238155dbdcab7985e9aa27d31aeb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/16e0023d514a238155dbdcab7985e9aa27d31aeb"}]},"branch":"refs/heads/master"},"3d5753446c31d06fe8652c82a941187c954d6620":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2019-04-23 13:34:59.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/80/654580/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-zuul-jobs","ref":"refs/changes/80/654580/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/80/654580/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/80/654580/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/80/654580/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/80/654580/4"}}},"commit":{"parents":[{"commit":"86925f88717c4c141707cdc502ccef378f9dead6","subject":"Use py36 instead of py35 for periodic master jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/86925f88717c4c141707cdc502ccef378f9dead6"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-04-22 17:12:46.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-04-23 13:34:05.000000000","tz":120},"subject":"Remove openstack-tox-py35-with-neutron-lib-master","message":"Remove openstack-tox-py35-with-neutron-lib-master\n\nThe job is not needed anymore, we use instead now the py36 variant.\nRemove the job.\n\nDepends-On: https://review.opendev.org/654574\nDepends-On: https://review.opendev.org/655146\nDepends-On: https://review.opendev.org/655149\nDepends-On: https://review.opendev.org/655150\nDepends-On: https://review.opendev.org/655153\n\nChange-Id: Id338db706d9bf3edeb3af31d5e5a665687a8e7b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/3d5753446c31d06fe8652c82a941187c954d6620"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/3d5753446c31d06fe8652c82a941187c954d6620"}]},"branch":"refs/heads/master"},"38e75a3d4973ad92dffd292993c8339a72e577b2":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2019-04-23 13:49:06.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/80/654580/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-zuul-jobs","ref":"refs/changes/80/654580/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/80/654580/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/80/654580/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/80/654580/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/80/654580/5"}}},"commit":{"parents":[{"commit":"86925f88717c4c141707cdc502ccef378f9dead6","subject":"Use py36 instead of py35 for periodic master jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/86925f88717c4c141707cdc502ccef378f9dead6"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-04-22 17:12:46.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-04-23 13:48:16.000000000","tz":120},"subject":"Remove openstack-tox-py35-with-neutron-lib-master","message":"Remove openstack-tox-py35-with-neutron-lib-master\n\nThe job is not needed anymore, we use instead now the py36 variant.\nRemove the job.\n\nDepends-On: https://review.opendev.org/654574\nDepends-On: https://review.opendev.org/655146\nDepends-On: https://review.opendev.org/655149\nDepends-On: https://review.opendev.org/655150\nDepends-On: https://review.opendev.org/655153\nDepends-On: https://review.opendev.org/655161\nDepends-On: https://review.opendev.org/655162\nDepends-On: https://review.opendev.org/655163\nDepends-On: https://review.opendev.org/655164\n\nChange-Id: Id338db706d9bf3edeb3af31d5e5a665687a8e7b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/38e75a3d4973ad92dffd292993c8339a72e577b2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/38e75a3d4973ad92dffd292993c8339a72e577b2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
