)]}'
{"id":"openstack%2Fopenstack-zuul-jobs~659735","triplet_id":"openstack%2Fopenstack-zuul-jobs~master~Ia14b417d5cf549f761a38ff6b3ac133caf8644ad","project":"openstack/openstack-zuul-jobs","branch":"master","hashtags":[],"change_id":"Ia14b417d5cf549f761a38ff6b3ac133caf8644ad","subject":"Update filelist for gerrit check","status":"MERGED","created":"2019-05-17 07:38:02.000000000","updated":"2019-05-17 12:18:04.000000000","submitted":"2019-05-17 12:18:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"659735-1558095484026-4f1731f4","meta_rev_id":"5deb80be278976160430fdb755e5a888e4ffe671","_number":659735,"virtual_id_number":659735,"owner":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":4162,"name":"Paul Belanger","email":"pabelanger@redhat.com","username":"pabelanger"},{"value":2,"date":"2019-05-17 12:18:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"all":[{"value":2,"date":"2019-05-17 12:08:31.000000000","_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":2,"date":"2019-05-17 12:09:47.000000000","_account_id":4162,"name":"Paul Belanger","email":"pabelanger@redhat.com","username":"pabelanger"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4162,"name":"Paul Belanger","email":"pabelanger@redhat.com","username":"pabelanger"},"all":[{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":1,"date":"2019-05-17 12:09:47.000000000","_account_id":4162,"name":"Paul Belanger","email":"pabelanger@redhat.com","username":"pabelanger"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"_account_id":4162,"name":"Paul Belanger","email":"pabelanger@redhat.com","username":"pabelanger"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-17 12:08:31.000000000","updated_by":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"reviewer":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"state":"REVIEWER"},{"updated":"2019-05-17 12:09:47.000000000","updated_by":{"_account_id":4162,"name":"Paul Belanger","email":"pabelanger@redhat.com","username":"pabelanger"},"reviewer":{"_account_id":4162,"name":"Paul Belanger","email":"pabelanger@redhat.com","username":"pabelanger"},"state":"REVIEWER"},{"updated":"2019-05-17 12:18:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"652849ab9200056354f3a990dfa6333cb90f2584","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-05-17 07:38:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"873c7601d819a4e7819b34f7f226b050b9a73300","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-17 07:46:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tox-docs http://logs.openstack.org/35/659735/1/check/tox-docs/94dd6e1/html/ : SUCCESS in 3m 14s\n- openstack-infra-extra-integration-xenial http://logs.openstack.org/35/659735/1/check/openstack-infra-extra-integration-xenial/1938204/ : SUCCESS in 2m 36s\n- openstack-infra-extra-integration-bionic http://logs.openstack.org/35/659735/1/check/openstack-infra-extra-integration-bionic/7626459/ : SUCCESS in 2m 10s\n- openstack-zuul-jobs-linters http://logs.openstack.org/35/659735/1/check/openstack-zuul-jobs-linters/b780a14/ : SUCCESS in 4m 25s","accounts_in_message":[],"_revision_number":1},{"id":"3e948574bfe0c777bc00a446bdc458464af473f1","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2019-05-17 12:08:31.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"8a58d1e00f93b5f14b50f85778d6c77a798444d2","author":{"_account_id":4162,"name":"Paul Belanger","email":"pabelanger@redhat.com","username":"pabelanger"},"date":"2019-05-17 12:09:47.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"79a57088bc04f5355aedf2728d11b28200164d25","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-17 12:10:03.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d7114e0a3c96db8368337095c7690550ed997f70","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-17 12:18:03.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tox-docs http://logs.openstack.org/35/659735/1/gate/tox-docs/66f2239/html/ : SUCCESS in 2m 53s\n- openstack-zuul-jobs-linters http://logs.openstack.org/35/659735/1/gate/openstack-zuul-jobs-linters/6364669/ : SUCCESS in 4m 36s","accounts_in_message":[],"_revision_number":1},{"id":"5deb80be278976160430fdb755e5a888e4ffe671","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-17 12:18:04.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ccc4606f418928e148c724c743e8b00970ccc38e","revisions":{"ccc4606f418928e148c724c743e8b00970ccc38e":{"kind":"REWORK","_number":1,"created":"2019-05-17 07:38:02.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/35/659735/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-zuul-jobs","ref":"refs/changes/35/659735/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/35/659735/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/35/659735/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/35/659735/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/35/659735/1"}}},"commit":{"parents":[{"commit":"9e633d92fbeb4d474b10fcc5db11267c28dd8b2d","subject":"Merge \"Remove unused jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/9e633d92fbeb4d474b10fcc5db11267c28dd8b2d"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-05-17 07:36:59.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-05-17 07:36:59.000000000","tz":120},"subject":"Update filelist for gerrit check","message":"Update filelist for gerrit check\n\nUpdate project-config-gerrit to check zuul/main.yaml as well.\n\nChange-Id: Ia14b417d5cf549f761a38ff6b3ac133caf8644ad\nNeeded-By: https://review.opendev.org/659728\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/ccc4606f418928e148c724c743e8b00970ccc38e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/ccc4606f418928e148c724c743e8b00970ccc38e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
