)]}'
{"id":"openstack%2Fopenstack-zuul-jobs~659948","triplet_id":"openstack%2Fopenstack-zuul-jobs~master~I045046ee0c0a28471385fdcdbcfa0516c2fac397","project":"openstack/openstack-zuul-jobs","branch":"master","topic":"docs","hashtags":[],"change_id":"I045046ee0c0a28471385fdcdbcfa0516c2fac397","subject":"Fix docs jobs in periodic template","status":"MERGED","created":"2019-05-19 08:49:23.000000000","updated":"2019-05-20 00:58:20.000000000","submitted":"2019-05-20 00:58:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"659948-1558313900376-8079ff73","meta_rev_id":"ed262fec13eb7dab45320978a4985d933d623831","_number":659948,"virtual_id_number":659948,"owner":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":2,"date":"2019-05-20 00:58:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"all":[{"value":2,"date":"2019-05-19 09:07:04.000000000","_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-05-20 00:50:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"all":[{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-05-20 00:50:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-19 09:07:04.000000000","updated_by":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"reviewer":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"state":"REVIEWER"},{"updated":"2019-05-20 00:50:44.000000000","updated_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"reviewer":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"state":"REVIEWER"},{"updated":"2019-05-20 00:58:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6cb08a711e67ea14be5d29315ef480bf6298dac9","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-05-19 08:49:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"eee5aaad5078f6764418e2b99b3bac210918ca7c","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-05-19 08:53:07.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"32c8d2f82ae6b652613f16a2278f26d9fde9fb27","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-19 09:01:45.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tox-docs http://logs.openstack.org/48/659948/2/check/tox-docs/379faec/html/ : SUCCESS in 3m 33s\n- openstack-infra-extra-integration-xenial http://logs.openstack.org/48/659948/2/check/openstack-infra-extra-integration-xenial/6f61cc7/ : SUCCESS in 2m 15s\n- openstack-infra-extra-integration-bionic http://logs.openstack.org/48/659948/2/check/openstack-infra-extra-integration-bionic/a1b9666/ : SUCCESS in 1m 37s\n- openstack-zuul-jobs-linters http://logs.openstack.org/48/659948/2/check/openstack-zuul-jobs-linters/412ee02/ : SUCCESS in 3m 57s","accounts_in_message":[],"_revision_number":2},{"id":"1b544620fc7bbda0a757ede0e4029a840e087d57","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2019-05-19 09:07:04.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"1988e9180f8cbdd3fbd77b3e637013ce022c25f0","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2019-05-20 00:50:44.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"6075525d12fc6079bfd809c6fd145e66ae7d28fe","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-20 00:50:59.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"fc1941d5181d42a4d1bc88c8b2516eba2d2739d5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-20 00:58:20.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tox-docs http://logs.openstack.org/48/659948/2/gate/tox-docs/80252b0/html/ : SUCCESS in 2m 49s\n- openstack-zuul-jobs-linters http://logs.openstack.org/48/659948/2/gate/openstack-zuul-jobs-linters/c76641c/ : SUCCESS in 3m 40s","accounts_in_message":[],"_revision_number":2},{"id":"f999614aca8791f746d5b9d39dd4b5d003933e8f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-20 00:58:20.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"6ad1636060d2e61f8386ecd3ffb2e45312f959e3","revisions":{"c92057173b944ac031f6ea9164b328fbf2582fe9":{"kind":"REWORK","_number":1,"created":"2019-05-19 08:49:23.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/48/659948/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-zuul-jobs","ref":"refs/changes/48/659948/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/48/659948/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/48/659948/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/48/659948/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/48/659948/1"}}},"commit":{"parents":[{"commit":"2419270663642019e947851effe66c7941490fc1","subject":"Remove unused legacy telemetry-dsvm playbook","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/2419270663642019e947851effe66c7941490fc1"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-05-19 08:47:35.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-05-19 08:47:35.000000000","tz":120},"subject":"Fix docs publishing in periodic jobs","message":"Fix docs publishing in periodic jobs\n\nstein uses the new PTI job openstack-tox-docs for doc building, not the\nolder build-sphinx.\n\nChange I43adc20ef8a51748784b3adb1b65d66550586ecc added this for the main\nperiodic-stable template, add it to the other ones.\n\nAlso, add stable/stein to the ceilometer specific template, it was\nmissing there.\n\nChange-Id: I045046ee0c0a28471385fdcdbcfa0516c2fac397\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/c92057173b944ac031f6ea9164b328fbf2582fe9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/c92057173b944ac031f6ea9164b328fbf2582fe9"}]},"branch":"refs/heads/master"},"6ad1636060d2e61f8386ecd3ffb2e45312f959e3":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-05-19 08:53:07.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/48/659948/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-zuul-jobs","ref":"refs/changes/48/659948/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/48/659948/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/48/659948/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/48/659948/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/48/659948/2"}}},"commit":{"parents":[{"commit":"2419270663642019e947851effe66c7941490fc1","subject":"Remove unused legacy telemetry-dsvm playbook","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/2419270663642019e947851effe66c7941490fc1"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-05-19 08:47:35.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-05-19 08:52:37.000000000","tz":120},"subject":"Fix docs jobs in periodic template","message":"Fix docs jobs in periodic template\n\nstein uses the new PTI job openstack-tox-docs for doc building, not the\nolder build-sphinx.\n\nChange I43adc20ef8a51748784b3adb1b65d66550586ecc added this for the main\nperiodic-stable template, add it to the other ones.\n\nAlso, add stable/stein to the ceilometer specific template, it was\nmissing there.\n\nChange-Id: I045046ee0c0a28471385fdcdbcfa0516c2fac397\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/6ad1636060d2e61f8386ecd3ffb2e45312f959e3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/6ad1636060d2e61f8386ecd3ffb2e45312f959e3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
