)]}'
{"id":"openstack%2Fopenstack-zuul-jobs~662402","triplet_id":"openstack%2Fopenstack-zuul-jobs~master~Ifafc060c6e82b5760adcb44d88390d427386fc39","project":"openstack/openstack-zuul-jobs","branch":"master","topic":"stadium-migration","hashtags":[],"change_id":"Ifafc060c6e82b5760adcb44d88390d427386fc39","subject":"Add neutron-tempest-plugin to bgpvpn-bagpipe job","status":"ABANDONED","created":"2019-05-31 08:33:42.000000000","updated":"2019-05-31 09:53:56.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"fc8c7537253b55962f98872c827a821360460f92","_number":662402,"virtual_id_number":662402,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2019-05-31 09:09:29.000000000","_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":1,"date":"2019-05-31 08:41:48.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-31 08:41:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-05-31 09:09:29.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"}],"messages":[{"id":"f45d6c804353d4636694eb79146916f79e489a71","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-05-31 08:33:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"addfc7e8988dba1350a8b9216238cf6a7264c350","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-31 08:41:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tox-docs http://logs.openstack.org/02/662402/1/check/tox-docs/9136c0c/html/ : SUCCESS in 3m 00s\n- openstack-infra-extra-integration-xenial http://logs.openstack.org/02/662402/1/check/openstack-infra-extra-integration-xenial/8e7b382/ : SUCCESS in 2m 15s\n- openstack-infra-extra-integration-bionic http://logs.openstack.org/02/662402/1/check/openstack-infra-extra-integration-bionic/50f70cc/ : SUCCESS in 2m 15s\n- openstack-zuul-jobs-linters http://logs.openstack.org/02/662402/1/check/openstack-zuul-jobs-linters/8b70bd1/ : SUCCESS in 4m 06s","accounts_in_message":[],"_revision_number":1},{"id":"e87b1184e69d86ee9adc75e03082b3fbc4be108e","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-05-31 09:09:29.000000000","message":"Patch Set 1:\n\nI propose to move the job instead into the bgpvpn repo and update it there.\n\nthat allows to do changes per branch as well...","accounts_in_message":[],"_revision_number":1},{"id":"ee4c9d5f471fb394925dcb899caf0d586e1d60c6","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-05-31 09:53:56.000000000","message":"Abandoned\n\nIt\u0027s not needed","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f161cf5ba6c545b426615e2fe6ee26e5ac9c3ace","revisions":{"f161cf5ba6c545b426615e2fe6ee26e5ac9c3ace":{"kind":"REWORK","_number":1,"created":"2019-05-31 08:33:42.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/02/662402/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-zuul-jobs","ref":"refs/changes/02/662402/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/02/662402/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/02/662402/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/02/662402/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/02/662402/1"}}},"commit":{"parents":[{"commit":"a0697b63b681c0d215d940c5c9566c6ce8445096","subject":"Update puppet jobs to default to puppet 4","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/a0697b63b681c0d215d940c5c9566c6ce8445096"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2019-05-31 08:31:59.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2019-05-31 08:31:59.000000000","tz":120},"subject":"Add neutron-tempest-plugin to bgpvpn-bagpipe job","message":"Add neutron-tempest-plugin to bgpvpn-bagpipe job\n\nIn commit [1] we removed tempest plugin code from networking-bgpvpn\nrepo as it was moved to neutron-tempest-plugin.\n\nThis commit adds neutron-tempest-plugin repo to required projects\nfor legacy-tempest-dsvm-networking-bgpvpn-bagpipe job\n\n[1] https://review.opendev.org/#/c/657793/\n\nChange-Id: Ifafc060c6e82b5760adcb44d88390d427386fc39\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/f161cf5ba6c545b426615e2fe6ee26e5ac9c3ace"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/f161cf5ba6c545b426615e2fe6ee26e5ac9c3ace"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
