)]}'
{"id":"openstack%2Fopenstack-zuul-jobs~677481","triplet_id":"openstack%2Fopenstack-zuul-jobs~master~Icf7700f37c717c542c12217817083ad54da5dac5","project":"openstack/openstack-zuul-jobs","branch":"master","topic":"tox-docs-revert","hashtags":[],"change_id":"Icf7700f37c717c542c12217817083ad54da5dac5","subject":"Go back to normal publishing for tox-docs","status":"ABANDONED","created":"2019-08-20 15:15:01.000000000","updated":"2019-08-20 17:44:02.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"d6a24f6658586e2ea052a334e47fe097123c98a1","_number":677481,"virtual_id_number":677481,"owner":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2019-08-20 15:32:51.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-20 15:32:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c9bdcdc230ba9e4073230d43e477e59a7069945c","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-08-20 15:15:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"98870d7756169d22c58b7b17e8923b121fdfae3a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-20 15:32:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tox-docs https://zuul.opendev.org/t/openstack/build/10508823778f407bafbf9cd98000cb0f : SUCCESS in 4m 11s\n- openstack-infra-extra-integration-xenial https://zuul.opendev.org/t/openstack/build/79435032692c4ec88fc0cece4ffbb890 : SUCCESS in 2m 20s\n- openstack-infra-extra-integration-bionic https://zuul.opendev.org/t/openstack/build/968da6d94ab745c28bcc58ef98a04cb5 : SUCCESS in 2m 41s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/a15360d132f54bd3960014cd33f39e22 : SUCCESS in 6m 35s","accounts_in_message":[],"_revision_number":1},{"id":"7d1091735aee2de4fb2474284baaf0952366273a","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-08-20 17:44:02.000000000","message":"Abandoned\n\nnot needed","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"bb54cb87d43c959e83655bcd5a9b58e2cb5e89c5","revisions":{"bb54cb87d43c959e83655bcd5a9b58e2cb5e89c5":{"kind":"REWORK","_number":1,"created":"2019-08-20 15:15:01.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/81/677481/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-zuul-jobs","ref":"refs/changes/81/677481/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/81/677481/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/81/677481/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/81/677481/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/81/677481/1"}}},"commit":{"parents":[{"commit":"9af13d4fce89bb979b5d205fd24ea3bf67d3b817","subject":"Merge \"Use promote jobs for openstack-tox-docs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/9af13d4fce89bb979b5d205fd24ea3bf67d3b817"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-08-20 15:14:26.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-08-20 15:14:26.000000000","tz":120},"subject":"Go back to normal publishing for tox-docs","message":"Go back to normal publishing for tox-docs\n\nThis is a partial revert of I7d5ba35733be78b9d734776ff8edd7020ec6b694\nuntil we figure out whether everything is working.\n\nChange-Id: Icf7700f37c717c542c12217817083ad54da5dac5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/bb54cb87d43c959e83655bcd5a9b58e2cb5e89c5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/bb54cb87d43c959e83655bcd5a9b58e2cb5e89c5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
