)]}'
{"id":"openstack%2Fopenstack-zuul-jobs~677657","triplet_id":"openstack%2Fopenstack-zuul-jobs~master~Idc5e94a581359bcfd3a570e085aa13999c26eb86","project":"openstack/openstack-zuul-jobs","branch":"master","topic":"promote-docs","hashtags":[],"change_id":"Idc5e94a581359bcfd3a570e085aa13999c26eb86","subject":"Use promote job for infra","status":"MERGED","created":"2019-08-21 07:38:29.000000000","updated":"2019-08-22 07:23:42.000000000","submitted":"2019-08-22 07:21:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"677657-1566458478832-f2959db3","meta_rev_id":"6e49a4a5940a50a9b32ae38674b0299cb72c3e7e","_number":677657,"virtual_id_number":677657,"owner":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":2,"date":"2019-08-22 07:21:18.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"value":0,"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2019-08-21 16:44:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-08-22 07:08:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"value":1,"date":"2019-08-21 08:36:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-08-22 07:08:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"value":0,"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-21 08:36:22.000000000","updated_by":{"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"},"reviewer":{"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"},"state":"REVIEWER"},{"updated":"2019-08-21 16:44:24.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2019-08-22 07:08:45.000000000","updated_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"reviewer":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"state":"REVIEWER"},{"updated":"2019-08-22 07:21:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ac6fa7ccd70c8dd59a7a70fb91ee1bd296b4f7eb","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-08-21 07:38:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9329d9ef8aabf5bb39e2c8ce58c55a30f35ad3fc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-21 07:50:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/c886659f8c3249d594f317853eb578a7 : SUCCESS in 5m 24s\n- openstack-infra-extra-integration-xenial https://zuul.opendev.org/t/openstack/build/625cda1f15d442d5a8ea39e9321e1b80 : SUCCESS in 3m 36s\n- openstack-infra-extra-integration-bionic https://zuul.opendev.org/t/openstack/build/964f365709d5475484972993fca91ee1 : SUCCESS in 3m 28s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/c1162c6687c64ab8a566281e75732215 : SUCCESS in 5m 39s","accounts_in_message":[],"_revision_number":1},{"id":"5c92382f5fb9f0a5e37258d220173d9bf0a78bcf","author":{"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"},"date":"2019-08-21 08:36:22.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"81c48d7a8a98d4773b49b75eb576d7efdf2fde9a","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2019-08-21 16:44:24.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"528de021019199483d419a16d4a1651f9907245e","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2019-08-22 07:08:45.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"23b8babaf1860fdb6d6df98639362c84f1909dc2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-22 07:09:56.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d61cdc62649c6fe149ab9ac4287806d21a6cf996","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-22 07:21:18.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"ded806a54f9d67c7ed08a6b71a5e4c842d896d17","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-22 07:21:18.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/3bbde1d6b68a4559be96ea8480c4f122 : SUCCESS in 3m 58s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/e31e3e27d5a046adb61deb05a306387e : SUCCESS in 5m 12s","accounts_in_message":[],"_revision_number":1},{"id":"6beb8ee3ff249a17ebe8beedfc36b5e8d9387f50","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-22 07:23:42.000000000","message":"Patch Set 1:\n\nBuild failed (promote pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- promote-tox-docs-infra https://zuul.opendev.org/t/openstack/build/661e7ac8281b4d2fb6b3f03e95cbeb85 : FAILURE in 1m 01s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"68fd529761a4c01d6713f112f58fe9a561ac220d","revisions":{"68fd529761a4c01d6713f112f58fe9a561ac220d":{"kind":"REWORK","_number":1,"created":"2019-08-21 07:38:29.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/57/677657/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-zuul-jobs","ref":"refs/changes/57/677657/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/57/677657/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/57/677657/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/57/677657/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/57/677657/1"}}},"commit":{"parents":[{"commit":"af626666510fdbba1df488c234be7806912e9fa3","subject":"Revert \"Use promote for publish-tox-docs-infra\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/af626666510fdbba1df488c234be7806912e9fa3"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-08-21 07:36:20.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-08-21 07:36:20.000000000","tz":120},"subject":"Use promote job for infra","message":"Use promote job for infra\n\nFor using the promote jobs for publishing docs, change the docs job to\nopendev-tox-docs so that the tarball gets generated that is needed for\npublishing.\n\nUpdate  publish-tox-docs-infra to use the promote job, update\nrelease-tox-docs-infra to have the same jobs but no change of publish\njob (promote job does not work for release pipeline).\n\nChange-Id: Idc5e94a581359bcfd3a570e085aa13999c26eb86\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/68fd529761a4c01d6713f112f58fe9a561ac220d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/68fd529761a4c01d6713f112f58fe9a561ac220d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
