)]}'
{"id":"openstack%2Fopenstack-zuul-jobs~686366","triplet_id":"openstack%2Fopenstack-zuul-jobs~master~I46bf3185972d1d1605b6365dc16b9ebe3d4bebe5","project":"openstack/openstack-zuul-jobs","branch":"master","topic":"mypy-integration","hashtags":[],"change_id":"I46bf3185972d1d1605b6365dc16b9ebe3d4bebe5","subject":"Create openstack-tox-mypy job","status":"ABANDONED","created":"2019-10-03 09:46:01.000000000","updated":"2019-10-03 15:25:07.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"24bd44876f524481e897c12f14fb4045bd412c74","_number":686366,"virtual_id_number":686366,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2019-10-03 15:22:07.000000000","_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:check","value":1,"date":"2019-10-03 09:54:36.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":-1,"date":"2019-10-03 15:20:57.000000000","permitted_voting_range":{"min":-1,"max":0},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-03 09:54:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-10-03 15:22:07.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"}],"messages":[{"id":"49247b76b308e1a3dc93e22227d117c87ee5ea25","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-10-03 09:46:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c92bc1e34e08f5946b60b60dc51968ab2e8cc1e4","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-10-03 09:46:45.000000000","message":"Patch Set 1:\n\nI was planning to do this just for nova [1], but it seems other projects [2] might benefit from this so...\n\n[1] https://review.opendev.org/#/c/676210/","accounts_in_message":[],"_revision_number":1},{"id":"e37ae468ac78b695a75bc724ab552f624bc7beb3","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-10-03 09:46:53.000000000","message":"Patch Set 1:\n\n[2] https://review.opendev.org/#/c/686249/","accounts_in_message":[],"_revision_number":1},{"id":"9643f8d57a87fc267d0ac687d89d7be1a04da086","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-03 09:54:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/76e1b0791d714c3a8429b45336c175b0 : SUCCESS in 4m 06s\n- openstack-infra-extra-integration-xenial https://zuul.opendev.org/t/openstack/build/ee7c60bda18d49b0960cfa48e4063d95 : SUCCESS in 3m 37s\n- openstack-infra-extra-integration-bionic https://zuul.opendev.org/t/openstack/build/8db54f353a0f4f118b69a613c6b7a058 : SUCCESS in 3m 24s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/f415d93827dd4fe4ac4a8be68839072c : SUCCESS in 6m 29s","accounts_in_message":[],"_revision_number":1},{"id":"2a304824073eea16c0911ebc8295eb6a86d7159b","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-10-03 15:00:51.000000000","message":"Patch Set 1:\n\n(1 comment)\n\ndo you need a template as well?\n\nHave an example use in mind?","accounts_in_message":[],"_revision_number":1},{"id":"f4923b08b1f6ef6000302f2ab90eb99ab1e40b35","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-10-03 15:20:57.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"23ed4a446bfe3d7431f487bc556cf531642db2d9","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-10-03 15:22:07.000000000","message":"Patch Set 1:\n\nyes, doing it in pep8 sounds better - and avoids spinning up another VM...","accounts_in_message":[],"_revision_number":1},{"id":"533110d3d52d549f6cdbbf7f5d3e9bb6c1aad6fc","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-10-03 15:25:07.000000000","message":"Abandoned\n\nYeah, let\u0027s just do it in pep8","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"58a9a92ccbcf0c68ff38396e9239359dee87e71d","revisions":{"58a9a92ccbcf0c68ff38396e9239359dee87e71d":{"kind":"REWORK","_number":1,"created":"2019-10-03 09:46:01.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/66/686366/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-zuul-jobs","ref":"refs/changes/66/686366/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/66/686366/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/66/686366/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/66/686366/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/66/686366/1"}}},"commit":{"parents":[{"commit":"6e136c8abe70de66745bd387efd46d59f479e89e","subject":"Merge \"Bye, bye readthedocs-failing\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/6e136c8abe70de66745bd387efd46d59f479e89e"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2019-10-03 09:44:41.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2019-10-03 09:44:41.000000000","tz":60},"subject":"Create openstack-tox-mypy job","message":"Create openstack-tox-mypy job\n\nAdd a centralized job for running mypy. We assume it will be run on the\nlatest version of Python we currently support, Python 3.7.\n\nChange-Id: I46bf3185972d1d1605b6365dc16b9ebe3d4bebe5\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/58a9a92ccbcf0c68ff38396e9239359dee87e71d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/58a9a92ccbcf0c68ff38396e9239359dee87e71d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
