)]}'
{"id":"openstack%2Fopenstack-zuul-jobs~699679","triplet_id":"openstack%2Fopenstack-zuul-jobs~master~Ibc3aba9e5adffe7a3efab1f4b5bfe7ff48a179b8","project":"openstack/openstack-zuul-jobs","branch":"master","topic":"no-constraints","hashtags":[],"change_id":"Ibc3aba9e5adffe7a3efab1f4b5bfe7ff48a179b8","subject":"Add no-constraints job for py37","status":"MERGED","created":"2019-12-18 14:03:50.000000000","updated":"2020-01-02 13:31:38.000000000","submitted":"2020-01-02 13:28:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"699679-1577971735001-a23e576b","meta_rev_id":"8a92eedd596c7276b0c562f861baab99f954ad2a","_number":699679,"virtual_id_number":699679,"owner":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-02 13:28:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"all":[{"value":2,"date":"2020-01-02 13:07:41.000000000","_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":2,"date":"2020-01-02 13:11:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":1,"date":"2020-01-02 13:11:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-02 13:07:41.000000000","updated_by":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"reviewer":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"state":"REVIEWER"},{"updated":"2020-01-02 13:11:16.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2020-01-02 13:28:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b3d1ab95c35a75a4338d46d74863a8f71bdfa1f9","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-12-18 14:03:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"87a535601f5927426b2c1f6e09dc54e5799c88bc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-18 14:14:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/ea0e6d2e90e64301a831775ad072c5e8 : SUCCESS in 4m 16s\n- openstack-infra-extra-integration-xenial https://zuul.opendev.org/t/openstack/build/b81597e356db4d268b68509aac12a2f5 : SUCCESS in 2m 31s\n- openstack-infra-extra-integration-bionic https://zuul.opendev.org/t/openstack/build/5c2f8a1f08b7426d8807c07c54665e90 : SUCCESS in 2m 39s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/2a5e8f34b4bf4ae5a71144b9124ebb0e : SUCCESS in 5m 05s","accounts_in_message":[],"_revision_number":1},{"id":"0e3e50ba3e1c256dd6693b76e198b8533bc1ab4a","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-12-18 14:16:37.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"61c1caa539ae10a71fc1c96a8f47d84f19504e1d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-18 14:27:12.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/bf4a68ff090540f49766661f573297a3 : SUCCESS in 4m 42s\n- openstack-infra-extra-integration-xenial https://zuul.opendev.org/t/openstack/build/db7889351398487388910cc41d0fb13d : SUCCESS in 3m 39s\n- openstack-infra-extra-integration-bionic https://zuul.opendev.org/t/openstack/build/00a55eaeefc549b089c253a2feb36449 : SUCCESS in 4m 25s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/9c748baefbfd4f0292393758d9c9d366 : SUCCESS in 6m 50s","accounts_in_message":[],"_revision_number":2},{"id":"baf210c8bae621246c4034a9fb705f8cc191c198","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-12-18 14:37:37.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"83d721fa86131d9b3a0799cfaf2ec13139cc179c","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-01-02 13:05:50.000000000","message":"Patch Set 2: -Code-Review\n\nI wonder whether we should add pep8 and publish-openstack-python-branch-tarball to it so that this can replace the python2 template...","accounts_in_message":[],"_revision_number":2},{"id":"79fbacaa80f7ed0e84fbd3231b99fe095762d5ba","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2020-01-02 13:07:41.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"669d2e1d5c3019150745dd678ea6d9065f41515e","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-01-02 13:11:16.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nArgh, misremembered change - all fine.","accounts_in_message":[],"_revision_number":2},{"id":"c7954b9030298037e8812039f85797e45d1e226c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-02 13:11:32.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"86901856292a3f3378ad871288a0ced5433a8bdf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-02 13:28:54.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/c1bb2e4108ff49d5a1c6b27ab244cdf3 : SUCCESS in 3m 45s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/ccaeb09c6da44af3977c00855e5d6957 : SUCCESS in 6m 05s","accounts_in_message":[],"_revision_number":2},{"id":"e4b38b1896a454b8d43fc772ecf38b620d10a5e5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-02 13:28:55.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"f5c6371a2c08739666290c560cb7228828ac1166","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-02 13:31:38.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-tox-docs-infra https://zuul.opendev.org/t/openstack/build/bf4e6dff38c34d838a32fdfc8228fe3b : SUCCESS in 1m 21s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"f821782ad616886a535f7f5845e5cb8157f8c387","revisions":{"ddc8c61039bf7d676c3f545a09bdfe70064d50a9":{"kind":"REWORK","_number":1,"created":"2019-12-18 14:03:50.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/79/699679/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-zuul-jobs","ref":"refs/changes/79/699679/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/79/699679/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/79/699679/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/79/699679/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/79/699679/1"}}},"commit":{"parents":[{"commit":"b41cad775627dff675cf7cbd858fc5f3da03bc4d","subject":"Remove legacy-storyboard-js-integration jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/b41cad775627dff675cf7cbd858fc5f3da03bc4d"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-12-18 14:02:16.000000000","tz":-360},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-12-18 14:02:16.000000000","tz":-360},"subject":"Add no-constraints job for py37","message":"Add no-constraints job for py37\n\nSome repos, such as reviewstats, use no-constraints jobs since they are\nnot bound by the same constraints as official OpenStack projects.\n\nThis adds a variant using py37 to allow moving beyond the current max\nversion of py35.\n\nChange-Id: Ibc3aba9e5adffe7a3efab1f4b5bfe7ff48a179b8\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/ddc8c61039bf7d676c3f545a09bdfe70064d50a9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/ddc8c61039bf7d676c3f545a09bdfe70064d50a9"}]},"branch":"refs/heads/master"},"f821782ad616886a535f7f5845e5cb8157f8c387":{"kind":"REWORK","_number":2,"created":"2019-12-18 14:16:37.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/79/699679/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-zuul-jobs","ref":"refs/changes/79/699679/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/79/699679/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/79/699679/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/79/699679/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/79/699679/2"}}},"commit":{"parents":[{"commit":"b41cad775627dff675cf7cbd858fc5f3da03bc4d","subject":"Remove legacy-storyboard-js-integration jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/b41cad775627dff675cf7cbd858fc5f3da03bc4d"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-12-18 14:02:16.000000000","tz":-360},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-12-18 14:16:33.000000000","tz":-360},"subject":"Add no-constraints job for py37","message":"Add no-constraints job for py37\n\nSome repos, such as reviewstats, use no-constraints jobs since they are\nnot bound by the same constraints as official OpenStack projects.\n\nThis adds a variant using py37 to allow moving beyond the current max\nversion of py35.\n\nChange-Id: Ibc3aba9e5adffe7a3efab1f4b5bfe7ff48a179b8\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/f821782ad616886a535f7f5845e5cb8157f8c387"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/f821782ad616886a535f7f5845e5cb8157f8c387"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
