)]}'
{"id":"openstack%2Fopenstack-zuul-jobs~710112","triplet_id":"openstack%2Fopenstack-zuul-jobs~master~I6998d80228fd16911b660aa04f40e710bae1df99","project":"openstack/openstack-zuul-jobs","branch":"master","topic":"opendev-puppet-forge","hashtags":[],"change_id":"I6998d80228fd16911b660aa04f40e710bae1df99","subject":"Remove puppet-forge jobs","status":"MERGED","created":"2020-02-26 18:47:34.000000000","updated":"2020-02-27 10:25:29.000000000","submitted":"2020-02-27 10:22:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"710112-1582798920953-b95df229","meta_rev_id":"5d119d1ca6bc36dd37e849d5dfda2100fb26628f","_number":710112,"virtual_id_number":710112,"owner":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-02-26 19:48:16.000000000","_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-27 10:22:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":2,"date":"2020-02-27 08:46:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-02-27 10:07:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-02-27 10:07:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-27 08:46:50.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2020-02-27 10:07:13.000000000","updated_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"reviewer":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"state":"REVIEWER"},{"updated":"2020-02-27 10:22:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7b478eda1a0fae0d3cea39ea40ebcdd580ee59fe","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-02-26 18:47:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b26107da1b307f237934711faedb741d1623ad58","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-26 18:48:06.000000000","message":"Patch Set 1: Verified-1\n\nThis change depends on a change to a config project.\n\nThe syntax of the configuration in this change has been verified to be\ncorrect once the config project change upon which it depends is\nmerged, but it can not be used until that occurs.","accounts_in_message":[],"_revision_number":1},{"id":"0014daf7f0946651304232a573bd6405ed3ad0fc","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-02-26 19:48:16.000000000","message":"Patch Set 1:\n\nrecheck - dependency merged","accounts_in_message":[],"_revision_number":1},{"id":"2fa2f552c427729b83394a7b18a2f7a6d8eae578","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-26 20:01:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/ed9b737f4063486f8407b3402b140f1f : SUCCESS in 5m 19s\n- openstack-infra-extra-integration-xenial https://zuul.opendev.org/t/openstack/build/f17499dc0e1b43e6bc45b97211030893 : SUCCESS in 5m 12s\n- openstack-infra-extra-integration-bionic https://zuul.opendev.org/t/openstack/build/e65a783be3fe43159167ef38834c9307 : SUCCESS in 3m 20s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/67c80b018ee345deaec489a170177f7b : SUCCESS in 6m 30s","accounts_in_message":[],"_revision_number":1},{"id":"aaf5f4ec6eab7c0024deb8510cd794cd4f716c4a","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2020-02-27 08:46:50.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"1453f76729c46f93608e36f5454681bd1a704f08","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2020-02-27 10:07:13.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d7a6348edc3db871bb53a4419c3338c49393d8b1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-27 10:07:43.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7ac0d4322e02c7b33064d72b5ba866e339588747","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-27 10:22:00.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/a710a9b9815c4280b3f20951d5e708c1 : SUCCESS in 5m 14s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/be5181887c294d3599afb367e77dd79c : SUCCESS in 7m 17s","accounts_in_message":[],"_revision_number":1},{"id":"a8afd4d3561fd6459b34057c43292c28d1b9de06","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-27 10:22:00.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"637d66c2fe2367dc8d3b8cd88c07886f2737c8cb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-27 10:25:29.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-tox-docs-infra https://zuul.opendev.org/t/openstack/build/379c00b7cb2f4ff6b3d8ca12ff8d1f2f : SUCCESS in 1m 18s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b9842e3bc6dc14455163fa427e4f37b437231165","revisions":{"b9842e3bc6dc14455163fa427e4f37b437231165":{"kind":"REWORK","_number":1,"created":"2020-02-26 18:47:34.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/12/710112/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-zuul-jobs","ref":"refs/changes/12/710112/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/12/710112/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/12/710112/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/12/710112/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/12/710112/1"}}},"commit":{"parents":[{"commit":"8f1859f4a32e7786c3948e4f435dbf993919d57f","subject":"Use python3 in release-notes-jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/8f1859f4a32e7786c3948e4f435dbf993919d57f"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-02-26 18:45:59.000000000","tz":60},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-02-26 18:47:15.000000000","tz":60},"subject":"Remove puppet-forge jobs","message":"Remove puppet-forge jobs\n\nRemove the now unused jobs legacy-puppet-httpd-forge-upload and\nlegacy-puppet-storyboard-forge-upload.\n\nAlso, remove parent legacy-puppet-openstack-integration, it\u0027s not used\nanymore.\n\nDepends-On: https://review.opendev.org/710110\nChange-Id: I6998d80228fd16911b660aa04f40e710bae1df99\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/b9842e3bc6dc14455163fa427e4f37b437231165"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/b9842e3bc6dc14455163fa427e4f37b437231165"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
