)]}'
{"id":"openstack%2Fopenstack-zuul-jobs~714728","triplet_id":"openstack%2Fopenstack-zuul-jobs~master~Ic17ec65b2229d46b799e0c41824e6e368723098b","project":"openstack/openstack-zuul-jobs","branch":"master","topic":"drop-py27-support","hashtags":[],"change_id":"Ic17ec65b2229d46b799e0c41824e6e368723098b","subject":"Update python-charm-jobs template description","status":"MERGED","created":"2020-03-24 17:42:28.000000000","updated":"2020-04-02 17:42:36.000000000","submitted":"2020-04-02 17:31:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"714728-1585848670433-9ed293ab","meta_rev_id":"9b6f58ebdefc51b401b38bd760e9a4638f867cde","_number":714728,"virtual_id_number":714728,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"date":"2020-04-01 21:44:22.000000000","_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-02 17:31:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"recommended":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"all":[{"value":1,"date":"2020-04-02 08:33:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},{"value":2,"date":"2020-04-02 15:41:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":2,"date":"2020-04-02 04:56:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},{"value":1,"date":"2020-04-02 12:30:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":0,"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},{"value":1,"date":"2020-04-02 15:41:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-24 19:54:44.000000000","updated_by":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"reviewer":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"state":"REVIEWER"},{"updated":"2020-03-24 23:57:30.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-04-02 04:56:29.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2020-04-02 08:33:32.000000000","updated_by":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"reviewer":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"state":"REVIEWER"},{"updated":"2020-04-02 12:30:23.000000000","updated_by":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"reviewer":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"state":"REVIEWER"},{"updated":"2020-04-02 15:41:26.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2020-04-02 17:31:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"18a50b1f357bd269f1b95216a88f0c8875111762","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-03-24 17:42:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4471e6f023569c082ec7e499af2cad0c0badc68f","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-03-24 17:42:41.000000000","message":"Topic set to drop-py27-support","accounts_in_message":[],"_revision_number":1},{"id":"b40a954299f35fa49b288bab78a94df80b7a32aa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-24 17:55:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/37258620df414b47b90065bf6aa57174 : SUCCESS in 3m 17s\n- openstack-infra-extra-integration-xenial https://zuul.opendev.org/t/openstack/build/6fb167c416fc4724a6bc70666068009f : SUCCESS in 4m 25s\n- openstack-infra-extra-integration-bionic https://zuul.opendev.org/t/openstack/build/8a6cd29b80364590b52c7bc8c89e7bf8 : SUCCESS in 2m 41s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/86fed196330a4cd19aa749224479f7b3 : SUCCESS in 5m 15s","accounts_in_message":[],"_revision_number":1},{"id":"9c86d082d872746d529129b7b276b04f8a4ee9ed","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-03-24 19:57:20.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nI don\u0027t think this is right, there\u0027re other python35 jobs.\n\nI agree that action is needed.","accounts_in_message":[],"_revision_number":1},{"id":"0d7c0105d2bb0370cd3c0f635aa14afba7fa5c5b","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-03-24 23:24:51.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0920d808c4885cb751b09b93dc5f1fa342181e78","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-03-24 23:57:12.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"33b474a7b62a55cba9a6802653a720abb62c51de","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-24 23:57:23.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n\nWarning:\n  Dependency cycle detected","accounts_in_message":[],"_revision_number":2},{"id":"bb4a867c03a6c3897dadcd4239d2c8ab82f87085","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-24 23:57:23.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n\nWarning:\n  Dependency cycle detected","accounts_in_message":[],"_revision_number":2},{"id":"105b9e66cbc3494abd6b92850fee3287f5421a80","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-24 23:57:24.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).\n\n\nWarning:\n  Dependency cycle detected","accounts_in_message":[],"_revision_number":2},{"id":"99df15e79f5c4fc66ad6ab192358e7ac2c33edb6","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-03-24 23:57:29.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n\nWarning:\n  Dependency cycle detected","accounts_in_message":[],"_revision_number":2},{"id":"da1f20dd18bcd0a6303689ab8df98b78b7e03ed3","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-03-24 23:57:30.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n\nWarning:\n  Dependency cycle detected","accounts_in_message":[],"_revision_number":2},{"id":"312753b6a98bc8194454afd0cacfc34f1db8e1d6","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-03-25 06:15:28.000000000","message":"Patch Set 2:\n\nI doubt this will succeed since there are stable branches that use the job. So, we can retire it in a couple of years...\n\nYou might want to update the text instead to say \"Should be used for older stable releases only\"","accounts_in_message":[],"_revision_number":2},{"id":"3bc3c94f8dae4408e27d2ff668661d00ee58edd4","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-03-25 06:15:38.000000000","message":"Patch Set 2:\n\nrecheck - cycles broken","accounts_in_message":[],"_revision_number":2},{"id":"27ee627428135eed9cdcc89d66a63294753ef847","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-25 06:17:16.000000000","message":"Patch Set 2:\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/charm-ceph-fs on branch stable/18.08.  The error was:\n\n  The project template \"python-charm-jobs\" was not found.\n\nThe error appears in the following project stanza:\n\n  project:\n      templates:\n        - python-charm-jobs\n        - openstack-python35-jobs\n\n  in \"openstack/charm-ceph-fs/.zuul.yaml@stable/18.08\", line 1, column 3","accounts_in_message":[],"_revision_number":2},{"id":"2f489faf39d88deb7d0ad33be08ac5eaeabeab1f","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-03-27 03:11:41.000000000","message":"Patch Set 2:\n\n\u003e I doubt this will succeed since there are stable branches that use\n \u003e the job. So, we can retire it in a couple of years...\n \u003e \n \u003e You might want to update the text instead to say \"Should be used\n \u003e for older stable releases only\"\n\nyeah, I thought all charm repos are brnahcless. updating.","accounts_in_message":[],"_revision_number":2},{"id":"c4d3657f1dfc035df1c842d04996633339152ad4","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-03-27 03:16:52.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"35ea67e4c6c8388ffd8dbd8c817fdcbd3108a472","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-27 03:20:31.000000000","message":"Patch Set 3: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":3},{"id":"ad1e75c81b6622f35a62d656115b61a83722231f","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-03-27 08:44:38.000000000","message":"Patch Set 3:\n\nYou could remove the depends-on, this can merge now on its own.","accounts_in_message":[],"_revision_number":3},{"id":"28ac6cb4145a0a81ccb814dd30f993cc5d1f3d2e","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-04-01 21:44:13.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"b6c398c2bf72e0400bf9b5dd05f0109120b5b1b4","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-04-01 21:44:22.000000000","message":"Patch Set 4:\n\n\u003e You could remove the depends-on, this can merge now on its own.\n\ndone","accounts_in_message":[],"_revision_number":4},{"id":"09ad06d4979806a585225ae6b33ca0784d9d1f9a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-01 22:00:22.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/0aaac890dd9a4378bb519b956093eb09 : SUCCESS in 3m 20s\n- openstack-infra-extra-integration-xenial https://zuul.opendev.org/t/openstack/build/662baad460064396814559a2166bc9f7 : SUCCESS in 2m 33s\n- openstack-infra-extra-integration-bionic https://zuul.opendev.org/t/openstack/build/9d3b425f34254901871fedf3e44d98b3 : SUCCESS in 3m 36s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/62b0c278ff9a4fa9919e5be627ceb50a : SUCCESS in 4m 40s","accounts_in_message":[],"_revision_number":4},{"id":"3424cc40e9036814ad375a931d83b7e09ce0716c","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-04-02 04:56:29.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"fe94e926e148a274bcc6f2588320065e8833fc90","author":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"date":"2020-04-02 08:33:32.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"c9dc41122f51394164231b5c591ff1f118f60c48","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2020-04-02 12:30:23.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"8ef7bf2c87e9414500329b5336256c565e0705e6","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2020-04-02 15:41:26.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"ffea1eaeb12e0654bea3d1a526d37e0ab793fbb4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-02 16:20:27.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"ee884a9f5c72da139baedc9f1ee4eed8bcb027eb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-02 17:31:10.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/62931e9674aa499194eae218fb5463b1 : SUCCESS in 4m 28s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/5d5a64b349e7498bb74d8094076f385f : SUCCESS in 6m 55s","accounts_in_message":[],"_revision_number":4},{"id":"1b59c2d4f787356c00f504d91f5549d5ed869562","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-02 17:31:10.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"9b6f58ebdefc51b401b38bd760e9a4638f867cde","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-02 17:42:36.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-tox-docs-infra https://zuul.opendev.org/t/openstack/build/a3b37efe47c641a58371674a6c567e1c : SUCCESS in 2m 23s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"200121f9a71fcda8ae09a1be80ee2b3dac623508","revisions":{"01895c64d3d66d118939baaa9f1ce53ecf999a81":{"kind":"REWORK","_number":1,"created":"2020-03-24 17:42:28.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/28/714728/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-zuul-jobs","ref":"refs/changes/28/714728/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/28/714728/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/28/714728/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/28/714728/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/28/714728/1"}}},"commit":{"parents":[{"commit":"b66c46434f4190ccc483b8f4d1357239386d9eb2","subject":"Don\u0027t run infra-puppet-apply on docker and ansible","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/b66c46434f4190ccc483b8f4d1357239386d9eb2"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-03-24 17:41:03.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-03-24 17:42:25.000000000","tz":-300},"subject":"Move python-charm-jobs template to py3","message":"Move python-charm-jobs template to py3\n\npython-charm-jobs template use py2.7 tox job which\nrun on various charm repos. Moving that to py3.7\n\nChange-Id: Ic17ec65b2229d46b799e0c41824e6e368723098b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/01895c64d3d66d118939baaa9f1ce53ecf999a81"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/01895c64d3d66d118939baaa9f1ce53ecf999a81"}]},"branch":"refs/heads/master"},"9aeda9f9178329775f91b91fa0ae5fbd2cfe9fc4":{"kind":"REWORK","_number":2,"created":"2020-03-24 23:57:12.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/28/714728/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-zuul-jobs","ref":"refs/changes/28/714728/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/28/714728/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/28/714728/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/28/714728/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/28/714728/2"}}},"commit":{"parents":[{"commit":"b66c46434f4190ccc483b8f4d1357239386d9eb2","subject":"Don\u0027t run infra-puppet-apply on docker and ansible","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/b66c46434f4190ccc483b8f4d1357239386d9eb2"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-03-24 17:41:03.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-03-24 23:57:08.000000000","tz":0},"subject":"Remove python-charm-jobs py2 template","message":"Remove python-charm-jobs py2 template\n\npython-charm-jobs template is for py2.7 testing which\nwas used by various charm repos. They have been changed to\nrun py3 jobs template to test pep8 and unit tests.\n\nDepends-On: https://review.opendev.org/#/c/710578/\nDepends-On: https://review.opendev.org/#/c/710374/\nDepends-On: https://review.opendev.org/#/c/710377/\nDepends-On: https://review.opendev.org/#/c/709386/\nDepends-On: https://review.opendev.org/#/c/709385/\nDepends-On: https://review.opendev.org/#/c/714796/\n\nChange-Id: Ic17ec65b2229d46b799e0c41824e6e368723098b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/9aeda9f9178329775f91b91fa0ae5fbd2cfe9fc4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/9aeda9f9178329775f91b91fa0ae5fbd2cfe9fc4"}]},"branch":"refs/heads/master"},"42c3585fb86c55a1d1b53df591444276c8e2c1b4":{"kind":"REWORK","_number":3,"created":"2020-03-27 03:16:52.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/28/714728/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-zuul-jobs","ref":"refs/changes/28/714728/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/28/714728/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/28/714728/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/28/714728/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/28/714728/3"}}},"commit":{"parents":[{"commit":"b66c46434f4190ccc483b8f4d1357239386d9eb2","subject":"Don\u0027t run infra-puppet-apply on docker and ansible","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/b66c46434f4190ccc483b8f4d1357239386d9eb2"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-03-24 17:41:03.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-03-27 03:16:48.000000000","tz":0},"subject":"Update python-charm-jobs template description","message":"Update python-charm-jobs template description\n\npython-charm-jobs template is for py2.7 testing which\nwas used by various charm repos and they stable branch.\n\nThose repo moved to py3 template but stable branch still\nneed to keep using the py2 template so let\u0027s update the\ndescription of this template.\n\nDepends-On: https://review.opendev.org/#/c/710578/\nDepends-On: https://review.opendev.org/#/c/710374/\nDepends-On: https://review.opendev.org/#/c/710377/\nDepends-On: https://review.opendev.org/#/c/709386/\nDepends-On: https://review.opendev.org/#/c/709385/\nDepends-On: https://review.opendev.org/#/c/714796/\n\nChange-Id: Ic17ec65b2229d46b799e0c41824e6e368723098b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/42c3585fb86c55a1d1b53df591444276c8e2c1b4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/42c3585fb86c55a1d1b53df591444276c8e2c1b4"}]},"branch":"refs/heads/master"},"200121f9a71fcda8ae09a1be80ee2b3dac623508":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2020-04-01 21:44:13.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/28/714728/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-zuul-jobs","ref":"refs/changes/28/714728/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/28/714728/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/28/714728/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/28/714728/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/28/714728/4"}}},"commit":{"parents":[{"commit":"b66c46434f4190ccc483b8f4d1357239386d9eb2","subject":"Don\u0027t run infra-puppet-apply on docker and ansible","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/b66c46434f4190ccc483b8f4d1357239386d9eb2"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-03-24 17:41:03.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-04-01 21:44:10.000000000","tz":0},"subject":"Update python-charm-jobs template description","message":"Update python-charm-jobs template description\n\npython-charm-jobs template is for py2.7 testing which\nwas used by various charm repos and they stable branch.\n\nThose repo moved to py3 template but stable branch still\nneed to keep using the py2 template so let\u0027s update the\ndescription of this template.\n\nChange-Id: Ic17ec65b2229d46b799e0c41824e6e368723098b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/200121f9a71fcda8ae09a1be80ee2b3dac623508"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/200121f9a71fcda8ae09a1be80ee2b3dac623508"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
