)]}'
{"id":"openstack%2Fopenstack-zuul-jobs~715928","triplet_id":"openstack%2Fopenstack-zuul-jobs~master~I058c89ef18f98c54ab3f1431c0dc522688253813","project":"openstack/openstack-zuul-jobs","branch":"master","topic":"pep8-inline-comments","hashtags":[],"change_id":"I058c89ef18f98c54ab3f1431c0dc522688253813","subject":"DNM: test inline pep8 (should not fail)","status":"ABANDONED","created":"2020-03-30 12:14:04.000000000","updated":"2020-04-01 15:18:41.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"966a0d095c0c27d8712cce5b6ddda59213093da7","_number":715928,"virtual_id_number":715928,"owner":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2020-03-30 12:26:41.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-30 12:26:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5b8e44381afc66682f6f6e4e35ae8128386aa197","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2020-03-30 12:14:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"370f49b27ddb2649ce1e8d6d65ad553af73bcd4c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-30 12:26:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/50176bf8630e4e98b77efc6b9e0c9485 : SUCCESS in 3m 26s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/4ca416e1469745f594be3f5363b74c6e : SUCCESS in 5m 03s","accounts_in_message":[],"_revision_number":1},{"id":"f71e5c86952a848c9d49c44abb00f119390bfe57","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2020-03-30 12:33:29.000000000","message":"Topic set to pep8-inline-comments","accounts_in_message":[],"_revision_number":1},{"id":"966a0d095c0c27d8712cce5b6ddda59213093da7","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2020-04-01 15:18:41.000000000","message":"Abandoned\n\nnot needed anymore","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3683c915724f82535bde25b052aa1d255324a463","revisions":{"3683c915724f82535bde25b052aa1d255324a463":{"kind":"REWORK","_number":1,"created":"2020-03-30 12:14:04.000000000","uploader":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"ref":"refs/changes/28/715928/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-zuul-jobs","ref":"refs/changes/28/715928/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/28/715928/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/28/715928/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/28/715928/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/28/715928/1"}}},"commit":{"parents":[{"commit":"0a7e479381b8e7ee5b1cdc7e18062ca99ab34087","subject":"Removed unused infra spec helper job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/0a7e479381b8e7ee5b1cdc7e18062ca99ab34087"}]}],"author":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2020-03-30 12:13:36.000000000","tz":-240},"committer":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2020-03-30 12:14:02.000000000","tz":-240},"subject":"DNM: test inline pep8 (should not fail)","message":"DNM: test inline pep8 (should not fail)\n\nDepends-On: https://review.opendev.org/715727\nChange-Id: I058c89ef18f98c54ab3f1431c0dc522688253813\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/3683c915724f82535bde25b052aa1d255324a463"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/3683c915724f82535bde25b052aa1d255324a463"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
