)]}'
{"id":"openstack%2Fopenstack-zuul-jobs~727976","triplet_id":"openstack%2Fopenstack-zuul-jobs~master~Ie51711e05e323aebaab67f766eb7189fb16fd322","project":"openstack/openstack-zuul-jobs","branch":"master","topic":"nodejs4-template","hashtags":[],"change_id":"Ie51711e05e323aebaab67f766eb7189fb16fd322","subject":"Use new npm jobs in nodejs[468]-jobs templates","status":"MERGED","created":"2020-05-14 08:56:05.000000000","updated":"2020-05-15 16:37:31.000000000","submitted":"2020-05-15 16:30:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"727976-1589560249170-6f122f09","meta_rev_id":"55e726941896bfa8feff63d25bbf30a3d68554b2","_number":727976,"virtual_id_number":727976,"owner":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"date":"2020-05-14 09:44:08.000000000","_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-15 16:30:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"all":[{"value":2,"date":"2020-05-15 16:23:58.000000000","_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":2,"date":"2020-05-14 08:57:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"all":[{"value":1,"date":"2020-05-15 16:23:58.000000000","_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-14 08:57:15.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2020-05-15 16:23:58.000000000","updated_by":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"reviewer":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"state":"REVIEWER"},{"updated":"2020-05-15 16:30:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7e960f86c6bb2b9ea1e02809e5d53a1edc553bd0","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2020-05-14 08:56:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4de40412f33016d1a0fe60311eb6fdaca934c2ca","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-05-14 08:57:15.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"a14953168329f148830a023d1e4dbdae09499766","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2020-05-14 09:00:49.000000000","message":"Patch Set 1:\n\nnodejs4-jobs template is being verified via https://review.opendev.org/#/c/726712/ patch set 2 in horizon stable/rocky.","accounts_in_message":[],"_revision_number":1},{"id":"2a7e98476d0b86929866e239fec150617777bf9c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-14 09:10:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/64646b2a43e34b38ac309fb49325ea71 : SUCCESS in 4m 40s\n- openstack-infra-extra-integration-xenial https://zuul.opendev.org/t/openstack/build/4d51f4c5e4074e90b0b66e6c3dfea273 : SUCCESS in 4m 16s\n- openstack-infra-extra-integration-bionic https://zuul.opendev.org/t/openstack/build/06c35e0fdac44946a9e3ded72fd54b7e : SUCCESS in 2m 30s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/40065b85945d4ec69d469be39c88b6e0 : SUCCESS in 8m 11s","accounts_in_message":[],"_revision_number":1},{"id":"a909ccbfe60f436dd6f35ce016c13c8598bd3912","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2020-05-14 09:44:08.000000000","message":"Patch Set 1:\n\nnodejs-run-lint and nodejs-run-test-browser both succeeded in https://review.opendev.org/#/c/726712/","accounts_in_message":[],"_revision_number":1},{"id":"87d55161e2d98a06babf495ed06f6f30bbbc9561","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2020-05-15 16:23:58.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"eeaf14e440808bb8743ce18a6781c3df1273de7c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-15 16:24:15.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f50cc0ca2e00b2b484b8edd9616a9f8a5f8ef1d1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-15 16:30:48.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/0fda6c738fbf4c419708d3c992b1986a : SUCCESS in 4m 00s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/2b115f1719f8409caf1faf9953a32353 : SUCCESS in 5m 27s","accounts_in_message":[],"_revision_number":1},{"id":"f6bb9ae2c73031e9d5d368f14baef2af08a107ca","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-15 16:30:49.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"b1b88db3b89a167817c57f4ce3ba81e7ac6a715a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-15 16:37:31.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-tox-docs-infra https://zuul.opendev.org/t/openstack/build/fc16f44e5e134d40b8c01770892a77af : SUCCESS in 3m 15s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2238b2dc616e736fd1b35a45e19e7208f0f53359","revisions":{"2238b2dc616e736fd1b35a45e19e7208f0f53359":{"kind":"REWORK","_number":1,"created":"2020-05-14 08:56:05.000000000","uploader":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"ref":"refs/changes/76/727976/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-zuul-jobs","ref":"refs/changes/76/727976/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/76/727976/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/76/727976/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/76/727976/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/76/727976/1"}}},"commit":{"parents":[{"commit":"f7f8fbb03296bc8e3ecc618382fa99a6b37d20f2","subject":"Also trigger Github mirroring on tags","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/f7f8fbb03296bc8e3ecc618382fa99a6b37d20f2"}]}],"author":{"name":"Akihiro Motoki","email":"amotoki@gmail.com","date":"2020-05-14 08:45:21.000000000","tz":540},"committer":{"name":"Akihiro Motoki","email":"amotoki@gmail.com","date":"2020-05-14 08:55:51.000000000","tz":540},"subject":"Use new npm jobs in nodejs[468]-jobs templates","message":"Use new npm jobs in nodejs[468]-jobs templates\n\ncommit 202cce830e4e0d63684fefc74dfc8d6aaa9248c6 introduced\nnew npm jobs. npm jobs on ubuntu-xenial do not work now\nas yarn does not work with nodejs \u003c10, so I believe this change\nhas no negative impact. It is required to fix horizon jobs\nin stable/rocky or older.\n\nChange-Id: Ie51711e05e323aebaab67f766eb7189fb16fd322\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/2238b2dc616e736fd1b35a45e19e7208f0f53359"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/2238b2dc616e736fd1b35a45e19e7208f0f53359"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
