)]}'
{"id":"openstack%2Fopenstack-zuul-jobs~731634","triplet_id":"openstack%2Fopenstack-zuul-jobs~master~Ica59634eec533db20fe8b1dd148ee9b4d041fd3c","project":"openstack/openstack-zuul-jobs","branch":"master","topic":"wheel-gate","hashtags":[],"change_id":"Ica59634eec533db20fe8b1dd148ee9b4d041fd3c","subject":"Allow other projects to run wheel cache jobs","status":"MERGED","created":"2020-05-29 00:56:12.000000000","updated":"2020-05-29 18:34:04.000000000","submitted":"2020-05-29 18:25:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"731634-1590776749371-06556da8","meta_rev_id":"991bb73831a63490a2b3aae95eb5c5716d0040fc","_number":731634,"virtual_id_number":731634,"owner":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-29 18:25:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2020-05-29 17:03:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":2,"date":"2020-05-29 05:17:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":1,"date":"2020-05-29 17:03:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":1,"date":"2020-05-29 18:19:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-29 17:03:48.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2020-05-29 18:19:35.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2020-05-29 18:25:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3e6b8e7e0fef2ba0ee5d828aec3d24c3968443cd","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2020-05-29 00:56:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6d92fdb64a000035e4f7b6fbdb758ce01f18604c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-29 01:04:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/55387a39d44c47f582343debbf2424f3 : SUCCESS in 3m 31s\n- openstack-infra-extra-integration-xenial https://zuul.opendev.org/t/openstack/build/ce98a64d37a84fe986988d913af37490 : SUCCESS in 3m 38s\n- openstack-infra-extra-integration-bionic https://zuul.opendev.org/t/openstack/build/dee6dc036cd14dde9d4ec2604c778517 : SUCCESS in 2m 44s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/6aa43516fb7e4e47a52db64f083bf71f : SUCCESS in 4m 47s","accounts_in_message":[],"_revision_number":1},{"id":"cdc52d0374adf4e229254bc0eba9e10a66e694a5","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2020-05-29 01:14:36.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"86c1f304f6b64fd02f6f942795c21423f29b6f47","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2020-05-29 01:22:50.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"7fe1c18f054697123f94f3dc841c4eb2717baec7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-29 01:31:21.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/0e806ac7e36d47eaa40ce11c7245f127 : SUCCESS in 3m 01s\n- openstack-infra-extra-integration-xenial https://zuul.opendev.org/t/openstack/build/37b595087b8c40f4841b8c42d8ee1323 : SUCCESS in 3m 16s\n- openstack-infra-extra-integration-bionic https://zuul.opendev.org/t/openstack/build/77830f8c9cf641e99b51a958825bee3b : SUCCESS in 3m 22s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/8e5a026c51444819a5db751069d3bd16 : SUCCESS in 5m 53s","accounts_in_message":[],"_revision_number":3},{"id":"23cebc71328b9431cda46bb9e5f5db85ccca37e8","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-05-29 05:17:20.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"1b82025df4ae0aafdeb9de250b1531f5ab033edf","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2020-05-29 17:03:48.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"bcf8ca69a3f9323cc72f5008c99d17dbdcf822f1","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-05-29 18:19:35.000000000","message":"Patch Set 3: Workflow+1\n\ndependency merged","accounts_in_message":[],"_revision_number":3},{"id":"5bb02294ac3bd818c8720b214d4d735c86468ea8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-29 18:19:50.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"54bf29e066ddcd95defea713678db217f576457c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-29 18:25:49.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"f2dd099696fa102452641d0a090a8d0681f9cf30","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-29 18:25:49.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/48d37a47a406493aa8a0b2d27112ab70 : SUCCESS in 3m 27s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/33d60370640d4ee886390285705598da : SUCCESS in 5m 40s","accounts_in_message":[],"_revision_number":3},{"id":"56db699e667eafc3d02fe77560d349eabfd3e907","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-29 18:34:04.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-tox-docs-infra https://zuul.opendev.org/t/openstack/build/71b9f0b09614457e93f33708b82f0a89 : SUCCESS in 1m 14s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"8f9feb21a8b101cea1f4f6f01acb729bd70310d5","revisions":{"63add496cd1e03f4872b105bac6ed14f9a6aa47f":{"kind":"REWORK","_number":1,"created":"2020-05-29 00:56:12.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/34/731634/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-zuul-jobs","ref":"refs/changes/34/731634/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/34/731634/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/34/731634/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/34/731634/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/34/731634/1"}}},"commit":{"parents":[{"commit":"0913b3ed7ed4a45fec4e6d70b86869d2b918d03c","subject":"build-wheel-cache: update README to be more accurate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/0913b3ed7ed4a45fec4e6d70b86869d2b918d03c"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2020-05-29 00:47:10.000000000","tz":600},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2020-05-29 00:54:10.000000000","tz":600},"subject":"Allow other projects to run wheel cache jobs","message":"Allow other projects to run wheel cache jobs\n\nRemove the job restriction so we can run these jobs in this gate when\nwe update the roles/playbooks.  The restriction is moved to the\npublish job with the dependent change.\n\nDepends-On: https://review.opendev.org/731632\n\nChange-Id: Ica59634eec533db20fe8b1dd148ee9b4d041fd3c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/63add496cd1e03f4872b105bac6ed14f9a6aa47f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/63add496cd1e03f4872b105bac6ed14f9a6aa47f"}]},"branch":"refs/heads/master"},"1dfa0efe02bbe8644711a18e1cc8b5acd350cabd":{"kind":"REWORK","_number":2,"created":"2020-05-29 01:14:36.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/34/731634/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-zuul-jobs","ref":"refs/changes/34/731634/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/34/731634/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/34/731634/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/34/731634/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/34/731634/2"}}},"commit":{"parents":[{"commit":"0913b3ed7ed4a45fec4e6d70b86869d2b918d03c","subject":"build-wheel-cache: update README to be more accurate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/0913b3ed7ed4a45fec4e6d70b86869d2b918d03c"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2020-05-29 00:47:10.000000000","tz":600},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2020-05-29 01:12:37.000000000","tz":600},"subject":"Allow other projects to run wheel cache jobs","message":"Allow other projects to run wheel cache jobs\n\nRemove the job restriction so we can run these jobs in this gate when\nwe update the roles/playbooks.  The restriction is moved to the\npublish job with the dependent change.\n\nAdd dependency on requirements so it is available outside that\nrepository.\n\nDepends-On: https://review.opendev.org/731632\n\nChange-Id: Ica59634eec533db20fe8b1dd148ee9b4d041fd3c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/1dfa0efe02bbe8644711a18e1cc8b5acd350cabd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/1dfa0efe02bbe8644711a18e1cc8b5acd350cabd"}]},"branch":"refs/heads/master"},"8f9feb21a8b101cea1f4f6f01acb729bd70310d5":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2020-05-29 01:22:50.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/34/731634/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-zuul-jobs","ref":"refs/changes/34/731634/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/34/731634/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/34/731634/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/34/731634/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/34/731634/3"}}},"commit":{"parents":[{"commit":"0913b3ed7ed4a45fec4e6d70b86869d2b918d03c","subject":"build-wheel-cache: update README to be more accurate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/0913b3ed7ed4a45fec4e6d70b86869d2b918d03c"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2020-05-29 00:47:10.000000000","tz":600},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2020-05-29 01:21:38.000000000","tz":600},"subject":"Allow other projects to run wheel cache jobs","message":"Allow other projects to run wheel cache jobs\n\nRemove the job restriction so we can run these jobs in this gate when\nwe update the roles/playbooks.  The restriction is moved to the\npublish job with the dependent change.\n\nAdd dependency on requirements so it is available outside that\nrepository.\n\nThe system-config roles aren\u0027t required.  I think this was from when\nwe used to directly copy to afs from the build nodes and required the\nopenafs-client install roles.  Now the executor copies the results via\nthe publish jobs.\n\nDepends-On: https://review.opendev.org/731632\n\nChange-Id: Ica59634eec533db20fe8b1dd148ee9b4d041fd3c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/8f9feb21a8b101cea1f4f6f01acb729bd70310d5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/8f9feb21a8b101cea1f4f6f01acb729bd70310d5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
