)]}'
{"id":"openstack%2Fopenstack-zuul-jobs~733748","triplet_id":"openstack%2Fopenstack-zuul-jobs~master~I64a9307a4ffc6b03ff9f15443fffa54b6ee2140f","project":"openstack/openstack-zuul-jobs","branch":"master","topic":"openafs-tar-download","hashtags":[],"change_id":"I64a9307a4ffc6b03ff9f15443fffa54b6ee2140f","subject":"openafs-rpm-package-build: make artifact directories","status":"MERGED","created":"2020-06-05 06:00:10.000000000","updated":"2020-06-05 06:46:09.000000000","submitted":"2020-06-05 06:42:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"733748-1591339368352-5027401d","meta_rev_id":"96fc02d1564d878e247a1ede0657268ff210897a","_number":733748,"virtual_id_number":733748,"owner":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-05 06:42:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-06-05 06:23:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-06-05 06:23:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-05 06:42:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7ef640068072b9a2b1dfdd966bcb93ec47f17d82","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2020-06-05 06:00:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ba06dff771ce6e79d81989cc1d3ad62e5d332a19","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-05 06:22:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/754b66297e0649c680a07f8988bbf080 : SUCCESS in 5m 24s\n- openafs-rpm-package-build-centos-7-x86 https://zuul.opendev.org/t/openstack/build/aa7d85951e8048d99611008b0042df9c : SUCCESS in 11m 07s\n- openafs-rpm-package-build-centos-8-x86 https://zuul.opendev.org/t/openstack/build/567f7e427c1a4bb1b6ebbcd016b7be15 : SUCCESS in 14m 37s\n- openafs-rpm-package-build-centos-8-arm64 https://zuul.opendev.org/t/openstack/build/4ed2b1f43d05475187a499e7fb0e436e : SUCCESS in 15m 38s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/163cd87ffc10404e9e78686149c27d80 : SUCCESS in 8m 47s","accounts_in_message":[],"_revision_number":1},{"id":"2d15dabe0e0e31c5a177891b95b9daf9c06f52bd","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2020-06-05 06:23:25.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nmerging this as it\u0027s only in the promote pipeline ...","accounts_in_message":[],"_revision_number":1},{"id":"c3255cb23796bb0ac3a163063784a7181d541c0c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-05 06:23:36.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e12155802596ce5c5f4f65007a36f25acddf6109","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-05 06:42:47.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/3e254150ee4d465aba05daf341339b67 : SUCCESS in 5m 39s\n- openafs-rpm-package-build-centos-7-x86 https://zuul.opendev.org/t/openstack/build/650495bb23a74fc6855af32ab66a2b32 : SUCCESS in 10m 24s\n- openafs-rpm-package-build-centos-8-x86 https://zuul.opendev.org/t/openstack/build/463a97a5909d49cea0e8f3dac35c5efa : SUCCESS in 14m 42s\n- openafs-rpm-package-build-centos-8-arm64 https://zuul.opendev.org/t/openstack/build/d54b48bbbb764389b55a2554540c21ac : SUCCESS in 13m 51s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/5fc5670092cc43b793a5d94ac5016dc6 : SUCCESS in 7m 32s","accounts_in_message":[],"_revision_number":1},{"id":"441e0af0c44bd7de760423f58d09020f381a5ee3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-05 06:42:48.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"7e65efcb1f81334738133e9451092d7b9d122721","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-05 06:46:09.000000000","message":"Patch Set 1:\n\nBuild failed (promote pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- promote-tox-docs-infra https://zuul.opendev.org/t/openstack/build/d725221deb68483ca8db31ff8c3c9e9a : SUCCESS in 2m 33s\n- openafs-rpm-package-build-promote https://zuul.opendev.org/t/openstack/build/3d6255f0acf14eff86e210041fbd457b : FAILURE in 2m 52s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3b93422450f9f454787e39f706ddcb14df48a5d8","revisions":{"3b93422450f9f454787e39f706ddcb14df48a5d8":{"kind":"REWORK","_number":1,"created":"2020-06-05 06:00:10.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/48/733748/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-zuul-jobs","ref":"refs/changes/48/733748/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/48/733748/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/48/733748/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/48/733748/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/48/733748/1"}}},"commit":{"parents":[{"commit":"5fc86d109b07504e3685dbe297ba81b5c6540eb5","subject":"openafs-rpm-package-build: trigger publish run","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/5fc86d109b07504e3685dbe297ba81b5c6540eb5"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2020-06-05 05:58:50.000000000","tz":600},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2020-06-05 05:58:50.000000000","tz":600},"subject":"openafs-rpm-package-build: make artifact directories","message":"openafs-rpm-package-build: make artifact directories\n\nIn the publish step, the unarchive role does not create these\ndirectories.\n\nChange-Id: I64a9307a4ffc6b03ff9f15443fffa54b6ee2140f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/3b93422450f9f454787e39f706ddcb14df48a5d8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/3b93422450f9f454787e39f706ddcb14df48a5d8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
